)]}'
{"id":"openstack%2Fswift~781093","triplet_id":"openstack%2Fswift~master~I0d3c4cae88ed50a09e892e6d77b5aaa96e6195a5","project":"openstack/swift","branch":"master","topic":"p-get-partition-for-hash","hashtags":[],"change_id":"I0d3c4cae88ed50a09e892e6d77b5aaa96e6195a5","subject":"Make more use of get_partition_for_hash in unit tests","status":"MERGED","created":"2021-03-17 12:52:06.000000000","updated":"2021-03-17 19:40:30.000000000","submitted":"2021-03-17 19:34:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781093-p-get-partition-for-hash","meta_rev_id":"cc1a9c024606103205946be0290ff4a81b5d7925","_number":781093,"virtual_id_number":781093,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-17 19:34:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-17 17:43:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-17 17:43:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-17 14:01:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-17 17:43:51.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"2901be84701fb60587235aa8e00c0b1042527f6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-17 12:52:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5cca957d4fe13d7ec734b7b76b79330fe86277d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 14:01:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa68fa7a125d4b21ba82b3bd6ac26bea : SUCCESS in 9m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4a92d2dc979541e4822bfe710f17b78a : SUCCESS in 8m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c0c50bde2a8741dea8b6a89c8181d39e : SUCCESS in 8m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/28ed49a42ccd4a4bbdfdbc67116c23cc : SUCCESS in 11m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2780aabefe3b43efbe6b240aa758907a : SUCCESS in 16m 40s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6f8c8bd1d5684aea94e4d2f91831ff5c : SUCCESS in 17m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c51beeff7d2140719936ddf04ea4023d : SUCCESS in 17m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/55c36d1fe2e548abb54d0b1286348bdb : SUCCESS in 12m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bbce7485fca84d10b4c72a7b0686c2cf : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4f806b965aef495c8687738173f5acee : SUCCESS in 19m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fc6ef12912cb4c96824aecb8cf08c0d3 : SUCCESS in 17m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e6b9ea270b02464fa5840a2a648a494c : SUCCESS in 1h 07m 37s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9838af79ee2740c480ec2b3563e1878e : SUCCESS in 19m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2ea123f1443a46d3845d8d42ebcb17a5 : SUCCESS in 22m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/896487649eee4abe9bbba45f3f5127f9 : SUCCESS in 44m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c332c53a84f14fad8a2bf9c189521158 : SUCCESS in 42m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c436065b2d82426083b14e7af897e4a5 : SUCCESS in 17m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/000a9aa8960c4eec9f0e8905e8d52b84 : SUCCESS in 6m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/16786ce3d6274a63af8a62d530df12a5 : FAILURE in 12m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9acd53469605229bdb91d4c0ac03b8f2eb52fffc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-17 17:43:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"80c85a7563f7a3fbad357f1bf04947ac70cb6845","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 17:44:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"91d75455d9d9b7fa7bd839ddfac875d010c25c3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 19:34:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f2d134e756b47dbb56d8eecc475b87e : SUCCESS in 9m 03s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0320c07f1f024192a2bcb8696403d93c : SUCCESS in 11m 53s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1a0f32ba210b4ccdb6b6f81708a309a4 : SUCCESS in 8m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4b2204546e3c4821b5ee25c2445dda25 : SUCCESS in 21m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/27150cf659ac435081688909ad2e58e7 : SUCCESS in 30m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b8a53f63802d44ce97e44025d08839ba : SUCCESS in 18m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f1bc8ae63b1649959d099addf6732591 : SUCCESS in 15m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c62580216bc941b084358e8c2d31d856 : SUCCESS in 31m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/26b2809808eb47bdbd1d9aca6e4108a8 : SUCCESS in 17m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4280144b8b7e41dda004e67d81e451b7 : SUCCESS in 22m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c38f8ec4331d44f6a758a53f73d2f0f7 : SUCCESS in 20m 59s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f849ba5011704db89a9d44880c6b41bf : SUCCESS in 31m 27s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7b12dea2ddfc44dd98d9d9f05cb64ec1 : SUCCESS in 19m 43s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/011e612e1aec425c8153ef34837273c4 : SUCCESS in 20m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4cee60825f954bf6810234cc4662cac7 : SUCCESS in 6m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a3bbf29fb24e466c8481904c3675f7b4 : SUCCESS in 58m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/61f9a8a86c604416a3b8a51013653140 : SUCCESS in 42m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1f2b1268708413598fda12c2e39fbf6 : SUCCESS in 15m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e171b1105b444e97a50a2cc49d0a175e : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":1},{"id":"9a80b9b1da2b80c150852ae6e2e1d7556d451c44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 19:34:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cc1a9c024606103205946be0290ff4a81b5d7925","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 19:40:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a1d8042e0a546a9a1f4844aea7b984b : SUCCESS in 1m 11s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c4135bdc61bc4f97b74884dd39cdb475 : SUCCESS in 48s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/997696546e55452eb0d583003a6b9ee7 : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"acf72e88030901473470934be9e1ea060684f946","revisions":{"acf72e88030901473470934be9e1ea060684f946":{"kind":"REWORK","_number":1,"created":"2021-03-17 12:52:06.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/93/781093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/781093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/781093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/781093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/781093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/781093/1"}}},"commit":{"parents":[{"commit":"354a86a80ee6de462c240c5684bf982c806bf7a7","subject":"Merge \"py36: Fix syslog fallback to UDP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/354a86a80ee6de462c240c5684bf982c806bf7a7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-17 12:43:53.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-17 12:51:54.000000000","tz":0},"subject":"Make more use of get_partition_for_hash in unit tests","message":"Make more use of get_partition_for_hash in unit tests\n\nRefactor to make more use of the utils.get_partition_for_hash() helper\nfunction that was added in [1].\n\n[1] Related-Change: I5fbf8671f2ee63467d46f651e2c24529a2697626\n\nChange-Id: I0d3c4cae88ed50a09e892e6d77b5aaa96e6195a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/acf72e88030901473470934be9e1ea060684f946"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/acf72e88030901473470934be9e1ea060684f946"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
