)]}'
{"id":"openstack%2Fswift~781495","triplet_id":"openstack%2Fswift~stable%2Fwallaby~I7b1a91726d0de5dad82a47e277c14af00cfcadcf","project":"openstack/swift","branch":"stable/wallaby","topic":"create-wallaby","hashtags":[],"change_id":"I7b1a91726d0de5dad82a47e277c14af00cfcadcf","subject":"Update .gitreview for stable/wallaby","status":"MERGED","created":"2021-03-18 14:47:33.000000000","updated":"2021-03-18 18:06:33.000000000","submitted":"2021-03-18 17:47:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781495-create-wallaby","meta_rev_id":"89f6eb6f070ebc92c68628fc902a0bc17c601589","_number":781495,"virtual_id_number":781495,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-18 17:47:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-18 15:14:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-18 15:14:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 15:14:32.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-03-18 16:14:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8c7241bb4bbb156dcf14ea9398b2cb530a05ddc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-03-18 14:47:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3dfb4d372f8341a3042d18430051f9bdfa5123da","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-18 15:14:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e21fb94e4d837a7ff0f082fb40e4bebd425f3b3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 16:14:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/988ec13bbc494b82869557ef430dec96 : SUCCESS in 10m 39s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/32445a5bea16404e9c154fdf9628a081 : SUCCESS in 11m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/514e2cdb201f4405b7890677a4ebc127 : SUCCESS in 12m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4b825ce3a5a54acdae350e46611f2e42 : SUCCESS in 17m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7fb5dbc8029c4e41a1fcd024e5586067 : SUCCESS in 18m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7469d4e760bd4ba792713830caadbfd1 : SUCCESS in 15m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ad7f80d7d0b14839b5c09de33e454f2c : SUCCESS in 14m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d03036a23f9347338524da38eeccedb7 : SUCCESS in 19m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61802160cc084877b376a067d6ec44b0 : SUCCESS in 12m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05ba047d5bfa48dda82a311f216c5ef3 : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":1},{"id":"2f55067cc7940f44f61d59e6b0aeed340e7c1d7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 16:28:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7538fa86654739ae691a3974c02fd35d5e186a39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 17:47:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cce7c8e3b124db6bc35fbb6eb7eac84 : SUCCESS in 9m 11s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/76c167a6c4e24247853b046bbf16d9cf : SUCCESS in 12m 11s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/7e90d7393fd94b07b52a91f87e111724 : SUCCESS in 9m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0f6bdda6e27a4cf6b8276d729bb33c69 : SUCCESS in 15m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/84210b8ed2ab4c75995cb8b86c132c89 : SUCCESS in 16m 28s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/59919076cd9f4fdd89392235494ad15a : SUCCESS in 17m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d57de4dcd4a84927894404a2c7b0baa5 : SUCCESS in 13m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9c9591b7726a4efc8f3c3f0134723c22 : SUCCESS in 13m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6db8eb149a8d4eb8aea23d11cab0e0cf : SUCCESS in 19m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/760f75d3d28a4368958b58bb67b0726f : SUCCESS in 20m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c38cf9f91e2947ed9d58931ec3858c3d : SUCCESS in 19m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c3784fb101754b41ad37179f4eb5212f : SUCCESS in 21m 45s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cc39a1f486f34373933c797da1efe993 : SUCCESS in 20m 28s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8bb12c993f0f4765aaaaf4fd956065b9 : SUCCESS in 21m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1f5ab9272bf0466ea97ec745206d40f4 : SUCCESS in 7m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07b585ef211e44f99146bf7de3aa1b96 : SUCCESS in 16m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d6d4477aef849ef9ed09cf36bff4bba : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":1},{"id":"70fe326647535073fdf5418f453daf3fd07bd25f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 17:47:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"89f6eb6f070ebc92c68628fc902a0bc17c601589","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 18:06:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6dc12f329b34bbe9cd7422671cb07b1 : SUCCESS in 1m 21s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/1634a41544a14b598cd79f28bc971bf7 : SUCCESS in 1m 41s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0190245b6f9549a091e51339197997ad : SUCCESS in 1m 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69ac74fe50026238cb5dd9638270f8f6fc46feed","revisions":{"69ac74fe50026238cb5dd9638270f8f6fc46feed":{"kind":"REWORK","_number":1,"created":"2021-03-18 14:47:33.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/95/781495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/95/781495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/781495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/781495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/95/781495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/95/781495/1"}}},"commit":{"parents":[{"commit":"c6a64036651e18f3d7d0715c5d4876257aaa0e6f","subject":"AUTHORS/CHANGELOG for 2.27.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6a64036651e18f3d7d0715c5d4876257aaa0e6f"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-18 14:47:29.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-18 14:47:29.000000000","tz":0},"subject":"Update .gitreview for stable/wallaby","message":"Update .gitreview for stable/wallaby\n\nChange-Id: I7b1a91726d0de5dad82a47e277c14af00cfcadcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69ac74fe50026238cb5dd9638270f8f6fc46feed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69ac74fe50026238cb5dd9638270f8f6fc46feed"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
