)]}'
{"id":"openstack%2Fswift~781496","triplet_id":"openstack%2Fswift~stable%2Fwallaby~I35df9e06ed5a7ec0393b3e1368f13da6fcddb23e","project":"openstack/swift","branch":"stable/wallaby","topic":"create-wallaby","hashtags":[],"change_id":"I35df9e06ed5a7ec0393b3e1368f13da6fcddb23e","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","status":"MERGED","created":"2021-03-18 14:48:03.000000000","updated":"2021-03-18 19:03:21.000000000","submitted":"2021-03-18 19:00:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781496-create-wallaby","meta_rev_id":"c2e7ad284df475f844b43cea1b1570eb5fdc4c91","_number":781496,"virtual_id_number":781496,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-18 19:00:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-18 15:15:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-18 15:15:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 15:15:38.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-03-18 16:49:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a41f267b9c41aa9ae63234d52b8eaab5d0d029e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-03-18 14:48:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ce2d7a739c8e9ae0f4b257719d0389b4f9cd4c2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-18 15:15:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"87f4d58c80468597b0ec4a8c357ae9971e452e2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 16:49:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35d83baa287846e19597950d8f98c227 : SUCCESS in 10m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73e57322f4e84fa99ebcbd3a682ec8a3 : SUCCESS in 5m 58s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1e52ab5ad3b04a5dbe71180b22cfcc1a : SUCCESS in 13m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1e69607e60cd4dc7bb8bbd9aa683c21a : SUCCESS in 13m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f0f7b3b74c4245d2b2c2fb28b738f57a : SUCCESS in 22m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/090357ccb21545e3acc2e746c19e7d82 : SUCCESS in 17m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e9416d9e28264e08a313de75f4d49633 : SUCCESS in 19m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d9d8cfdbde924f128a0018d810610a59 : SUCCESS in 19m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/36471cd2f57b4b72b08457d82025b9e2 : SUCCESS in 18m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a8a665cd18bf423e8f7ffdb5a1681435 : SUCCESS in 18m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c5df7df493ad4336bf38d4ca5d2d699b : SUCCESS in 21m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dbf518d03374435b88409f3fd10c18db : SUCCESS in 21m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a0971db94163447abd5805c0730197c2 : SUCCESS in 24m 54s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/76c06cf71a7549bb9485ce01e8f41cbf : SUCCESS in 19m 37s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fe78833c55964cc093c6a0c29d2cadb1 : SUCCESS in 19m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e8249638468d443e9d49b2222f53ece2 : SUCCESS in 6m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/87f9f6877c474821878940367e4a3c1f : FAILURE in 18m 05s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/66fbdbfc06804c02a62eaae7393c4fff : SUCCESS in 51m 39s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/a15260bc495a463db757a12e37a40e4e : SUCCESS in 41m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c328d417ddc7493d88b284625700fba8 : SUCCESS in 16m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2277e95c92d141798c794b6c52ac5bfb : SUCCESS in 6m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8a9d26a36ff9423ebcb385dfb3738524 : SUCCESS in 24m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb8b225527267a3e21c94b00a0c108f6650517bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 16:50:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f6d30041c7298716e2a1bc7c3ad2b23e3d7d3865","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 19:00:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d517ccf8add444118caa119ef01abf8b : SUCCESS in 18m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb5609f8ec5e449895317cd0c79960d6 : SUCCESS in 8m 22s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4281abaf75c449f59c6dccab95fc520a : SUCCESS in 14m 20s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/299bcbc510d649dbad987c975830f9db : SUCCESS in 11m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ec867dc9d600472cbd5feb93ed05d50b : SUCCESS in 16m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/abfdf58694374510a45ac9595fe2a01c : SUCCESS in 16m 55s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9d70e4e7c7d64f8fa20971cf44aa5eba : SUCCESS in 18m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/be040bda67b2435b9c9a9adce429f6b1 : SUCCESS in 18m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/967a9a90219346b7bbeac8c6489018b4 : SUCCESS in 24m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d8bc282dfa194c4389cda42acd7243ef : SUCCESS in 19m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/59d443be28004b64804fb11abcc9bb4a : SUCCESS in 21m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0f835f09c9f144839c45593712552316 : SUCCESS in 19m 44s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f83a50100f4746fb8136bce58d5cbf6b : SUCCESS in 26m 30s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/45c546bc740b4f0aaeb6016168f12af7 : SUCCESS in 21m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6a8d10fe83d8427da12c907b43ebd25e : SUCCESS in 29m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a4097ea70b754af3812ff4ae8c86a851 : SUCCESS in 9m 43s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5248f33560a24925b2195acb6bccf63d : SUCCESS in 55m 26s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/c2bf3df2ed8e4d65b01c95f672bfc40c : SUCCESS in 50m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42539ca57b564c1c8fccaf86277c471f : SUCCESS in 24m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/276f4cba34a2445a8b24b72c4f7b553d : SUCCESS in 9m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f10d04a8c75684ff4cb0594b8168e0b98e5d12bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 19:00:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c2e7ad284df475f844b43cea1b1570eb5fdc4c91","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 19:03:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78c5ffe5ef4a4990ac4bca4ff6758644 : SUCCESS in 1m 32s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a7628a525ba4028813b945f72ba3235 : SUCCESS in 1m 13s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8d2fcac805604a13b4d789d627e67dc6 : SUCCESS in 40s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/c564c63ed1cc41f494325f83d29067b5 : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d47724e403448bb7e3e02b2df89a7876c20bc4c1","revisions":{"d47724e403448bb7e3e02b2df89a7876c20bc4c1":{"kind":"REWORK","_number":1,"created":"2021-03-18 14:48:03.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/96/781496/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/96/781496/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/781496/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/781496/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/96/781496/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/96/781496/1"}}},"commit":{"parents":[{"commit":"69ac74fe50026238cb5dd9638270f8f6fc46feed","subject":"Update .gitreview for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69ac74fe50026238cb5dd9638270f8f6fc46feed"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-18 14:47:58.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-03-18 14:47:58.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","message":"Update TOX_CONSTRAINTS_FILE for stable/wallaby\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/wallaby branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I35df9e06ed5a7ec0393b3e1368f13da6fcddb23e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d47724e403448bb7e3e02b2df89a7876c20bc4c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d47724e403448bb7e3e02b2df89a7876c20bc4c1"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
