)]}'
{"id":"openstack%2Fswift~781550","triplet_id":"openstack%2Fswift~master~I1ec6d8c62ca47c2a039672331551a62c7fdab5cc","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I1ec6d8c62ca47c2a039672331551a62c7fdab5cc","subject":"relinker: Tolerate missing files when cleaning up partitions","status":"MERGED","created":"2021-03-18 19:26:29.000000000","updated":"2021-03-29 21:13:12.000000000","submitted":"2021-03-29 21:01:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"781550","meta_rev_id":"7a803a3530f26123c856ed7aa6d1ad4391f82269","_number":781550,"virtual_id_number":781550,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-29 21:01:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2021-03-29 15:35:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2021-03-29 15:35:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 20:33:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-19 15:23:25.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2021-03-19 17:43:35.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"d5aea7720f76056085d0db6ef5486905b2e923f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-18 19:26:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51e79a5a7587fe8a7f3e040c88a8dd13e9b7c721","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 20:33:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92c0b8e9d14e4bab902a4ce6dd7c44d6 : SUCCESS in 10m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/51224db21c6b4b8d8840be82a6372741 : SUCCESS in 1h 04m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/00139f58c7294dc1b932492db8155f47 : SUCCESS in 47m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7a7117cad61b48279e70d899b259d782 : SUCCESS in 8m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/94a7a8e8bd22409a8ecb8b2a1aae2a94 : SUCCESS in 8m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7643e56ede08422c897d08b94d3addcb : SUCCESS in 27m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3a89ec7533ef496db97eb8fa82d3399b : SUCCESS in 16m 38s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7103db4bc51246d2a90b96372536793b : SUCCESS in 28m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d6e695f7ac4c42ed882bdd4dc40da9ba : SUCCESS in 13m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d3f51da7e8e24d7aa357ea5f32e5b5e8 : SUCCESS in 15m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/02a3578131504e5ebab5b4362d6f7a91 : SUCCESS in 18m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d77bc317284646f89bb93cf2967687dc : SUCCESS in 17m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/46183952a3d84c31ae9b8a5f37e42411 : SUCCESS in 19m 04s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/860bbcb1ae9548718899d5c0001ade4d : FAILURE in 20m 17s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/361484d45ba5448ab8a732ccacaca474 : FAILURE in 17m 01s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5163d61bd54740e28115e0ebea397bc4 : SUCCESS in 19m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5d0dece1c96e4306afcb63107e4fc60e : SUCCESS in 6m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4d6a66f099154fa99c9a139c4385c5ad : SUCCESS in 21m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/031b7cfe106944fc88da6de0c1f333b0 : SUCCESS in 52m 17s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/92ed9aa4099c49e1ba98594136527243 : SUCCESS in 45m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6c661df933fc478c82dc0160c2e534f2 : SUCCESS in 33m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9e53a6431f7345b1810cbf44974bee59 : SUCCESS in 31m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27525a28f0004334960227a4c8698388 : SUCCESS in 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9803414ca9542669f30d7f6c21d1d7e : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c6288d184657437baee19cf7e0048b9f : SUCCESS in 23m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f658f22c94ef4b0a9dde59fe611b4578 : SUCCESS in 59m 31s","accounts_in_message":[],"_revision_number":1},{"id":"c2cde44c3b93dd7bee853736fa68aeef94eee9bc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-18 20:47:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c28a6cf991720973117924e844b2aa76f43c914a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 21:53:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d56681a3c8c347da8cd7490443477c4d : SUCCESS in 10m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/361bdf73398247319bd651f4df657cad : SUCCESS in 1h 04m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1ce4f75aae194107b36a65917f70c388 : SUCCESS in 38m 54s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5d994616db54448886b04d1e38cb5a2c : SUCCESS in 7m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/edeedf118ee24346ba481cce20c22adc : SUCCESS in 6m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ab4434af5ce541199868d3a2d15186fe : SUCCESS in 18m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/375d27ea15ff405492395758be070264 : SUCCESS in 20m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cf37d9b551cc4182ba658db1e769965a : SUCCESS in 18m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/639191ed9e8e4c1cb686825636320d09 : SUCCESS in 20m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3001b87fcb214f989bd331c97563c033 : SUCCESS in 21m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4902931be311403785d39d68d9d9dc84 : SUCCESS in 21m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/abc7153054a84d3597411bd2f9c8321f : SUCCESS in 26m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fe59aa3469804362a0f7c9989dec64af : SUCCESS in 27m 13s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/325a8679dbf848f8bcaff86f54b9d91b : SUCCESS in 26m 39s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/655c47df7b324e86814b0f52500a6488 : SUCCESS in 26m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e3eff4536590417697af2225e6ffc7d5 : SUCCESS in 20m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0b4a0a0758a74a06bc5920916a4f32a4 : SUCCESS in 6m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e6e021d5a0ff4bc285e87566af781cc0 : SUCCESS in 26m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cd613f383132444f81280db816bd1f05 : SUCCESS in 49m 35s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/e83f30c8e54d49f4a69c95ab65c8877a : SUCCESS in 50m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e5c1a1c2c8294823b35695ef9c7ab499 : SUCCESS in 41m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1bee83f8207a4f159550c8f6c16e373d : SUCCESS in 39m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7f1b30dc67b418a93360d9aad6cdcec : SUCCESS in 24m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5298a6e15754e05801b530d32028c06 : SUCCESS in 6m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/68bf04729b8149d4ad323d2cac4e07a9 : SUCCESS in 28m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6eb3062c28064ce1a9713242abc61add : SUCCESS in 55m 21s","accounts_in_message":[],"_revision_number":1},{"id":"8eb72687afbdabca7f843ace0905a9ac4f98defa","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-03-19 15:23:25.000000000","message":"Patch Set 1: Code-Review-1\n\nChange seems quite reasonable; needs tests.\n\nI think test_cleanup_deleted (test.unit.cli.test_relinker.TestRelinker) can be adapted to demonstrate this change - let me know if you need some help.","accounts_in_message":[],"_revision_number":1},{"id":"cfdd1e08fe591516313a04fba0ca13c4586f698b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-19 17:43:35.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f975dff0e93e85a7bd868e3d688ff02c0d7b71c7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-19 23:14:40.000000000","message":"Patch Set 1:\n\nSo it\u0027s strange -- I *know* I saw some nearly-empty partitions looking like\n\n $ ls -la /srv/node/d731/objects/104122\n total 12\n drwxr-xr-x  2 swift swift   27 Mar  4 22:06 .\n drwxr-xr-x 83 swift swift 8192 Mar  5 00:31 ..\n -rwxr-xr-x  1 swift swift    0 Mar  3 14:43 .lock\n\nwhich actually would have been in the *new, upper* half (and should have been rehashed during relinking) -- but any call to get_hashes should have at least written hashes.pkl.\n\nWhich makes it hard to write a test to demonstrate the problem that isn\u0027t contrived -- in that test, I see none of those files present in the old partition before running cleanup, but all three are present immediately before the unlinking. *Maybe* I could get somewhere with a test that links into that second quartile, then running relink *again*? Or maybe it needs a relink -\u003e cleanup -\u003e back-dated-relink flow...\n\nStill, this seems like good hygiene, and if I had just done it this way to begin with, I bet no one would\u0027ve batted an eye ;-)","accounts_in_message":[],"_revision_number":1},{"id":"11b05c110ba5a763e4b24505ea70316b1b347e2e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-03-22 15:36:37.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nI think this is not needed now that https://review.opendev.org/c/openstack/swift/+/781576 has merged","accounts_in_message":[],"_revision_number":1},{"id":"919d7186203e535ae502880f6b5bb2774e7e887c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-22 18:00:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b5286124826a706fe49382ad949fc2f3bc80bf92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 20:45:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e3c95ef67884db38d4ebc758a48398f : SUCCESS in 11m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/0f46c24784d74454bfe8daa901bbd377 : SUCCESS in 57m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8307fc7d0d6c41bca38151abb5f2ee56 : SUCCESS in 42m 17s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2aa78ed505764068a2ed4c4b6cf8f608 : SUCCESS in 11m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/33cef363f260466cb5b1c2d37e089b44 : SUCCESS in 8m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b28f27a495a44463862e05bbb8b60c3f : SUCCESS in 18m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/24d8e7af8199450186fd31be03851e4e : SUCCESS in 18m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c37cbdd785ea4808a8a2578dac29b8e2 : SUCCESS in 17m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d34292083c944c7ab030251b9020f34e : SUCCESS in 16m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/87091c103a924a37bf09be5df0104c4f : SUCCESS in 14m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3b7b6b5be9a6453d87789ecc2849b1b9 : SUCCESS in 16m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aa1c0a1c9f6442c9abfd354a7e47c441 : SUCCESS in 21m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/58d5ca6f89f845088981622bd87a8569 : SUCCESS in 21m 41s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/db19460a0a874017afe4f061a6e51d1a : SUCCESS in 17m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/15079b6e0e9b4f3e8bafaadf8cb19d58 : SUCCESS in 40m 30s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a4266c948cec49bcba48905307430611 : SUCCESS in 17m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/db83a0627b2c4e4290e7d9d2f27ef5ce : SUCCESS in 8m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ab1a070e44834d3d8f07a588ae4d9ff7 : FAILURE in 18m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/840c3744985e40f0b4004d2619d98176 : SUCCESS in 52m 02s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/8f4b6392fb4442a7bcd9b2e811408649 : SUCCESS in 53m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9440d714557d479a915f452dc5a1e313 : SUCCESS in 43m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f3475b0bb08549e6bf02285bb04304fb : SUCCESS in 48m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fe4981273594213b72654111ce47a90 : SUCCESS in 17m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6469732556e34bd683b2852669588585 : SUCCESS in 6m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7cfa6e960d034b3f93fa84d7efae7647 : SUCCESS in 29m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7eaaa7c33c264089811aada20dcf9180 : SUCCESS in 1h 08m 35s","accounts_in_message":[],"_revision_number":2},{"id":"590fa7d6dd06bdee37f30922a9609f3afb63c3b4","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-03-29 15:35:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"16b1c2fc7902e5997a47945caf8ca8a62ff2b5a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 15:36:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fb9dced65b72d55b2d665a7f1f03bf2001aac584","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 21:01:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ddf039a799d44a7af93a8c90dccb035 : SUCCESS in 9m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/52aa5bf50c004423b2980f0acc4d9b96 : SUCCESS in 1h 16m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f980f2fb23ed413ebcf9283212e710c7 : SUCCESS in 41m 31s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/45607b5701ce4809b266702f83da1f08 : SUCCESS in 8m 46s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c2ecb441957943d0849a88378618ec1b : SUCCESS in 10m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bed05109680d48e58007e11c8a798d6b : SUCCESS in 12m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4a32cbb6b61c4f0ab4dea7eeb9c18474 : SUCCESS in 15m 54s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/05b3cce9a802492e9834d1792dfcbec3 : SUCCESS in 13m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/52f8aff4f6e8437db90bbe1934232478 : SUCCESS in 15m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7a8b68e24b8b4dde947cae1cc6f553d0 : SUCCESS in 12m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/006cfcab1711490188561bf180fff461 : SUCCESS in 18m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9eec25b3485c4a60937d2775213b8e96 : SUCCESS in 14m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b5915380688d4daf8f64409793279221 : SUCCESS in 18m 01s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b7e8b58810cb4c8f92c61d884cd6a54f : SUCCESS in 15m 20s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/70cf2cb8ce374328b8559fbea97b67dc : SUCCESS in 18m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/efe1aa67807a40eaad29ceaa7c1614e4 : SUCCESS in 16m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2b7cc5b4dbba489c94333cac44761855 : SUCCESS in 8m 29s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bf91b2bbb913484dae13c47fd5e814ea : SUCCESS in 35m 40s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/a17de1e5eac04f7a8531cc428fe68b50 : SUCCESS in 56m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/deca2865cff649b9b98fe968afec7250 : SUCCESS in 49m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8cb1b4ff8b9540f497d7b13e2aade1f3 : SUCCESS in 40m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dafaaaa2da854bc29fb74b757dd9714a : SUCCESS in 12m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/718b7355c2a244e78c029b58123f04c2 : SUCCESS in 6m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/02b7e0ecdbda4b51af7cedc2ee2fe31a : SUCCESS in 26m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe0421608c3445a5b5fac8e821356b9e : SUCCESS in 1h 03m 15s","accounts_in_message":[],"_revision_number":2},{"id":"06f0b30e9faa86359da5adddec7acef00901f639","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 21:01:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7a803a3530f26123c856ed7aa6d1ad4391f82269","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 21:13:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f46a6fd988949ffaebc8182c17cddb9 : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/065c9f51c3e24836b811fc80a2b70816 : SUCCESS in 44s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/50178511b5364c8bb03d09bff422f4f7 : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4ef7b10e0f6e9753a7d65de15ca156f7ea6f036d","revisions":{"a97b48d9d148bb48469c17d79b4ce45b3c9262da":{"kind":"REWORK","_number":1,"created":"2021-03-18 19:26:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/50/781550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/781550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/781550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/781550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/781550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/781550/1"}}},"commit":{"parents":[{"commit":"c6a64036651e18f3d7d0715c5d4876257aaa0e6f","subject":"AUTHORS/CHANGELOG for 2.27.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6a64036651e18f3d7d0715c5d4876257aaa0e6f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-18 19:26:08.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-18 19:26:08.000000000","tz":-420},"subject":"relinker: Tolerate missing files when cleaning up partitions","message":"relinker: Tolerate missing files when cleaning up partitions\n\nChange-Id: I1ec6d8c62ca47c2a039672331551a62c7fdab5cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a97b48d9d148bb48469c17d79b4ce45b3c9262da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a97b48d9d148bb48469c17d79b4ce45b3c9262da"}]},"branch":"refs/heads/master"},"4ef7b10e0f6e9753a7d65de15ca156f7ea6f036d":{"kind":"REWORK","_number":2,"created":"2021-03-22 18:00:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/50/781550/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/50/781550/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/781550/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/781550/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/50/781550/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/50/781550/2"}}},"commit":{"parents":[{"commit":"7e278291f88afe1ac78d930404bd4dbbfdb4f53a","subject":"relinker: Move filters \u0026 hooks to be methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e278291f88afe1ac78d930404bd4dbbfdb4f53a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-18 19:26:08.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-22 17:59:52.000000000","tz":-420},"subject":"relinker: Tolerate missing files when cleaning up partitions","message":"relinker: Tolerate missing files when cleaning up partitions\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I1ec6d8c62ca47c2a039672331551a62c7fdab5cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ef7b10e0f6e9753a7d65de15ca156f7ea6f036d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ef7b10e0f6e9753a7d65de15ca156f7ea6f036d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
