)]}'
{"id":"openstack%2Fswift~781576","triplet_id":"openstack%2Fswift~master~If906513ed47f203ca97ef08d4375b01c76f812bb","project":"openstack/swift","branch":"master","topic":"p-relinker-refactor","hashtags":[],"change_id":"If906513ed47f203ca97ef08d4375b01c76f812bb","subject":"relinker: Move filters \u0026 hooks to be methods","status":"MERGED","created":"2021-03-18 22:20:55.000000000","updated":"2021-03-22 15:41:31.000000000","submitted":"2021-03-22 14:16:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"781576-p-relinker-refactor","meta_rev_id":"425bbe4722c4194b1bf0f772eb2fa98a85ce0ccd","_number":781576,"virtual_id_number":781576,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-22 14:16:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2021-03-22 10:34:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2021-03-22 10:34:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 22:21:18.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2021-03-18 23:45:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-19 14:55:22.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"64903f0eacfe95dc7496b42fe453e6423ad64df0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-18 22:20:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f33a321f7986b9ed60cda349c1aa42edca2d6203","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 23:45:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b832a6890bb94014985a1bffb0cef024 : SUCCESS in 9m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/a883905a6b2f4b90b46690e255b87c2c : SUCCESS in 1h 16m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8fa233e14417495db14b987427005cb6 : SUCCESS in 50m 15s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/69c2f6c2ef574e4f8111572c16a039a8 : SUCCESS in 10m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/557a74a3f3034394aa6b0b05a2ac8314 : SUCCESS in 6m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9facc8643fe0485da03dc2c800917d91 : SUCCESS in 14m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/14f4abfed1d643ecbb6533ee891239bc : SUCCESS in 16m 21s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/876848d02d84474e980dfa3c27d4dd53 : SUCCESS in 14m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/851617fdedcb4c95a6cf6fcc38cf6abb : SUCCESS in 15m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/843485471b704594b0aa28cf454606d3 : SUCCESS in 14m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6e77f81f2cd1466fa2978258fa7d5da5 : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0a78b6c25b624828a4e77cc9b3592479 : SUCCESS in 16m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/63742b1da0d54037a5067b8d27453be8 : SUCCESS in 16m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/95b372a0f1af477ca0bca2bd55dc3011 : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/fd164ee6c40c4e2aab42313907098c07 : SUCCESS in 19m 16s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3b12573827f3468888786d4d4e0edc95 : SUCCESS in 19m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dd8e5c43568048c2befba3b61c3daf7d : SUCCESS in 6m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d0eafe7d952747f181e4e100f7998240 : SUCCESS in 19m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bfd893e489904673b2f80cfcbfa99692 : SUCCESS in 55m 41s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/043e510a8c5e4e99a5934f129eb26256 : SUCCESS in 44m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f8adda60e9b748c68ec07f24c506f905 : SUCCESS in 41m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1ca2eef24cf6413b9d123ed082d0f51c : SUCCESS in 41m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99e5d6b09bf34189b1360ded85ccc003 : SUCCESS in 13m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/833259c0a4764d5d89ab69a3161d396a : SUCCESS in 5m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9ecece3e49454e78ba7b88d850a96c56 : SUCCESS in 20m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1a8d9c3af744096bdf2ef7198bf43a2 : SUCCESS in 1h 08m 24s","accounts_in_message":[],"_revision_number":1},{"id":"9d76f5a78d9079371d9696adda4bd72c4cc71443","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-03-19 14:55:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)\n\nThis seems pretty obvious to me","accounts_in_message":[],"_revision_number":1},{"id":"5675bc2d3aaba341891d841665539c00c6a379c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 14:55:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5eb3918fc5cd992d330865fda9f03295fac1e5a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 21:03:17.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc57cbf284b8448694e2fa507f9b705c : SUCCESS in 9m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/3a7efc1874844ebca37f4f98791607a5 : SUCCESS in 1h 02m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a27065329eaa4136b32c3d26abde102e : SUCCESS in 44m 22s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f7e68401541c4ce9a18cee70241a0951 : SUCCESS in 12m 12s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/fd3245ac813648c98559929922e39d2b : SUCCESS in 9m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/46985bdc6f73420183170b5439d89cd5 : SUCCESS in 18m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/056c7893ada24912aea29e63d4bfe73f : SUCCESS in 54m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f95cdf42f82d476a91a54473dfe68729 : SUCCESS in 18m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/caa45a765896442f8ed79313c5ba956d : SUCCESS in 18m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7e025104fcff4d6b9dc604dbf28b4722 : SUCCESS in 19m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8e23cfe1afdd4268b0f05ea6f4bb82ac : SUCCESS in 43m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ead0f5e0a8ad43c49c909e7ae56c9728 : SUCCESS in 24m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1218629701674780931f3d68fabfebd5 : FAILURE in 41m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2ef7685aeb0f4b7f997ee63e8ade7bb4 : SUCCESS in 20m 04s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9f193916dc1447e097ac2ed951f48dcb : SUCCESS in 18m 13s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d50677f247164db882600f305c709635 : SUCCESS in 21m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9d682e2c93df46c4a6b318236a080154 : SUCCESS in 7m 53s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e3b23806c30641749d820bce6f2f5877 : SUCCESS in 52m 49s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/4ee4fd9ff9114ef295608fe0fb74fe5f : SUCCESS in 58m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e1dfde814cca4349b9778668700ae601 : SUCCESS in 41m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8a2fb3fb1aee467d8a7005c5c0329d8f : SUCCESS in 35m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a43b33605644f6b9fbe7d721725f3c9 : SUCCESS in 18m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd67744b123349db97b0a77a0caf57b3 : SUCCESS in 9m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7ca728d09cd643dfb9ce42128c35225d : SUCCESS in 30m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f32867bc082b4f49b0e477c7f89a2e62 : SUCCESS in 57m 24s","accounts_in_message":[],"_revision_number":1},{"id":"61c749f74268c2b01029c6639fa8f5e2bfd62fcd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-20 00:20:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5d059bf4e52df73348344db9a1fb1699de379932","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 01:25:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/618e960153c741219ce56a888ce6cf34 : SUCCESS in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/8d6fe449b2cc4efdb7c8c4a06ec02d75 : SUCCESS in 1h 03m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5b22e3850d6647369855a40afaf300c5 : SUCCESS in 43m 17s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/62c738964c4d4ab98a2967255cc1c57f : SUCCESS in 8m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/aced5c10eb3f4ba694a57037209e84a4 : SUCCESS in 6m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5f64a7c92411407e9992ed63e77c3718 : SUCCESS in 15m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dcf39c20b176413fabd33fa86949cf0a : SUCCESS in 18m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d3f8e86465a7470c8f0157a76b7e7fd4 : SUCCESS in 15m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8d1b8c0c228547788217a210a15169d0 : SUCCESS in 15m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/82e9b07d0d0741bc9aec34a0e1f8351c : SUCCESS in 18m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0952d69459dc42258b7525b3a690901a : SUCCESS in 20m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6eafcf714abe4cec9e580e82a91bd022 : SUCCESS in 20m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6679be36f1de49f89521521cf388f811 : SUCCESS in 22m 07s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bdc9fd0ac98147ec9b0dd42f9f542aea : SUCCESS in 20m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4802450da6264e4ca61eb08d166c5361 : SUCCESS in 20m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8532cc96b35f4906b3650e365dfc627c : SUCCESS in 20m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7ef9477e1a384d4eab66be90c7a84cde : SUCCESS in 6m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7cba53e2ab324e739104394b8af4ffc0 : FAILURE in 26m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/457e56aa6c154b45b68b46238626b49a : SUCCESS in 51m 42s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/17dbc90b697b4c36bff0b08c8e07f32c : SUCCESS in 49m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d7ba2f5177ca42a7a3b9f2f63dd9e471 : SUCCESS in 41m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f8f32ad14143409d89e3baa20862ea96 : SUCCESS in 41m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d0e4b9e34e741809b8cb6bc9cee1580 : SUCCESS in 18m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b328b0ca00045da971f57066d16ddcf : SUCCESS in 6m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a798faa2bba244c59da98363e890df59 : SUCCESS in 26m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2b3dd2fa79c423ab79a28e71dd86bae : SUCCESS in 1h 01m 12s","accounts_in_message":[],"_revision_number":1},{"id":"104179d66a7f6f1c2112857e1f78983e7b6bfa48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-20 03:22:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"25c079307050c085c0d2c1501ad54eb7dafe5732","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-20 03:23:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nJust rebased to resolve conflict with --policy patch. Merging based on prior approval.","accounts_in_message":[],"_revision_number":2},{"id":"6138a35ccecfd32b1a7f1196985806e535ebbbc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 04:48:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9096acd3406a4982b430078ace3edc39 : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/c624e1efc0544b08843e164332f51423 : SUCCESS in 50m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/196118c0939b4c90a39fa4f6df799db7 : SUCCESS in 49m 18s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6ea99e3331a943dab69adf1f804e1ecb : SUCCESS in 7m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fddc257402274d49a7936a3265698fca : SUCCESS in 6m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/83beea99b8814c0dab1047b3e42eb7c6 : SUCCESS in 18m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/20d26dc491a34f879a5bc8e6a4c36a48 : SUCCESS in 17m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8e198b25603b41dbbf12ddceb9d93064 : SUCCESS in 18m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8050596bc4ba4fb4976edf4b8ed5a25c : SUCCESS in 24m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2911b0009b154d3e8e6c8d95c04f0f80 : SUCCESS in 34m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/942bfbb72a724f7cb9b205f8cf5c427e : FAILURE in 1h 17m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/34dac9d44c3b46b88aec1dca1d9b1983 : SUCCESS in 26m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b354a730b9e942619fbdcf7ed77f8606 : SUCCESS in 39m 01s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b94b3c94cbe14067a5de41361096ab41 : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/65a4119f815041208995dfc111d12a87 : SUCCESS in 39m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c52eca9cf2b94a2099bd032a07f6834c : FAILURE in 1h 23m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a35c7bba2581488bb7d3b160d8727557 : SUCCESS in 8m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/194b9757d41a4faaa320a350560b2972 : FAILURE in 41m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/18c72ab72e404f7f94bab3d12dbab9a4 : FAILURE in 50m 24s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/001fb6084aa74a7b98803dde0534b4a6 : SUCCESS in 46m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bbb58639708f4abe99478417937d84e6 : SUCCESS in 40m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cdd8f7b6f4264c02acb6f6d622013fb4 : SUCCESS in 42m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82bcab0d5315460da6c1871fb8df6c01 : TIMED_OUT in 40m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adee9e25ed4341f1a1685b0f79772af5 : SUCCESS in 6m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3159a27a6b074c81a51070a58d892505 : FAILURE in 32m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f2a8922746941c0a6b4a64c360389c1 : SUCCESS in 49m 19s","accounts_in_message":[],"_revision_number":2},{"id":"b76a7b95260a869ef577e6de12d52e2f6288298a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-20 05:00:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c864b43526f0569eff585934cac757c6baac200d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 06:24:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24c8e7d42e754c6a8a0bb732ef13fe56 : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/a8903b1c45d549d484a33bd6ce45205d : SUCCESS in 1h 22m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6ff8553aa13e421caf31b2b687e5801d : SUCCESS in 1h 14m 41s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4dfa96c4a9bb43bd9c001c3db9c632f4 : SUCCESS in 8m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4c0b42b7ad064a6c9d0ff3af29dee4bc : SUCCESS in 6m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cf7c8293f42a48bdb50b8b87b3f3e2f8 : SUCCESS in 19m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/012bd9ee4978405ab312e6a1017132bb : SUCCESS in 27m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4888a5d295cc4b58994be1c1b68cd1cb : SUCCESS in 17m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b72783ab1d1e462e89f8d8a0a1ab3fb1 : SUCCESS in 22m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6c330652685f4685a03b4f46e74cd0e4 : SUCCESS in 21m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fbe7bf969f3b408ebe6ab26e61968d3e : SUCCESS in 19m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/52b344e9d6b740ebb589eee6a3ebd785 : SUCCESS in 58m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e28094c871c642d59d873f8681cb009d : SUCCESS in 30m 25s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/902d953d2ab14fa2acbc80e87e833033 : SUCCESS in 26m 16s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/67588caf6f0145b793130ad021e7b0f4 : SUCCESS in 22m 09s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/674c4d6481514359af952161fdf5b054 : SUCCESS in 59m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/78a442182e294e69be3fb734e28f082f : SUCCESS in 6m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0396a6af8e34471b82d1a28299e87e1e : SUCCESS in 1h 16m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4371c5b4e2ad4242a8a62a62e9d4520d : SUCCESS in 56m 25s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/c9e129b1e7bd41928a4a35b1113f354d : SUCCESS in 45m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/444c7064352c4f0bb963b151a288ed4f : SUCCESS in 42m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/11996f59d5c54388a6475f12173ccbdc : SUCCESS in 38m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b93cf8410904703b5a31bc4aef73a0b : TIMED_OUT in 40m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f6d4a70761545e8a7a75198be4afe97 : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/15479210134644f0a5fcec4997893b78 : SUCCESS in 26m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbf650f9191b44cdb9a3910495bd9c64 : SUCCESS in 48m 31s","accounts_in_message":[],"_revision_number":2},{"id":"48c20096455c48cf4896cea1f609d50899d8d194","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-20 17:12:55.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"760979575ce5d4b82959ad595347b90530acf1bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 18:14:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bad8f7657bc4498b8e09457f68ab39c3 : SUCCESS in 12m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/49d0864331004990ab283a5a9fcf4517 : SUCCESS in 1h 00m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6a141e0d00c643f5bdd6e9992dc4cb46 : SUCCESS in 41m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0f8a53a9e95048e39a0c5ecf64b650e5 : SUCCESS in 8m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5ab62666dec649a8953bcf2ef571c1d7 : SUCCESS in 6m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ed319dae2935419998c2c070455eea39 : SUCCESS in 16m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d995634ffd5f48468844d5d36a6bffcf : SUCCESS in 15m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fb008a83d861444ea9fe794d482374b6 : SUCCESS in 20m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c15702aa1c13419a99a227dd7cd2b40e : SUCCESS in 15m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/70bb9a39836046f8842ee61f4e9f21ad : SUCCESS in 17m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d860669441414df3999e172761195821 : SUCCESS in 18m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ca3063a3101a4d498c7bd5e866cfc0ac : SUCCESS in 20m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8a62ad731dd04a99a9ecb389f7cebe0f : SUCCESS in 22m 30s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8952f951b1cb4db9a65e4e74c052330a : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8974a18bc0a94a5bae7af87487eadbc4 : SUCCESS in 16m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/be3aa3f79ee944d3bbbb1bf11eeb34bd : SUCCESS in 19m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/95e347b57f2d424db2bc8a71dde5100c : SUCCESS in 5m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2b571e716a9a432b950011078ac16abe : SUCCESS in 26m 30s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0a8eac04c7f143a0b7f49069b0ce4a04 : SUCCESS in 36m 17s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/6837c44ceedc4c1e9662014a132e3be3 : SUCCESS in 48m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2dd02f4948404dd98fa087654ecdbf57 : SUCCESS in 31m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/614bfcbbd14144d6b6d0cee9f1e93c4e : SUCCESS in 40m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5238def9c2684902addeeb1cf9981a09 : SUCCESS in 18m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c9394418f454a37aee45302b6721bea : SUCCESS in 5m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4d33a6baa9284d0e8acd4e50694161d5 : SUCCESS in 19m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9baf8399fbc54af69cc8ba801a5ae7a4 : SUCCESS in 53m 26s","accounts_in_message":[],"_revision_number":2},{"id":"0677e485337d56c55da9415f6a76fb1b26868b33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 18:14:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ead8a03f4225025ca7fdbee5dc2273fa37fcd6e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 19:16:29.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8e774858cc646f9932c26f5664f519d : SUCCESS in 7m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/c96a4da442a94ff8bb3ca75717f224eb : SUCCESS in 1h 00m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/88dc67bb8b394cad8c78a9e255d89af8 : SUCCESS in 44m 54s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b11f154e73974d96be497b8023dabb16 : SUCCESS in 8m 52s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/a737172253064e0d8baede64874908f7 : SUCCESS in 7m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0100e9779b79453daaa0c5572626c817 : SUCCESS in 15m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5661f3ee39444f81b6b6d620e3b0d2ef : SUCCESS in 16m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/be242fd8dd494ebc9a4cd6f1f8bfeae6 : SUCCESS in 18m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a4db1d70160d4b95ac7a42eecaef0ba4 : SUCCESS in 15m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1d6cba9a733d4d86818bc8cbf1d9c42e : SUCCESS in 18m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6c8d1a5751ec419eaec05dd6597bd2da : SUCCESS in 17m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4e292c2cf5c44d4baa3f9b6587316b65 : SUCCESS in 20m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ba62937f581347a5b3c70db9a76c03cd : SUCCESS in 21m 28s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5e66751344a54006a927522606bdc96f : SUCCESS in 16m 10s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/59117e003c2c402f8a607ba858e5a37a : SUCCESS in 21m 12s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9a76263600294722b55db23a069c12d9 : SUCCESS in 18m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d9d2977e0aca40c2b88af0bd66f554e1 : SUCCESS in 5m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/78b0a7230b144c908eca929de23ac9bf : SUCCESS in 52m 31s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/343a92cccc464850b30e6b3f9c86d913 : SUCCESS in 46m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2bce5e36ee54487abf69b52177f56728 : SUCCESS in 43m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/70ccb94223224695a19e6791de61c742 : SUCCESS in 33m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4edf5dd23bd4c289f55c6358e18e83c : SUCCESS in 15m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9c2bf423657408cae58d346c87de402 : SUCCESS in 6m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/49067f5683c941ae90311cdf1c5b0e7e : FAILURE in 12m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5d3125f665f4d65a80d51198b4f5f10 : SUCCESS in 56m 55s","accounts_in_message":[],"_revision_number":2},{"id":"7a98d4cc121a382bbb7001df71ea2d413afa9d05","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-20 20:54:15.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b91f8e22dadccdcb48ff21469d5c50e1578c9652","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 22:07:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5212d4c7e5854e39abf37c14ee75cb33 : SUCCESS in 9m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/429246f6d5c84c7aa1dd01a3e51d7980 : SUCCESS in 58m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f620b7e2d8ca48269bf229eee92a4627 : SUCCESS in 40m 23s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/164255c105584eb7970fa7bc87e6dea3 : SUCCESS in 9m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/34e680216e06418b9bea89a4127b093d : SUCCESS in 7m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/380e155380b449b88ba08fc721852a6b : SUCCESS in 15m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/508947b7ae5244cbbfc8936ade15d021 : SUCCESS in 15m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7e7747c4bd384fa78b6b81ed415db96e : SUCCESS in 14m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/56cd418816504c448976272b429f3f1a : SUCCESS in 14m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7e525a6015124a4d935ede5235d2e359 : SUCCESS in 14m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/91d5a705c73e4aefb45c5d746571e570 : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eae88eea602749468df5c4c729e70dfc : SUCCESS in 16m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9c3de82985f8486996eacaec06da0386 : SUCCESS in 19m 14s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/bf3a1f4d684b442e9c7071d2ed0e5cf2 : SUCCESS in 17m 54s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/59a27c17b6fb43da98ebb1fb2083af93 : SUCCESS in 17m 37s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/32241c78bd9b456b80c006106b56816a : SUCCESS in 20m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d34064bae9db4e26b7e44b22b3cb659d : SUCCESS in 5m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d557232ee90640ccb06b16d5d48fed2a : SUCCESS in 22m 02s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9f35ce4881d24dc9a97c72a96343df3c : SUCCESS in 47m 44s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/9c02bbcf7d43409ab3ba542d7f7b59fc : SUCCESS in 48m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2fbafec23b754fecb28f8b8cccbb9380 : SUCCESS in 43m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ee82912d772b443f9e4ec5133907b288 : SUCCESS in 45m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4190c7509e1d481aac5df3ce1ed17889 : SUCCESS in 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59ebb33bdbe845f29e066ab8f90b977b : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/20cc6d61a64544bfa49e67cd8d223de3 : SUCCESS in 22m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9437ae65f1d45beb61fb7dea28023d6 : SUCCESS in 1h 06m 48s","accounts_in_message":[],"_revision_number":2},{"id":"ca99dd009bddefb786ffd64fe5bf039f90bd2513","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 22:53:33.000000000","message":"Patch Set 2: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"f96458d6771c51ab11da0a01c2180739806327a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-21 03:53:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"51d24f8c0c7cec39ddbf9f4b1f97028ad1b886d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-21 04:54:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71e57ab57024498f88de3ef3e0f72b22 : SUCCESS in 8m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/65690d26eadd4fdabb550106139b926b : SUCCESS in 59m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f9752891ec0d43839fcb493c6edb7836 : SUCCESS in 38m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3a68e9f74f394b3f9a4816548fec131a : SUCCESS in 9m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3f36f8212d6242a68f04e98c50bfb789 : SUCCESS in 6m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cffe10c1ec0c446eb86498eb9bb1eddc : SUCCESS in 13m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7f77aac75450465a8c09d2cd5b216be8 : SUCCESS in 14m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/25e7ae0d53424772b8881c294ccb4994 : SUCCESS in 13m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/52cd5a6df97b499f8833907d3ce382d1 : SUCCESS in 14m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/21c5d446606f4872b9db244b1a5a28e5 : SUCCESS in 15m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9cb0c773302b43aeb69cf57036e4721b : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4a5bef2ed7d546b2941f97223794debe : FAILURE in 18m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/40305369d0cc43bc9d2e5e3397b38bbb : SUCCESS in 17m 03s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5569eacef1b44aea9ccd79cff169eb7f : SUCCESS in 17m 44s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/35dc698daa3a48b0b3c1c5e5892fbfff : SUCCESS in 17m 52s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3cc40ffd8c754022895f6371a1cfa99b : SUCCESS in 20m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d5da33ce43c84262a40154ef4aef72dc : SUCCESS in 6m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3bacb23b22cc4c30a2206c24cf011655 : SUCCESS in 22m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4dce80f895104699bed8193cf9b333dc : SUCCESS in 39m 59s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/9654e91df028439483099d5ea8a33041 : SUCCESS in 47m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/14f51fa847664b60bcf07b015af0427e : SUCCESS in 31m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0891e45b62fc484cbf90d73ea2ca6580 : SUCCESS in 38m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f939cc44d58241239955c0aaf83c862f : SUCCESS in 13m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91c7fe260a9745ebb624603258c6999e : SUCCESS in 6m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6f104506d313433a9a3e81d8efe44877 : SUCCESS in 22m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c549a81cce034957b0352f2c3315f28e : SUCCESS in 53m 36s","accounts_in_message":[],"_revision_number":3},{"id":"2f1a2190010824156fcc4cf0d9b1e8088ff9ced8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-22 04:35:36.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"00102f5f532205bdf3959cbfe32e6702b756e17a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 05:53:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6a773ff8b244e56aaf54cbdfa73bc65 : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/547534c3462d411c99001cc218c78064 : SUCCESS in 1h 12m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/42d950100859473992942ed58d5af7dd : SUCCESS in 52m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f7e94eaca9a74910a1b90fe666453e74 : SUCCESS in 9m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b3bdac143e6f4fd1b1ede16996dc0ca1 : SUCCESS in 6m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/58eee12e11574beda480a5dd2a8bfe56 : SUCCESS in 15m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6311732c7ce34a0dbd19826bc0ef8a36 : SUCCESS in 14m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b14b1b21c1b44282adaca87b97b4ce00 : SUCCESS in 13m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e9d7153969d64122afdd6db5be0654e6 : SUCCESS in 14m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/72d4e627f1d04bd3ad9d76950d75a14e : SUCCESS in 13m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/30bce745b2e845ff9fbb19b624a709e5 : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/49bbd432bd3a4637bf557e8c55d207ba : SUCCESS in 16m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e23ffdc22ff34cacbd64a2dc7919454d : SUCCESS in 19m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/91c934c0f22f4ef49cb2786d9af93dce : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4b961be8763943f1ba7ec701735366a9 : SUCCESS in 16m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/21b6150a0f1143df80077d36d7c3607a : SUCCESS in 17m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/370c5c40cfb644fab5a122ce587beaf6 : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c2c1a8db80be412e8a50eaec4e57d0b8 : SUCCESS in 22m 30s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cc5e4ad04cea47588386d833c7c1528c : SUCCESS in 44m 50s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/dd4d82dd204243cebba84adbcf786602 : SUCCESS in 43m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4970933879ff4d13b9536de49e773cf2 : SUCCESS in 36m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7a56428f7ea241c38a3d6025a25e1cdc : SUCCESS in 36m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10e1ba5facc744b780179b67d6b831df : SUCCESS in 14m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98dde79750304ed68a1598a3bf5ba2c5 : SUCCESS in 5m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2e3196344d6d46f6ad3183bcabadbef1 : SUCCESS in 21m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a20988eb8d6477e8ecfff48ac9d3570 : SUCCESS in 57m 06s","accounts_in_message":[],"_revision_number":3},{"id":"2eaeb0d0aad53e234f1ae1de41c719e23bad369e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-22 10:34:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"29459546e0ebaaa09619084442e09501b8206d74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 10:38:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b4fe2eca1c3cac6066a7ba2669749521480d4447","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 14:16:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65589d02e76d4b5188292120ff85d97d : SUCCESS in 9m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/439fa108e07140f0b9e0cb43016b77e1 : SUCCESS in 1h 12m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/789f6149b806475c933c451857be00ce : SUCCESS in 47m 38s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/08b2c83042584fefa48acd6caf9db90f : SUCCESS in 9m 53s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/82aede434dd646c8993b5b30dc1b7a5b : SUCCESS in 11m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/49aa312a6ab24eaf882fc8a60246e070 : SUCCESS in 15m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0056f39a45154793a938d1adf1b1d244 : SUCCESS in 17m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c7a50bfa1ceb4656a014b35cd86dab80 : SUCCESS in 16m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/106659e38f0d495593b77e35e4185c7e : SUCCESS in 15m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/74de6d16993f488381b0df8e1fd221dc : SUCCESS in 16m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/24cc48f36ed04c6287e7583d188e8760 : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e7936bbc07ea4f3ea048405ea3b8b31c : SUCCESS in 18m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dddc3c42e21649f79a059a1d77e08916 : SUCCESS in 20m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8b6ae394e5754fb4a1a9dc0476271fd5 : SUCCESS in 52m 30s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4b82ee5f17dd4b01b6fecec569e3d33f : SUCCESS in 16m 37s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d17ea51be24e45418c5f14fa2c07487e : SUCCESS in 19m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a36f894edb1c47ad953f1fa4b1581f88 : SUCCESS in 6m 05s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1dcdd76cc8bf4495839f782b36f58a94 : SUCCESS in 52m 22s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/74284d58129841e3afc68e5de8b5ecaf : SUCCESS in 48m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1b46d0726d1243b9a24eedeff21d09a9 : SUCCESS in 40m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/465ffc5ad214421d8f0e52524a31ee90 : SUCCESS in 1h 07m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b2555218ced48c7b69db7b6c5354f8f : SUCCESS in 15m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f791d92b49784e53bc6614c19a400ea0 : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/145a83d2be3e491b98d81d78b09225f2 : SUCCESS in 33m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc864a430f4642de9392495e5d7a0c10 : SUCCESS in 52m 42s","accounts_in_message":[],"_revision_number":3},{"id":"002c82b2ae009399efb4371963a107ac9daf99cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 14:16:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"bba85b86128ba20c5d6a49b1d2733fc5ea73c135","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 14:18:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/793f424491a443aaad340a83850911c5 : SUCCESS in 1m 13s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/128efebae07a49c6a69c380522ddfb76 : SUCCESS in 52s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/82bbb94c647a42f7b5cea8270d4b1c39 : SUCCESS in 1m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f361583436fdeeba372037d25b90d8227b72c779","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-03-22 15:36:21.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nI just noticed this refactor included some cleanup that conflicts when I tried to add https://review.opendev.org/c/openstack/swift/+/782281 ontop of https://review.opendev.org/c/openstack/swift/+/781550","accounts_in_message":[],"_revision_number":3},{"id":"425bbe4722c4194b1bf0f772eb2fa98a85ce0ccd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-22 15:41:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7e278291f88afe1ac78d930404bd4dbbfdb4f53a","revisions":{"8af6e69e94abc4d8eb3a505cf27b9e0fd5f29747":{"kind":"REWORK","_number":1,"created":"2021-03-18 22:20:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/76/781576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/781576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/781576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/781576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/781576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/781576/1"}}},"commit":{"parents":[{"commit":"0b129509c93f9190aeb375ef503ac24c8852000d","subject":"relinker: refactor to share common code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b129509c93f9190aeb375ef503ac24c8852000d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-18 22:20:43.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-18 22:20:43.000000000","tz":-420},"subject":"relinker: Move filters \u0026 hooks to be methods","message":"relinker: Move filters \u0026 hooks to be methods\n\nShouldn\u0027t be any functional changes, though.\n\nChange-Id: If906513ed47f203ca97ef08d4375b01c76f812bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8af6e69e94abc4d8eb3a505cf27b9e0fd5f29747"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8af6e69e94abc4d8eb3a505cf27b9e0fd5f29747"}]},"branch":"refs/heads/master"},"96e5363f639e3bb93cb709fe3dfd9a191e53bf0a":{"kind":"REWORK","_number":2,"created":"2021-03-20 03:22:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/76/781576/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/781576/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/781576/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/781576/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/781576/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/781576/2"}}},"commit":{"parents":[{"commit":"8b07e5e732fe8e2187711fe303b5386ba6a38531","subject":"Merge \"relinker: add --policy option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b07e5e732fe8e2187711fe303b5386ba6a38531"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-18 22:20:43.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-20 03:22:08.000000000","tz":-420},"subject":"relinker: Move filters \u0026 hooks to be methods","message":"relinker: Move filters \u0026 hooks to be methods\n\nShouldn\u0027t be any functional changes, though.\n\nChange-Id: If906513ed47f203ca97ef08d4375b01c76f812bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96e5363f639e3bb93cb709fe3dfd9a191e53bf0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96e5363f639e3bb93cb709fe3dfd9a191e53bf0a"}]},"branch":"refs/heads/master"},"7e278291f88afe1ac78d930404bd4dbbfdb4f53a":{"kind":"REWORK","_number":3,"created":"2021-03-21 03:53:56.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/76/781576/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/781576/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/781576/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/781576/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/781576/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/781576/3"}}},"commit":{"parents":[{"commit":"f8265c18bfff0ce268a1b091cacf50a68e59b38c","subject":"Merge \"relinker: make relink step a subset of cleanup step\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8265c18bfff0ce268a1b091cacf50a68e59b38c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-18 22:20:43.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-21 03:53:40.000000000","tz":-420},"subject":"relinker: Move filters \u0026 hooks to be methods","message":"relinker: Move filters \u0026 hooks to be methods\n\nShouldn\u0027t be any functional changes, though.\n\nChange-Id: If906513ed47f203ca97ef08d4375b01c76f812bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e278291f88afe1ac78d930404bd4dbbfdb4f53a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e278291f88afe1ac78d930404bd4dbbfdb4f53a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
