)]}'
{"id":"openstack%2Fswift~781584","triplet_id":"openstack%2Fswift~master~I9a1ebec645ced755aece8e28452f51488c2b776c","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I9a1ebec645ced755aece8e28452f51488c2b776c","subject":"Don\u0027t require swift be installed to have passing manager tests","status":"MERGED","created":"2021-03-18 23:35:22.000000000","updated":"2021-03-22 07:38:32.000000000","submitted":"2021-03-19 19:00:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781584","meta_rev_id":"c43d9c004b1df48bb0b9a06ff8d6f61ddeda2e15","_number":781584,"virtual_id_number":781584,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-19 18:59:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2021-03-19 11:31:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-22 07:38:32.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2021-03-19 11:31:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 00:18:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-19 11:31:52.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2021-03-22 07:38:32.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"REVIEWER"}],"messages":[{"id":"8aa0468d2dd3c1f7b23668f767a0efd5ef935aeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-18 23:35:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4510d0e03978e1aab1426350708aca57c01d6fbe","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-18 23:38:22.000000000","message":"Patch Set 1:\n\nFWIW, I could repro the failures zigo was seeing by running something like\n\n PATH\u003d /usr/local/bin/nosetests test/unit/common/test_manager.py","accounts_in_message":[],"_revision_number":1},{"id":"defc8c0246ddec3ab688b182b98811473f5de6bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 00:18:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bd2ce01f42449c5af4f1aab146ddc2a : SUCCESS in 9m 09s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/857098cd0ad1406b898cef983d9939b8 : SUCCESS in 11m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/10d7bf2038fe4615b503e5f419fba748 : SUCCESS in 6m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/927fe2053f5f4a1da7a1a1d38606c78a : SUCCESS in 17m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f7ece1dcf6424b129d5d1ff638f3c9cb : SUCCESS in 15m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3cc52ea46e5647189f6e042983421d7c : SUCCESS in 20m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8f1ed281e02f4e66ad1b13a5d10c3338 : SUCCESS in 20m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c92535c9a8df4cbaba1ed3cdd70ad7f7 : SUCCESS in 15m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ed13ec7741964727944b1a23e414cd00 : SUCCESS in 22m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/30ea661be5e44b338db2c000bcbb5bdc : SUCCESS in 19m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a73195056d0f4a3ab5973b3e42697119 : SUCCESS in 22m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/343a88071f6949998aff63211f171af2 : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/eda386787ecd49a3a9094f1c614fdfd8 : SUCCESS in 17m 09s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6db24a84fa674e66af3cba235183007c : SUCCESS in 18m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/48889d7bfa1446db9b6001fe0434ba46 : SUCCESS in 39m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/265e0db3b4834f2db26c99693c5044cc : SUCCESS in 37m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e1e86b3521e451eb8e26f2bb89e8663 : SUCCESS in 14m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27dab7f5d0154a37bbac5ee8e423df17 : SUCCESS in 5m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0cd489f59df74421be85757a5e06ce61 : SUCCESS in 22m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a1e5fc9bdabf04aa2bec48c11a16ca490536d2a1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-19 11:31:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nGreat!\n\nFor me the repro was to rm swift-proxy-server from my venv bin dir","accounts_in_message":[],"_revision_number":1},{"id":"505b678f4f90924426d57239788a3845fb4577d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 11:32:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1221afe0f9263d4909c41d973b1c7c5f782aec53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 18:59:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1722645e7ff24eb699cd2f2e69d20367 : SUCCESS in 10m 36s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c6bf25f231cf45a3a55014ecb5ba49eb : SUCCESS in 11m 44s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/5d7f2faeebd14b1e80b2d27ff61738fd : SUCCESS in 9m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f74ad1dc2ca94739b197555a48f36822 : SUCCESS in 13m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3e5fd1a7e8fd4eff9b0a1774ef71711c : SUCCESS in 14m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7898df5ac01b4c47a047c5982bd01a8e : SUCCESS in 20m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/16a6fa9ed27844c79df85f19db232532 : SUCCESS in 14m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c482d7adff5543d2804d2eec11fb320f : SUCCESS in 19m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c69b4746e58d4f48b8eeb3b3d13f86f8 : SUCCESS in 21m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bd50cda84759448aadb01ef26adec1fb : SUCCESS in 17m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1ac6ecf5d31147f08265b4ac5cb5131c : SUCCESS in 21m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/30689df8317146e8892e6d85286a16e2 : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f20f83f22e6e4b1a8fdd9858367b506c : SUCCESS in 22m 01s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/37747c813fbb43c99d8c47c9ed0ddb24 : SUCCESS in 22m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/99885455b2534da3906292297e5c2ad9 : SUCCESS in 7m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e9bf5c91a01d463b84e5a3a6d1c1bf3b : SUCCESS in 41m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/60b90e6f04d549e191c91ce47e11d18a : SUCCESS in 44m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8e8bf39846140b4a415dfa0921cd075 : SUCCESS in 15m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0ffaf29bcb34c698205e38d7309a063 : SUCCESS in 8m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9adf4f452e784f77837d6ccd3e5bc030 : SUCCESS in 29m 14s","accounts_in_message":[],"_revision_number":1},{"id":"a9b1bb6ab225b0cdf8f823a53c39dd215c494104","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 19:00:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5ba0d03aceef346f4847c2afaa4dfbd0ce3309d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 19:05:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00b4a2d6cc9e4473ae45dc399c85517c : SUCCESS in 1m 36s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b8ca08bd7ac249dbafc234f33860351b : SUCCESS in 1m 22s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/10abc64c89a549238cdb5532dd9bdd8b : SUCCESS in 1m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c43d9c004b1df48bb0b9a06ff8d6f61ddeda2e15","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-03-22 07:38:32.000000000","message":"Patch Set 1: Code-Review+1\n\nThis fixes unit testing in Debian.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d54083771f90ef96b7dae4c4fc2a268b6b375394","revisions":{"d54083771f90ef96b7dae4c4fc2a268b6b375394":{"kind":"REWORK","_number":1,"created":"2021-03-18 23:35:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/84/781584/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/781584/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/781584/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/781584/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/781584/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/781584/1"}}},"commit":{"parents":[{"commit":"ec37893b82672ec35a20eea9fffa91f9d48026fd","subject":"Only test with \u0026-delimited query strings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec37893b82672ec35a20eea9fffa91f9d48026fd"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-18 23:35:05.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-18 23:35:05.000000000","tz":-420},"subject":"Don\u0027t require swift be installed to have passing manager tests","message":"Don\u0027t require swift be installed to have passing manager tests\n\nChange-Id: I9a1ebec645ced755aece8e28452f51488c2b776c\nRelated-Change: I09143ae20bd6249083c0b80cdaa9d561e6abb301\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d54083771f90ef96b7dae4c4fc2a268b6b375394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d54083771f90ef96b7dae4c4fc2a268b6b375394"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
