)]}'
{"id":"openstack%2Fswift~781586","triplet_id":"openstack%2Fswift~master~I04dc249365bbdb511b036c7f433778ff30285591","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I04dc249365bbdb511b036c7f433778ff30285591","subject":"gate: Make swift-multinode-rolling-upgrade voting again","status":"MERGED","created":"2021-03-18 23:48:02.000000000","updated":"2021-03-19 02:28:33.000000000","submitted":"2021-03-19 02:25:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781586","meta_rev_id":"1d2cab8fc5daaa9b6e267193510529b9f87c2c57","_number":781586,"virtual_id_number":781586,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-19 02:25:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-18 23:48:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-18 23:48:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 01:11:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1473480194c51c441afb79b3a5a305fed82b788b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-18 23:48:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73f8be90883e573c422f71ea67c5ecbc91c4acee","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-18 23:48:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nNew tag with the requirements fixes, so it actually passes again!","accounts_in_message":[],"_revision_number":1},{"id":"3134d0630f6333afb8c5408cf2bb534c78f10dd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 01:11:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7053db70df4e49a5990903cc9885f2de : SUCCESS in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/af01bb412de840aeaf4e7f225d1f9668 : SUCCESS in 1h 16m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5e93b3489e8c4e2b910de3cb434ab5e4 : SUCCESS in 47m 04s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b8999b8efc7e45f68ac3d7edd86f2495 : SUCCESS in 9m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/19607f2e7b904c0fafdc59b5951c652c : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d577eb01ae1341f3823590fab9a7d743 : SUCCESS in 15m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/493e2bd634e847c1abcfa9c40d7e3fea : SUCCESS in 18m 02s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3d579f2fc7314ddc89119d3000f77175 : SUCCESS in 14m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cc710098b96b4ecfaeac70b2c8431eee : SUCCESS in 15m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e9370869c9734cc3bf4528a2680e1a48 : SUCCESS in 15m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/32ce5dde80ad4505a4abac01a1a174c3 : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/124a5fdc7d624225be2e4d695ec596aa : SUCCESS in 18m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/731b0f011f674d9dbf93fb503d39c4a5 : SUCCESS in 16m 24s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ffc1be124bcb4b879da63dced12d4b29 : SUCCESS in 20m 39s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/320039c4fc8048358ec6341842ceb77e : SUCCESS in 23m 40s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/391110301caa4e66977668111968e079 : SUCCESS in 22m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/62446908462d44a584613a15219cd721 : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a0df4ad6a6794988adb89786ea26ad5b : FAILURE in 25m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/58da7eff669e40f981b005cfc378ffe6 : SUCCESS in 54m 34s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/e70a7ec900db49e2916d47dee94edf31 : SUCCESS in 49m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cadf102a12e14181b9bfd2c7467bde72 : SUCCESS in 41m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/418dbdeed71b4cb1935b0338d50a2b68 : SUCCESS in 39m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7fb16af4375e4f099c79b6b2858e43a1 : SUCCESS in 16m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f89ae80a26e84888a1e2048314739a81 : SUCCESS in 6m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/845222b4b3c74e4aa275b61d912e0858 : SUCCESS in 23m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a6c5e0d7f9845f6a5fdacdb0536cec0 : SUCCESS in 53m 36s","accounts_in_message":[],"_revision_number":1},{"id":"8b7999c6ffeb905752e8a77f78d33400308406cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 01:11:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9a9234f96c923578e49a8bd10b9cf2777b31631","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 02:25:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84dadb9c9a5e4aedaea4da55426e5869 : SUCCESS in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/036990ed81ab4888ae7d5f5d8ff60e1c : SUCCESS in 59m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/607b5447b51440a9bc30f110c9666bfe : SUCCESS in 38m 34s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4f2466da5cad458aa1436168ae3fa776 : SUCCESS in 8m 38s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/2d5f6a7efcd5461686c499db2ee5108d : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1b8251d011164a4d8542c1a2cb51e23e : SUCCESS in 16m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3d5191a6bdbb4d1dbe4530e1ac69f704 : SUCCESS in 18m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b4187d2702e04af3b989b5949693740d : SUCCESS in 18m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/599d567e43574b22bb30dcada2277e72 : SUCCESS in 14m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ac5cc4248d7747ce894c9c6df1910d7a : SUCCESS in 17m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/99c2ff8908d0414799020ec49e42fa80 : SUCCESS in 18m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6c3763e9341d4f329f4fb0c955beb4cc : SUCCESS in 29m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/83addc36592e43be9b37d150926f46d2 : SUCCESS in 18m 18s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7f5beb7dbc15446d88ff21b75b971d86 : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bd1fb2fa89f8470989c73ca4ca8ef3c2 : SUCCESS in 21m 04s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d87188190be047bba4cd6072f3534892 : SUCCESS in 22m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dcb7276ea7744fbcae78e39216b7c99f : SUCCESS in 6m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/503ba26d5d934062bfa431334f87b857 : SUCCESS in 48m 23s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/274bf7e5a630472591f768c3ce3a43e6 : SUCCESS in 45m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c8d4305a289546da81e6f945242aa626 : SUCCESS in 40m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0a9bbc69fa744d0186825ba7772b0ef7 : SUCCESS in 39m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6d11407fa754988a666d152241afbba : SUCCESS in 15m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db01594fdfec4e8b86b99d3b08338a20 : SUCCESS in 6m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aa8063819cf644339fa15bc923a658d4 : SUCCESS in 24m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f46ae6eca9e454388db3440a7ecfcee : SUCCESS in 57m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d427d8ba1d4882e034bbc98e6a26c5fbdb124ca4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 02:25:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1d2cab8fc5daaa9b6e267193510529b9f87c2c57","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 02:28:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f95a7bc4beb14c2aa983b625604c44e1 : SUCCESS in 1m 01s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6b8cb42940c0422984ad07923741812a : SUCCESS in 50s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/73bb4acd386244e1b1159271fd72a862 : SUCCESS in 38s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ab9acb03f2b86d7e49bfd565c36e6f49b192bf6","revisions":{"9ab9acb03f2b86d7e49bfd565c36e6f49b192bf6":{"kind":"REWORK","_number":1,"created":"2021-03-18 23:48:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/86/781586/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/86/781586/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/781586/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/781586/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/781586/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/86/781586/1"}}},"commit":{"parents":[{"commit":"c6a64036651e18f3d7d0715c5d4876257aaa0e6f","subject":"AUTHORS/CHANGELOG for 2.27.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6a64036651e18f3d7d0715c5d4876257aaa0e6f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-18 23:47:57.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-18 23:47:57.000000000","tz":-420},"subject":"gate: Make swift-multinode-rolling-upgrade voting again","message":"gate: Make swift-multinode-rolling-upgrade voting again\n\nChange-Id: I04dc249365bbdb511b036c7f433778ff30285591\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ab9acb03f2b86d7e49bfd565c36e6f49b192bf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ab9acb03f2b86d7e49bfd565c36e6f49b192bf6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
