)]}'
{"id":"openstack%2Fswift~781597","triplet_id":"openstack%2Fswift~master~I9919a9690db3d1682d36303b40226420bdd2c060","project":"openstack/swift","branch":"master","topic":"p-shard-repair","hashtags":[],"change_id":"I9919a9690db3d1682d36303b40226420bdd2c060","subject":"swift-manage-shard-ranges: sharded shard test + weighted object count","status":"NEW","created":"2021-03-19 05:44:23.000000000","updated":"2021-03-22 06:01:47.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"40ed8472f74570cb74a77de504ee7a8c937e8d12","_number":781597,"virtual_id_number":781597,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-22 06:01:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":-1,"date":"2021-03-22 04:54:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 06:33:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f0b8546ae84306375c6593b0d556eb6ee8c0b83b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-03-19 05:44:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3f2ee12accad1f6777a4b1ce58291060e3ed72c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 06:33:20.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e36f1df75424961a9456ea0b54f8fb4 : SUCCESS in 9m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1aced9a138e6499dbc3e581cb79d8567 : SUCCESS in 8m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a90c7c44f83e49c494492360e2aef738 : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9d1d2134c10345d2ad4918ce6b3a0c43 : SUCCESS in 17m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/534069bafc1b44f08b8d97594beb21d6 : SUCCESS in 17m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c40bb6d409204053aed8c48a1f56f104 : SUCCESS in 18m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7346286286824b74a5d5fdda7623d244 : SUCCESS in 15m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f12cf0a28bfa4a2ba3cdf21148591736 : SUCCESS in 16m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/33f909160f3c43b2b98cde9fa11f6bef : SUCCESS in 22m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/66fe6e971ff1433f9e5fdcc168918668 : SUCCESS in 18m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4712d8513dec41e7808b066c9cf728fd : SUCCESS in 17m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9165723b9a494a96b968985c5545aeac : SUCCESS in 17m 25s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f390b0230a044b0ca95a9777c500136c : SUCCESS in 20m 46s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1671bc9af974421ca967a255fd0940b8 : SUCCESS in 24m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4a557fc8a8074e18a471f6a089730af3 : SUCCESS in 41m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9a1506c250824222b0488bdaf757f4c1 : SUCCESS in 47m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a8a35911bc134271a5aa23b677984465 : SUCCESS in 14m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8be7cc90a8a4bab854826bc5363960a : FAILURE in 5m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bbe8689b2d51422c84405a2b68bd361c : SUCCESS in 26m 21s","accounts_in_message":[],"_revision_number":1},{"id":"5248eaa53ba5c122d2f885bd02dd567b8427fcec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-03-22 04:53:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"946221d4400824d5c5c5155814074eabca9cab42","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-03-22 04:54:43.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"40ed8472f74570cb74a77de504ee7a8c937e8d12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 06:01:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/345ed65145bb4032af01a4a688f66e38 : SUCCESS in 11m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/b39bc069033d48d788c41472db0373e2 : SUCCESS in 1h 00m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2f157d69545a4976810709483b04c0ac : SUCCESS in 43m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f4f7939bf59a41b2a876b98b205005af : SUCCESS in 8m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d2bd78ee36484c8782d58e95d95febe0 : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0960590d2e1b45648a1c2eb10e63ee6c : FAILURE in 19m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1553ef0ee96c41e6809924f919b625c6 : FAILURE in 19m 13s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/13b23db3c0c54e3b82e44c06d19994c9 : FAILURE in 13m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/727c2d03fd5b43459d4afc30213dd3c7 : FAILURE in 16m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0f09015349f74c33a4a8aeaa532ee088 : FAILURE in 13m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f956136a61964a809e6ace313555e3f7 : SUCCESS in 20m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8c1de9c02c7849d085ff0fc534d2c87d : SUCCESS in 20m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e0e0783519504d43b70a536db545fe6b : SUCCESS in 18m 49s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9ef5bc498f5e4653961006b0d82ef54f : SUCCESS in 21m 12s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3b1cf46573ae4650b1b115ba4a406eee : SUCCESS in 34m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/dd17d8fe6fda4a30acbec464ffd231b7 : SUCCESS in 20m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/678bf8eb83854c73a0916d707f648048 : SUCCESS in 6m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/83cbb100ded944f0865e4f205418ecc4 : FAILURE in 18m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/da33b21d86e94969b24e619fd0f88c8a : FAILURE in 48m 38s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/bdf3cba1779e4ef1b51f0577415e29fc : FAILURE in 1h 06m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/431eb64b2f4c4e219566fb196e835bd3 : SUCCESS in 40m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d1ce409b1d3c472686a2fbcc5797de98 : SUCCESS in 36m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/288b01e4fc4f49adb604ac3a459cbe87 : FAILURE in 15m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9701bfd6ca344526ac10748fedfb4b99 : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/35d01f50fe4e4c259e1039c5d9f56431 : SUCCESS in 28m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b841a305833452788962dfbe1dbe2f0 : SUCCESS in 53m 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e20f48675363aeaee141c03b763019c0ceba2c0e","revisions":{"c37594c9c8eb0fcc7dafc1421066ac65d7d98c91":{"kind":"REWORK","_number":1,"created":"2021-03-19 05:44:23.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/97/781597/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/781597/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/781597/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/781597/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/781597/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/781597/1"}}},"commit":{"parents":[{"commit":"a9757c879d4f7868a9e3e57ce720087af875f186","subject":"swift-manage-shard-ranges: add repair and analyze commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9757c879d4f7868a9e3e57ce720087af875f186"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-03-19 05:39:21.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-03-19 05:39:21.000000000","tz":660},"subject":"swift-manage-shard-ranges: Follow up sharded shard test","message":"swift-manage-shard-ranges: Follow up sharded shard test\n\nI know as it currently stands the repair doens\u0027t understand our\nthreshold limits for containers. But the test in this patch\nshows why we may need to think about adjusting the \"weight\" to take into\naccount our thresholds. As it shows that because of same object count\nand more timestamps in the sharded path, it actually wants to repair to\na pre-sharded state, which would mean we\u0027d need to shard it again.\n\nChange-Id: I9919a9690db3d1682d36303b40226420bdd2c060\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c37594c9c8eb0fcc7dafc1421066ac65d7d98c91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c37594c9c8eb0fcc7dafc1421066ac65d7d98c91"}]},"branch":"refs/heads/master"},"e20f48675363aeaee141c03b763019c0ceba2c0e":{"kind":"REWORK","_number":2,"created":"2021-03-22 04:53:51.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/97/781597/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/781597/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/781597/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/781597/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/781597/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/781597/2"}}},"commit":{"parents":[{"commit":"a9757c879d4f7868a9e3e57ce720087af875f186","subject":"swift-manage-shard-ranges: add repair and analyze commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a9757c879d4f7868a9e3e57ce720087af875f186"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-03-19 05:39:21.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-03-22 04:51:49.000000000","tz":660},"subject":"swift-manage-shard-ranges: sharded shard test + weighted object count","message":"swift-manage-shard-ranges: sharded shard test + weighted object count\n\nI know as it currently stands the repair doens\u0027t understand our\nthreshold limits for containers. But the test in this patch\nshows why we may need to think about adjusting the \"weight\" to take into\naccount our thresholds. As it shows that because of same object count\nand more timestamps in the sharded path, it actually wants to repair to\na pre-sharded state, which would mean we\u0027d need to shard it again.\n\nI\u0027ve gone added the weighted object count code I hacked together in the\nPOC last year to see how it\u0027ll look.\n\nChange-Id: I9919a9690db3d1682d36303b40226420bdd2c060\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e20f48675363aeaee141c03b763019c0ceba2c0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e20f48675363aeaee141c03b763019c0ceba2c0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
