)]}'
{"id":"openstack%2Fswift~781598","triplet_id":"openstack%2Fswift~master~I19f56a873d2461e3a9837c4fec75ea9a64081883","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I19f56a873d2461e3a9837c4fec75ea9a64081883","subject":"Rewrite get_partition_for_hash to use int","status":"ABANDONED","created":"2021-03-19 05:53:04.000000000","updated":"2023-11-28 17:49:12.000000000","total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"c7de2a26ec1884160ace83ba651abc321af550bb","_number":781598,"virtual_id_number":781598,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-19 17:25:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-03-29 20:27:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 07:10:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-19 14:51:53.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2021-03-26 20:27:25.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"}],"messages":[{"id":"efd26de8662607f3daf103f2d06f39240d9607c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-19 05:53:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96f3259acfafdae68502a5a031369ab51d9d0974","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 07:10:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68ba4fa0268140fdb99b8dbfe642ae73 : SUCCESS in 9m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/296e5c4ddd114a7ab0daffb81fd206eb : SUCCESS in 1h 11m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/da39c434aacf469c848f3c08b08bfc32 : SUCCESS in 50m 32s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/85f575984bc34cc7ab57ddeecd3bc24a : SUCCESS in 10m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0764d223ea8a4b6ca1cf211869ea8407 : SUCCESS in 6m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/51e7992fc37e4db18a1b2d47f6f1e07f : FAILURE in 21m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/87379d9a88114cc2898c455c7f2e1c8e : FAILURE in 20m 34s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e6b0fda6407b4adc83f7fa42d0b6feeb : FAILURE in 16m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5f8f9246452b478289e82522f7d10848 : FAILURE in 20m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ec7b9b8eb64441ddb4a4753ddd3e96d3 : FAILURE in 15m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1c57b9aa3b42492d9bff0799e5ae7724 : SUCCESS in 16m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4cbabd070f7d43599650bb80dffd97cd : SUCCESS in 19m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/386bedfce0b743598d3de299a4ba62ae : SUCCESS in 24m 01s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/eda8f9cba9884f2fa595ac7c20049f59 : SUCCESS in 21m 48s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/dc7e7a43c6c94f92a9bb81ad1aee71c0 : SUCCESS in 18m 27s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d7dd350144dd419ab83843134d5282cd : SUCCESS in 23m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5375f723d29d41e094bd2b6ddeb33759 : SUCCESS in 9m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8b0d05057c364698a989fc4de3b44f1a : FAILURE in 25m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9ea8f2ab34ab4967be573cda42c1ab2b : FAILURE in 1h 00m 26s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/a7b22d519228443b9c1e255e4ae38bad : FAILURE in 50m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bb071464b6b94a65ad61fac10c995071 : SUCCESS in 45m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f90e00ff93104fc2bfeec1ce070b4a72 : SUCCESS in 36m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a04de674fcd45c98d29ed0f1038b389 : FAILURE in 16m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da77aa7aabc841cca2af4e7d6e3585e0 : SUCCESS in 6m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7b8ed020ee77456cb03a68c4c4d94f7e : FAILURE in 35m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/642cf800514248dfbae5da40172ba3e8 : SUCCESS in 1h 08m 08s","accounts_in_message":[],"_revision_number":1},{"id":"a0a884f120fdcd1949b80b92ac77024a2b7bb83a","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-03-19 14:51:53.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nNot sure what\u0027s going on with py39\n\nFAIL: test_PUT_next_part_power_races_around_makedirs_enoent (test.unit.obj.test_server.TestObjectController)\n----------------------------------------------------------------------\nTraceback (most recent call last):\n  File \"/home/zuul/src/opendev.org/openstack/swift/test/unit/obj/test_server.py\", line 2968, in test_PUT_next_part_power_races_around_makedirs_enoent\n    make_request(ts_1)\n  File \"/home/zuul/src/opendev.org/openstack/swift/test/unit/obj/test_server.py\", line 2943, in make_request\n    self.assertEqual(resp.status_int, 201)\nAssertionError: 500 !\u003d 201\n-------------------- \u003e\u003e begin captured stdout \u003c\u003c ---------------------\ntest-object-controller ERROR: ERROR __call__ error with PUT /sda1/389/a/c/o : \nTraceback (most recent call last):\n  File \"/home/zuul/src/opendev.org/openstack/swift/swift/obj/server.py\", line 1338, in __call__\n    res \u003d getattr(self, req.method)(req)\n  File \"/home/zuul/src/opendev.org/openstack/swift/swift/common/utils.py\", line 2015, in _timing_stats\n    resp \u003d func(ctrl, *args, **kwargs)\n  File \"/home/zuul/src/opendev.org/openstack/swift/swift/obj/server.py\", line 1047, in PUT\n    writer.put(metadata)\n  File \"/home/zuul/src/opendev.org/openstack/swift/swift/obj/diskfile.py\", line 2941, in put\n    super(DiskFileWriter, self)._put(metadata, True)\n  File \"/home/zuul/src/opendev.org/openstack/swift/swift/obj/diskfile.py\", line 1912, in _put\n    tpool.execute(self._finalize_put, metadata, target_path, cleanup)\n  File \"/home/zuul/src/opendev.org/openstack/swift/test/unit/obj/test_server.py\", line 155, in \u003clambda\u003e\n    tpool.execute \u003d lambda f, *args, **kwargs: f(*args, **kwargs)\n  File \"/home/zuul/src/opendev.org/openstack/swift/swift/obj/diskfile.py\", line 1861, in _finalize_put\n    new_target_path \u003d replace_partition_in_path(\n  File \"/home/zuul/src/opendev.org/openstack/swift/swift/common/utils.py\", line 5833, in replace_partition_in_path\n    part \u003d get_partition_for_hash(path_components[offset + 4], part_power)\n  File \"/home/zuul/src/opendev.org/openstack/swift/swift/common/utils.py\", line 5811, in get_partition_for_hash\n    return int(hex_hash[:8], 16) \u003e\u003e (32 - part_power)\nTypeError: unsupported operand type(s) for -: \u0027int\u0027 and \u0027str\u0027\ntest-object-controller INFO: - - - [19/Mar/2021:06:21:43 +0000] \"PUT /sda1/389/a/c/o\" 500 1480 \"-\" \"-\" \"-\" 0.0056 \"-\" 8375 0","accounts_in_message":[],"_revision_number":1},{"id":"ebcf1373324c6b648ab02a84d83d61edf154d751","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-19 15:39:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"06aa66e96257d2febc503c9f22c666b7ceb0823c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 17:25:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/335884e1cf154d93be56de3047495675 : SUCCESS in 9m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/882c64f11f1943c5afc75e2e7dce75dc : SUCCESS in 1h 02m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/015dc8c2fba147e48b49443c7e7013c6 : SUCCESS in 54m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/03c4bede4be0490fbd31474cf0ffc5c6 : SUCCESS in 9m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9a32fd66a8ea4ab988a3fcac95d4e93a : SUCCESS in 9m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/068c84e55c4445c7becf41da9023b4d4 : SUCCESS in 16m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6aaa4ec5f473414bbc8f4a70cc204eaa : SUCCESS in 17m 08s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b0b90e710ee14740a87d105fccea56f4 : SUCCESS in 18m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d896b044223e4532afa0970f16dd3cf7 : SUCCESS in 19m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/69237026703748599e21dff15c060b26 : SUCCESS in 16m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0c55b800d2fd4b3daf6f3a2b70938c5d : SUCCESS in 17m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d51862435bb14c249c8e4d4345269904 : SUCCESS in 21m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c3b7ba0ad0d944fdad71d2af1a5c4f64 : SUCCESS in 20m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/67a85dbcb38e4f5f9fb1f7795743d73f : SUCCESS in 21m 55s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1b8feebb391740f39d91eff6fa4bce1b : SUCCESS in 19m 00s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/70cd3e0dac3b447bbbaa3c9b82ea1062 : SUCCESS in 21m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/534916a5c49b4e89b7f400f1bca406ba : SUCCESS in 7m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a72af98e159f40148325016276cef6b3 : FAILURE in 26m 15s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fcbd01048a4f47beb6f7f98f410dbdb3 : SUCCESS in 55m 03s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/e3a7a64a2c6e45c0913026e3950d6298 : SUCCESS in 53m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/07bea7518ebb4c0ba462134eeb464479 : SUCCESS in 42m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eed772ca1bce41fe84de718eadaa4923 : SUCCESS in 41m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b16f920450241ca991f4a1fab4f05e6 : SUCCESS in 16m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef7afecf4b20456b8c3aa9d77239507f : SUCCESS in 6m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4cbdd13f71ef4965b99b5868dfcb78eb : SUCCESS in 27m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18c2cd499f0a43e19a671ddc6cf61bdf : SUCCESS in 1h 02m 00s","accounts_in_message":[],"_revision_number":2},{"id":"33d317d10faee0568a03df189d532bdc4aa151ce","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-03-26 20:27:25.000000000","message":"Patch Set 2:\n\nThis looks okay, but I\u0027m curious how does it crop up?","accounts_in_message":[],"_revision_number":2},{"id":"dd767005161d65146852d1d398d1e1cc8cb2f562","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-29 20:27:02.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\n\u003e This looks okay, but I\u0027m curious how does it crop up?\n\nIt was mostly a gut-reaction when I was reviewing the refactor at https://review.opendev.org/c/openstack/swift/+/778540 -- but maybe the current dance *is* best...","accounts_in_message":[],"_revision_number":2},{"id":"c7de2a26ec1884160ace83ba651abc321af550bb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-11-28 17:49:12.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"44e7e12504e1aa74a9d27fa8e3d7d9141cc99672","revisions":{"e92cdedc3417f99e31e63a265b499dc2da96caf8":{"kind":"REWORK","_number":1,"created":"2021-03-19 05:53:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/781598/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/781598/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/781598/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/781598/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/781598/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/781598/1"}}},"commit":{"parents":[{"commit":"04154cff636ad36d67e7c654e8f3625e385a8844","subject":"Merge \"gate: Make swift-multinode-rolling-upgrade voting again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04154cff636ad36d67e7c654e8f3625e385a8844"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-19 05:52:49.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-19 05:52:49.000000000","tz":-420},"subject":"Rewrite get_partition_for_hash to use int","message":"Rewrite get_partition_for_hash to use int\n\nChange-Id: I19f56a873d2461e3a9837c4fec75ea9a64081883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e92cdedc3417f99e31e63a265b499dc2da96caf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e92cdedc3417f99e31e63a265b499dc2da96caf8"}]},"branch":"refs/heads/master"},"44e7e12504e1aa74a9d27fa8e3d7d9141cc99672":{"kind":"REWORK","_number":2,"created":"2021-03-19 15:39:35.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/98/781598/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/781598/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/781598/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/781598/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/781598/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/781598/2"}}},"commit":{"parents":[{"commit":"04154cff636ad36d67e7c654e8f3625e385a8844","subject":"Merge \"gate: Make swift-multinode-rolling-upgrade voting again\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04154cff636ad36d67e7c654e8f3625e385a8844"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-19 05:52:49.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-19 15:39:03.000000000","tz":-420},"subject":"Rewrite get_partition_for_hash to use int","message":"Rewrite get_partition_for_hash to use int\n\nChange-Id: I19f56a873d2461e3a9837c4fec75ea9a64081883\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44e7e12504e1aa74a9d27fa8e3d7d9141cc99672"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44e7e12504e1aa74a9d27fa8e3d7d9141cc99672"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
