)]}'
{"id":"openstack%2Fswift~781819","triplet_id":"openstack%2Fswift~stable%2Fwallaby~I4510a126d8e5501d2ff977c8b1d9027b446feb96","project":"openstack/swift","branch":"stable/wallaby","hashtags":[],"change_id":"I4510a126d8e5501d2ff977c8b1d9027b446feb96","subject":"Only test with \u0026-delimited query strings","status":"MERGED","created":"2021-03-19 16:43:05.000000000","updated":"2021-03-20 03:01:21.000000000","submitted":"2021-03-20 02:59:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781819","meta_rev_id":"dd49e57fb1c1deb8f3f2a783e134c0fe12c22f0a","_number":781819,"virtual_id_number":781819,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-20 02:59:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-19 19:16:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-19 19:16:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 18:26:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"efb6c88032b4546f097bc309ea2404de78ea7619","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-19 16:43:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4922f9eede99fc311a629f7903b8c8b254e86f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 18:26:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f69234cf6924e94987c00c508ada96a : SUCCESS in 8m 10s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/eec0da713f7e4f7884bf9ed34e21f594 : SUCCESS in 12m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/26a00474cb8842a48c4dc5537f85feaa : SUCCESS in 8m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a643708a010842d9a162974781f8ea63 : SUCCESS in 14m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3647fc7bacce45fb8226ff34d25002e6 : SUCCESS in 14m 30s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3eeea9b253aa445781350d6270ef7386 : SUCCESS in 16m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dfe499b199434224a4e5ee48fa712c82 : SUCCESS in 22m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/35567fffcdd84d4b88681c5f5c25263c : SUCCESS in 24m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a2c15a8309a0485f93c637b651c362e4 : SUCCESS in 19m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c4f1d61992534c069b37965c38cb4575 : SUCCESS in 18m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f1a1c179f89846ad8455b19ae9403a56 : SUCCESS in 23m 36s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/52490aded54d4574850d9101b49fadf1 : SUCCESS in 22m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/668e576444724d4d86e4c1bc546aa35d : SUCCESS in 20m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2c6037cd699f4db7bdce9f77b6d58f46 : SUCCESS in 19m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b27c266066347168148a5645e88d588 : SUCCESS in 14m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a5e8b231ba34c92862b361a21889dc5 : SUCCESS in 6m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4c9fe41b6e914eff94495294fbf858b7 : SUCCESS in 25m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"145afb0f5ab0acf32a5e069e24d196b2186d9ce2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-19 19:16:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nOnly a matter of time until our stable gate breaks without this.","accounts_in_message":[],"_revision_number":1},{"id":"05b3b6b41c4c8d6cb7bc8ce90d5ac21c5d9e7e7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 21:00:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2f2619d5c0615bd21e137572db24199c5900e3ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 02:59:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3327f99d40824981b59d4f49d3d312e7 : SUCCESS in 8m 57s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d97daf41dcf847ab88e867be4068d183 : SUCCESS in 8m 42s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/878200210336457eb3a60a81c55a0bd4 : SUCCESS in 6m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ca31fd3fdd2840368a3bb74a0ac9dd80 : SUCCESS in 26m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d004c5eaa9b344cc99f14fe724e91b86 : SUCCESS in 23m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/463091a55a9b4acfa9daa5ee437a411c : SUCCESS in 16m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ef8b3b8815244b85ae9da86d8656c00a : SUCCESS in 21m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ce8d4fb169d14f21b865eff9f2dd9fe8 : SUCCESS in 17m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/eea6abb3f6a44fdb960bcbb9b3f08c3e : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aeb8fa8ffbe04fc0af7c9494821f45de : SUCCESS in 18m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1daff8eeda0548d68a1c49870f449772 : SUCCESS in 24m 33s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d57d5793284347259865de732791cb95 : SUCCESS in 19m 39s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/88968eff31a043b2a470bb6d044fa351 : SUCCESS in 29m 11s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0f5b12dbc1164a15964be6e9c45b229e : SUCCESS in 23m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1a384c34ca9e4d2d946022a1ab87b1b1 : SUCCESS in 10m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77ed1d684b4c4852ba8cd5bd0eed9d1f : SUCCESS in 16m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28933a45ce59460fb302509ff8432548 : SUCCESS in 8m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6b64b7f3ee767acd94e9107c93b01222ad67febb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 02:59:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dd49e57fb1c1deb8f3f2a783e134c0fe12c22f0a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 03:01:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d53431e14cc405ea0305e8207a6b2f2 : SUCCESS in 54s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a947cbd4d5634fa8a8a33dbc6e8a0afc : SUCCESS in 38s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/320e67c9a7024708a1ac3f2f5042c74d : FAILURE in 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"223bdd73677ef7313e1511b710fbc531067db8cd","revisions":{"223bdd73677ef7313e1511b710fbc531067db8cd":{"kind":"REWORK","_number":1,"created":"2021-03-19 16:43:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/19/781819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/781819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/781819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/781819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/781819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/781819/1"}}},"commit":{"parents":[{"commit":"d47724e403448bb7e3e02b2df89a7876c20bc4c1","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d47724e403448bb7e3e02b2df89a7876c20bc4c1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-18 23:23:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-19 16:41:37.000000000","tz":-420},"subject":"Only test with \u0026-delimited query strings","message":"Only test with \u0026-delimited query strings\n\nChange-Id: I4510a126d8e5501d2ff977c8b1d9027b446feb96\nPartial-Bug: 1920080\n(cherry picked from commit ec37893b82672ec35a20eea9fffa91f9d48026fd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/223bdd73677ef7313e1511b710fbc531067db8cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/223bdd73677ef7313e1511b710fbc531067db8cd"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
