)]}'
{"id":"openstack%2Fswift~781820","triplet_id":"openstack%2Fswift~stable%2Fwallaby~I9a1ebec645ced755aece8e28452f51488c2b776c","project":"openstack/swift","branch":"stable/wallaby","hashtags":[],"change_id":"I9a1ebec645ced755aece8e28452f51488c2b776c","subject":"Don\u0027t require swift be installed to have passing manager tests","status":"MERGED","created":"2021-03-19 16:43:05.000000000","updated":"2021-03-20 03:02:27.000000000","submitted":"2021-03-20 02:59:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781820","meta_rev_id":"17c81ff1ee710e7faf7ff3b76c41071e8675ab29","_number":781820,"virtual_id_number":781820,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-20 02:59:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-19 19:16:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-19 19:16:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 18:25:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8f7ddabeaed06c6fc0c425c5010c6dd108be6dd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-19 16:43:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5fd97914d008b845602329cef41f346ae42cfda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 18:25:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddbfeb4018a74af8a9ca2c96955093b5 : SUCCESS in 10m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7d1a0850373e442b8faf1eabd7b27868 : SUCCESS in 9m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/379be8820f6b4e25baee1542236eeeb9 : SUCCESS in 8m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/90620cfd7c02432bb809e55baacd6431 : SUCCESS in 20m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dbde76c8f09c46c0ad67028aba101602 : SUCCESS in 17m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6d247088a0e34fc4986f8eb2ad8e0e12 : SUCCESS in 21m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1348d45383164ad18396e409b19618c8 : SUCCESS in 16m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/28501d4cc3ce4af4b4bbd47705bc4461 : SUCCESS in 16m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/172c3765e54345e8b2b257b7bf5c3cae : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/77636f2afa264d7eb6a86d8fe33fca93 : SUCCESS in 19m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/049fd181d1424c6f8203149f407064e8 : SUCCESS in 17m 32s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3004838664b943de810827955f295599 : SUCCESS in 24m 46s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f33d4aedd07a43498dee3d579cbbfe31 : SUCCESS in 18m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/deb50e9aa4e24c92838c0d02640a1890 : SUCCESS in 18m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c73c1046bf04b328e276bd7967e006b : SUCCESS in 15m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7275cd6eed740378788be211acd226f : SUCCESS in 6m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3dc426518bd147c2bfbdede8070d9f3d : SUCCESS in 25m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4055cc6a6a0435495d9095f5c6ab845cce16201f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-19 19:16:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1e35ec1147746aef1e62075276fcbebab8fee89f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 22:10:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0beb662b6d02e144de67727d4054ed7719155835","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 02:59:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa84425c5c7942428a1c128c00daad99 : SUCCESS in 8m 35s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ebf73ea7044c43f3bcdcb5cc3570e4a1 : SUCCESS in 8m 14s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/205adea0532148e58d51d1a990a4faa0 : SUCCESS in 9m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0cd894b389e54fe09b22da06d0d478a5 : SUCCESS in 26m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/11beb6f2ee534ab5aa95032862f80c45 : SUCCESS in 23m 12s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f9253aca5ea84342978d97dfe57d548a : SUCCESS in 21m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/81e7a8effdbd4c4c99e251aca72e3abd : SUCCESS in 23m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cfe9a6609b144243aa27e788cd1b6154 : SUCCESS in 18m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/759b190e04f0411fb2fe7e2c5255458c : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b6a3458e81214b9988ac9722cb02c086 : SUCCESS in 14m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c1e51098b94140d597898653ea53cfdb : SUCCESS in 14m 13s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1b5e917d2972449fa80fb2b406104bae : SUCCESS in 24m 08s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/72d0520b19d043c39b6097c4b629c62e : SUCCESS in 19m 05s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/29dedd02f13247a49981cde816de0fea : SUCCESS in 25m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/512f593a9ec74596ac183c1b15410ed5 : SUCCESS in 7m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8cfeb26753a4825a3d39298ef421ce0 : SUCCESS in 16m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02a70610de8a4b3ea43d389b59151a88 : SUCCESS in 7m 18s","accounts_in_message":[],"_revision_number":1},{"id":"16ccb5350ea9306bbc5a968a608fa0f3000f5b63","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 02:59:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"17c81ff1ee710e7faf7ff3b76c41071e8675ab29","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 03:02:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c971d9ee77e4af6826bbe50ed587558 : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/406244a700c14951a9f65501e1aefe40 : SUCCESS in 55s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/7088d4173aba4c09b7a3417b27477440 : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"056859bb039a41c3719d0c8c607471c83d79c29b","revisions":{"056859bb039a41c3719d0c8c607471c83d79c29b":{"kind":"REWORK","_number":1,"created":"2021-03-19 16:43:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/20/781820/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/20/781820/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/781820/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/781820/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/20/781820/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/20/781820/1"}}},"commit":{"parents":[{"commit":"223bdd73677ef7313e1511b710fbc531067db8cd","subject":"Only test with \u0026-delimited query strings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/223bdd73677ef7313e1511b710fbc531067db8cd"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-18 23:35:05.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-19 16:41:47.000000000","tz":-420},"subject":"Don\u0027t require swift be installed to have passing manager tests","message":"Don\u0027t require swift be installed to have passing manager tests\n\nChange-Id: I9a1ebec645ced755aece8e28452f51488c2b776c\nRelated-Change: I09143ae20bd6249083c0b80cdaa9d561e6abb301\n(cherry picked from commit d54083771f90ef96b7dae4c4fc2a268b6b375394)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/056859bb039a41c3719d0c8c607471c83d79c29b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/056859bb039a41c3719d0c8c607471c83d79c29b"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
