)]}'
{"id":"openstack%2Fswift~781861","triplet_id":"openstack%2Fswift~master~Icf1517bd930c74e9552b88250a7b4019e0ab413e","project":"openstack/swift","branch":"master","topic":"p-relinker-refactor","hashtags":[],"change_id":"Icf1517bd930c74e9552b88250a7b4019e0ab413e","subject":"relinker: Accept policy names, too","status":"MERGED","created":"2021-03-19 18:54:36.000000000","updated":"2021-03-25 21:58:39.000000000","submitted":"2021-03-25 21:52:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"781861-p-relinker-refactor","meta_rev_id":"e35308150c18254161dd70439fae3d4315222780","_number":781861,"virtual_id_number":781861,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-25 21:52:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2021-03-25 12:53:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2021-03-25 12:53:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 20:15:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 05:18:03.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2021-03-22 09:17:06.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"4230f60ceb01d57ebe820b9c2adce295ef156107","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-19 18:54:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de3ac2f0376e39c74ccc4d98eef6b7d4cb94b02f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 20:15:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60510d8a9f754f68bfdd7446ff92b32b : SUCCESS in 9m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/26f25a02081a4dcf986c28443e7e42b4 : SUCCESS in 1h 04m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f7c51f6c39864b389606d193e585fe8f : SUCCESS in 49m 03s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3940f229a5c74c15950f9ea030e225dd : SUCCESS in 10m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/563daa0a71a348e38938386dee09c19e : SUCCESS in 7m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/47e23424704b4626be6876d5c0e6385a : SUCCESS in 17m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4df529c3f028474aa1165bd38b34b719 : SUCCESS in 18m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/07b76f1410944115997aa2b4f0d08a6d : SUCCESS in 18m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8253be002a084412a380f151a86ae606 : SUCCESS in 17m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fb070e6f2d7f452498aa82fad005b281 : SUCCESS in 20m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6a8ab2e1c39c40f8ab759d4d73da6bc7 : SUCCESS in 19m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8c936268f6c5455faf1b12e7ebfbb666 : SUCCESS in 21m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8be4ff6ffac64c729310c5dee72ee49e : SUCCESS in 20m 32s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2232b56765e142f2999364300271c7b6 : SUCCESS in 19m 03s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2649ee2d9abc4f0ca7a41cfc162504be : SUCCESS in 24m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fb0f3ee4f5ac4ef0b7908fbfd18894da : SUCCESS in 19m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/69ee43f9acb84c498b747c343e84171f : SUCCESS in 9m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a3aef69006cd440dab4d78bba1c3bbf9 : SUCCESS in 21m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a513c299b3ee4b2b9e1948c45c3286b4 : SUCCESS in 57m 22s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/10f619c07b9a4a0d898cc82ed6722872 : SUCCESS in 51m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0758d90e4da849a8967dc76d2bb6aebe : SUCCESS in 40m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/06ca69a056d849099048f2e42875520c : SUCCESS in 39m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3dcd21fd12064ca6b3347dae646c8e01 : SUCCESS in 16m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69d1c421e7f942db81661b058f821399 : SUCCESS in 10m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9fff8ccd94bb49769bbf3556a9a69e5c : SUCCESS in 27m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e9cb88e36964f97bb37fe4947421161 : SUCCESS in 1h 10m 29s","accounts_in_message":[],"_revision_number":1},{"id":"9550ca55427e24334092f54f38653ca052ccfcb8","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-03-22 05:18:03.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea34408941ab943a40a458d92e2d24936af490c7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-22 09:17:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f16f3b7c60d900f224c38a797963ae86e196dfa","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-22 17:37:10.000000000","message":"Patch Set 1: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"1bd0e218f4002f8a17b6db6b46f6cd8aca7b38eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-22 18:47:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9c151b56623f44ce007af9be2deb13d58452e587","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-22 20:28:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f9c03c41c41e9a6a9bfd50dd65c7a4f60e2cc739","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 22:58:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1629118b8c3a4ba498c4012d84a311c9 : SUCCESS in 15m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/fd2fa15024944ad4bf602077714d043b : SUCCESS in 56m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ebd734aa850e46dab72f0870892e645b : SUCCESS in 45m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/095d176f8f9e4d498772deec32d807b9 : SUCCESS in 9m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f8d51abeebd042679e2eccd37686025d : SUCCESS in 6m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3896ca261e1d4cadb086d58d785c11a9 : SUCCESS in 15m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f7d337fc909b48c09818ad9422910e34 : SUCCESS in 15m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/143043656f8e407487cc080356f51d85 : SUCCESS in 14m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cd39cfe0ee8e422ba675ecf35d0f2c8a : SUCCESS in 16m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5aab99ee648447b2a1d35da71a421f73 : SUCCESS in 16m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5d9712561a8f48978b6fae642a9f5057 : SUCCESS in 17m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bb79f9d71c2e421caece87ead36eb6cf : SUCCESS in 24m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/25ff8da651b5404e8f5105641d4007e9 : SUCCESS in 24m 49s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e693de351d8a485a882ac4e9472d4f33 : SUCCESS in 19m 30s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/47a760c0d5b04b7eba26f0bdf239f9d5 : SUCCESS in 17m 49s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4735c0ad492d4eea991cc706361fe040 : SUCCESS in 20m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/721c5d16cdf44e298e6b18f8e87d041e : SUCCESS in 7m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/064811cb2dbe48a796930b03fab7ef1f : FAILURE in 20m 20s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0737413e6d984cfbbec6661adb78501f : SUCCESS in 48m 33s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/d87e2d8df9e24f5dbc3dbf4517ef75fe : SUCCESS in 52m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/be30e403d0284c2cb6604249f795f2c8 : SUCCESS in 54m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4aa190240afe4d58bf43e13a4ca08c4c : SUCCESS in 49m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9fd74a4d5cf64785babc7bb8db580d88 : SUCCESS in 22m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/971130298234485db7a8d606a047d3cf : SUCCESS in 6m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1ba35305f380402ba94eb7ae958d598b : SUCCESS in 23m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e333d508511f43fa8854c8705f782f17 : SUCCESS in 51m 03s","accounts_in_message":[],"_revision_number":3},{"id":"36f4c9fcf915080292c29f9d6e627047d4e0daee","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-03-25 04:49:27.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8ecc2009d6e225f0db6c37d0bb881c8284a98f71","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-25 10:25:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)\n\nfollow up doc change: https://review.opendev.org/c/openstack/swift/+/782944","accounts_in_message":[],"_revision_number":3},{"id":"46297fecba7577d61557e7953c77cb2764fb27bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 10:26:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"49ffd4e9c6af88aeb6b1f97ff82f16cec3279956","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 11:29:24.000000000","message":"Patch Set 3: Verified-2\n\n(1 comment)\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4fca59744ef41a7a6be5c68154487d2 : SUCCESS in 10m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/e753d2420d1f479d912f151a2aec4405 : SUCCESS in 59m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3ea25c095e9a4eb99b2ec3f6662f8d51 : SUCCESS in 46m 49s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/30f66607bbc946b7a5936c09d7e4aab6 : SUCCESS in 11m 25s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/b1c974f9b264492a9d39cb85aeaf7b09 : SUCCESS in 8m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/736e1ab856774b80830f9bd6c9fb04ab : FAILURE in 19m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/864ec4cfeed44e6fa756e661da4573e4 : FAILURE in 20m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0e0e314200f341baa9d52faea97cb8cb : FAILURE in 18m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6ce65a403a9b408e94b0dbb3e0f27708 : FAILURE in 18m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4c083f75c3c945388daa884417f639c6 : FAILURE in 20m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/aeb43be4306d43f594109d6e746d25ab : SUCCESS in 21m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d9fcccfb8bf44e75993247e56bfed569 : SUCCESS in 21m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f173669a6fe44d9ba588c06a8bcf9f7c : SUCCESS in 22m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/19f57163fbc24139b17d3d185e32b17d : SUCCESS in 22m 35s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b9c411bb84f04ca0af2a5d1424a52606 : SUCCESS in 18m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a4773b48cb9d4493acc9f1831ce8d416 : SUCCESS in 23m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/377cbe2aedf34515bc4a8959e19c5bd3 : SUCCESS in 7m 46s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e42999c54e7345f296e1083f2e1cd230 : FAILURE in 52m 57s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/6a100a2e7cf74890b7f9072ae68ac6e8 : FAILURE in 54m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3cd98474bd4c4feebb7edb19ef0da071 : SUCCESS in 42m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4d81ad9a8cc94c48898c925fda339c8a : SUCCESS in 43m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/603fd692479c4d9faa2c39b12cd1e70d : FAILURE in 19m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6040adebf1e74dcea1fd55b2fd6ccf2d : FAILURE in 7m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cd30e796704542ef8633437612859638 : SUCCESS in 30m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58664ededa7b4a98aaa8d670c0946060 : SUCCESS in 56m 44s","accounts_in_message":[],"_revision_number":3},{"id":"b1bd04abe54c3de66c4e92aa7664cb4b7a1262dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-25 12:51:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"16c19b3899261ca8c743fe5b68f9b1faf08d755a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-25 12:52:35.000000000","message":"Patch Set 3:\n\ntests failed with \n\n  NameError: name \u0027non_negative_int\u0027 is not defined\n\nbecause this patch removes it from imports but it is used on master","accounts_in_message":[],"_revision_number":3},{"id":"8de00124b2bd5a99d088b62a89e21dbf0d971295","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-25 12:53:25.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"251239f730df16703e0906c0d622df73b25f0f4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 14:11:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c229e9deaaa4cdf92dd68184765ad3a : SUCCESS in 9m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/17b5949885e44679925078f6c1345f44 : SUCCESS in 1h 07m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9cec334a1b8f42c8bdb543eb463c7d81 : SUCCESS in 59m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/65268cceffba40e49cd1babc8f137c69 : SUCCESS in 9m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/95dc910f4b9842eeb60b9abe43a39f66 : SUCCESS in 6m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/108f79d83346428299f6696914cba328 : SUCCESS in 14m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a1235f84a80b4868b1b7acfae9bfc84e : SUCCESS in 17m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5f594288d2544e2ea831dfc6a5ceadd1 : SUCCESS in 18m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7ee5ef1da6e340fcb256ddd1a9a6cb7a : SUCCESS in 15m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/66c1843711c441248eb23615ac789352 : SUCCESS in 18m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c7932b0fd2284c2980cba639cd1941f2 : SUCCESS in 17m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/89b115581e46480a9fada350c0d6bb1b : SUCCESS in 16m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d0171d1f18f046188671411a52c34fc0 : SUCCESS in 17m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5f24285cfbfe424da536c22b3cdf8227 : SUCCESS in 19m 03s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d6d4c21c67ba4e8ebd558f2ded5c9bf2 : SUCCESS in 18m 10s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2e7aed0c7a334c8fa1538972a5a0e71a : SUCCESS in 19m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0cbb0fb4b05d42ef8543d19ce2407b29 : SUCCESS in 6m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2e7c5f9ab0db4fb6ac999dc4e22e9489 : FAILURE in 22m 01s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/71051733800f494284dfee4099667eb3 : SUCCESS in 51m 58s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/27176eb8c850499bbf1fd8a4e257350c : SUCCESS in 49m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5e49d340820b47e4af982028d9b04396 : SUCCESS in 40m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4bf1a878fec84a0c943483ce3a363538 : SUCCESS in 42m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1256daed304f4d969dcd38a46999a4ac : SUCCESS in 15m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5c66643a9944f45a5d20b7108e365ff : SUCCESS in 5m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4f151d2ef3e74004abe9465176afd9c8 : SUCCESS in 32m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8fe3c9c7efb41d9bda707915f6dd662 : SUCCESS in 1h 11m 14s","accounts_in_message":[],"_revision_number":4},{"id":"c26338d0c9e83f6c0624537b21afa983372770cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 14:12:16.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"75c2ce193a70593deca9c0e8e9ea063be4b72dbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 16:19:35.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06a3696ce45c433cad70a47b69a37af9 : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/e8749a8e70724b1b954680cb26dc25ce : SUCCESS in 1h 08m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b0823e9137a14f179b43bca42c5bd002 : SUCCESS in 45m 53s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0cd1425263034d2bae00d60736e2abbe : SUCCESS in 11m 07s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/41c347d69a0e4dfe9205c9a5fb1b3e63 : SUCCESS in 8m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dda21549ba5840b6812de0f92b3c9fee : SUCCESS in 15m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7deff5154e3945e390516ee88c545806 : SUCCESS in 15m 51s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/81d3a910c21341759ed1f4affed04d46 : SUCCESS in 14m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1a1b4e308c4445ba8e1e65c9872784b8 : SUCCESS in 27m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/05ad6e62678f4982b93f20036d1d5870 : SUCCESS in 26m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/291b4046eca746ec9a3eaa61a0d3284f : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9d7faffeecc9438c8802397f6e5b4c62 : SUCCESS in 20m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8140535e5d6e4ad6b504d2d2b1733243 : SUCCESS in 25m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a4a94bff537c4e48ac8a9f676f4a279d : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/925531e45a61450da9138f7e8cb75525 : SUCCESS in 18m 16s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/25d1617e3261477393e36546a2e0a9f0 : SUCCESS in 19m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e549f3bd0deb4a569ed39f6d543e201b : SUCCESS in 6m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c2cfa30d5cf94ebd8062b7c2720d725b : SUCCESS in 1h 25m 27s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/d1afd3f4ea82462eb72a64abf7059ce2 : SUCCESS in 58m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ddcd60861aba4a018f985c3292529b81 : SUCCESS in 42m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/83af15288f01402db2d69e53aedd36f6 : SUCCESS in 41m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1279ff06b4b4e0c934654442ca9f087 : TIMED_OUT in 41m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c0956c1a7ea4dbca007986fe3735b84 : SUCCESS in 7m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a3380f1d336e4482a2181e3a1e1e25b2 : SUCCESS in 31m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5189d6f427524fb49a61d98771e76399 : SUCCESS in 1h 18m 51s","accounts_in_message":[],"_revision_number":4},{"id":"0cda646a84a87701513a1b7158bedfa0b5738f85","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-25 16:40:02.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"336f242f89e78ddb913db25693ea827aef41050e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 17:47:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6474f3f8f0384efca0bf53b4b02c8d00 : SUCCESS in 11m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/056401f2b81f464e8ec1c378a0a56c62 : SUCCESS in 57m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ad27b9c485b04a72b50f214a739f76c2 : SUCCESS in 39m 55s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a5a40fbfaf5f46d7aa53fef873fec156 : SUCCESS in 9m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fc86f0117fa64471b71d4d137887b784 : SUCCESS in 10m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/84f24ef37b504f0b8837042cf7567e86 : SUCCESS in 14m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6595d578827e4713a3eba8bfb670701c : SUCCESS in 20m 17s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f98ceb5baf1248eebdf06b80d00dd2cc : SUCCESS in 15m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/72efd72e24044ebcb44b813a942d3600 : SUCCESS in 19m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9251e8044a7f4840a47fb73dcdac19b4 : SUCCESS in 24m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bb7729379b9142909ec788db9be228f8 : SUCCESS in 23m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5899965e19b74ac5a6e169cedc48eae5 : SUCCESS in 16m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aeb12d85acf44a0d8998727d40c1deb5 : SUCCESS in 22m 47s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/26b638cd64a74fcbbe82738230048fc8 : SUCCESS in 18m 30s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/455f1badb8e04d15b1674da3cebfa098 : SUCCESS in 28m 28s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c1468eefb4434ed8b63e2c91375de6a9 : SUCCESS in 19m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4101089933c74a0ba77f3357a506fb44 : SUCCESS in 7m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/294dbe313b2f438eae480bafc02a09d3 : SUCCESS in 23m 03s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8dbd41e02ae343cabaa52ff97ad1f6eb : SUCCESS in 1h 05m 17s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/bf6a4428623c4bbb836aa7f5e3ccfeb1 : SUCCESS in 51m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0beef9279ce24b969273ddaec0509bee : SUCCESS in 38m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fc81fcfb81dc48459bedfd0756995215 : SUCCESS in 38m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f80e3b2d04541b79dcd1c1a6de35767 : SUCCESS in 14m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8d3489423194a92b19a90755d58d426 : SUCCESS in 7m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7bfaf17767624ed6bb57de252284dd57 : SUCCESS in 31m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a748b37a5d142bfa8c5073b66d20a05 : SUCCESS in 1h 00m 13s","accounts_in_message":[],"_revision_number":4},{"id":"38ac68585d2606aef04f507474bb5ab53bbfee2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 17:51:56.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7348b5eb63dda39578451400fe4f5e975043d65c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 21:52:00.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a2776640c6b40ad84c8afb07be6e6f4 : SUCCESS in 8m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/a383f72c5b5f451b89dd6e2048705300 : SUCCESS in 1h 17m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/936ca80aad144e4b8fb352d99ae4ec89 : SUCCESS in 42m 54s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/955f034162804683a119f1c5f3b2e525 : SUCCESS in 10m 39s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/2d78ec0b8276409fb046f4a1f32443d5 : SUCCESS in 6m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f49efd81c53c45f7bd61a5e4285ac80f : SUCCESS in 15m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5d6b9a017ef649679960597b24819b9d : SUCCESS in 16m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0de96f1f923342509f4b0ba936e3ad5d : SUCCESS in 16m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4925d9f731c943bd849ff77bf1b06e24 : SUCCESS in 16m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2151407b57984491ac00332a13b2d18d : SUCCESS in 13m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/63c173a434824e46bfda2641d13e4d47 : SUCCESS in 18m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/375f16673871435eb9b70036b54c450a : SUCCESS in 21m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/56f8fdacadb04516a888a8e7ffbc8b19 : SUCCESS in 21m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/589f17ab961e4a5099c2bac6253e8d56 : SUCCESS in 21m 00s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/de5a39ddf7b347678331129d137b51ae : SUCCESS in 15m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/dbd8b00a72684d73a3642e96d476cc8c : SUCCESS in 20m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/54f639ac732940c6853bb97cdb353d3d : SUCCESS in 7m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/17def045fc6949759714b30292fa8e29 : SUCCESS in 49m 56s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b67d71e84ed84c3c8cf22aaeacd0bc40 : SUCCESS in 50m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d0ea0974d02a45a385c3939619a0271b : SUCCESS in 42m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6ff5c776a306458d8cb7ed7ab55c5ec2 : SUCCESS in 29m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a8043dd9f5d4d3bbcb6ede13a826aa9 : SUCCESS in 15m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64f3e278cd0f492fbc8a41b7bd89785d : SUCCESS in 7m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8b949c6740f6430c9dd3a9bc18f0f931 : SUCCESS in 25m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0862e4d1c281480cb3ff0076b03314a9 : SUCCESS in 44m 50s","accounts_in_message":[],"_revision_number":4},{"id":"dfa50173d99f2ef1330f35b3f440749a6b054e85","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 21:52:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"e35308150c18254161dd70439fae3d4315222780","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 21:58:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaf2dc701cd1478c95b9d6d9e3b8d308 : SUCCESS in 1m 23s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/173dfabdf53e467fa56635cf5a4ad549 : SUCCESS in 51s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/730e837809114dedafb1d2b68747eeab : SUCCESS in 1m 12s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a967d472957598374937650da222c731d9e30704","revisions":{"e3c84fb9257436c6af5c544a26b0a0527652db5a":{"kind":"REWORK","_number":1,"created":"2021-03-19 18:54:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/61/781861/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/781861/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/781861/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/781861/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/781861/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/781861/1"}}},"commit":{"parents":[{"commit":"2b4ec5a45b31fad6884b75a23a229e8b1dfb5776","subject":"relinker: add --policy option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b4ec5a45b31fad6884b75a23a229e8b1dfb5776"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-19 17:24:59.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-19 17:24:59.000000000","tz":-420},"subject":"relinker: Accept policy names, too","message":"relinker: Accept policy names, too\n\nChange-Id: Icf1517bd930c74e9552b88250a7b4019e0ab413e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3c84fb9257436c6af5c544a26b0a0527652db5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3c84fb9257436c6af5c544a26b0a0527652db5a"}]},"branch":"refs/heads/master"},"4f2a739b762cd863cc0aa8b876f86d34ecdd1bde":{"kind":"REWORK","_number":2,"created":"2021-03-22 18:47:20.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/61/781861/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/781861/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/781861/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/781861/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/781861/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/781861/2"}}},"commit":{"parents":[{"commit":"2b4ec5a45b31fad6884b75a23a229e8b1dfb5776","subject":"relinker: add --policy option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b4ec5a45b31fad6884b75a23a229e8b1dfb5776"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-19 17:24:59.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-22 18:45:31.000000000","tz":-420},"subject":"relinker: Accept policy names, too","message":"relinker: Accept policy names, too\n\nChange-Id: Icf1517bd930c74e9552b88250a7b4019e0ab413e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f2a739b762cd863cc0aa8b876f86d34ecdd1bde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f2a739b762cd863cc0aa8b876f86d34ecdd1bde"}]},"branch":"refs/heads/master"},"047eb37193d5628c420c7dd0322dfac0cd7ced5f":{"kind":"REWORK","_number":3,"created":"2021-03-22 20:28:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/61/781861/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/781861/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/781861/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/781861/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/781861/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/781861/3"}}},"commit":{"parents":[{"commit":"788bf852345f28bd5a96523de52ae953c1d96021","subject":"Merge \"Add test for relinking when the object-server failed to do it\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/788bf852345f28bd5a96523de52ae953c1d96021"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-19 17:24:59.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-22 20:27:22.000000000","tz":-420},"subject":"relinker: Accept policy names, too","message":"relinker: Accept policy names, too\n\nChange-Id: Icf1517bd930c74e9552b88250a7b4019e0ab413e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/047eb37193d5628c420c7dd0322dfac0cd7ced5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/047eb37193d5628c420c7dd0322dfac0cd7ced5f"}]},"branch":"refs/heads/master"},"a967d472957598374937650da222c731d9e30704":{"kind":"REWORK","_number":4,"created":"2021-03-25 12:51:29.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/61/781861/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/781861/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/781861/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/781861/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/781861/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/781861/4"}}},"commit":{"parents":[{"commit":"8cc1ef92558bd3f7c345c955f386498b936bace1","subject":"Merge \"relinker: Add policy to relinker progress output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cc1ef92558bd3f7c345c955f386498b936bace1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-19 17:24:59.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-25 12:50:13.000000000","tz":0},"subject":"relinker: Accept policy names, too","message":"relinker: Accept policy names, too\n\nChange-Id: Icf1517bd930c74e9552b88250a7b4019e0ab413e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a967d472957598374937650da222c731d9e30704"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a967d472957598374937650da222c731d9e30704"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
