)]}'
{"id":"openstack%2Fswift~781862","triplet_id":"openstack%2Fswift~master~I6805f0466f0200a58abe79cbb8ed3e6d2fd82d62","project":"openstack/swift","branch":"master","topic":"p-relinker-refactor","hashtags":[],"change_id":"I6805f0466f0200a58abe79cbb8ed3e6d2fd82d62","subject":"relinker: Allow multiple policies to be specified","status":"MERGED","created":"2021-03-19 18:54:36.000000000","updated":"2021-03-23 13:22:33.000000000","submitted":"2021-03-23 13:19:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"781862-p-relinker-refactor","meta_rev_id":"8046b6bfc734b34f6651507f0c520ae8616cc3e9","_number":781862,"virtual_id_number":781862,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-23 13:19:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2021-03-23 10:44:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2021-03-23 10:44:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 20:13:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 09:28:04.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"23b15f858032136c2678220fa5174427ea82db0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-19 18:54:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d29b5e0ed10d5665131a22805a4d22faa4426f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 20:13:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d7f6ccbe174340960d5327130b160a : SUCCESS in 11m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/b942c65fdacf4aa6a6136a9fd02ddf09 : SUCCESS in 1h 11m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/98d20386f6ba4584b1f2f41416cdaf45 : SUCCESS in 51m 20s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/72b5d170f1f9407ab8d3cb1e9ead877a : SUCCESS in 12m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3e6965fe34db4fedac2cbcce98169e66 : SUCCESS in 9m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8e8e64e9ba844c3b852f482f2c1edad5 : SUCCESS in 17m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e4b1a0b866bf4d32b6b4b09e3a1cac91 : SUCCESS in 18m 24s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1bbbcdc558a24fb78b6a1b01c8c6637c : SUCCESS in 16m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a3be49aa34e9498ebf89a1f5fc4a82bc : SUCCESS in 15m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b00a5d2f6cca47c98f62ddf21686e890 : SUCCESS in 17m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c1f5023285da44e8a61e036384d42f0a : SUCCESS in 19m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fd1aae85b4fd4590b9b386c1a7ff5586 : SUCCESS in 18m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5f8911682d8245cda9bef2154bd6762a : SUCCESS in 27m 44s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d443377a4aaf4fb48b64e83e614c59ec : SUCCESS in 19m 15s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/40709bfe70d0402eb414e12162fe52f8 : SUCCESS in 21m 05s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/97523dc76cdb42758f69917ba5fbe055 : SUCCESS in 21m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/479715c1b11342f9b5151565e269f161 : SUCCESS in 7m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4c347e36e9694a139ada3d2955e3b8d7 : SUCCESS in 23m 55s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/aa16ed42a31a48268905666c086513c8 : SUCCESS in 48m 37s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/7d25dcecf2174d3d8beb33da7708d4b4 : SUCCESS in 44m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e80192cbef9547aaa52e413ee93f1583 : SUCCESS in 39m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aed5e6bb3ea44c6ca40daa9a8727b773 : SUCCESS in 40m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfb26059a45841f1b547d40c217f5eaf : SUCCESS in 17m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c9d1d3086884dcf8d7bcb9defbd107c : SUCCESS in 9m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0ba2a5fdc6864df69f813424735e840a : SUCCESS in 28m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d93f4417e94f4cc8b0fd866a2cac0d2c : SUCCESS in 1h 03m 03s","accounts_in_message":[],"_revision_number":1},{"id":"43f0a2efc40d8fcec785f79bf195b9128af42421","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-22 09:28:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)\n\nOK!","accounts_in_message":[],"_revision_number":1},{"id":"2032f48939c6081bd0c09d0809a00d6d5252b5ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-22 17:38:34.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"9e0f5b754834455e3a192504bcaec4d0490b1a9e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-22 17:39:12.000000000","message":"Patch Set 2: Workflow+1\n\nJust rebased to pop off the chain with https://review.opendev.org/c/openstack/swift/+/781861","accounts_in_message":[],"_revision_number":2},{"id":"c2a3475167175ba5a9759495f89a2a4a95bab358","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-22 17:40:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d32c6be1c5a37d2865c20174685f505b5b30e179","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 20:33:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de78de2e441f4cc7bd7f99bee2818818 : SUCCESS in 11m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/bc5a9593892b4995a88055560973a06a : SUCCESS in 1h 02m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1a2e76492e6b4189be36654b6f268168 : SUCCESS in 47m 45s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/445a72c0183049a099dfd0e138a0e676 : SUCCESS in 11m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8a132ec2986e4fbf8dbeb00f82292ca7 : SUCCESS in 7m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d826fa0079d94b4c8d773f7de615dd1a : SUCCESS in 18m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/76c766d5d3c44567967e6c7cb74b0d0a : SUCCESS in 18m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bfce239a552d453e967469c9096e68f6 : SUCCESS in 17m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5b1cf3004055423e9493d3fd4a5b369e : SUCCESS in 14m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c0a4359818d74f7eb49a5e580d88fc1e : SUCCESS in 18m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b56850a16d334a41a93ff15aa788fb5c : SUCCESS in 25m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c4ebb852c8f545f792e9e7fbe51fa663 : SUCCESS in 27m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3fce6d0afb4a40a0b700600c4e53d6a3 : SUCCESS in 32m 25s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e72396da8b18490d818b94ecac3ef981 : SUCCESS in 21m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/bee9cb3d78f0482a9e8f36b0afd04257 : SUCCESS in 26m 52s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/fd3907deee944466a4f453e0a26adc2e : SUCCESS in 23m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f5e962b897374ed4a7c3270962a7a0ee : SUCCESS in 8m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a09f2219f6ee434ca9dd0b7e9f1ea7ba : SUCCESS in 24m 35s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e6a4212cb63c493c92f228e99c757c3d : SUCCESS in 50m 25s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/4fc4f767cfef490a84e6cccb52db1d95 : SUCCESS in 47m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dab929d490ca4d56903c459e494ef1d4 : SUCCESS in 48m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3dacecb0c6e748dfa2bf76eaf2d34142 : SUCCESS in 49m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/569d9206d7a14712b1ba9caedab15921 : SUCCESS in 23m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4226ab80e45f4c2196553ef49d5a5da9 : SUCCESS in 9m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/da1cf747bd9f4d3bb34b6e38a60e4791 : SUCCESS in 33m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f195d98bef54a0495af13cca8cf082b : SUCCESS in 1h 03m 38s","accounts_in_message":[],"_revision_number":3},{"id":"b182e1d92860ef3d9967e4e914a4743b63b4b899","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-23 00:59:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ae00049e47afbe02e747821b7d915de30a899cd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 01:52:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea4c576a9c86411fa497f593e7c3e34b : SUCCESS in 8m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/2d87ef7d721e4add9e5f6cb2a3826488 : SUCCESS in 48m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2d812462faad4fa7bacabc182e52a12d : SUCCESS in 37m 49s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1f1f4b6340be478a8499a12170765d6e : SUCCESS in 7m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d5c6edb1a8fc4baea6ca0dfac740224b : SUCCESS in 8m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cc4d632edccd4e73b7d28aaa4fcb3d34 : SUCCESS in 16m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/277de495143e43d290e5cc9f16911434 : SUCCESS in 15m 24s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/45494eec8f1d46b887668a0160ac5693 : SUCCESS in 17m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/db6fdcc40563441c81afbe59da132ed7 : SUCCESS in 10m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0656496a6dfc4bb0ae5b51c7a6772b65 : SUCCESS in 21m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/351803b4a0d8427faa14a1050a35cc0f : SUCCESS in 17m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8060e1bc65da41c4841f8985aeb00036 : SUCCESS in 17m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/320adcce8d0d497db5cbbac5391b3d3c : SUCCESS in 13m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/78f4be4e01d24e5bb34e635446c4b434 : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2d7c539cad714d6e96fb837ca17f7539 : SUCCESS in 18m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/850cb4b744d644e58e51a5dc80612b55 : SUCCESS in 14m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5fab27c8833c43118ebf453fafbd8e6e : SUCCESS in 7m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/47ee6dc9368d46c9a3d661fa8cf0eb4d : FAILURE in 20m 49s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d1b7c1c3bd874f26b7007714557b5955 : SUCCESS in 48m 55s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/90dd8fe327524e7e81c023fc1a2e942a : SUCCESS in 39m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/854e0aa4b704446aa193cd3b63da1f8a : SUCCESS in 36m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3beea70dce034f22938b41f4a2649eff : SUCCESS in 36m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e9ca534cef7d47ba8826992d514df24b : SUCCESS in 21m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3846b9027364fffa76be8b9c825b3bc : SUCCESS in 6m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e5fa3d2f11c7430dae3c310da8a5c611 : SUCCESS in 26m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/813e61e0855549a9a3b47c2094dc02b5 : SUCCESS in 44m 59s","accounts_in_message":[],"_revision_number":4},{"id":"3442f7233a8601477f5b4fbb2a7b81cee7bf2963","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-23 10:44:26.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2bce10064fc0cc76d13adce9b7f7790d41201cb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 10:45:05.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1f4d0e4f3d38b54775019336537d7f241afa01a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 13:19:11.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e033304960bc4b62b929a936b050b781 : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/d5fb8c2efe8d44d389f3b8681c37fa36 : SUCCESS in 1h 01m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eba8ab92570a4138b5b0f0c41aba3e5c : SUCCESS in 34m 54s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d0f2d6ad65484295b38b7ac263726a06 : SUCCESS in 10m 02s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/065e8627ef824c1d90cde1023918daf0 : SUCCESS in 7m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/aa077690c05f484ca5fcd5563d5ec0aa : SUCCESS in 22m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d3acca269dfc4716a7c0974de672c319 : SUCCESS in 13m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bb16e65b10594e6492be991bab832afd : SUCCESS in 14m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2fc9dfda1b8446fcb0f63011e94d3fb1 : SUCCESS in 21m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7b292a323e974150b9d200c51f458b36 : SUCCESS in 14m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c4f35c6ae92b475db45cb25ceee1f885 : SUCCESS in 13m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/78fb7d12ae1f446c8354fc6d89d0689b : SUCCESS in 15m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eaf90c11cec8404094a27b6276ca2d54 : SUCCESS in 14m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/93f41c8c63ed449eb128198dd49576a2 : SUCCESS in 20m 25s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2e69de5dbf9b488ba8808cba925d1e73 : SUCCESS in 16m 16s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f127a64ba85f4925afc9f0733051b432 : SUCCESS in 21m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c136abe55b1440058fbbe58eb7c2e681 : SUCCESS in 6m 42s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3072f24d91e2489d81735c7422c488e4 : SUCCESS in 50m 14s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b598d4f7f51b4002838c93d4c759cdca : SUCCESS in 49m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/40379cb7a352444c90be2b2ff65c2ea1 : SUCCESS in 40m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e5e050cdca124b10b5b0f785e8209d1d : SUCCESS in 40m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17dd7055157a4b9f8b3d047108556883 : SUCCESS in 16m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b09fa4e3d1bb41878ba7b9a611fb6b6d : SUCCESS in 6m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b3df4b79757e47869854f71d0d6d14cd : SUCCESS in 31m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0c064dfec4844fd8b299ea84d40fb47 : SUCCESS in 53m 29s","accounts_in_message":[],"_revision_number":4},{"id":"ca4a764ee1b50388b2ed09542487a22613ce55c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 13:19:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"8046b6bfc734b34f6651507f0c520ae8616cc3e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 13:22:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cef7dd10171b4941906f304d60fc6af5 : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/fdfdd4de2dcc4161b6aa3780091dddbb : SUCCESS in 43s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/107843c9d1614576ad672887d393f033 : SUCCESS in 1m 11s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3a41cbe67027e0abe2816a7b659a8fd01f95fd14","revisions":{"15944b228f0a6ee6929cb50cb037d45c8a5e8db4":{"kind":"REWORK","_number":1,"created":"2021-03-19 18:54:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/781862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/781862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/781862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/781862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/781862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/781862/1"}}},"commit":{"parents":[{"commit":"e3c84fb9257436c6af5c544a26b0a0527652db5a","subject":"relinker: Accept policy names, too","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3c84fb9257436c6af5c544a26b0a0527652db5a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-19 17:33:12.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-19 17:41:00.000000000","tz":-420},"subject":"relinker: Allow multiple policies to be specified","message":"relinker: Allow multiple policies to be specified\n\nChange-Id: I6805f0466f0200a58abe79cbb8ed3e6d2fd82d62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/15944b228f0a6ee6929cb50cb037d45c8a5e8db4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/15944b228f0a6ee6929cb50cb037d45c8a5e8db4"}]},"branch":"refs/heads/master"},"db48b2167207479d163bc6ebf755339e6810d653":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-22 17:38:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/781862/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/781862/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/781862/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/781862/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/781862/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/781862/2"}}},"commit":{"parents":[{"commit":"7e278291f88afe1ac78d930404bd4dbbfdb4f53a","subject":"relinker: Move filters \u0026 hooks to be methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e278291f88afe1ac78d930404bd4dbbfdb4f53a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-19 17:33:12.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-22 17:38:25.000000000","tz":-420},"subject":"relinker: Allow multiple policies to be specified","message":"relinker: Allow multiple policies to be specified\n\nChange-Id: I6805f0466f0200a58abe79cbb8ed3e6d2fd82d62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db48b2167207479d163bc6ebf755339e6810d653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db48b2167207479d163bc6ebf755339e6810d653"}]},"branch":"refs/heads/master"},"d8cab2ce28c436d7fbda3bd64689aaf05b2bec9a":{"kind":"REWORK","_number":3,"created":"2021-03-22 17:40:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/781862/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/781862/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/781862/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/781862/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/781862/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/781862/3"}}},"commit":{"parents":[{"commit":"7e278291f88afe1ac78d930404bd4dbbfdb4f53a","subject":"relinker: Move filters \u0026 hooks to be methods","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e278291f88afe1ac78d930404bd4dbbfdb4f53a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-19 17:33:12.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-22 17:40:38.000000000","tz":-420},"subject":"relinker: Allow multiple policies to be specified","message":"relinker: Allow multiple policies to be specified\n\nChange-Id: I6805f0466f0200a58abe79cbb8ed3e6d2fd82d62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8cab2ce28c436d7fbda3bd64689aaf05b2bec9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8cab2ce28c436d7fbda3bd64689aaf05b2bec9a"}]},"branch":"refs/heads/master"},"3a41cbe67027e0abe2816a7b659a8fd01f95fd14":{"kind":"REWORK","_number":4,"created":"2021-03-23 00:59:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/62/781862/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/781862/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/781862/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/781862/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/781862/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/781862/4"}}},"commit":{"parents":[{"commit":"31e27a168ac8b6d073863da2c03d272c29474c5c","subject":"Merge \"relinker: add --partition option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/31e27a168ac8b6d073863da2c03d272c29474c5c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-19 17:33:12.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-23 00:59:42.000000000","tz":-420},"subject":"relinker: Allow multiple policies to be specified","message":"relinker: Allow multiple policies to be specified\n\nChange-Id: I6805f0466f0200a58abe79cbb8ed3e6d2fd82d62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a41cbe67027e0abe2816a7b659a8fd01f95fd14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a41cbe67027e0abe2816a7b659a8fd01f95fd14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
