)]}'
{"id":"openstack%2Fswift~781886","triplet_id":"openstack%2Fswift~master~I729fdafbeef5bcb48526f5b7e131647fb5f49727","project":"openstack/swift","branch":"master","topic":"p-relinker-refactor","hashtags":[],"change_id":"I729fdafbeef5bcb48526f5b7e131647fb5f49727","subject":"Add test for relinking when the object-server failed to do it","status":"MERGED","created":"2021-03-19 19:44:50.000000000","updated":"2021-03-22 17:49:21.000000000","submitted":"2021-03-22 17:49:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781886-p-relinker-refactor","meta_rev_id":"1f007da4397d7ff24298b3c5e8d82cf88aad30ec","_number":781886,"virtual_id_number":781886,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-22 17:49:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2021-03-22 12:00:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2021-03-22 12:00:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 20:42:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 12:00:50.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"b3d0cdfeb141c09f3c0d0ddf85d79aebbf99603b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-19 19:44:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"323c658539abe829c912416e5639adc10227fae1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 20:42:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dee22dd332a840d6a9a4b7c15753b5ed : SUCCESS in 9m 11s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5ffbd22bae3e4e2c8c2bd6b1bc1b4a8b : SUCCESS in 9m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/492ec6e51f5548fd933ad86032f49370 : SUCCESS in 8m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/38a78be2c22448a2bb6ca27586e8d403 : SUCCESS in 15m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c15fb0ca3834481895104dbdc87b575d : SUCCESS in 13m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/85b5ba974ea84263a73d5538768202c4 : SUCCESS in 16m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/87422e4da7d34243b009971df1bb1d80 : SUCCESS in 13m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/06111deec442460a80c358fe270d9d5f : SUCCESS in 16m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/547bdae1e3264cf4a0f43ff7140c2521 : SUCCESS in 20m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/163ab9e15d1f4c8ab3f8ac4c4d2a2bb7 : SUCCESS in 15m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/acee1d152f154b7a80150d9f69f11615 : SUCCESS in 22m 05s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8373b3256548469cbe8eac8fb2294816 : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/21ce9714e9fe4eeea1452f2bd9ee45c8 : SUCCESS in 18m 43s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8c2a82d3fb46489390c2aa0cfa060022 : SUCCESS in 22m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/95ab80f625494fc7af2a5f2122720796 : SUCCESS in 33m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/792e6a6c0da949be9f3bdc9313017fb6 : SUCCESS in 43m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28851946a4cd4ed4a30b8beacfce677a : SUCCESS in 17m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1f578d4941a4a59892d1ff3f7d2371a : SUCCESS in 7m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ff153d9942ea4323a93e59bd8ec4035e : RETRY_LIMIT in 8m 54s","accounts_in_message":[],"_revision_number":1},{"id":"44877e249d7f37eb77948ccc81803afcff4ba3fd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-19 20:51:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b923f8654d1ccbf266b841fa287bfc204af193f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 21:43:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e092d3c28dd84d08949ab961798f1bd3 : SUCCESS in 8m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/abe75c75f702427b8e544ff4400ef5e2 : SUCCESS in 8m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/441ffb4e6a5848f79eb2b4577234faa0 : SUCCESS in 7m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fa3b83e95caf48ea8db7f6a76d73c9ac : SUCCESS in 27m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b4509d0eb88a41768b9c0e03394e2a91 : SUCCESS in 28m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/02d36e1f161849759f310b271f0b8c4f : SUCCESS in 15m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d1f21ededa6d4153bd3c16af7ee935e2 : SUCCESS in 16m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/720e818d845e4ab49cc6affc89a890f8 : SUCCESS in 24m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ea3b2a685d1d4bac8ad0043a5ab66eee : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d0ff562ad5e34573a3ee8d97758bd641 : SUCCESS in 18m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/324ed6aae506437e9e7026025d7593d1 : SUCCESS in 25m 54s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9e0bed6937c64fa3a0d4075e606e2e6a : SUCCESS in 19m 09s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0b02ce6664d24e69937cf580a9b302e2 : SUCCESS in 20m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/12cc48888407426dbbaff2f1b0520a84 : SUCCESS in 18m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/509fe2b84c0e4a0cae30ebc82ac484f1 : SUCCESS in 40m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d989141745634fbeb8c8803b4ff903c9 : SUCCESS in 45m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8dcfc35a0e4149cdba83b59637ff0dd1 : SUCCESS in 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d592ead1e6a541a78faaf9f007823704 : SUCCESS in 7m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/299bf2b7cecf43b49bb2a23f7d534248 : SUCCESS in 27m 42s","accounts_in_message":[],"_revision_number":1},{"id":"5fd9665bbe9ec4ae86d1363c47cca242a0d60c4d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-22 12:00:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"04e34d50a7d7a41fdbe9986ba778d01af2d6dc03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 12:06:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d987f332158d831c60ed13d2858d2ef986fdfe3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 14:42:33.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9f293f7de981eed8aac1c2be0aee0e4bd77dbb18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 17:49:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80d5c9ef02b34e328eccc262a9db5eb3 : SUCCESS in 11m 36s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4a305d195d494bb39b7a428f0b288c1b : SUCCESS in 11m 21s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/eb7346ef213441c9a7aa4ce72932ae0a : SUCCESS in 9m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/39fbef4b09eb44a790996f8129cf92b5 : SUCCESS in 16m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f01f23f732fd43b6b8df3f2fe17973f3 : SUCCESS in 16m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/196a6a0868d54500ba757fc764643496 : SUCCESS in 16m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e315b2de136f4a4e84984bb8194f789e : SUCCESS in 16m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1e9ab98187334432bec8bc37401f67f2 : SUCCESS in 14m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8bdf1780c53747e6904371650c2c0e63 : SUCCESS in 22m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2dfd6d9d3cc4458aa44c57d43f06d299 : SUCCESS in 25m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ac03ae41d4214aca97b34d8178ee654f : SUCCESS in 23m 18s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f23a6c04084b4b81804922ea1a49fcf4 : SUCCESS in 20m 44s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/44463407dcbd46338ba61e818c5c7e39 : SUCCESS in 27m 27s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ccc9a208769941cbac90c87bb01eb306 : SUCCESS in 28m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f85ef06e2ff640e6a4de903dccba193d : SUCCESS in 7m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f6b35870b8674bdaa3f37fef80f7b9f7 : SUCCESS in 43m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2e99c521ed4b43aea9f5a49bb33f9121 : SUCCESS in 41m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a44be43a09a5444182ac2d024afbc731 : SUCCESS in 19m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d07f8a1ab374d8394515a17e0e8b2bb : SUCCESS in 7m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/55bd1baa4621452aa2fad8d6c3d1b486 : SUCCESS in 32m 14s","accounts_in_message":[],"_revision_number":1},{"id":"1f007da4397d7ff24298b3c5e8d82cf88aad30ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 17:49:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"edc3f4d97dd7e8d4dbd3903d677a2f4ba0ac8ed4","revisions":{"edc3f4d97dd7e8d4dbd3903d677a2f4ba0ac8ed4":{"kind":"REWORK","_number":1,"created":"2021-03-19 19:44:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/86/781886/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/86/781886/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/781886/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/781886/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/86/781886/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/86/781886/1"}}},"commit":{"parents":[{"commit":"a1350a2c21166ea980690258ce921ef1d3876594","subject":"relinker: make relink step a subset of cleanup step","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1350a2c21166ea980690258ce921ef1d3876594"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-19 19:44:02.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-19 19:44:02.000000000","tz":-420},"subject":"Add test for relinking when the object-server failed to do it","message":"Add test for relinking when the object-server failed to do it\n\nIt\u0027s a little duplicative with test_relink_existing_data_meta_ts_files\nbut seems different enough to be worth testing.\n\nChange-Id: I729fdafbeef5bcb48526f5b7e131647fb5f49727\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/edc3f4d97dd7e8d4dbd3903d677a2f4ba0ac8ed4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/edc3f4d97dd7e8d4dbd3903d677a2f4ba0ac8ed4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
