)]}'
{"id":"openstack%2Fswift~781916","triplet_id":"openstack%2Fswift~master~I31d11af3fdba844c5cd632925b8fd774b109dbd1","project":"openstack/swift","branch":"master","topic":"includes","hashtags":[],"change_id":"I31d11af3fdba844c5cd632925b8fd774b109dbd1","subject":"Add --includes to s-m-s-r","status":"MERGED","created":"2021-03-19 21:44:29.000000000","updated":"2021-03-23 00:08:09.000000000","submitted":"2021-03-23 00:06:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"781916-includes","meta_rev_id":"236fc6892b45c73a8ced14999d3e7fbc8de30a6d","_number":781916,"virtual_id_number":781916,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-23 00:06:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2021-03-22 11:58:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2021-03-22 11:58:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-19 22:46:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-22 11:58:42.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2021-03-22 17:53:52.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"a73b506a6adc9070c36f1e38d47c3ea39b8cfcfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-03-19 21:44:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f9af966cbbb7680936c2db4f8535854b1bc7386","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 22:46:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6245cad46b5b493bbe4ae366d0f3afd0 : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/2e854159e4ea498d8813c4d0ff4ba713 : SUCCESS in 58m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bb1a0596657742ff8e64d78725d1e83b : SUCCESS in 47m 45s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/77eda2caabf74ca3a94598855e7dc5ed : SUCCESS in 9m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/89a47437b5aa4bddbeb3968bae79ad79 : SUCCESS in 6m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/da13043bbf794b2b99003dec54c6b584 : SUCCESS in 15m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/41913f624b5f4346b47781d112792fb2 : SUCCESS in 22m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c3016d3672c94294aaf8964094b2a550 : SUCCESS in 18m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b6e39fcaa19b49b1be5653d1da39fcdf : SUCCESS in 16m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8b2a0ff2c32f400ab09780bc8e2042e0 : SUCCESS in 15m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6eb090a85616479b960d665536ef46cc : SUCCESS in 26m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8772a154e2984014b4dd8d39e7aa515a : SUCCESS in 21m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6351c0df1bdd43e8a853207b72a78570 : SUCCESS in 29m 02s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/576524d541af4ca09affb38f83f01f0d : SUCCESS in 20m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a6e46c1f2a4245858959665ae58aa6b4 : SUCCESS in 20m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/33fe570d7b9e4247bf96d11bda50c49a : SUCCESS in 22m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9fe0705b65c944a3b87309fd2c519bee : SUCCESS in 6m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6768954a570a4da598dafb4b8e590874 : FAILURE in 26m 59s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4c02f325fc064605bcebf26b6b95248c : SUCCESS in 49m 22s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/7bd58e10065e465b891cb0b90d573e87 : SUCCESS in 48m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dbccd61e30434faabcd5d14367a4702e : SUCCESS in 41m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ad082f52e83c4c96b49e3559a4361128 : SUCCESS in 38m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8042a0b3bb7942a7a3e58fa289faa57c : SUCCESS in 16m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b371d7bd8d734e6abd8e1f993880b0a2 : SUCCESS in 7m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/28f01a51a0c14838bce50da4b79abb26 : SUCCESS in 30m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b52a6fa3dc0f4e109cafb14b9d81d713 : SUCCESS in 56m 33s","accounts_in_message":[],"_revision_number":1},{"id":"234f0b96bf4a0782770d0c6ddb1d24624a12575a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-22 11:58:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)\n\na useful addition","accounts_in_message":[],"_revision_number":1},{"id":"2870e71ac72a912c05604d5e9d05aab7c2faf406","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 12:02:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"58b367144b905bcc4a0888389114ecf075f95d77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 14:42:22.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5fa87d7654ebc0ad2bb9f7211f52dd60fbd0d157","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 17:49:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62c8b9a7108d4d8e8ed2b0ed6b0ca460 : SUCCESS in 9m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/d12f7a99fb50497ab88d7e10636a5b33 : SUCCESS in 1h 13m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1e8213198af746f790281417c0301011 : SUCCESS in 39m 05s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4b091b532ca5409ea9a8e5a2933d6fbb : SUCCESS in 10m 03s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/4dacd2f334d04b4aab765e2fc0b701bc : SUCCESS in 9m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/48cf8964f22d42e09e53349acbe63806 : SUCCESS in 11m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/63a0e24010c04a5b8330a00fac8d0a44 : SUCCESS in 12m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d6dca204d1f341acacf1c32d73122d4b : SUCCESS in 13m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d50099bfcfb54f938d785da569502768 : SUCCESS in 12m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d88082a975ed472586a123c84fe857fb : SUCCESS in 12m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/70862b5b9d5a471388a4cb0890a2f8a9 : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a50943cd3afa4babaff1e49662359b39 : FAILURE in 31m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ee7c71ce1caa4ffda389579726f16c27 : SUCCESS in 26m 00s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f100ead068dc4b7287f0b1ffa64cca25 : SUCCESS in 1h 09m 45s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/95b97716212948d480d943b48537d4b3 : SUCCESS in 34m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/848518092aef4267a8d772c974c27349 : SUCCESS in 25m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1fd8c7acb7c84273adefbe1631b0bf24 : SUCCESS in 8m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/266537b8daf94352905e9ea2dad6e576 : TIMED_OUT in 2h 01m 01s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/92f4ea4010024a1d96b18c34d50184b3 : SUCCESS in 1h 09m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/130fbad770ae4e0bb40e664e9c000fda : SUCCESS in 51m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5b81cd16533c484ebd7b8243b5163513 : SUCCESS in 35m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59548518c9be40d1b70be01acb99378d : SUCCESS in 16m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03ada3605c61464bb93b8dc50543991c : SUCCESS in 7m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4811323f53c74852810d3ef7aec39952 : SUCCESS in 24m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9eb365cdefe74d089d74133fa843ff1e : SUCCESS in 1h 01m 51s","accounts_in_message":[],"_revision_number":1},{"id":"35dd3318802e38f6da249c3d0440db2930620bb1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-22 17:53:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9a5b6619e7f31a4ca759ef2dc7516b6454e422e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 20:47:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b8406973d094f29817de10b56f64559 : SUCCESS in 12m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/18eb272970204a1d8ef3ed3bbd534541 : SUCCESS in 1h 14m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f3b10be950214e2cb86112136f726c79 : SUCCESS in 54m 19s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dddeb0bde2244a2cb13b964707d431cf : SUCCESS in 9m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0e6a8315e49d4d9481ec2d2242505fc1 : SUCCESS in 11m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f65fd7f5c3204a0c9c863d424720bd9d : SUCCESS in 16m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/548d43e74f904bb2a65b6580e302fe9a : SUCCESS in 15m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4a33f1d89958497a9b9ad96bb461aecd : SUCCESS in 16m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9a8b009ced9f426b8b5fa3109a46d457 : POST_FAILURE in 16m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f88b4b895d894dc8b3a2e998687f41de : SUCCESS in 27m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/23369fcbd3da41e582ab72d4785ef5f8 : SUCCESS in 27m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b428809070eb4ed7b56cb33f68f6f785 : SUCCESS in 24m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3967aef1861b4122ba47b1177c69bcb7 : SUCCESS in 22m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5268711b0d07440390a16c56515f5570 : SUCCESS in 22m 01s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b93173f789a846c98e2d0338fb7647a1 : SUCCESS in 20m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8140311c578f40a89430097d40646d03 : SUCCESS in 19m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/65462f993a8b4b0da0ab9300054c6c51 : SUCCESS in 7m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/78e29ed17d37418fa0612efc49fc7687 : FAILURE in 25m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0e284ebe50f54dafa56fd1dddb943d0b : SUCCESS in 49m 09s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/0f59ea9978814207aac87f4009531dac : SUCCESS in 49m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7e646c4ee13f474e99661a7449cc388e : SUCCESS in 1h 03m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/27a3659d502544e5befc0095cb0f63fd : SUCCESS in 58m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d3f247d53b64cc3b7c9bcb2624c9d3e : SUCCESS in 16m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fb1575239364078b9c2a9763a9d0992 : SUCCESS in 9m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/71b891cd9781412e81b9a7b7e2052dbe : SUCCESS in 26m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/941f505009204ef59a7195e50c09abb3 : SUCCESS in 1h 02m 03s","accounts_in_message":[],"_revision_number":1},{"id":"a85810f1c0c5281f3807247689326fb79d2bcbfc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-22 21:12:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"821588fa396f24873bea022dce30b58d66676d77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 23:07:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a46d24cdfca34c22bd8d85e87e9c1f09 : SUCCESS in 9m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/5e349ce0fe0845079afab3330a98c237 : SUCCESS in 1h 03m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7efb1fbb25c64db5a09a18e98ec6fb19 : SUCCESS in 44m 53s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/621a776115be4499adde2cae59fa8b68 : SUCCESS in 8m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/35c751d0faa844d4a4e203ed75e53c26 : SUCCESS in 6m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2ea62107be084e5cbca5650efa9abb7d : SUCCESS in 15m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8c924b0a42424ea1a024f6186da775e7 : SUCCESS in 15m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/da93911365fa4639aeac4e6cd773767a : SUCCESS in 18m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e784d24f450341dbb55636dbb5f2f4a9 : SUCCESS in 18m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/53386fd324c94145bb925fff0096d0b7 : SUCCESS in 18m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/41b93021da4848cf80c9041cedb380f6 : SUCCESS in 17m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9de55d3fb670406894998aa477a93a23 : SUCCESS in 19m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8cca8dfa2d944a53b72d8af57d6ab42d : SUCCESS in 17m 40s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6b9394d9f1ba48b9960762b3e8418c78 : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cc7e53505fb0400e888287295cab4f89 : SUCCESS in 20m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6af44169c1604b8a889387e9fec7455c : SUCCESS in 19m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2c2ff111fb6546f8b0310d4749936110 : SUCCESS in 6m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f8bd4a1b548d4f2ca233bab3991085fb : SUCCESS in 18m 07s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fe42644f349040c0a5faf7f905e08c97 : SUCCESS in 51m 31s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/6041a960d5aa4bfea99c85714eecca81 : SUCCESS in 49m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cdf68088108a47c49bcc00c32594e8a6 : SUCCESS in 39m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4915598e6a484848bb5d6e7cf7e14f82 : SUCCESS in 41m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/709c3e1fe315446abe8d18d52941640d : SUCCESS in 15m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/581047a3f08f485d909689f33e432632 : SUCCESS in 7m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a4c8b20440f744349468144eab67cb8a : SUCCESS in 30m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/270f3ac32485462ea5207234a48230ad : SUCCESS in 1h 02m 43s","accounts_in_message":[],"_revision_number":1},{"id":"1be29835500bb61f9a52aecc301d1bd14e85e942","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 23:07:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01c2b14e8a867b1c137ce59142f557117e2ecf55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 00:06:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b514451f753146cda1f773c84040115a : SUCCESS in 7m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/1d14f62d57614f72b3a71766768d2a71 : SUCCESS in 44m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0921919869c84b94ba4acc0c2086a16b : SUCCESS in 45m 56s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/cb44e66129dd443193d945aeb9c4d9ee : SUCCESS in 12m 11s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/914d1e6d5ac74ef7be9f583700a32f4a : SUCCESS in 8m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0a246883a42f42d9aa2ea86b5394e9a1 : SUCCESS in 12m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6f4cda084da148f59bde97aa09ab809a : SUCCESS in 18m 57s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8ea240e6eb0f462ea34a48d1f57f4378 : SUCCESS in 15m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/83e997b0d2fe4bb08a73a4cab3203ba0 : SUCCESS in 13m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0b524b568c724101add7b86a37ac835a : SUCCESS in 14m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f68f58c8266f42f0a09cffb6bca8a9b0 : SUCCESS in 14m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/51132105b1b84ae9b9fce858f55a7667 : SUCCESS in 22m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/81f66d74ba6846ceb2e53f55cbd0cc71 : SUCCESS in 17m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/294b7bd155cc4c758a1c2afcbcf55249 : SUCCESS in 15m 17s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/88c5e6ee8c4443e3b925ad1491cb6c80 : SUCCESS in 18m 07s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/948d97c47e724a4c9f0352ff7a570824 : SUCCESS in 22m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/33ffae40ad0e4a7ba5926ce221d62edb : SUCCESS in 7m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a311e0b67cd44b29bcfb0f77788990e7 : SUCCESS in 36m 14s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b55d805c772d4e6eabeb04e43a30c932 : SUCCESS in 46m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/12f7dc7db8db40eb9ce7353abdb0b195 : SUCCESS in 40m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4c94c7a578004cc4a4da0231fd83f117 : SUCCESS in 39m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8dc95202d6a45c58599b55118d93090 : SUCCESS in 12m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/347ebc5d16ea431d9ee16d8ace9ccab0 : SUCCESS in 6m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/37ed2fdbaa3d429b96c45f36616dbdf2 : SUCCESS in 33m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1e53e81d72a49878e6aea05e13a8be3 : SUCCESS in 56m 23s","accounts_in_message":[],"_revision_number":1},{"id":"8bfe4f153bc418e96b21a65d5edf1b5c4634b960","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 00:06:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"236fc6892b45c73a8ced14999d3e7fbc8de30a6d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 00:08:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12bc0a7a19054d1d93884d57fe4cdebd : SUCCESS in 1m 02s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/93f0698da63842078866db3a2d0ff71e : SUCCESS in 42s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/3954694316e9415ba07ce6eef66d68a0 : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"55386c836dc1e9fc466955aa4af5fce1728a5eb6","revisions":{"55386c836dc1e9fc466955aa4af5fce1728a5eb6":{"kind":"REWORK","_number":1,"created":"2021-03-19 21:44:29.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/16/781916/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/781916/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/781916/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/781916/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/781916/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/781916/1"}}},"commit":{"parents":[{"commit":"a7b9d1ea6953c4e8bcedb10392ced3a2b9cc286c","subject":"Merge \"Don\u0027t require swift be installed to have passing manager tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7b9d1ea6953c4e8bcedb10392ced3a2b9cc286c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-03-19 21:43:26.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-03-19 21:43:26.000000000","tz":-300},"subject":"Add --includes to s-m-s-r","message":"Add --includes to s-m-s-r\n\nChange-Id: I31d11af3fdba844c5cd632925b8fd774b109dbd1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55386c836dc1e9fc466955aa4af5fce1728a5eb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55386c836dc1e9fc466955aa4af5fce1728a5eb6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
