)]}'
{"id":"openstack%2Fswift~781931","triplet_id":"openstack%2Fswift~master~I758e7cd59855d468401511010f36823eb34edf27","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I758e7cd59855d468401511010f36823eb34edf27","subject":"Give functests another chance to succeed","status":"ABANDONED","created":"2021-03-20 04:57:33.000000000","updated":"2021-04-01 14:56:40.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d34e3c85b0441e60094aeaecaec5506bbab60d5e","_number":781931,"virtual_id_number":781931,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-20 06:41:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-20 06:41:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0890325736e22674f13c97d92f86c54d00bca278","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-20 04:57:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"535b57b5b9c42991792e68f27447c16a3724fbca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 06:41:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/73fb67cd199c4a45a213c4353676e5ba : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/587b72c6b757424f970af13b55484fde : SUCCESS in 8m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f3d2959ff601455bb5994799679e568f : FAILURE in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d0b298668c94425845a2eac3bd1e563 : SUCCESS in 5m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/9cb48800937b45c5a1b8bb5692b83d7f : SUCCESS in 1h 06m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6276fae0e91341388398565657222fd8 : SUCCESS in 48m 25s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c318682fe5014b019db01acaa5fa03da : SUCCESS in 8m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/56ec6d788eeb4153883efda50b345a94 : SUCCESS in 6m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/361967acf19f46649fe9b3c2279ba08e : SUCCESS in 14m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/387468ecbde94881aa57b362a050ab33 : SUCCESS in 14m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/694acfd9dbf34138bc80e187ff0b9915 : SUCCESS in 14m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/41b23ea1bf5544f8a5889f474771a99d : SUCCESS in 14m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e63b7f1ea2964756838dc0ceebec127e : SUCCESS in 22m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d908765faf884f29be11434491f02e15 : SUCCESS in 13m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0b736d460c53406eadeb84fc22269712 : SUCCESS in 12m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/02e1a9d07890428baecd6184543042ff : SUCCESS in 13m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/62d84884afd74be3a8d0f8d8bd90dcaf : TIMED_OUT in 1h 31m 28s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3067fa574d0a4dc694ce549643067087 : TIMED_OUT in 1h 31m 24s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/25838bf2514a46f6bcd5dfaa98a08b96 : TIMED_OUT in 1h 30m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7fdcbc0aa6604a6fa00b741241c50f96 : SUCCESS in 5m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2e16f51ebf6f45d58ecc11931ff34f97 : FAILURE in 33m 43s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3ec0c691a41e47cb9605a537ffa3287a : SUCCESS in 1h 43m 05s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/bf71b5c32403470fbc337ea3edf87225 : SUCCESS in 45m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bc46074157af4b7a8bb8b023994437e1 : FAILURE in 28m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/94ca16452cb046a582c12acb4b476296 : FAILURE in 23m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b09b2beb6b144e7dbe7bf4bca1a7bc27 : SUCCESS in 25m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9964649e5dc54db4a97eaec47feabde0 : SUCCESS in 5m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6f9297b4ad6c4696ae2d29053aa89e76 : FAILURE in 43m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6900c6234d454d3eb5d3ef12ec34b31b : SUCCESS in 51m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d34e3c85b0441e60094aeaecaec5506bbab60d5e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-01 14:56:40.000000000","message":"Abandoned\n\nGoing with https://review.opendev.org/c/openstack/swift/+/783091 -- using stestr wasn\u0027t too bad of a transition.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cb40e4f9561c6730636222426654d97b60b3c2cd","revisions":{"cb40e4f9561c6730636222426654d97b60b3c2cd":{"kind":"REWORK","_number":1,"created":"2021-03-20 04:57:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/781931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/781931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/781931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/781931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/781931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/781931/1"}}},"commit":{"parents":[{"commit":"8b07e5e732fe8e2187711fe303b5386ba6a38531","subject":"Merge \"relinker: add --policy option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b07e5e732fe8e2187711fe303b5386ba6a38531"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-20 04:54:05.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-20 04:54:09.000000000","tz":-420},"subject":"Give functests another chance to succeed","message":"Give functests another chance to succeed\n\nSide-effect: change the func test running to use pytest.\n\nos-testr is deprecated and while we could just back out\nhttps://github.com/openstack/swift/commit/9d7f71d5, nose is basically\ndead -- last commit was over 5 years ago. Our requirement that func\ntests run serially seems to pretty well negate any benefit of moving to\nstestr, so here we are.\n\nChange-Id: I758e7cd59855d468401511010f36823eb34edf27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb40e4f9561c6730636222426654d97b60b3c2cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cb40e4f9561c6730636222426654d97b60b3c2cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
