)]}'
{"id":"openstack%2Fswift~782757","triplet_id":"openstack%2Fswift~master~I21292e7991e93834b35cda6f5daea4c552a8e999","project":"openstack/swift","branch":"master","topic":"p-shard-repair","hashtags":[],"change_id":"I21292e7991e93834b35cda6f5daea4c552a8e999","subject":"Make ShardRange.OuterBound a proper singleton","status":"MERGED","created":"2021-03-24 14:16:06.000000000","updated":"2021-03-25 00:54:26.000000000","submitted":"2021-03-25 00:52:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"782757-p-shard-repair","meta_rev_id":"6086a183272bbea3cf1d91bed9efa5135be9a33d","_number":782757,"virtual_id_number":782757,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-25 00:52:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-24 18:14:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-24 18:14:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-24 15:41:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-24 18:14:25.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"a7efcd11abf7f4e251a27ac07e3164bff824361a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-24 14:16:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ba2fe746e26749eceb1eee4685855743959e7911","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 15:41:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/effbc4d4758b44218f479d2ee3325dac : SUCCESS in 11m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/65240fb0fa3c4339bc383a528d2051d9 : SUCCESS in 1h 05m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/755f120b81ee43c9b68eabbdc85a96eb : SUCCESS in 40m 58s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/009d689d5c1943ee814ebc2348ebe9a7 : SUCCESS in 11m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bb219c6ccad44dc2b7ea6ec123519e08 : SUCCESS in 9m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/78956334eebb4c8cb6f184d4b20fe369 : SUCCESS in 26m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2a039bdc2fe349b682dc4f0fcc2a08e2 : SUCCESS in 16m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5b9731c11b854c93bce8cad833e6791f : SUCCESS in 19m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3a019288b90b4755aafc673cb9fb3752 : SUCCESS in 15m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7018152d1cb04587b2b41384b6ed8f37 : SUCCESS in 15m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c19ba1aabfe841f4b758fe59063453e2 : SUCCESS in 24m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9c50fe60cbc44f39866dd65bb140f56d : SUCCESS in 22m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ffcd0b39b9e34129b00e3831e0492fd1 : SUCCESS in 23m 10s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e93283b289af44b79206e0f8206c9063 : FAILURE in 17m 46s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cacec3a33cf0456aa23dd29b83ad0c72 : FAILURE in 15m 32s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/78fe8fae079a4ef89bd11fc6f87981a1 : FAILURE in 20m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9275e815c79c49018b9ce79a9ffa70e7 : SUCCESS in 6m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3a6dc90d13c4453295a3a2b13489b3a8 : FAILURE in 17m 16s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/96fa9d37b7094e7c84f53ee28ca6f106 : SUCCESS in 45m 13s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/96f8319ff3f04982b05a047d054073ca : SUCCESS in 51m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b03001ad54a341ab95f32baa53674b6d : SUCCESS in 40m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/13cbf15a4b6f4a32b6fb2a0bd204fb81 : SUCCESS in 45m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f596b5d6c0d40899bc196ed9b269e68 : SUCCESS in 22m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14b285d73ee0402189fa48a03f6d16c6 : SUCCESS in 6m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dfd755f7c6614ccdb137dfcbacb10908 : SUCCESS in 28m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfa0652e6c9a4eab9dc98ae9f516d31f : SUCCESS in 56m 38s","accounts_in_message":[],"_revision_number":1},{"id":"a213587d4d484c69634c8d4ef113023a1a6ebd2d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-24 18:14:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f5a1a8fc2eca08ca978204419f8502ae4539b615","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 19:47:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a5ae15b527945d28cc7abdfb446cab2 : SUCCESS in 8m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/646c39129533478eb13591c0f56cc705 : SUCCESS in 1h 21m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8e78ad71094a4a4a88d304eed839020c : SUCCESS in 52m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0d0fa8621c2845b08765ae34cdf461dd : SUCCESS in 9m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/64b0fa323832441c8708c9ca2def1f96 : SUCCESS in 6m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c190c75cce2042829690cd1fdd693d73 : SUCCESS in 16m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a00ec4aaae3a470c85e73f92f5e58055 : SUCCESS in 29m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6bd69e893dd04cf4ae1b909a71e8d3d0 : SUCCESS in 29m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/88ec04b8f6534fc68abddb99123e7a45 : SUCCESS in 17m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/53677eede60b44229dde004b5ae292bf : SUCCESS in 28m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c670f9d6a2eb48a2a4aeff4b2d89a164 : SUCCESS in 20m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b180acf8b64145e9923bd42af2605626 : SUCCESS in 21m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cbe89ff6b21a40139f4165243ccd8b08 : SUCCESS in 17m 40s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/585fa0dbedda435fbcff94041230c014 : SUCCESS in 18m 07s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f2e825821e79477996b80b23aa878ab9 : SUCCESS in 17m 50s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b04c1d9d8c634df5bb68a635a72b30d4 : SUCCESS in 19m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bbe0427c41c746fb908af88ce7bdcae2 : SUCCESS in 9m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/22330f94df6b419eb7547e33b950c370 : SUCCESS in 24m 27s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2adc8cf2d2374c2ab2851e8d91cf138d : SUCCESS in 53m 44s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/0eae358998644962a7ad49a7976a85ee : SUCCESS in 45m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c3b32de7049f42c39e8b9108881aafee : SUCCESS in 51m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b83aa90b36ec4d80908fe5a465018908 : SUCCESS in 40m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30afe7b659c347cf8e1848d6c63e5301 : SUCCESS in 16m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb05c26604dd47e79bcdb42495d0451d : SUCCESS in 6m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ca0937dc3d954c4b9a8f4c41160527cd : SUCCESS in 51m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9dc49ca6f5a4c61971a8d53dd3bf42d : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":1},{"id":"2b761c54a99afb08d5b3c7ff5b420b46eeb45f21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 19:48:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ddb236c347a721267fdf876664f67f0f7caef2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 00:52:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5480d87adc364628ae2b14a10bedd63b : SUCCESS in 9m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/4240dab1ad2d4d6280355263fc4122ca : SUCCESS in 1h 06m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e4c1e25b7b524415aa4c9b678bdf84be : SUCCESS in 42m 49s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3de1e56ea81f40f2ab5e5c1b50f2537f : SUCCESS in 9m 51s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/bbb9b53aa5204e0491e43dda3b21d860 : SUCCESS in 8m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d87e24a065bc478c8f6557896450c6ac : SUCCESS in 19m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1266ad9746ad4230b01be7075aaccab7 : SUCCESS in 14m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/acad6fc99d904544a4a155c68bc3e839 : SUCCESS in 17m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1fba7dd5af634202bfb3b5452ff14c0d : SUCCESS in 14m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0becbfbde8ec4e60844a5c012acb0aa8 : SUCCESS in 19m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d81186d62ac84e21a117001c5863db4c : SUCCESS in 20m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0971977df4f64bcabc78101fd7489b08 : SUCCESS in 18m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/af1dce668a8b45b585197579dcea336e : SUCCESS in 19m 44s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f83af812cdb84b5cb12e49f01e5019d0 : SUCCESS in 21m 41s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/19969b03d3ef4f37a71b8bdb9d83945a : SUCCESS in 20m 31s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e9a5d1d590984ffcae83d34a152d349e : SUCCESS in 20m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/08139f6e383d41138a3311fc0d99ae26 : SUCCESS in 6m 09s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/25e9ebc402db49d3b4fcd3b2c0235658 : SUCCESS in 53m 19s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/12b98fc16d614bf699f92b2047a9a078 : SUCCESS in 54m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c74323c161e84881886bcd7be09afd3d : SUCCESS in 42m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d18150310c15432d91179ac0898d293d : SUCCESS in 44m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96112b384c4742998e7ba875c8fb3460 : SUCCESS in 15m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/020f138fdd454ae886f2da35ed5f2f4e : SUCCESS in 6m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/25d7597b2c374d27a0323b104e4578a5 : SUCCESS in 26m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7b1465b263344ab83e4226516d738e7 : SUCCESS in 1h 05m 14s","accounts_in_message":[],"_revision_number":1},{"id":"425240d90f38f942063a8598611ac25c430b2c37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 00:52:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6086a183272bbea3cf1d91bed9efa5135be9a33d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 00:54:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c787c05c865043fba6455f259f6821e4 : SUCCESS in 1m 03s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/857196f9d5b34b88867e2721f965b628 : SUCCESS in 42s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/d79775cc739940e0bf6589caf4105ba0 : SUCCESS in 47s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b17dd7ec75eaa4af6db62dd55f8a94d4c9f972bc","revisions":{"b17dd7ec75eaa4af6db62dd55f8a94d4c9f972bc":{"kind":"REWORK","_number":1,"created":"2021-03-24 14:16:06.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/57/782757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/782757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/782757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/782757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/782757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/782757/1"}}},"commit":{"parents":[{"commit":"3a41cbe67027e0abe2816a7b659a8fd01f95fd14","subject":"relinker: Allow multiple policies to be specified","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a41cbe67027e0abe2816a7b659a8fd01f95fd14"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-24 12:00:55.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-24 12:00:55.000000000","tz":0},"subject":"Make ShardRange.OuterBound a proper singleton","message":"Make ShardRange.OuterBound a proper singleton\n\nShardRange.OuterBound had the singleton-like property of all instances\nof the same class being equal, but did not prevent multiple instances\nbeing created.  Instead, ShardRange.MIN and ShardRange.MAX are used to\nhold single instances of each outer bound.\n\nThe ShardRange.OuterBound equality property was achieved by overriding\nthe default __eq__ method, but as a consequence the type was not\nhashable, preventing it being used in lists or as a dict key without\nalso overriding the default __hash__ method.\n\nThis patch makes ShardRangeOuterBound a proper singleton, removing the\nneed to override __eq__ or __hash__.\n\nChange-Id: I21292e7991e93834b35cda6f5daea4c552a8e999\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b17dd7ec75eaa4af6db62dd55f8a94d4c9f972bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b17dd7ec75eaa4af6db62dd55f8a94d4c9f972bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
