)]}'
{"id":"openstack%2Fswift~783292","triplet_id":"openstack%2Fswift~master~I160ba0172756ee79aa2551c16ac1a0b08b848515","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I160ba0172756ee79aa2551c16ac1a0b08b848515","subject":"diskfile: rewrite get_hashes and supplement docstring","status":"ABANDONED","created":"2021-03-26 12:20:23.000000000","updated":"2021-04-01 10:08:58.000000000","total_comment_count":5,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"b1a654da6b9b346770804b6c52f049e228cf1312","_number":783292,"virtual_id_number":783292,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-26 14:09:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-26 14:09:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-26 21:15:10.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"604e0395300e752e3b18f653c896696dd7be93c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-26 12:20:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"969a8ca3e7ba889f72a43c4b5513ef660e1dad5e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-26 12:28:09.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nI\u0027m most bothered about the diskfile unit tests getting back their assumptions. The relinker test assertion seems worth having. The get_hashes rewrite isn\u0027t crucial.","accounts_in_message":[],"_revision_number":1},{"id":"7612314a514f55f67545926bb2ee159de3ac2ffa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 14:09:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7561d500a6a1436093160fe1739bb66a : SUCCESS in 9m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/e6338f093b824c3791b511ed81d5a566 : SUCCESS in 1h 15m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3b2f50a04da24210ad69873ba99ea3b3 : SUCCESS in 46m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e916e2621f7d444eb88fdc4781a32899 : SUCCESS in 9m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d16f86eb0104410096644150b6b77887 : SUCCESS in 7m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fe34e03d07524c7a9e5fb1a3e30c19ef : SUCCESS in 19m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4ebc1da5cacc44b4ae8d09b391d3f281 : SUCCESS in 16m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8d04af1d9add47d7a7b1bdb4bd9fb7b2 : SUCCESS in 16m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4c522df1fbd949dabd3437759251c769 : SUCCESS in 13m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8f51d17b25d84684b22f01851baaeec9 : SUCCESS in 13m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/754e6e6b7f714e1f898edfdc7267a5a3 : SUCCESS in 19m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/edc5ab8692d74c8f8ea710ed55ea084c : SUCCESS in 17m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d92f235433b148b0b47531d9aa0e7a02 : FAILURE in 27m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1d3fc232192841298e6fc0104aa96bad : SUCCESS in 18m 39s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/dfc30b04b41b49a09c7896565ea99eef : SUCCESS in 17m 03s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/23d7f816148746cbb6f77f0e05f0b9af : SUCCESS in 19m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6bdc1bd8a5384451a1caa086c867a658 : SUCCESS in 6m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9270fd59cdcf4f78a40e9e8740f6792b : SUCCESS in 25m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/67f7e3a11f9143ed90d0db6287c481f5 : SUCCESS in 54m 21s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/1a301c91996f44e09eb3594ed7b45f28 : SUCCESS in 52m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8cc32bbdde4f4b43b092bd369656d9e5 : SUCCESS in 42m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c4e13e4a4cb849e0adb13c2347042dfd : SUCCESS in 38m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c54ee9302d334c9b832b17448464612f : SUCCESS in 14m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97dd00ae2fd547fbb6f96e05cda78f5e : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3bac6d6a8c434c7e89ea07a7aa06e160 : SUCCESS in 26m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83c8b68d22c04ac5a4589061b7fe41ec : SUCCESS in 54m 29s","accounts_in_message":[],"_revision_number":1},{"id":"bc55627074f8714fbe8ab4d75dcd2aea6ce19e25","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-26 21:15:10.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b1a654da6b9b346770804b6c52f049e228cf1312","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-01 10:08:58.000000000","message":"Abandoned\n\nbits squashed into https://review.opendev.org/c/openstack/swift/+/781908/4","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a057f532de0d73ff70fe1e3415dd835342c30b2e","revisions":{"a057f532de0d73ff70fe1e3415dd835342c30b2e":{"kind":"REWORK","_number":1,"created":"2021-03-26 12:20:23.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/92/783292/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/783292/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/783292/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/783292/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/783292/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/783292/1"}}},"commit":{"parents":[{"commit":"94cd4684e5a7f4ffda7a228da0df01e5915a7dc3","subject":"diskfile: Prevent get_hashes from creating missing partition dirs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94cd4684e5a7f4ffda7a228da0df01e5915a7dc3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-26 11:06:23.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-26 12:15:00.000000000","tz":0},"subject":"diskfile: rewrite get_hashes and supplement docstring","message":"diskfile: rewrite get_hashes and supplement docstring\n\nAdd create_partition optional arg to support the previously\n(often assumed) behaviour.\n\nFix unit tests that assumed get_hashes() did create the partition.\n\nAlso, add test assertion that relinker does not create otherwise empty\npartitions.\n\nChange-Id: I160ba0172756ee79aa2551c16ac1a0b08b848515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a057f532de0d73ff70fe1e3415dd835342c30b2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a057f532de0d73ff70fe1e3415dd835342c30b2e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
