)]}'
{"id":"openstack%2Fswift~783342","triplet_id":"openstack%2Fswift~master~I92d8fcb910b9ce05982f8d12070b92adbe6843b1","project":"openstack/swift","branch":"master","topic":"p-relinker-eexist-tombstone","hashtags":[],"change_id":"I92d8fcb910b9ce05982f8d12070b92adbe6843b1","subject":"WIP: relinker: tolerate EEXIST due to existing tombstones","status":"ABANDONED","created":"2021-03-26 15:51:49.000000000","updated":"2021-04-06 11:43:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8e12f1ef188d8ae3aace4d6c9d83e9b51ca8efd1","_number":783342,"virtual_id_number":783342,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-26 17:11:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-2,"date":"2021-03-26 20:06:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-26 17:11:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"96d899ce0d35998190c40b2e20740637d546a9c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-26 15:51:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"916e4f5f9b85cc3926c25e31df105bbb4441e6a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 17:11:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4bf04a71ce44a4c9f9f5d2b516d7e32 : SUCCESS in 11m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/4ac9c82436424e40a246644e9bbd40d4 : SUCCESS in 1h 02m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b0d77ba6c894417a91f55580f8114f9e : SUCCESS in 48m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c4e1ed569037491284fafee6d939a893 : SUCCESS in 10m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/934d54189aeb4d48998f36ecf9e226c2 : SUCCESS in 8m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/96ed0ba4505148368ed5c23efb28ba1d : FAILURE in 18m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/95a3c4cd714644b3b3f386c0e6ee9daa : FAILURE in 17m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b3c162b3303f4a989d24e86623285132 : FAILURE in 18m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/df5e8e2338554aac9d1e2891627b9766 : FAILURE in 15m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0eeb377d36614baba4f9a66200b95cfd : FAILURE in 31m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d7c26152b26849b98f960ab91ffb6f15 : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ceab98f968344946bee81c214da1ab4d : SUCCESS in 17m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6a0ffd8c44494c3bb17236362f903405 : SUCCESS in 17m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6bd6348463e9461fbe6e4442113d59d0 : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c03ba608f1b94acf8e62eb609f85fc18 : SUCCESS in 16m 51s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5f376beebdcf4475ba4294aa2adcba41 : SUCCESS in 18m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/17547b6d3ed64b5185803c45c68ac7ae : SUCCESS in 7m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/caabd3eda93c494fa54fa002f28ad2bd : SUCCESS in 20m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7b87e6ae9cc145adaee4ede490b70676 : SUCCESS in 44m 59s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/9530e5e0d853433eae4fc7355981d518 : SUCCESS in 44m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ebb7d7d400114a089835efbd726501ba : SUCCESS in 37m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a830962218af4e5c92b128741f239e13 : SUCCESS in 41m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f155673494b472e9a0dcbbeb87cbe9d : FAILURE in 14m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/758e65ea0dda41a9b492bc1e3d995b86 : SUCCESS in 6m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ff6948e2ed0d4779ad44522b34d55039 : SUCCESS in 25m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3340c2e54c53481685ddb11e3bdcc4da : SUCCESS in 1h 06m 58s","accounts_in_message":[],"_revision_number":1},{"id":"54fcc91e3fb351592fd85b4262d921b07de5a79a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-26 20:06:18.000000000","message":"Patch Set 1: Code-Review-2\n\nI currently prefer this solution https://review.opendev.org/c/openstack/swift/+/783467","accounts_in_message":[],"_revision_number":1},{"id":"8e12f1ef188d8ae3aace4d6c9d83e9b51ca8efd1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-06 11:43:03.000000000","message":"Abandoned\n\nalternative merged https://review.opendev.org/c/openstack/swift/+/783467","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a0c02e7922bc936be19047df02b86f9ab89cecb3","revisions":{"a0c02e7922bc936be19047df02b86f9ab89cecb3":{"kind":"REWORK","_number":1,"created":"2021-03-26 15:51:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/42/783342/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/783342/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/783342/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/783342/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/783342/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/783342/1"}}},"commit":{"parents":[{"commit":"810e4c0cdd4f8461eb6962faa20adbdecfa94495","subject":"Merge \"Get rid of strip_self\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/810e4c0cdd4f8461eb6962faa20adbdecfa94495"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-26 15:43:27.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-26 15:51:34.000000000","tz":0},"subject":"WIP: relinker: tolerate EEXIST due to existing tombstones","message":"WIP: relinker: tolerate EEXIST due to existing tombstones\n\nIf a previous partition power increase failed to cleanup all files in\ntheir old partition locations, then during the next partition power\nincrease the relinker may find the same file to relink in more than\none source partition. This currently leads to an error log due to the\nsecond relink attempt getting an EEXIST error.\n\nThis patch makes the relinker more tolerant of EEXIST errors by\nchecking if the existing file is a valid diskfile that is at least as\nnew as the file being linked, and not logging an error if that is the\ncase.\n\nThis relaxed error logging is only applied to tombstone files.\n\nChange-Id: I92d8fcb910b9ce05982f8d12070b92adbe6843b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0c02e7922bc936be19047df02b86f9ab89cecb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0c02e7922bc936be19047df02b86f9ab89cecb3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
