)]}'
{"id":"openstack%2Fswift~783467","triplet_id":"openstack%2Fswift~master~If9beb9efabdad64e81d92708f862146d5fafb16c","project":"openstack/swift","branch":"master","topic":"p-relinker-eexist-tombstone-v2","hashtags":[],"change_id":"If9beb9efabdad64e81d92708f862146d5fafb16c","subject":"relinker: retry links from older part powers","status":"MERGED","created":"2021-03-26 20:04:29.000000000","updated":"2021-04-06 08:39:58.000000000","submitted":"2021-04-02 00:27:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":27,"unresolved_comment_count":13,"has_review_started":true,"submission_id":"783467-p-relinker-eexist-tombstone-v2","meta_rev_id":"c27d9132689e359725025443fe2ae56e7bb8f0f4","_number":783467,"virtual_id_number":783467,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-02 00:27:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-01 21:34:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-01 21:34:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-26 21:07:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-28 23:20:50.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2021-04-01 05:32:15.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2021-04-01 21:34:02.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"c7cad4c16a88d2887551a8455c025dfaedd0c5c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-26 20:04:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cddf34df7a10f803c382c188944e8263497608bc","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-26 20:05:31.000000000","message":"Patch Set 1: Code-Review-1\n\nGetting this off my laptop before the weekend. I need to check over my work and tidy up tests.","accounts_in_message":[],"_revision_number":1},{"id":"8a2f9563433be29b9dca08f02e86933c150854ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 21:07:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc7c1dcf83a34e46bd17f0abf1f8f3a4 : SUCCESS in 10m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/f88bd15fcc9543eba509c4484ad1ff13 : SUCCESS in 1h 00m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9dbc66038d2749faa19b39f4966b75a5 : SUCCESS in 43m 55s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2cf001d2a659499fb331c92cfabc4f9b : SUCCESS in 10m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/36e111fa40fe4e5baebb9525b6bbb017 : SUCCESS in 10m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/02528f52c8a94f89b56861172871125c : SUCCESS in 16m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d2e188456a664185b28619177138352b : SUCCESS in 18m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fcda35b00f1245dd961dfb9662dc9ce5 : SUCCESS in 16m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/56beadadb5ec4545bd402aa3da2ebfdf : SUCCESS in 14m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f1c493dc8ab34519bc5bc5b37a73836a : SUCCESS in 14m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d9a9fd3c3c2047409508d63c7f73ef5d : SUCCESS in 20m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c3e2481d2a9047428d2e9a134d636783 : SUCCESS in 16m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/607e0689a843411d851f83f79e2938d7 : SUCCESS in 16m 34s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b986b13f2c054374bdb67dad9377ccc8 : SUCCESS in 18m 47s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5c9f7fb256514b94b86505c31b6736c9 : SUCCESS in 19m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ff644d2192c6473788870e65fe49dc3d : SUCCESS in 18m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b231cd1ef3c5419f9cc7f32d67602b7c : SUCCESS in 7m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2a1b9bad53884a51abb2528d0e9182d8 : SUCCESS in 20m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4b5f7f0bc4d8408e992de22f0b2a8509 : SUCCESS in 46m 31s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/25a7afa8532b4476b1da2c02eda4cf1d : SUCCESS in 47m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/50934613fc7d4b0480ff2104a25e183b : SUCCESS in 40m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/70650b37bc814c27aa2e754b4f41d3de : SUCCESS in 48m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9fcee60fc11645f0bdf9a57cdd879cf9 : SUCCESS in 15m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17f10627dda44266b33166d6cfca1ee3 : SUCCESS in 6m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/efefb155c49443588d64ac38513bc677 : SUCCESS in 29m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0514260158ca45d280531d398051d374 : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":1},{"id":"2113df21ef3d3d12de0e68789960b25eb3666f40","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-03-28 23:20:50.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"50a5ef9ea8aa1ac5ff3530d6d3ecb19f191809d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-29 16:05:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b68fc6bc0720073927e43a54006d21005cea7ebb","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-29 16:06:45.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d527ee0334fe2bf6fe43ad453c845b940412d6e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 18:45:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47f5e8f76c8f4048b2f7fe68090c5513 : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/800ce277573f4f638e556fbf496409ef : SUCCESS in 58m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e2d6c39dffef4607af15927719e7ac03 : SUCCESS in 43m 04s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1a43711c6cbd4c54a48307482dd9a098 : SUCCESS in 12m 02s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4449b1d1d77440a39c1e49d871507866 : SUCCESS in 6m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/55af387570d743649307aed0c5934099 : SUCCESS in 17m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c3462ce3e467451fb890e78c88d0c09e : SUCCESS in 18m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/42cdcaa7ae6744b6b56e9d49d7e6ac44 : SUCCESS in 22m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c696d8f9174046ecb466913be52f99df : SUCCESS in 17m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4e6fc8b2e8b7431e8a149e9808fe4d26 : SUCCESS in 22m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/67510dc708474c1cb6095cfb4e4e3b58 : SUCCESS in 20m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2add23fe3a40447ba1768fdf741e437b : SUCCESS in 21m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2a4c37b2d8a14c7db7e282cf8a26791a : SUCCESS in 22m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a85d88899749496c9f97398ffbbb9a12 : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/dcbd9a806cbb46c99b58b87801f6e5e4 : SUCCESS in 21m 07s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/43761b415dfb4416b98a10254771913b : SUCCESS in 20m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bf8e0ec21aaa4fc2846d1d1cfa2930f9 : SUCCESS in 6m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5d958940c0f24513aa3cdaec7b5f89de : SUCCESS in 26m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4ec513aa878747f2b755b2f19607cb31 : SUCCESS in 43m 47s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/6e481d5c4e5043339a42a0fab2fd4232 : SUCCESS in 48m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/78a7608baa5349c2b9ccfb12791a3df9 : SUCCESS in 46m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bbdc912ba1cf4fb29c48224e4ba71de7 : SUCCESS in 41m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4893569556754e3fb02f96d000db9edb : SUCCESS in 14m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45adc4688aa34648875e96fa85d128d6 : SUCCESS in 6m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/343c1270fec7455182324f7bbe560a9c : SUCCESS in 29m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/131b5d89f8cb4bf2bade0d7a45227dc7 : SUCCESS in 1h 23m 47s","accounts_in_message":[],"_revision_number":2},{"id":"400106ceaeae610f506f9a85b67c511dd4faa9b2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-01 05:32:15.000000000","message":"Patch Set 2:\n\n(9 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e691715df7c4d6cc81ffdbe03f24d905688b7d13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-01 17:44:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"08efb67d31fbd0cfa41efd0a5d102cf15323252d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-01 17:44:41.000000000","message":"Patch Set 3:\n\n(9 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c10cb1323812e81a230769e03bca9b720cb95e9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-01 17:57:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"89287108369733119d97e98503f0d8f7b15dbc07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 19:01:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef577867c44f43e7ac6ba87cc3fcface : SUCCESS in 9m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/fff8004548174e6a89ffdb01c1fb48fa : SUCCESS in 53m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5d43cc3924e943a1b61ae39691d222e8 : SUCCESS in 34m 58s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fafc0c7b37e541a493d2fb138544f0be : SUCCESS in 9m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/36d5687101404e0abdd7691612f9bf96 : SUCCESS in 7m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5285a3c804dc4307986f0910b265dba4 : SUCCESS in 18m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ecf821731ae64a488f7c91af4291dcac : SUCCESS in 15m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2b0cf2cb2309489399983298b80235cf : SUCCESS in 19m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/06c70c59eb7a4b53ad126833f268971f : SUCCESS in 14m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/914c62ec9d5e4a2d8f81da0d441ce43a : SUCCESS in 15m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bd5f1fc02c384823afb21b5151dabca1 : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0194a73c4a6b450ea84db34212c8fc47 : SUCCESS in 19m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/24a4aa717d1a413fbb40b9cf7a02b9ae : SUCCESS in 19m 38s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c1222c01ff3549f5859b8e2fdd17b97e : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/53a2e4194ce14a5cad47bd13d5e592cd : SUCCESS in 19m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8c1740051ce541e8b7009b175663c9a9 : SUCCESS in 17m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8f2c0b0e0a00492e8a96b2baed9a168f : SUCCESS in 6m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/52b199d796874df58ff23d1b3b48ee24 : FAILURE in 25m 25s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d86f0b8d2c7b4fb2a2ccf730f62d685c : SUCCESS in 48m 46s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/944cba1786f94dd7969eb5612f5e92c2 : SUCCESS in 43m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/641644d5f22c47a79eb186e82a7ebefe : SUCCESS in 38m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5b5528076f74495686cb6fabf0c78f8b : SUCCESS in 36m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eeaf600710b64d5daeac4fe009b336e0 : SUCCESS in 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d88a53c3f074d869d9dd249543494b0 : SUCCESS in 6m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6de340ba63f2421796705cac0d6540cb : FAILURE in 32m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19b99a54ce264822be1893f9060e51a4 : SUCCESS in 48m 19s","accounts_in_message":[],"_revision_number":4},{"id":"23eae7ee8b7db24fbb06e7eaeaa7f150fdb70577","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-01 21:34:02.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(4 comments)\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c0dea35f51227cac4e0fdb0921013378a3b2f262","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 22:50:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59a66d0cf5ce43318ee080a7bdf38a64 : SUCCESS in 8m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/fb39945345684dd3abc570ffdda00f0d : SUCCESS in 1h 12m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c7f8fcd39b69472c8ebb496d1c7cda01 : SUCCESS in 54m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c308fb248c134fb4801157f9c40f45b2 : SUCCESS in 8m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e6dc5818de654fb1a8f85401e2c42f06 : SUCCESS in 6m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/87a9e4b322284bcab3a2a78a265f12a6 : SUCCESS in 14m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cf86bcf047ca4c58825bece552d584b0 : SUCCESS in 26m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/79ed51f04f8041d68b7a9fe2b3125f4d : SUCCESS in 19m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/485d1ad225de4bcf83b09b055e7d90e6 : SUCCESS in 23m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/20103366b4194167b3b9846a91773f2a : SUCCESS in 17m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/19c2df44cd8246feab767d695875c72d : SUCCESS in 24m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/baea5a9172024db69669ab250f2b0827 : SUCCESS in 18m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/59b696ef193a4023b40bc57bd67ecefa : SUCCESS in 20m 06s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7c3840ecf62846faa93e7d953b4531c9 : SUCCESS in 17m 51s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2940c6e59c614bb3b0c5a429896a31d5 : SUCCESS in 27m 31s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a8e435b8a2294496aad86673ae7e4770 : SUCCESS in 18m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a68c9933b11440c8a0d7bc7310c6e21e : SUCCESS in 7m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8dbeb3537a0d4dceaa8bf99de3ccbb0e : SUCCESS in 23m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/da560eb6154348769b31403f8c7e77e6 : SUCCESS in 41m 34s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/cb5a3dadf19746d6838b25fe371175b3 : SUCCESS in 43m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6b00602f894d45c4ab7357856f8c8351 : SUCCESS in 33m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4c98c82ba4364b3c816c9fe0d7504df0 : SUCCESS in 39m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a1547ec1e404789b0b3da70c2b28b0f : SUCCESS in 15m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bd988d96b22452c997ecfdc95c4381e : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/db882f1461ee4bacb5be5fcdebd4fd47 : SUCCESS in 26m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b25709a469674bdd91e2ae8a0edf0de4 : SUCCESS in 47m 33s","accounts_in_message":[],"_revision_number":4},{"id":"8f056fb16bc0e3a9a299e2110fa925289ae8f964","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 22:50:34.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5224b67d77c76c758eb85d291dff5a95040a842e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 00:27:21.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9b0ae0779424436bbd1654d396253a3 : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/4472392b97624a7599da557289c45d3f : SUCCESS in 1h 31m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b5a993e9fcea4eff85d9557981b8d8ea : SUCCESS in 48m 53s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/98bd547d7dae4059b70fd91386cdfe6d : SUCCESS in 10m 26s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/b70084883a2b4454b6a2d1a8e6ed9402 : SUCCESS in 7m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/299a4f20aeb2497281fb4285385d360f : SUCCESS in 14m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/42f134fc82ad48a4a19212ec7680cb0f : SUCCESS in 17m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1c27b9910d7744d8b60f6affd08b8d3c : SUCCESS in 23m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b79866a0da1f4454a833991ed1b3cbf0 : SUCCESS in 15m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0d4063ee43a849c2a7a2e9ebe398f562 : SUCCESS in 17m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2bf1ac6f62964b38b9bef90c7236a620 : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eaa534ad8a6a4e398c1daab67385bf4b : SUCCESS in 17m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8047346c8fa64b0faf2ea05b60a28e29 : SUCCESS in 20m 38s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5fa9b5073fa54b76afae129781c2e1fc : SUCCESS in 20m 15s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e2c59c5660d844f796c35271fa8000bd : SUCCESS in 18m 32s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/009232bf600141bab8f712a919f8f913 : SUCCESS in 17m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b8c1733e3ab8481e9b273d07e72d33a0 : SUCCESS in 6m 26s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/85731db2dd194d02a2270855bed6e465 : SUCCESS in 47m 13s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/822e90a17164428bb9b006bf7227aeb3 : SUCCESS in 47m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/24f95f03a0a240cfbf85f096db0d8c9e : SUCCESS in 38m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e860a4fb915048dfb85c7705678d9eab : SUCCESS in 44m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/880c60cbbe774d3d9ba58ad1003d500d : SUCCESS in 15m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9926d4372544be7929056f8d9a49048 : SUCCESS in 5m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9d3667c2ab22463db2ec60eec7f7a223 : SUCCESS in 24m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b981b3d8b23343818e5c19576ea250c7 : SUCCESS in 1h 00m 24s","accounts_in_message":[],"_revision_number":4},{"id":"9a79105f2e069b488157ea9229946e6263e05271","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 00:27:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"0ab09b92b1779b8cbb03603fa8678cb5d747c725","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 00:28:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7493dd5a2baa4f58ae184f81a9cc4c1b : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e8d9a57a0e034cb2aa3eae97b62412c8 : SUCCESS in 43s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/21f48517091c44d5a1bcc72172aef191 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c27d9132689e359725025443fe2ae56e7bb8f0f4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-06 08:39:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3bdd01cf4a1c8ee9ac953850106da505d381aa44","revisions":{"6f712cd25892d5c834a881c8d32a6b274d4edf66":{"kind":"REWORK","_number":1,"created":"2021-03-26 20:04:29.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/67/783467/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/783467/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/783467/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/783467/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/783467/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/783467/1"}}},"commit":{"parents":[{"commit":"810e4c0cdd4f8461eb6962faa20adbdecfa94495","subject":"Merge \"Get rid of strip_self\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/810e4c0cdd4f8461eb6962faa20adbdecfa94495"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-26 13:41:36.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-26 20:01:37.000000000","tz":0},"subject":"WIP relinker: retry links from older part powers","message":"WIP relinker: retry links from older part powers\n\nIf a previous partition power increase failed to cleanup all files in\ntheir old partition locations, then during the next partition power\nincrease the relinker may find the same file to relink in more than\none source partition. This currently leads to an error log due to the\nsecond relink attempt getting an EEXIST error.\n\nWith this patch, when an EEXIST is raised, the relinker will attempt\nto create/verify a link from older partition power location to the\nnext part power location, and if such a link is found then suppress\nthe error log.\n\nThe error is only supressed for tombstones, and the relinker only\nlooks back over the previous two partition powers.\n\nChange-Id: If9beb9efabdad64e81d92708f862146d5fafb16c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f712cd25892d5c834a881c8d32a6b274d4edf66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f712cd25892d5c834a881c8d32a6b274d4edf66"}]},"branch":"refs/heads/master"},"4b83ae24dcbb95f07be6ec9e8b11dde4faffa5fe":{"kind":"REWORK","_number":2,"created":"2021-03-29 16:05:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/67/783467/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/783467/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/783467/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/783467/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/783467/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/783467/2"}}},"commit":{"parents":[{"commit":"810e4c0cdd4f8461eb6962faa20adbdecfa94495","subject":"Merge \"Get rid of strip_self\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/810e4c0cdd4f8461eb6962faa20adbdecfa94495"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-26 13:41:36.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-29 16:00:35.000000000","tz":60},"subject":"relinker: retry links from older part powers","message":"relinker: retry links from older part powers\n\nIf a previous partition power increase failed to cleanup all files in\ntheir old partition locations, then during the next partition power\nincrease the relinker may find the same file to relink in more than\none source partition. This currently leads to an error log due to the\nsecond relink attempt getting an EEXIST error.\n\nWith this patch, when an EEXIST is raised, the relinker will attempt\nto create/verify a link from older partition power locations to the\nnext part power location, and if such a link is found then suppress\nthe error log.\n\nDuring the relink step, if an alternative link is verified and if a\nfile is found that is neither linked to the next partition power\nlocation nor in the current part power location, then the file is\nremoved during the relink step. That prevents the same EEXIST occuring\nagain during the cleanup step when it may no longer be possible to\nverify that an alternative link exists.\n\nFor example, consider identical filenames in the N+1th, Nth and N-1th\npartition power locations, with the N+1th being linked to the Nth:\n\n  - During relink, the Nth location is visited and its link is\n    verified. Then the N-1th location is visited and an EEXIST error\n    is encountered, but the new check verifies that a link exists to\n    the Nth location, which is OK.\n\n  - During cleanup the locations are visited in the same order, but\n    files are removed so that the Nth location file no longer exists\n    when the N-1th location is visited. If the N-1th location still\n    has a conflicting file then existence of an alternative link to\n    the Nth location can no longer be verified, so an error would be\n    raised. Therefore, the N-1th location file must be removed during\n    relink.\n\nThe error is only suppressed for tombstones, and the relinker only\nlooks back over the previous two partition powers.\n\nCloses-Bug: 1921718\nChange-Id: If9beb9efabdad64e81d92708f862146d5fafb16c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b83ae24dcbb95f07be6ec9e8b11dde4faffa5fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4b83ae24dcbb95f07be6ec9e8b11dde4faffa5fe"}]},"branch":"refs/heads/master"},"d3cfc5fca7e11824041e4a59ca7a0f38a65c41a7":{"kind":"REWORK","_number":3,"created":"2021-04-01 17:44:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/67/783467/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/783467/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/783467/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/783467/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/783467/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/783467/3"}}},"commit":{"parents":[{"commit":"810e4c0cdd4f8461eb6962faa20adbdecfa94495","subject":"Merge \"Get rid of strip_self\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/810e4c0cdd4f8461eb6962faa20adbdecfa94495"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-26 13:41:36.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-01 17:37:44.000000000","tz":60},"subject":"relinker: retry links from older part powers","message":"relinker: retry links from older part powers\n\nIf a previous partition power increase failed to cleanup all files in\ntheir old partition locations, then during the next partition power\nincrease the relinker may find the same file to relink in more than\none source partition. This currently leads to an error log due to the\nsecond relink attempt getting an EEXIST error.\n\nWith this patch, when an EEXIST is raised, the relinker will attempt\nto create/verify a link from older partition power locations to the\nnext part power location, and if such a link is found then suppress\nthe error log.\n\nDuring the relink step, if an alternative link is verified and if a\nfile is found that is neither linked to the next partition power\nlocation nor in the current part power location, then the file is\nremoved during the relink step. That prevents the same EEXIST occuring\nagain during the cleanup step when it may no longer be possible to\nverify that an alternative link exists.\n\nFor example, consider identical filenames in the N+1th, Nth and N-1th\npartition power locations, with the N+1th being linked to the Nth:\n\n  - During relink, the Nth location is visited and its link is\n    verified. Then the N-1th location is visited and an EEXIST error\n    is encountered, but the new check verifies that a link exists to\n    the Nth location, which is OK.\n\n  - During cleanup the locations are visited in the same order, but\n    files are removed so that the Nth location file no longer exists\n    when the N-1th location is visited. If the N-1th location still\n    has a conflicting file then existence of an alternative link to\n    the Nth location can no longer be verified, so an error would be\n    raised. Therefore, the N-1th location file must be removed during\n    relink.\n\nThe error is only suppressed for tombstones. The number of partition\npower location that the relinker will look back over may be configured\nusing the link_check_limit option in a conf file or --link-check-limit\non the command line, and defaults to 2.\n\nCloses-Bug: 1921718\nChange-Id: If9beb9efabdad64e81d92708f862146d5fafb16c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3cfc5fca7e11824041e4a59ca7a0f38a65c41a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d3cfc5fca7e11824041e4a59ca7a0f38a65c41a7"}]},"branch":"refs/heads/master"},"3bdd01cf4a1c8ee9ac953850106da505d381aa44":{"kind":"REWORK","_number":4,"created":"2021-04-01 17:57:45.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/67/783467/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/783467/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/783467/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/783467/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/783467/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/783467/4"}}},"commit":{"parents":[{"commit":"810e4c0cdd4f8461eb6962faa20adbdecfa94495","subject":"Merge \"Get rid of strip_self\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/810e4c0cdd4f8461eb6962faa20adbdecfa94495"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-26 13:41:36.000000000","tz":0},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-01 17:56:57.000000000","tz":60},"subject":"relinker: retry links from older part powers","message":"relinker: retry links from older part powers\n\nIf a previous partition power increase failed to cleanup all files in\ntheir old partition locations, then during the next partition power\nincrease the relinker may find the same file to relink in more than\none source partition. This currently leads to an error log due to the\nsecond relink attempt getting an EEXIST error.\n\nWith this patch, when an EEXIST is raised, the relinker will attempt\nto create/verify a link from older partition power locations to the\nnext part power location, and if such a link is found then suppress\nthe error log.\n\nDuring the relink step, if an alternative link is verified and if a\nfile is found that is neither linked to the next partition power\nlocation nor in the current part power location, then the file is\nremoved during the relink step. That prevents the same EEXIST occuring\nagain during the cleanup step when it may no longer be possible to\nverify that an alternative link exists.\n\nFor example, consider identical filenames in the N+1th, Nth and N-1th\npartition power locations, with the N+1th being linked to the Nth:\n\n  - During relink, the Nth location is visited and its link is\n    verified. Then the N-1th location is visited and an EEXIST error\n    is encountered, but the new check verifies that a link exists to\n    the Nth location, which is OK.\n\n  - During cleanup the locations are visited in the same order, but\n    files are removed so that the Nth location file no longer exists\n    when the N-1th location is visited. If the N-1th location still\n    has a conflicting file then existence of an alternative link to\n    the Nth location can no longer be verified, so an error would be\n    raised. Therefore, the N-1th location file must be removed during\n    relink.\n\nThe error is only suppressed for tombstones. The number of partition\npower location that the relinker will look back over may be configured\nusing the link_check_limit option in a conf file or --link-check-limit\non the command line, and defaults to 2.\n\nCloses-Bug: 1921718\nChange-Id: If9beb9efabdad64e81d92708f862146d5fafb16c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3bdd01cf4a1c8ee9ac953850106da505d381aa44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3bdd01cf4a1c8ee9ac953850106da505d381aa44"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
