)]}'
{"id":"openstack%2Fswift~783476","triplet_id":"openstack%2Fswift~master~I8a48b4dd9ec7a85e61221d9a495837f0cca0a5cb","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I8a48b4dd9ec7a85e61221d9a495837f0cca0a5cb","subject":"WIP: Add an audit watcher to find orphaned s3api segments","status":"NEW","created":"2021-03-27 01:05:30.000000000","updated":"2022-01-19 22:04:35.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":8,"unresolved_comment_count":6,"has_review_started":true,"meta_rev_id":"4f694236396fd299e88028272b246609ab0a7b2d","_number":783476,"virtual_id_number":783476,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-12-22 22:31:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-27 02:05:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-27 02:32:57.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2021-03-29 05:56:44.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"5692a8d7dfe90d644583294628c5effa4f76ba20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-27 01:05:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c572fdff25bbba9cc4f0e7d98bb8069b05546eba","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-27 01:15:22.000000000","message":"Patch Set 1: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"bd7809925f3de3a56fe5eb7ea1cd6d86d2c36429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 02:05:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3e78a2195e4e47d0b969d5a01cf93150 : SUCCESS in 2m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aee6bba990f4fd7ae0fb77fd61725bc : SUCCESS in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/f49a53cb98f948e18df06d124d301e5f : SUCCESS in 57m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2e3be7067e134b4f82c2e4b367231505 : SUCCESS in 40m 25s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/103b44da304540558417d26c68170c7e : SUCCESS in 9m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bba8c916d94d43e09d242af1681c5466 : SUCCESS in 8m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0d67494cc16446748831cf792ae1be83 : SUCCESS in 14m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/98d3b648f15f4a168bdccf53675fbd25 : SUCCESS in 14m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f8e580e881f348b7b2a5c15203cd87f7 : SUCCESS in 19m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/eb347a50ecc04a44b53c42185733d404 : SUCCESS in 19m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f9a17da7f04b479ba9aabce59b1eedba : SUCCESS in 18m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ade9d05ef5184e1083c1e88aa5bb41d9 : SUCCESS in 17m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7e8e19eb1e2e4ddd9b5b941fadab6279 : SUCCESS in 20m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cd7f81ec165945c3a13d0eb61d9ef60c : SUCCESS in 20m 36s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c4b042ff8b2d4dc1ab8f66e3ae4e50d7 : SUCCESS in 17m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4801743a0d284242925df4cac663febe : SUCCESS in 18m 48s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ed95e625a71d4b069b2cd54f2ae7612e : SUCCESS in 17m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b01798f346bf42faada2967adef00730 : SUCCESS in 9m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0fdea4e3b10842f998422055d3fdd7c3 : SUCCESS in 26m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/aa9c7d81fe7c429d99e7fe4e47a2e301 : SUCCESS in 48m 38s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/2d8b4bf65b464454a6ec244ed20673bb : SUCCESS in 48m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/10105a4cb64e4f1b8d9d0e92768c736f : SUCCESS in 37m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8cd6036972114c3dadcf7578d3244d32 : SUCCESS in 37m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/490c8b172d54410ab82127834d1ad885 : SUCCESS in 16m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cc46e3fe4d243c985093cf9e83ab549 : SUCCESS in 6m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/620baccd1ce9458d97a2fa1c59d62822 : SUCCESS in 26m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cefd43ddff08457bbed2173e30f87782 : SUCCESS in 55m 16s","accounts_in_message":[],"_revision_number":1},{"id":"8d3edf851942d6a190bf8d5d86e474f5de04f531","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-03-27 02:32:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0256ae1571d5a2d2ad37d4288a0d60f181f66b5","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-03-29 05:56:44.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"339f739058fb152b617799271ae440cfeed7f021","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-30 19:38:09.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"59542193afe1c035576fae950fd78ffc6664009d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-12-22 21:28:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"21ae01d8b1b00e82dfb8e768ad01ecd2bd8d67b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 22:31:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/eec9a014bd4c4fc7ace8a71fde9e95f9 : SUCCESS in 2m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/450d7cb8a75c49988e9833f3e6efed13 : SUCCESS in 8m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/4f39e6f2bc0349029ee5a515dffc4bf3 : SUCCESS in 58m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9a5d79c6609c457db4a3706c39fc9305 : SUCCESS in 43m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e6b9272f3c14771a09a21e57bd7f13f : SUCCESS in 47m 04s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/960987313ac1470a9e0738eb496a80fb : SUCCESS in 8m 06s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9b549fec76c24d088438a0b441ae8ae3 : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/25b0cdb9552c41c0a5a57bd1ab663621 : SUCCESS in 15m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/978222a5d7dd408fa1b3c2a09a726e87 : SUCCESS in 15m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/447157407572432c9a20564bf3b1e41c : SUCCESS in 14m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7a99524eb1e449ec98ac4a1c0c6f6951 : SUCCESS in 26m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e66d58ebd51e4c8c8666f6a66276dafe : SUCCESS in 27m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cd6f5dc6c27f47a2b016dc971cc4dca7 : SUCCESS in 25m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6c3938b3dd594d5996106f6fd771d3e5 : SUCCESS in 14m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/05bb009f9d3d4eb3a420da6868817e5c : SUCCESS in 18m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/11f00e3d17c74a1486a50ff1af88a80e : SUCCESS in 15m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/48626011d3994fbf8baa00ebc6d28d01 : SUCCESS in 6m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b361e9e90807450ca8e801da4334b332 : SUCCESS in 23m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2dd3e48b865f4ad4aa91fdd769e8a651 : SUCCESS in 49m 20s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/b02a0b34c9d04f80bd217901d823f370 : SUCCESS in 52m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c1fc704eea264e0e98478181747279d4 : SUCCESS in 39m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f633afc03e444a5793124eb808eee541 : SUCCESS in 39m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f573130103404522aa9ba3320d707d90 : SUCCESS in 15m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05f5b2f932be413fae3652f175fd7ade : SUCCESS in 5m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ff54d3e9916444559d48c228a4dede2c : SUCCESS in 20m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1b836da93f3470c9c2fdf4a761c45d0 : SUCCESS in 51m 11s","accounts_in_message":[],"_revision_number":2},{"id":"e7f7fbc8b8efce37db90934a4be45198b0dea2f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 22:57:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/638cfda5436c448bbfb0a0d65febf64a : SUCCESS in 43m 26s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f5f412a94cba43ac8791089bb688bcf2 : SUCCESS in 43m 02s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/9ac571d68d0648fa905ec8be1a6ae368 : SUCCESS in 1h 23m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bb667db1e1254db0888c5e42a3cda3f4 : SUCCESS in 41m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/dfb280fd03654164b65c875a9fe884c5 : SUCCESS in 39m 24s","accounts_in_message":[],"_revision_number":2},{"id":"4f694236396fd299e88028272b246609ab0a7b2d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-01-19 22:04:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"da37f621fddd3bee8dd7048b24d976ab093757b8","revisions":{"0b70c44a2d97db68bd223bead65bb0edd025dccf":{"kind":"REWORK","_number":1,"created":"2021-03-27 01:05:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/76/783476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/783476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/783476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/783476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/783476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/783476/1"}}},"commit":{"parents":[{"commit":"8cc1ef92558bd3f7c345c955f386498b936bace1","subject":"Merge \"relinker: Add policy to relinker progress output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cc1ef92558bd3f7c345c955f386498b936bace1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-27 00:52:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-27 01:01:28.000000000","tz":-420},"subject":"WIP: Add an audit watcher to find orphaned s3api segments","message":"WIP: Add an audit watcher to find orphaned s3api segments\n\nOptionally, also delete them.\n\nThis doesn\u0027t do anything to stop the creation of orphaned segments, but\nat least it can help reclaim some space now and then.\n\nTODO: check old versions in versioning-ever-enabled containers\n\nChange-Id: I8a48b4dd9ec7a85e61221d9a495837f0cca0a5cb\nRelated-Bug: #1813202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b70c44a2d97db68bd223bead65bb0edd025dccf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b70c44a2d97db68bd223bead65bb0edd025dccf"}]},"branch":"refs/heads/master"},"da37f621fddd3bee8dd7048b24d976ab093757b8":{"kind":"REWORK","_number":2,"created":"2021-12-22 21:28:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/76/783476/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/783476/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/783476/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/783476/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/783476/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/783476/2"}}},"commit":{"parents":[{"commit":"1b8708d9c589b9d8c9b2e851ec9a5193d43ea57c","subject":"Merge \"CI: Move py37 and py38 jobs to experimental pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b8708d9c589b9d8c9b2e851ec9a5193d43ea57c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-27 00:52:48.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-12-22 21:27:31.000000000","tz":-480},"subject":"WIP: Add an audit watcher to find orphaned s3api segments","message":"WIP: Add an audit watcher to find orphaned s3api segments\n\nOptionally, also delete them.\n\nThis doesn\u0027t do anything to stop the creation of orphaned segments, but\nat least it can help reclaim some space now and then.\n\nTODO: check old versions in versioning-ever-enabled containers\n\nChange-Id: I8a48b4dd9ec7a85e61221d9a495837f0cca0a5cb\nRelated-Bug: #1813202\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da37f621fddd3bee8dd7048b24d976ab093757b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da37f621fddd3bee8dd7048b24d976ab093757b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
