)]}'
{"id":"openstack%2Fswift~783728","triplet_id":"openstack%2Fswift~master~I7be30c3247a53969a17fed63d0eba5ab4f4027e4","project":"openstack/swift","branch":"master","topic":"p-shard-repair","hashtags":[],"change_id":"I7be30c3247a53969a17fed63d0eba5ab4f4027e4","subject":"swift-manage-shard-ranges analyze: accept incomplete shard data","status":"MERGED","created":"2021-03-29 18:12:50.000000000","updated":"2021-04-01 18:36:18.000000000","submitted":"2021-04-01 18:30:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783728-p-shard-repair","meta_rev_id":"cb3f9b60859cc1ef031313629d77ab476d56eaf3","_number":783728,"virtual_id_number":783728,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-01 18:30:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2021-03-30 07:43:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2021-03-30 18:30:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2021-03-30 18:30:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-29 19:30:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-30 07:43:21.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2021-03-30 18:30:48.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2021-03-31 17:38:55.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"54c6814da97b1cae9bf37710868c2e26cea34929","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-29 18:12:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f62977d7787af694bb93dc10ec89a8ba6bf5c2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 19:30:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/202b26fe7afb49ccbce017f6b3d1ffa7 : SUCCESS in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/17eb42b202704948b103eb22020f0c35 : SUCCESS in 1h 11m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ed67ff0c90694e4db0e58fc5de91ac50 : SUCCESS in 48m 49s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/12fd99a97a674f8ca892fb5cd51a1363 : SUCCESS in 10m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/621450ed270f472f93ba258b927b714a : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/aedf19dc40184885b5ed3053f41a5d4f : SUCCESS in 16m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5a1acc8fe6ca46559bf1cafd21cd936a : SUCCESS in 15m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fcf662e40bc041d6a253339de1efb6e6 : SUCCESS in 16m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/68e82d66bf2f45559cbe6b25a8298c2b : SUCCESS in 15m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d7b1edc39d8d4d31a2037f4e3e70f468 : SUCCESS in 14m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bda786ebdc234b14a1de8cf722dd8f5e : SUCCESS in 17m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b18bdd14fdcb4f428cddc0c3d8719c96 : SUCCESS in 17m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/da8c78f7b61b40f1942d5a352f695ea0 : SUCCESS in 18m 33s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/48e034fe9bcf496b8885494163af4e10 : SUCCESS in 19m 02s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/84304a952d8545348e27920595990348 : SUCCESS in 22m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f15055c0dcd748f29bb7155ecc85f142 : SUCCESS in 17m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3e52b3851d9a40ca98fded1e52733281 : SUCCESS in 8m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/707c4f6c470d48ffb4010e70e8b55ff3 : SUCCESS in 22m 36s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/30b78d4bcd264b75a998aa5cb1801d5a : SUCCESS in 48m 31s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/cf06eeeee5ba41ec868c38eb966ed473 : SUCCESS in 55m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/302ebcc3d2124362897efd2b8515914e : SUCCESS in 38m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b0840747354c4b9991268df99a37b915 : SUCCESS in 41m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2e9aef9f33c4de4bfbb1b59625976d5 : SUCCESS in 19m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a9a1c0444144ed980b02e528acd45d0 : SUCCESS in 6m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f9a2aede7b314bd38936a3f60a24bdcb : SUCCESS in 22m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20515f2ec9994890a7857e115ac7c57f : SUCCESS in 1h 02m 33s","accounts_in_message":[],"_revision_number":1},{"id":"4755d439c9f3b9b8591c179641af27170391ff52","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-03-30 07:43:21.000000000","message":"Patch Set 1: Code-Review+2\n\nMakes sense and yeah not needed in analyze","accounts_in_message":[],"_revision_number":1},{"id":"1b6e4f7f54b2dec0e23c63feae8998d1d29b14c9","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-03-30 18:30:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nthis is a huge help for testing; I also found https://review.opendev.org/c/openstack/swift/+/783947 helped glue things together.\n\nRepairs necessary to remove overlapping shard ranges.\nChosen a complete sequence of 29 shard ranges with current total of 12110550 object records to accept object records from 9 overlapping donor shard ranges.\nOnce applied to the broker these changes will result in:\n    9 shard ranges being removed.\n    4598221 object records being moved to the chosen shard ranges.","accounts_in_message":[],"_revision_number":1},{"id":"ffa6f48cae718cdd311298ace3f9ad92aa77882a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 05:24:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d5c9fa42c6729b05516285ca0690585eed7b0a31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 09:42:33.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f066fb826d64686aeee47a4ec0f376f : SUCCESS in 10m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/ac5ce55356444c5687f40ab3fd24416f : SUCCESS in 1h 04m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7b1e2444ad9042c78a7a06ccd056229b : SUCCESS in 49m 18s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/33ef5ddef9d0457382b4e5755ccb0350 : SUCCESS in 11m 02s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/6c1f0e763db84e42a5325bd0187ef812 : SUCCESS in 9m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f00a4636d54c4d9e9b8f7ff90a4faa2f : SUCCESS in 16m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/51047f74ad38466ebea2a25ed55f3025 : SUCCESS in 16m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e4cada5b4473448da018ceb8b6a62a98 : SUCCESS in 18m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ede97f375b5b408eaeeed17d9b581f9c : SUCCESS in 18m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/67ea9d5ea55f4c2fa95b79ca4507e5cb : SUCCESS in 46m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/08e4cba9627945eb899338839808802c : SUCCESS in 24m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c8bd3c8888af40a9a4ab48471eca36fc : SUCCESS in 23m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d0a28fcca1914305897992cd60e5e040 : FAILURE in 55m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/64a16d4bf69d4bae851d90c352a7301e : SUCCESS in 24m 12s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/adc858cc8f6444aea90c6851ef2cee01 : SUCCESS in 24m 31s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0e1433d298f642ecaad2b1c75b7e8d20 : SUCCESS in 20m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d9efd22c5ad24b5b8984042118eddba1 : SUCCESS in 8m 03s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8dd09ed96bc14d139d1b383e29825974 : SUCCESS in 51m 38s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/59438334844a4cab967c08e7a94cb219 : SUCCESS in 53m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/811fce239a6649468aede6e82b7652a2 : SUCCESS in 46m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dabe4850bb3a4434919d8d2758c73438 : SUCCESS in 42m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e2297a964474125ab531717c9238a82 : SUCCESS in 15m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/931b8a7ca30e49d0acf76b4a69f16a53 : SUCCESS in 7m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dcef4161fba34d3996f6a0c1f06c5941 : SUCCESS in 24m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b7a07ee8d80423e8aa4381e38d02165 : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":1},{"id":"7fc416a5670190b720855a555c49d94dcf34d56a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-31 10:54:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2dd30be65a748b8660c285e595b65be52eecc63a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 13:08:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14065bd4b5b8424ba96f3afe27bb4275 : SUCCESS in 10m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/dafdd6969f274094a64327f28f669a9f : SUCCESS in 46m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/22385796160045159f2cc747ccbf635b : SUCCESS in 46m 28s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0a204a6af01c47c0a37fd31256f5787f : SUCCESS in 8m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8a36e32ab1a24ac6bbc0e5312b983c20 : SUCCESS in 7m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f15c56f4590346a88e15f29d8130a095 : SUCCESS in 18m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dc2542a2e2f34b86a703e73430418762 : SUCCESS in 16m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/428e04eef8774c579cace9d9032f3f6d : SUCCESS in 17m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/668e435164314923946c10db83be5cb0 : SUCCESS in 17m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/02e7d0e788b9473b8ea3bed5b330192d : SUCCESS in 16m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/59588a1790b244d1a0fa2e160b66f126 : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0f9a6245edfc4c7db90f17db1ed74383 : SUCCESS in 17m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a979cac608f5497c83c510e98241ef9b : SUCCESS in 21m 11s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/71579beac77d441eab23ee748703e329 : SUCCESS in 19m 26s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/987eeca178b6458e9dad4a46af0685c2 : FAILURE in 22m 06s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/99af9aa9362b4d609e5b33b5b5958349 : SUCCESS in 23m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ee7584ebbd5c42fd8072a817c8137fd4 : SUCCESS in 7m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/520b670a449741b4b6b89db0cd72d5ed : SUCCESS in 26m 16s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d0c41d51e239464ab48eb94812940f76 : TIMED_OUT in 2h 01m 37s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/bc00f28339074fbfb87ea143b2e3fd61 : SUCCESS in 46m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/75bc4ce69a0646e589411cc5fe080ca4 : SUCCESS in 42m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c4480b0f095a489eb4cce25c8ac9124c : SUCCESS in 39m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a591f4803d854ce9ad87af383750e6f1 : SUCCESS in 19m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bd9e52fdbd14caab3515d93b761649f : SUCCESS in 6m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/587faf4f79694a9ca5e988b8a31c7ba6 : SUCCESS in 32m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2806eb0ffb6b48419b41911e63dbf250 : SUCCESS in 52m 49s","accounts_in_message":[],"_revision_number":1},{"id":"a830ede3d67f21d45d63643db4e9dda2a5d11fd6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-31 14:05:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"493b39ec9ada3d538450fc85e5487cde979f90b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 15:45:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83f64a6cf5404d1d94d96718c9ad3a63 : SUCCESS in 9m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/5b3c81455a9848d6b0c2745a52f43b76 : SUCCESS in 1h 02m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/84eeae8d21734c1f8055e584c7388286 : FAILURE in 1h 29m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c8aad64d1cf742ddb63e14a00e5531b7 : SUCCESS in 9m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9203d6218910460b94875c899b4c653b : SUCCESS in 7m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d36c35c158424f8db4ce8276ce2acd77 : SUCCESS in 10m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/64bc8f2575fe4aedbe9b3a9e6b78b7ac : SUCCESS in 12m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9143537d6ca549c7bee79c355919917f : SUCCESS in 24m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1641d9d8d3594da0867de59296815b92 : SUCCESS in 15m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5531a0fd192344a48a9f5e9b77e89c41 : SUCCESS in 14m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/55f6ca85fb804a83b537cdaff70dd30c : SUCCESS in 15m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/73ab2fdfbac44e75a7fe761dc8b2ed65 : SUCCESS in 24m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a47bdf5d51a54317aea2b074452a5498 : SUCCESS in 17m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5b9008c9a9734faca5c1a94dc5148ed9 : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/71d9928ff77c474391772f264561464d : SUCCESS in 22m 09s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/19e9c9706027488b8dce6cd1b45b025f : SUCCESS in 22m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9d1783cb70dc4a7da5edf9c7fd1cad71 : SUCCESS in 6m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e15b75eab5754c1681694922d12e73fa : FAILURE in 24m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/51f81b1c537442fcaba8521357295cbf : SUCCESS in 55m 47s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/79657b4cfdf244e79c28b2425df03849 : SUCCESS in 50m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/47364b64b59048dd853f78b536e34742 : SUCCESS in 31m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7b697c8c260f456cac75c82cae11566f : SUCCESS in 36m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc3bd6f7d97741f091d7547ac9570e21 : SUCCESS in 15m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99a041b7900246a790a3733e40dc0dd7 : SUCCESS in 6m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6f0c4c79aa214dd1904b5d66ff13a941 : SUCCESS in 33m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ea0609480bd48419adbc2e50ab82ff9 : SUCCESS in 54m 08s","accounts_in_message":[],"_revision_number":1},{"id":"1fd4d0eb3fbc717315c6233c6aa28864545b4f18","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-31 17:38:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1bc7c8d3235f4c41c46cef5440547c58e89c1f3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 20:15:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4ab33a98a9440239b0aeb889f17b18d : SUCCESS in 10m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/3c502a97600d4078a9b188f0b33c2660 : SUCCESS in 1h 08m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/572f36b7e9de4e42933e56c02fd45056 : SUCCESS in 50m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aa1b1e2b38c24c8cab0b87f4bef07aa8 : SUCCESS in 10m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/04b5a2992d144fa69ed28ed1b82677c8 : SUCCESS in 9m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5aa9289a43514b24a65139f4d7834e53 : SUCCESS in 16m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ee5b989d02954e13973f28bd09431180 : SUCCESS in 16m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b054aa07c76740f1a777c2e9286c989c : SUCCESS in 16m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f654f2e4aaa04d2fb7591cf57a14a0c8 : SUCCESS in 15m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6669264f8bbc4227b912c19f65c0570e : SUCCESS in 17m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bd5a2328be444985b4965e5993742bb8 : SUCCESS in 20m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/06b47e129a244b199569691fff421223 : SUCCESS in 22m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/427b740deafc47f4a5da77979046f0e4 : SUCCESS in 20m 22s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/24ce26b0412e460dad3cf819f84cf7b5 : SUCCESS in 19m 46s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/aab9ff59ed0d45358506bc9f077d646f : SUCCESS in 18m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8c930cc2aca04a688ae910105d7dcc60 : SUCCESS in 21m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b3a41770e7c94ebabb8bd6b32e086e92 : SUCCESS in 10m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9626615f7791402b81ad83f5d36ed039 : SUCCESS in 24m 06s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/804e12916a2a435889847b5c827093a1 : SUCCESS in 56m 23s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/edbbc763638641559c2ad56be8d7e208 : SUCCESS in 57m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8bd0da4f283844e1993e41b7628cf1ad : SUCCESS in 46m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f42b5f5eddfc40c7adff1612315ea107 : SUCCESS in 45m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed21686e8c8c4474a7d93c79e6a14681 : SUCCESS in 16m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ed338d4bd654fe8a4a6ea8a1317a633 : SUCCESS in 7m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5d42ec08948a4443a2dcaff7f5cb8c17 : SUCCESS in 26m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79e391fd2b344d629a0c1a0890d26fa6 : SUCCESS in 1h 13m 39s","accounts_in_message":[],"_revision_number":1},{"id":"6519f6e307ff4f320d5efa384d387a802c4197fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 20:30:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ebb00ae31bbd3de1b06e99c833701a69a3f3e279","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-01 03:35:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8ecfd5c2017a6c0e286f16cecedb66ad800d76eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 04:40:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ddb805046fb496688cd1da173ce15c6 : SUCCESS in 11m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/43b81a99e84647ffbe37a7bd4f09d82e : SUCCESS in 1h 03m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/544c1470cb784e39960803ca5824094b : SUCCESS in 47m 24s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1202b8426f1d4c21a787c50cd07772d3 : SUCCESS in 8m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/080595b2a67f459d9c3d023aea655e43 : SUCCESS in 6m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a97ac10774f8478c814e7e40c738eb8f : SUCCESS in 16m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/86670cc1d54c486899d1c18e86a3c65b : SUCCESS in 15m 46s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/70b5434dec924f6688bf1557e189947a : SUCCESS in 15m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a4840b28895b479688b864a22fb19cf6 : SUCCESS in 14m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0c2f8677abcb4c8e9111c6a21f574639 : SUCCESS in 14m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a8e634491bc44282af65ea0804cc28e4 : SUCCESS in 17m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/be922f40adb04707849c1571b7c6fca6 : SUCCESS in 17m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8f703560c4484f34af62ef624cba226b : SUCCESS in 17m 54s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/724567f126c44b648476c562d711a461 : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/aee2acf8b4c045849af1d9b59fda35b3 : SUCCESS in 18m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/eaaf7fc18cd248af9bf73c21ae9fefd9 : SUCCESS in 19m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/64217e4daeda4902ae4d73c2f1978fd1 : SUCCESS in 6m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/772d5a0d7a694b4aa39b06dd3029afc6 : FAILURE in 20m 52s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d23643efbbd640a9b296cf1c29e023fd : SUCCESS in 56m 37s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/44c91afe29e44bcb95db021408d08dbe : SUCCESS in 46m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7ac981e7fd6d426da993faf5709174cd : SUCCESS in 48m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bd5cc59d888a44698a2a22cb8231640c : SUCCESS in 39m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42349bfe00814218bc5e0b42010a6243 : SUCCESS in 16m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5924101eb9c465e814967b2c44ff91f : SUCCESS in 6m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9e391e7691a24c2ea2d48212907117c4 : SUCCESS in 36m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85f54f086ad94bd7a4a5ec52192d5da8 : SUCCESS in 1h 02m 41s","accounts_in_message":[],"_revision_number":1},{"id":"5c4fc073e0f2459ac35797b63a29f1047f2e8d4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 04:41:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"323785e593607c02d663a3cdc07cd9ea26691703","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 08:18:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f973fe7a7a0b4df8b8286ab9bd49faeb : SUCCESS in 9m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/bdeccdcdd9eb429f9f66f03e8e2308b0 : SUCCESS in 1h 17m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c74935911e81492585651c0c0d527184 : SUCCESS in 46m 53s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/70e149f78b204ae88802b277755e02c6 : SUCCESS in 9m 10s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1cfeca1596ae4d048d08260d53537231 : SUCCESS in 7m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3bdd6fc17ba6426481a88c2946b67af9 : SUCCESS in 18m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/57399fd809be419b8ebc1414b3cff92b : SUCCESS in 19m 06s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f158e0d4f5334256be40109a9690d3e9 : SUCCESS in 19m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6f23b03c2b324525b05f8c0cf2d671ae : SUCCESS in 19m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/32e29da008a64d1196dd09b9680b51e2 : SUCCESS in 18m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2ee979ef5d85430e90d4326d19d5eb96 : SUCCESS in 20m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1dd915c434354b6cb6a5aeb375650b77 : SUCCESS in 20m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7f7c13c300f14690a089894650e437fb : SUCCESS in 21m 12s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/fbb053d7213b4f88a7b1d5de8324a114 : SUCCESS in 18m 46s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c0c53434c56748b58a0882eaa206a1b1 : SUCCESS in 19m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2fed0589b75f4c16b992ed31727f3617 : SUCCESS in 20m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/150a24335eca440e8ff6ad1dbdb84692 : SUCCESS in 6m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/486175baa5d04550847103fe558325a8 : FAILURE in 49m 57s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/51c5898ac4bb46c8b7b656da3bec50f1 : SUCCESS in 49m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c553d8a47b5e49539db796e645c49221 : SUCCESS in 40m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2203d6c553c241b3bb0ccef6b93f12a2 : SUCCESS in 41m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2facf8ad445049889bc2b16d61aece07 : SUCCESS in 15m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8273e532ec784ee180850c2c6c07dc86 : SUCCESS in 6m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ba5328f1156b4891843aa25c92d340d6 : SUCCESS in 25m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/927ea3e4a69549839909aee11b8323d3 : SUCCESS in 54m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a36a20f798f914d95697edbb9bc3d24591009e8c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-01 12:56:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"10a0067330338c7d9b4ff46507447a2a14ef9d38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 14:12:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cb902b57c2f405798b3becfc3c2647b : SUCCESS in 9m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/6378c8b9f5a640d289b385beace6de58 : SUCCESS in 1h 10m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/33fb8bf043124e3485a78ada5eb474bf : SUCCESS in 55m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9eda9f7dda0a4377a94e5447c0c2099d : SUCCESS in 9m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/757cb63500b3426196afdfd9d4a7e27c : SUCCESS in 7m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/90b554d2340647058061676ce867251b : SUCCESS in 15m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cd31333ee0ed4cd18a7f34084a14e0c6 : SUCCESS in 15m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4dd500cf0597455daa373af4ef6125a7 : SUCCESS in 14m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e1d20a46fe8f47009534ea1fb18952b1 : SUCCESS in 19m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b0d9705d270340259af4a2a434e6f79e : SUCCESS in 16m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d05d582561df421f888d7089aa94210c : SUCCESS in 17m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/69a4a2e10b5e4232a2b23f1889270c31 : SUCCESS in 18m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/213897325bb7415fa5e449c3e78e4ed7 : SUCCESS in 22m 13s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8541373d02894846ac442bab4832cf0f : SUCCESS in 18m 26s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/12325650c0cc46acb56af414a625fa46 : SUCCESS in 17m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/56dabb9a0ec54e029ae7f35d3b6ef339 : SUCCESS in 20m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0bb4319eb39643f29569e0b2ad6f2cba : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/65532840bc9d4d78a76cbe0a2dd411d6 : FAILURE in 20m 22s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/59a161995079428eaacbd96e691f0274 : SUCCESS in 49m 43s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/157320aa202e4ad6bd3adc74f567e896 : SUCCESS in 49m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e566897f14394074ba068e909a5daa17 : SUCCESS in 38m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5f57ad5c2d54478daad91c362ac3b2fe : SUCCESS in 41m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce1f602629df4a1995639a46dea31cce : SUCCESS in 16m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/156ea8a9e05b428086456573eb7faabf : SUCCESS in 6m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7c5ca4f99609464a9e03d446e9fe8e6d : SUCCESS in 30m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/92f2ad0c7cfd4112b83790d3188b7a54 : SUCCESS in 53m 44s","accounts_in_message":[],"_revision_number":1},{"id":"4b8efac51736b710dd1c5b284ed8529070485227","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 14:13:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"348568b9cb2da32e29093a17383bb93ee8d11fbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 15:41:21.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2f0f1ad5b4cb8a76330ce69e937504b88cf21e84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 18:30:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51906ed31e4d4067a26421aacc755c5b : SUCCESS in 10m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/72ff04022e964172b195692600fd658a : SUCCESS in 1h 02m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4cd8dd48c7bb463ab78dca8c8a6ceaa0 : SUCCESS in 42m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/bddfbc34cb6648c6a0ee746798a99b32 : SUCCESS in 10m 12s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/a7372001bde34eb793b4f8659bbf56e2 : SUCCESS in 9m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1379146ee638481d86b6d54729fe5e33 : SUCCESS in 18m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/703159d9f75e40ff996735167ba410f1 : SUCCESS in 19m 12s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8898be969dc74d24b45a176c000e5dc3 : SUCCESS in 19m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/223a8fd5001b4f80a13329a2b1f6ee23 : SUCCESS in 20m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d36a76042ad747f7a4207eba4b60cf6d : SUCCESS in 20m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0e0cf6f637ed471d84e7a56bdba0212a : SUCCESS in 25m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2aa56d3c249e4deb875b07cd4defcbc6 : SUCCESS in 25m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ee9d9ada04e44a51b59e0b4b554388e3 : SUCCESS in 21m 26s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cee069f4b109490a8152f139367cb8ef : SUCCESS in 19m 58s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e53c530a6efe4cf49997642262618d8c : SUCCESS in 23m 02s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/574b2b3ec1904318bec074be30128ebf : SUCCESS in 20m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c48a5ba5a7fc4a8f9f9290165652c4bf : SUCCESS in 8m 10s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/59beb8aed8364851b9de8b4c2257ce1f : SUCCESS in 56m 03s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/9e046b74f0cf45c0b795ade54834e6d5 : SUCCESS in 59m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a208eb10654a4bf992345acca2625d84 : SUCCESS in 48m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7dc21b05ec384627b56f2375c1770485 : SUCCESS in 42m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f14f8e2e0384708b966ea8e3a3124d5 : SUCCESS in 17m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fca2ffc20b146d484a812def5bbfde6 : SUCCESS in 7m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/02dc12b7039d45daa84fca1c5f8110e0 : SUCCESS in 24m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/911aa65364dc41c7abeab25535847a16 : SUCCESS in 57m 22s","accounts_in_message":[],"_revision_number":1},{"id":"d59cf8f3ebf546605b841af2cc05727543fdde86","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 18:30:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cb3f9b60859cc1ef031313629d77ab476d56eaf3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 18:36:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28f8cc91e91941fd97b16c8621c94b03 : SUCCESS in 1m 10s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6f6c7418cb9b4373bea15d03cc0e2ee6 : SUCCESS in 44s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/9714e2fc52ce4e7c95c9429d9c4199aa : SUCCESS in 49s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc7c0afbe4c151348af32a65ae47df8a3088ed7a","revisions":{"fc7c0afbe4c151348af32a65ae47df8a3088ed7a":{"kind":"REWORK","_number":1,"created":"2021-03-29 18:12:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/28/783728/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/28/783728/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/783728/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/783728/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/28/783728/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/28/783728/1"}}},"commit":{"parents":[{"commit":"c9c42c07c9c6d354bd60b2e2d64b3db628b8e8f3","subject":"swift-manage-shard-ranges: add repair and analyze commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9c42c07c9c6d354bd60b2e2d64b3db628b8e8f3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-29 17:56:18.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-03-29 18:12:33.000000000","tz":60},"subject":"swift-manage-shard-ranges analyze: accept incomplete shard data","message":"swift-manage-shard-ranges analyze: accept incomplete shard data\n\nswift-container-info does not currently include \u0027epoch\u0027 among the\nshard range info that it outputs, which prevents shard data scraped\nfrom that tool being read by the swift-manage-shard-ranges \u0027analyze\u0027\ncommand.  The \u0027epoch\u0027 key is not significant to the \u0027analyze\u0027 results\nso it is safe to add a default value when it is missing.\n\nChange-Id: I7be30c3247a53969a17fed63d0eba5ab4f4027e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc7c0afbe4c151348af32a65ae47df8a3088ed7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc7c0afbe4c151348af32a65ae47df8a3088ed7a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
