)]}'
{"id":"openstack%2Fswift~783731","triplet_id":"openstack%2Fswift~master~I3c91127e19156af7a707ad84c5a89727df87f2f1","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I3c91127e19156af7a707ad84c5a89727df87f2f1","subject":"relinker: Rehash the parts actually touched when relinking","status":"MERGED","created":"2021-03-29 18:24:09.000000000","updated":"2021-05-12 11:16:04.000000000","submitted":"2021-05-12 11:14:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":17,"has_review_started":true,"submission_id":"783731","meta_rev_id":"1a24204805c99f78d690ce560f6ea0f5ce27e13c","_number":783731,"virtual_id_number":783731,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-12 11:14:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2021-05-12 09:15:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2021-05-12 09:15:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-29 19:45:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-06 00:27:22.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2021-05-07 17:19:09.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"b78d1252559193b1277425a570e3c4183de6c060","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-29 18:24:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82d3bbd7e83cd0d8a61993b6e64905bcdfb2a096","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-29 19:06:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a168b5f9b977e729c4af4b50fc16e1da6c58b1f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 19:45:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9e048bd4ee2417398cc5c85c473c5f2 : SUCCESS in 10m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/661b99b41dfe4e0db344f4736fd7813c : SUCCESS in 1h 18m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b018e87168a749a38401caa9cfcf6e5b : SUCCESS in 37m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0e9f91b67d8345d0b3b5c3a56d643462 : SUCCESS in 9m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3b7eeff94b4e45819f1e4d7958f9de76 : SUCCESS in 9m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ccd8aed9cf6947508ca70a3c82af7ca2 : SUCCESS in 18m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e66bba1dc1cc4999943d32d8bb171fe5 : SUCCESS in 16m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2e8a4a292f6a4122b6ff4ac8b49f3a39 : SUCCESS in 28m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/68889c4270564905880495156ddb4460 : SUCCESS in 15m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e74dbf72151743638a01141be6a83f47 : SUCCESS in 28m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4853d5b5b25a45ac9330b3aa6c381c7b : SUCCESS in 17m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b27b50a1c594414f9f2aac0f89ee3d46 : SUCCESS in 15m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/47b7cf11c0634e2a9dab9afd39df8f0d : SUCCESS in 18m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/0c3d4919d4094fd19a7cbd8324fb27a9 : SUCCESS in 16m 00s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f6fac7583a134f17aede0943ac7e6a39 : SUCCESS in 18m 48s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3b54639117674b4c884b8e425b3c0311 : SUCCESS in 20m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a43031fdc685485999a6560183b2f0f5 : SUCCESS in 8m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/edf68f24b010448eb002f7af10ac486f : SUCCESS in 18m 42s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5306eb2d7d094da498117caca321f111 : SUCCESS in 50m 01s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/2fe39a419d2744f090950cf07c9fd003 : SUCCESS in 44m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/06565abaf85947338f068c0b00dea37a : SUCCESS in 41m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/98b30a7f5a9a42cd9b3595fc54e3800e : SUCCESS in 30m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3fbc6ad2a774943ab4f11cb0ec661eb : SUCCESS in 14m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acd2138b8dd7409285adb974412eae6d : SUCCESS in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/95826d8f67ef48e89b444ce14beb777e : SUCCESS in 26m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f6374a36c2b4002abff79473f78a1fb : SUCCESS in 47m 36s","accounts_in_message":[],"_revision_number":1},{"id":"21ff78d45e0ee35e40619b8a032b6754e1a9b97a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-26 22:21:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"32ad24abef0e561d176edbfc8076a1b9caafe33b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 23:51:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd01c2cbe605415ca27c517aeb363cf4 : SUCCESS in 9m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/3e88cbd0cfe144dbb2553fb8a2c77e71 : SUCCESS in 1h 08m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/71efd2aeaaf5400ea0b32cad9ec31490 : FAILURE in 43m 06s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/89ef6434177747ef95cebcd2392b6e6b : SUCCESS in 8m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/104261a83f044ce0ab80452d263345b2 : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8a6dfded456049ad9162c97a069689fa : SUCCESS in 15m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/14988b90c2474af989f8cc1aaffe165f : SUCCESS in 18m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/20ddd1a7fe924d4e91b5d445bc81d83e : SUCCESS in 14m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d3e86eee1d194a76aa0425b99d18fa7a : SUCCESS in 15m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f8e631446c684616abb2b0e73ec2a751 : SUCCESS in 15m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d22ee7a336064ed5865b18172d40e25c : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/63265d3ce0e2408085ef3515695fc6ab : SUCCESS in 17m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/238495747dff4d1895e971fdacab12a4 : SUCCESS in 17m 34s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4ee60b3df6af456abe96f4620f5d4067 : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f25de0cd4a514e18bfcb8d4ab340f8d5 : SUCCESS in 18m 30s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/681a9faa725a46738d48fe17183e90f0 : SUCCESS in 19m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6b9005423d2f414c9da58bbfd1351411 : SUCCESS in 6m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/787e6fb64d3c4e34b095b1f4b95499d8 : SUCCESS in 18m 52s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c6ddbefcf46246eaa13aaac7dd491a3d : SUCCESS in 52m 00s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/14200bbae9894cb2ad764b73a0babaab : SUCCESS in 50m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6beb26fb5c894788b1a190231734f92e : SUCCESS in 39m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a6d7f3dd610b488cbf4da0e9fe62b538 : SUCCESS in 30m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc3a5ccd49914248a7a002505e478ec5 : SUCCESS in 15m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/734676b7e92b44169a34f2e879a17998 : SUCCESS in 6m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2ff77263f6d9497a9fb2f8c3b9856eb9 : SUCCESS in 22m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/853496663ec0453c8dfec7d1e3109f58 : SUCCESS in 52m 46s","accounts_in_message":[],"_revision_number":2},{"id":"5881584c7536459423a277ac4c4dd54a647d8fb6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-27 22:38:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"acd6de0be19de0c558293a0c625d84170509ce1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 23:34:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a8e29a5c6a84a31879c57e032bbba79 : SUCCESS in 7m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/90767af4d73640da8c89d6e9c4136d63 : SUCCESS in 53m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/65f6d76744e64d87a406cb150409237d : SUCCESS in 31m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cf35c3c0678c4aae81edcfddf763a9bf : SUCCESS in 11m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f880b6201a6f4726a06ca75ee1ffa214 : SUCCESS in 11m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3147e6361d3842d59e30621eb28a575a : SUCCESS in 19m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/72a8de9b42f1417f816b0adc59459fc7 : SUCCESS in 13m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/34aa43bb5845405080935b6921e226f5 : SUCCESS in 19m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/064ce8ee9c534f3abc174d2a21d11f63 : SUCCESS in 16m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e397ede889e84b318a959d7be257e5ac : SUCCESS in 15m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bad0e7639d9e4ae8920d4bd436097a00 : SUCCESS in 24m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ed3d256683de42bc927dffd5b627bd21 : SUCCESS in 22m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/abdc8787961e4487a813bad7b1110973 : SUCCESS in 18m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/063af023191d4ecea1a7d8851b994090 : SUCCESS in 16m 18s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d9918193536248909a97b45b34d11e56 : SUCCESS in 20m 04s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8b3663d9f2834595aa7f817ace72c06b : SUCCESS in 23m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9d7383a9757a421eb2a2e9aab4fd4f6f : SUCCESS in 8m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1faeb0b7d0c244eeaf78590910a4330b : SUCCESS in 21m 12s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/387196e2b4bc4d8eb46a72709c005b52 : SUCCESS in 39m 55s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/644f6ab5ced4431696d02d73ef27977e : SUCCESS in 54m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/60b7ee11dbc9469ba8085a9f6cbf0def : SUCCESS in 41m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fb8976923d604eb2b66de0c16a854f22 : SUCCESS in 39m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c18cd880c5d24121b4a27747773268bf : SUCCESS in 18m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d6b863f7630456ca39181009d489f6c : SUCCESS in 7m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b5a9b31117a04861a62e21aa783c3f7e : SUCCESS in 29m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ad7c22e3a92494587e3f208a782f7a4 : SUCCESS in 51m 28s","accounts_in_message":[],"_revision_number":2},{"id":"8acbc0e69b259a5e987442d263f703f5d2b224f9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-28 23:10:27.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"a403919749f18deee5095f4e84ab06a4dd682474","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 23:44:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (experimental pipeline).\n\n- swift-tox-py27-centos-7 https://zuul.opendev.org/t/openstack/build/94cc540c56b64053ab719bb85790fd0a : SUCCESS in 16m 20s\n- swift-tox-func-py27-centos-7 https://zuul.opendev.org/t/openstack/build/6bf9a4a9405d49b29f2d9ba5ce4ed7df : SUCCESS in 22m 59s\n- swift-tox-func-encryption-py27-centos-7 https://zuul.opendev.org/t/openstack/build/22104bb38e254258b87b9bcd45980dbc : SUCCESS in 20m 13s\n- swift-tox-func-ec-py27-centos-7 https://zuul.opendev.org/t/openstack/build/621cc5b24ef24f1f814f3e8efcfe67fc : SUCCESS in 27m 41s\n- swift-tox-py36-centos-8 https://zuul.opendev.org/t/openstack/build/df3d0e48f9d644019da5b6ea9f3c2f48 : SUCCESS in 16m 34s\n- swift-tox-func-py36-centos-8 https://zuul.opendev.org/t/openstack/build/cba27548c50f4d448826a37c68e3bbd8 : SUCCESS in 17m 48s\n- swift-tox-func-encryption-py36-centos-8 https://zuul.opendev.org/t/openstack/build/f5569b2cff7e4fc2961a9e5fc0288b2e : SUCCESS in 18m 57s\n- swift-tox-func-ec-py36-centos-8 https://zuul.opendev.org/t/openstack/build/841c0ccbeb2948bf9e7b392d93a3174a : SUCCESS in 24m 09s\n- swift-multinode-rolling-upgrade-rocky https://zuul.opendev.org/t/openstack/build/cd2ca9c32acf4ef1b2fcd0db69cbcadf : SUCCESS in 18m 33s\n- swift-multinode-rolling-upgrade-stein https://zuul.opendev.org/t/openstack/build/051d980da0594db19a280ee3715e2f95 : SUCCESS in 17m 54s\n- swift-multinode-rolling-upgrade-train https://zuul.opendev.org/t/openstack/build/9b4f311ead1844cdbf087973e3e947aa : SUCCESS in 18m 25s\n- swift-multinode-rolling-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/5abb0850237b4215b07714776b7bc5dc : SUCCESS in 21m 19s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/28825eb7f07b4e0db5fd81bb4f92dc99 : SUCCESS in 21m 57s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/8281985aadcf4b3191488483cfc0081b : SUCCESS in 23m 55s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/e5c460defce94ecdb81cfd6f0cc1aeeb : SUCCESS in 22m 26s","accounts_in_message":[],"_revision_number":2},{"id":"6c9d6177dfe08a21bb6a5c650e7c4c44c981aa67","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-05-06 00:27:22.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8126521f4008906c93f0be8e17600f3b1a9fd416","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-06 18:41:42.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nI\u0027m gonna try getting rid of the self.do_cleanup condition in the post-partition hook.","accounts_in_message":[],"_revision_number":2},{"id":"cb07aae5feb750390b124fb71e7c3698511fd5f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-06 19:46:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cc7496bf8223ac9630395ab090df14b7faad84f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 21:35:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f6a3b8d25c74bf1ae8a155e0533385c : SUCCESS in 9m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/c58f2e4758284084b4b5addee41f00d9 : SUCCESS in 1h 13m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a8ad883cecb7480dbc14db4532375f76 : SUCCESS in 52m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/71e0cc991ec1402fa8a8151764b467fc : SUCCESS in 9m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fb303974af294a4ebdbb8c21f0c0f52f : SUCCESS in 6m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/eee34333fc5d48218e746b2921cca9c3 : SUCCESS in 15m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fc6505c746754713b9b599ca8ea6f78d : SUCCESS in 17m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1f0c05b7586d4056913de496df59b53e : SUCCESS in 15m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/60ff4d71939347fb8a2da6741c4e38a8 : SUCCESS in 22m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/413621fcc7b34c40a8f4f9998075aabc : SUCCESS in 13m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f3021df6b9a24f749f8103e701ef9ff7 : SUCCESS in 31m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7011e21f3a6c41a886965a0b13d1812d : SUCCESS in 14m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e09a7b96e39845b18b47e80e95c9b398 : SUCCESS in 15m 34s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/07492733c2a44580b2769eb3ce6cb964 : SUCCESS in 17m 55s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/83ea8185c4fb4fbd95f3acc50f193436 : SUCCESS in 19m 29s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/728172c5443e4024a6df3ef0d7021054 : SUCCESS in 32m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d9ec4a47e6ac49da9d838e8a88d70467 : SUCCESS in 5m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/50666f1d18d846a5b2d74ac0716038ef : SUCCESS in 21m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d3042adeb1e34d4b869c8d5277fad0e8 : SUCCESS in 1h 34m 17s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/ceb6a3fb8d2c4aefaefe3ed105d2a83e : SUCCESS in 48m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a5fc0a261a9344288be7aebe442494e2 : SUCCESS in 39m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8843743d474e42418bb4cfa3ed6e883e : SUCCESS in 37m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b27b5b6b5cf545f6a043e552b3a4a919 : SUCCESS in 16m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/534c934516634b1abbb0223349b4abd7 : SUCCESS in 6m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0076c549bc4242e89da8f6a75cd5d785 : SUCCESS in 25m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8d5077eaac943caa2c9a0158f8e0b8a : SUCCESS in 50m 20s","accounts_in_message":[],"_revision_number":3},{"id":"2e22ac180b55935f9aef3b3903f053f38779ad16","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-05-07 07:04:13.000000000","message":"Patch Set 3: Code-Review+1\n\n(3 comments)\n\nNice one Tim.\n\nI\u0027ll try and do some partpower tests with it after I pick up the girls. In the meantime +1.","accounts_in_message":[],"_revision_number":3},{"id":"f1b7c1bd2ad7612eb3dcf459787e9b664e52f196","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-05-07 07:05:42.000000000","message":"Patch Set 3:\n\nIf I can\u0027t get to it tonight I\u0027ll get to it on my Monday so it\u0027s ready by build time for Clay if no one else gets to it before!","accounts_in_message":[],"_revision_number":3},{"id":"a9270a53a7dae12a0487e544640440aa9f08d023","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-07 15:58:39.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"312bb907de473dbae4ecd5e736915aab211584be","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-05-07 17:19:09.000000000","message":"Patch Set 3: Code-Review+1\n\n(7 comments)\n\nnice! this makes sense, just a few nits and small gap in unit tests","accounts_in_message":[],"_revision_number":3},{"id":"0770e66c010434d67cea3fc820419ae113d4bcdd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-08 00:02:45.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3763e4bcacbce664faeb8b1b35e2b04631830037","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-05-10 16:16:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3737a7f86ff512a27fcece9b20113abf026c6237","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 17:13:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3045279669e541b88efebca10088762d : SUCCESS in 7m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/3cdfa0919b0c493381ca318bf0ab1380 : SUCCESS in 50m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7b7951c104c649a3ace5bab2dd675043 : SUCCESS in 34m 15s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b7ccee5223a94a8eb19f2c2bf6d98707 : SUCCESS in 8m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bfd9a1dcf2da4ba88306b4496b52c91e : SUCCESS in 6m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f072c75360d245c7a2985dc24c7273a9 : SUCCESS in 13m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/24c1092c168747e0bb84ac70a92988ad : SUCCESS in 11m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/59af3c39ea2542d4b90552025fecb380 : SUCCESS in 13m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/80df1e6356d34ab3a88c6959ab16209f : SUCCESS in 25m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f7ee30b620d048a9a8c14a036ee6a84e : SUCCESS in 29m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/78426ace2da647d79f908edebc7c831b : SUCCESS in 17m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7e946283f5ce4dbab5747ac7a7c6924c : SUCCESS in 19m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2185ab5cf9514449bf8bd9ce1d0b3084 : SUCCESS in 37m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/70da24b1827f4726a568db40185e4f05 : SUCCESS in 32m 53s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b4cfd3914afa434da3c682530f527797 : SUCCESS in 20m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/137d1b2855964ba2b719b7583db43c73 : SUCCESS in 31m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8f6d9696905740bfba3b5eea8dde2dd2 : SUCCESS in 7m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/09e64fd935c2452c946891906f65efd7 : SUCCESS in 18m 39s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/13b2c8d828d24c71b1cb22e4d26134b5 : SUCCESS in 52m 36s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/0b203897e1524185aebf658e0e9837f1 : SUCCESS in 42m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/388194c2a1e6468a92d14f27521332b1 : SUCCESS in 30m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f7853bdb56b34a1d8d38e6bec09999fd : SUCCESS in 32m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b8ce30fbcf94328b97f8c4ba2b73bfb : SUCCESS in 19m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebc408b312764c55bf0ee4b98c67d466 : SUCCESS in 5m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c6be67dbd1334a4c9e02cc8c3b68279a : SUCCESS in 23m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84e9bba6dd1d485da990d7430a4dcb0b : SUCCESS in 39m 08s","accounts_in_message":[],"_revision_number":4},{"id":"65a32e709e5d6a431c8632328272ec66e602ce09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-12 00:11:43.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f3fc59bf765a4af53e50ca510a7a56321c173624","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 01:20:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb405363cf484033ba3a5e38e1671dad : SUCCESS in 8m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/df810735502144ff8c6356c9fb05efa5 : SUCCESS in 1h 03m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/69dbe5bb5f1e4c1ab57b05571ff48b6d : SUCCESS in 43m 25s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7b9500f412cc4e009084cb86c57f729b : SUCCESS in 9m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6b4534fc37b24060ab4c99aec4134b34 : SUCCESS in 7m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4b7d2b6942d843b793bc8b6a57bf53c8 : SUCCESS in 15m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/359fbf626bed43d5a8b9389c13b42d09 : SUCCESS in 16m 47s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d24b1014989e47299f9aa2c3f4a6cf8e : SUCCESS in 17m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/eb443ef8c5c14a9b88d6a1b11d90b49a : SUCCESS in 15m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/88cf4d078e7d4fcaa926fb6a50661a1e : SUCCESS in 16m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d1d1b4a198d3404abc601af79fffc1ee : SUCCESS in 25m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/009ba452d2b64191ab90aca623fb25e7 : SUCCESS in 23m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c8d573b27769442c8852cf4fc4cfb7a4 : SUCCESS in 19m 08s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2a5a2a78ca92420fbde12b6e4eecf241 : SUCCESS in 26m 37s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/caf5fc8eae984b59b2d2ac31a1e23337 : SUCCESS in 27m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6b8ac5e060f24da99a5056f1707eb9d7 : SUCCESS in 27m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/514831aa22f944d394daea02dcdd5f7d : SUCCESS in 6m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/07ef32752b4344f29bb6e39e4bdf0c3d : SUCCESS in 22m 55s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/052964ab4b2a4826876d66a28db9c0d1 : SUCCESS in 50m 08s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/41abb51ac6aa4b68ab53a6ec21986a26 : SUCCESS in 47m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8c11116417df4437b6be1edb0db8bf24 : SUCCESS in 37m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f92921796a06403994f59feff256a6d0 : SUCCESS in 37m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7769a9c6a5e453f843db08075cd3a19 : SUCCESS in 16m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/449705dca2c944a5a116b8fcedd0551e : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/719edcfd40434de69f68563008eaebd3 : SUCCESS in 22m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/660b554ec1544c5faa4a01985c8a2be8 : SUCCESS in 51m 39s","accounts_in_message":[],"_revision_number":5},{"id":"8d06086c310573f46f5ed59ce8762c007c4cdbfd","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-05-12 06:35:24.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)\n\nLooks good to me! Nice work Tim","accounts_in_message":[],"_revision_number":5},{"id":"541c588540941271cf5f78d8f3b488550629c41a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-05-12 09:02:55.000000000","message":"Patch Set 5: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f398f97e52f13010c466780f0a32ea28cad839df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-05-12 09:03:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7354e6da059e0bf45143f3f57cc5640036fe102f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-05-12 09:15:23.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nI made some trivial comment fixes. Otherwise looks great.","accounts_in_message":[],"_revision_number":6},{"id":"c5f14da70b85843632130b2898860c3ebf58256e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 10:01:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e489b6dfbfbb4e13893b6a9f9cedadd1 : SUCCESS in 7m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/b74922bc9a3b424ab0782b6c1f4e6f15 : SUCCESS in 53m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7b731ee06cbd43388a44ff55eeaa1f1e : SUCCESS in 40m 27s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/22ba58e12492407098386d7b25bde681 : SUCCESS in 8m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/34f774a5b99d4e5d9f2cef7410fc7fed : SUCCESS in 6m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/90c3cf8b4f534eab91b1d7b0699600ec : SUCCESS in 19m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/190d18a394534affbfd2a7e45ce98040 : SUCCESS in 16m 21s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1efff08a1fcc45c8bdc9dae92460ed6d : SUCCESS in 20m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e89788bba62147a69b3d63cd21492b2a : SUCCESS in 15m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ccbf66ea08bc473291da1d34a8edb053 : SUCCESS in 17m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bead3d8aa72e4057b9750c1e5ab6d335 : SUCCESS in 20m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/22593b030eb34fec9ef4a59f5c50c21e : SUCCESS in 18m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0d6a9a6a80cc457f842ab3dfb64b5614 : SUCCESS in 21m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/812cad76633c4cb29a3ff080a2417121 : SUCCESS in 21m 51s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c77ccf7ab4634df2ba0387599a6f3f28 : SUCCESS in 23m 12s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e364122f88f14a92b17382de5e107452 : SUCCESS in 24m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/acf29865c6d34a06b10d8627d4fd6b75 : SUCCESS in 7m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a17ccb08c4f24f11b8535a7e2d240d10 : FAILURE in 25m 55s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7dcf9fd244a44c2baf87f2e471b6d353 : SUCCESS in 51m 23s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/7ca08c6a825440b1978bfa2a4a5fdd4a : SUCCESS in 48m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a10e21c396914fea82822839f07c15f1 : SUCCESS in 37m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/daae2bdd666e4f68bab26390058929de : SUCCESS in 40m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ac325ae16884b5c8c05de3cdfaf2002 : SUCCESS in 20m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69c010afcb0f459685413b07c6c026a2 : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e47aff971a764f099a9757d35ae991fd : SUCCESS in 29m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac593f9a35df466596a953d239f65ac5 : SUCCESS in 51m 12s","accounts_in_message":[],"_revision_number":6},{"id":"0a461ff4ac31ad743f9487ff47989d27b0e1927d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 10:02:23.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"d8b9e10696bda6b17467bbbbfd1b3ee61b6564a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 11:14:08.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ea35996938e4fb9bf53c9145046b63e : SUCCESS in 9m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/ee3c8827e3d54abaade3310df13696ac : SUCCESS in 1h 05m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8af50f36dde4436c989343323a4024e7 : SUCCESS in 47m 10s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1bdef6d423144f6f9746b2e6e881fbcc : SUCCESS in 10m 19s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/871a7b2a3d6f45348a4ddb86a59cd21c : SUCCESS in 7m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0c5896ce7b5e4682ab187aaa7cc76795 : SUCCESS in 16m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/30cbd924c6154400afc6287ad3b442b2 : SUCCESS in 16m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dcf5e8010d81435fb5d183ad2f8a5cd2 : SUCCESS in 18m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/433e660d4d394f378aab1d5f3f7b8094 : SUCCESS in 16m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/094748ff0b5c4327a9f31768e3daf98e : SUCCESS in 15m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3d7215dbc9794d9481f1ecdc59f333f9 : SUCCESS in 28m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f814e20729504ad688c5aecc6a034733 : SUCCESS in 22m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fee03c1c52e74013a22621e20a0a8a7d : SUCCESS in 21m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/68cfde08772742258fa8b0e6d6af428c : SUCCESS in 20m 21s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a1a5fcec0cac43ac9c7914ea84aab7ed : SUCCESS in 27m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e79bab3303f348fcb81b039e96fb9247 : SUCCESS in 31m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/38354c4942e24c0f82627a4926945fa9 : SUCCESS in 6m 40s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4b09957f04bc4606a4f6e991e49e2d71 : SUCCESS in 48m 35s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/518e3e0533d241ee98aec8db129576e9 : SUCCESS in 48m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f1ae0cbea024a108281b03164f8e1bc : SUCCESS in 38m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d774cd8f189344c992a84a96caa8af15 : SUCCESS in 44m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/308ebe16b6824d23be96d85c61a743cc : SUCCESS in 16m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68b50bc795774b6cb614a94916c2ff4d : SUCCESS in 6m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4e06d83983e84873a454d3df8814ba59 : SUCCESS in 23m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f99bdf3ade04903995b9ea0ce28ecb9 : SUCCESS in 54m 13s","accounts_in_message":[],"_revision_number":6},{"id":"9db6fa3fe6434504636a2101df9e99493224a5f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 11:14:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"1a24204805c99f78d690ce560f6ea0f5ce27e13c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 11:16:04.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86caaaaf5e6b482eb614267597b9d358 : SUCCESS in 1m 24s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/88a3b1480473488e8f2ff8e6e84a8d07 : SUCCESS in 1m 24s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/85817f2f7a2d4262821fa293d190a228 : SUCCESS in 56s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"1db2816119b3826ba11f72218284da362639d1cc","revisions":{"ff8caa0c8e5eb3c70c0ee0f0bbcf59fc8bcdeecf":{"kind":"REWORK","_number":1,"created":"2021-03-29 18:24:09.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/783731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/783731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/783731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/783731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/783731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/783731/1"}}},"commit":{"parents":[{"commit":"6dd3c7cd153bea7d579d480763e7c2e39f4b8caf","subject":"Merge \"Update docs to discourage policy names being numbers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6dd3c7cd153bea7d579d480763e7c2e39f4b8caf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-29 18:19:18.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-29 18:19:22.000000000","tz":-420},"subject":"relinker: Rehash the parts actually touched when relinking","message":"relinker: Rehash the parts actually touched when relinking\n\nThis has a few different benefits:\n\n* When re-running the relinker, we won\u0027t try to rehash partitions\n  outside the expanded range.\n* When running on a small, sparse cluster (like in dev or testing) we\n  won\u0027t try to rehash an empty new partition.\n* If we find old files from an earlier part power increase and link them\n  into the correct partition, we\u0027ll rehash the one we actually linked\n  to.\n\nRelated-Change: I9aace80088cd00d02c418fe4d782b662fb5c8bcf\nChange-Id: I3c91127e19156af7a707ad84c5a89727df87f2f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff8caa0c8e5eb3c70c0ee0f0bbcf59fc8bcdeecf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff8caa0c8e5eb3c70c0ee0f0bbcf59fc8bcdeecf"}]},"branch":"refs/heads/master"},"baa10f11d0ef7195b91427e702d73605951aaebc":{"kind":"REWORK","_number":2,"created":"2021-04-26 22:21:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/783731/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/783731/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/783731/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/783731/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/783731/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/783731/2"}}},"commit":{"parents":[{"commit":"8bf93f1f40f213428188d17216c0d3af864448fc","subject":"Merge \"Include sharding cycle time in recon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bf93f1f40f213428188d17216c0d3af864448fc"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-29 18:19:18.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-26 22:20:49.000000000","tz":-420},"subject":"relinker: Rehash the parts actually touched when relinking","message":"relinker: Rehash the parts actually touched when relinking\n\nThis has a few different benefits:\n\n* When re-running the relinker, we won\u0027t try to rehash partitions\n  outside the expanded range.\n* When running on a small, sparse cluster (like in dev or testing) we\n  won\u0027t try to rehash an empty new partition.\n* If we find old files from an earlier part power increase and link them\n  into the correct partition, we\u0027ll rehash the one we actually linked\n  to.\n\nRelated-Change: I9aace80088cd00d02c418fe4d782b662fb5c8bcf\nChange-Id: I3c91127e19156af7a707ad84c5a89727df87f2f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/baa10f11d0ef7195b91427e702d73605951aaebc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/baa10f11d0ef7195b91427e702d73605951aaebc"}]},"branch":"refs/heads/master"},"f6a74c03b3cc34ef1319b6ce47a57e521744c7a4":{"kind":"REWORK","_number":3,"created":"2021-05-06 19:46:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/783731/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/783731/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/783731/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/783731/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/783731/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/783731/3"}}},"commit":{"parents":[{"commit":"8bf93f1f40f213428188d17216c0d3af864448fc","subject":"Merge \"Include sharding cycle time in recon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bf93f1f40f213428188d17216c0d3af864448fc"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-29 18:19:18.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-06 19:43:58.000000000","tz":-420},"subject":"relinker: Rehash the parts actually touched when relinking","message":"relinker: Rehash the parts actually touched when relinking\n\nThis has a few different benefits:\n\n* When re-running the relinker, we won\u0027t try to rehash partitions\n  outside the expanded range.\n* When running on a small, sparse cluster (like in dev or testing) we\n  won\u0027t try to rehash an empty new partition.\n* If we find old files from an earlier part power increase and link them\n  into the correct partition, we\u0027ll rehash the one we actually linked\n  to.\n* If we relink a file during cleanup that was missed during relink,\n  we\u0027ll rehash it rather than waiting for the replicator to do it.\n\nRelated-Change: I9aace80088cd00d02c418fe4d782b662fb5c8bcf\nChange-Id: I3c91127e19156af7a707ad84c5a89727df87f2f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6a74c03b3cc34ef1319b6ce47a57e521744c7a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6a74c03b3cc34ef1319b6ce47a57e521744c7a4"}]},"branch":"refs/heads/master"},"1f9a0d7e86a8bb3b1f67606ea7ea6d896eebefe9":{"kind":"REWORK","_number":4,"created":"2021-05-10 16:16:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/783731/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/783731/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/783731/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/783731/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/783731/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/783731/4"}}},"commit":{"parents":[{"commit":"4ce907a4ae97b150fa994237ebffd73ee8ac694d","subject":"relinker: Add /recon/relinker endpoint and drop progress stats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ce907a4ae97b150fa994237ebffd73ee8ac694d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-29 18:19:18.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-05-10 16:13:29.000000000","tz":60},"subject":"relinker: Rehash the parts actually touched when relinking","message":"relinker: Rehash the parts actually touched when relinking\n\nThis has a few different benefits:\n\n* When re-running the relinker, we won\u0027t try to rehash partitions\n  outside the expanded range.\n* When running on a small, sparse cluster (like in dev or testing) we\n  won\u0027t try to rehash an empty new partition.\n* If we find old files from an earlier part power increase and link them\n  into the correct partition, we\u0027ll rehash the one we actually linked\n  to.\n* If we relink a file during cleanup that was missed during relink,\n  we\u0027ll rehash it rather than waiting for the replicator to do it.\n\nRelated-Change: I9aace80088cd00d02c418fe4d782b662fb5c8bcf\nChange-Id: I3c91127e19156af7a707ad84c5a89727df87f2f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f9a0d7e86a8bb3b1f67606ea7ea6d896eebefe9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f9a0d7e86a8bb3b1f67606ea7ea6d896eebefe9"}]},"branch":"refs/heads/master"},"1078223eb960ee7fa226fdae43050bc8758e1587":{"kind":"REWORK","_number":5,"created":"2021-05-12 00:11:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/783731/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/783731/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/783731/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/783731/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/783731/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/783731/5"}}},"commit":{"parents":[{"commit":"4ce907a4ae97b150fa994237ebffd73ee8ac694d","subject":"relinker: Add /recon/relinker endpoint and drop progress stats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ce907a4ae97b150fa994237ebffd73ee8ac694d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-29 18:19:18.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-12 00:11:35.000000000","tz":-420},"subject":"relinker: Rehash the parts actually touched when relinking","message":"relinker: Rehash the parts actually touched when relinking\n\nThis has a few different benefits:\n\n* When re-running the relinker, we won\u0027t try to rehash partitions\n  outside the expanded range.\n* When running on a small, sparse cluster (like in dev or testing) we\n  won\u0027t try to rehash an empty new partition.\n* If we find old files from an earlier part power increase and link them\n  into the correct partition, we\u0027ll rehash the one we actually linked\n  to.\n* If we relink a file during cleanup that was missed during relink,\n  we\u0027ll rehash it rather than waiting for the replicator to do it.\n\nRelated-Change: I9aace80088cd00d02c418fe4d782b662fb5c8bcf\nChange-Id: I3c91127e19156af7a707ad84c5a89727df87f2f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1078223eb960ee7fa226fdae43050bc8758e1587"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1078223eb960ee7fa226fdae43050bc8758e1587"}]},"branch":"refs/heads/master"},"1db2816119b3826ba11f72218284da362639d1cc":{"kind":"REWORK","_number":6,"created":"2021-05-12 09:03:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/31/783731/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/783731/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/783731/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/783731/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/783731/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/783731/6"}}},"commit":{"parents":[{"commit":"4ce907a4ae97b150fa994237ebffd73ee8ac694d","subject":"relinker: Add /recon/relinker endpoint and drop progress stats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4ce907a4ae97b150fa994237ebffd73ee8ac694d"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-29 18:19:18.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-05-12 09:02:35.000000000","tz":60},"subject":"relinker: Rehash the parts actually touched when relinking","message":"relinker: Rehash the parts actually touched when relinking\n\nThis has a few different benefits:\n\n* When re-running the relinker, we won\u0027t try to rehash partitions\n  outside the expanded range.\n* When running on a small, sparse cluster (like in dev or testing) we\n  won\u0027t try to rehash an empty new partition.\n* If we find old files from an earlier part power increase and link them\n  into the correct partition, we\u0027ll rehash the one we actually linked\n  to.\n* If we relink a file during cleanup that was missed during relink,\n  we\u0027ll rehash it rather than waiting for the replicator to do it.\n\nRelated-Change: I9aace80088cd00d02c418fe4d782b662fb5c8bcf\nChange-Id: I3c91127e19156af7a707ad84c5a89727df87f2f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1db2816119b3826ba11f72218284da362639d1cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1db2816119b3826ba11f72218284da362639d1cc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
