)]}'
{"id":"openstack%2Fswift~783947","triplet_id":"openstack%2Fswift~master~I874e9c65e6209d727fa089be9d31462faf0d0c0b","project":"openstack/swift","branch":"master","topic":"p-shard-repair","hashtags":[],"change_id":"I874e9c65e6209d727fa089be9d31462faf0d0c0b","subject":"s-m-s-r: read shard_ranges from stdin","status":"MERGED","created":"2021-03-30 18:28:33.000000000","updated":"2021-04-01 18:37:26.000000000","submitted":"2021-04-01 18:31:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"783947-p-shard-repair","meta_rev_id":"96a379f47ddbe470ebe26ec2cd6d2cebc8b1d410","_number":783947,"virtual_id_number":783947,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-01 18:31:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2021-03-30 22:40:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2021-03-31 07:49:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-30 23:23:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-30 23:23:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-30 22:40:02.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2021-03-30 23:23:55.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-03-31 00:48:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-31 07:49:47.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"330ea22c42ff3fd86f759ac632b20f0f0b17c1e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-03-30 18:28:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23634842e68bcd99d5a97181a5d1dd7fe20fb2e3","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-03-30 22:40:02.000000000","message":"Patch Set 1: Code-Review+2\n\nLove it, nice little addition. And being able to read things from STDIN can be super useful!","accounts_in_message":[],"_revision_number":1},{"id":"55c9f7e6bf01ff619bd4c99e33e94f3b96671479","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-30 23:23:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd30232b4fb1a155c179509c95ba59b7688ac836","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 00:48:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21bd3a88c54f415e8ea6f45adffda649 : SUCCESS in 9m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/eae0a37d0354438eb373869a972e7f60 : SUCCESS in 59m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d3482ba175a9461492c8a640d5047a7a : SUCCESS in 46m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/929459528cb74fdea92d0c2d11bd760e : SUCCESS in 10m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1f32f86d0caa401395a80f35c84a7fb7 : SUCCESS in 8m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fb8008e4c5904c96bcc6668b27e4eb1b : SUCCESS in 16m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0ee43f12cdbf4289b54645020badbe35 : SUCCESS in 16m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1df2a53f50034747b3c501d56aa61189 : SUCCESS in 20m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0dccfd11fff54b97b52e9eb9d7e42858 : SUCCESS in 16m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2adb96832a83424ea4d6166775fa5bc4 : SUCCESS in 16m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3ca7872259cb490bacfd09212a599e87 : SUCCESS in 18m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2aa64d9498784591bfd06a47df8364ea : SUCCESS in 17m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c12f7c24d77d4ec4aae09f698d526919 : SUCCESS in 21m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/cd48f622cecc43f2aede7cdcc5d1ef44 : SUCCESS in 26m 48s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2c5745ed4ca7487b9b24c5eab2ded818 : SUCCESS in 29m 10s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/22d6c3fef48745c8a9a42e16a97f19cc : SUCCESS in 19m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6c9b1602650a47b9bd9072613c29dac6 : SUCCESS in 7m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/eae013c9c5cb473998c1808d0b9fb825 : FAILURE in 20m 32s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3e22c1cc8f264439a75b9cd899a1ac33 : SUCCESS in 48m 38s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/7153797b70d74b928fcb38ce0e952838 : SUCCESS in 51m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a94f572fce494a4b916713757962adcf : SUCCESS in 46m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3e85f8523d9d4276b9295b6318ba9f07 : SUCCESS in 45m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30517570c2024eef8f5fb4736121cf9d : SUCCESS in 17m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cbc22397ab0407da8c1a60470e0a689 : SUCCESS in 6m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/60dfcc3da33c4723a864cadb7577a954 : SUCCESS in 34m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/10a368861446465eb77d6a2bb405359c : SUCCESS in 1h 04m 00s","accounts_in_message":[],"_revision_number":1},{"id":"73302b7a3b003cb814682cdb287a7af3e1b08f31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 07:29:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b4323c35c234eec29e77ced2a5114fbcf862701","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-03-31 07:49:47.000000000","message":"Patch Set 1: Code-Review+2\n\nthanks!","accounts_in_message":[],"_revision_number":1},{"id":"5a959ac8b8417728e5107c93e16677973332fbfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 09:43:08.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"1bc44ea470af9a22a1379e5282a1011d3a7b1dc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 15:41:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee133b7ac3cb7642002769829af5fbf3b2378a01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 18:31:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c0e10e713804641ba35335a0f5b60ef : SUCCESS in 9m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/eaa9ae3c04794ac9b5f3008b97d5115d : SUCCESS in 1h 08m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/689f2510a8f34d58b025a6d764b0551e : SUCCESS in 53m 04s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/80f6cdd0707648e9887e83137d140c27 : SUCCESS in 11m 57s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/aebd750a069c4a288abdca471f1e6be1 : SUCCESS in 8m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/311134e6c8ef47eabe3fbafdc12f5d8e : SUCCESS in 19m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d276044cdcfc489d980b6ee30b2f69d5 : SUCCESS in 18m 16s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9d2dce215c9b40beab4a38d2450d2ad2 : SUCCESS in 24m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8d61ee5addcc4c6e8d18de70a2a23156 : SUCCESS in 17m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b1a4d5a2190140019d8c5e14047e2aab : SUCCESS in 24m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/284d32cf7b24415896d98db87f9c907d : SUCCESS in 21m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e208e3428052499f8159e1bfe373a71f : SUCCESS in 21m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2f7aef17ab9048f0bad184f9884722e6 : SUCCESS in 20m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4d2eea8f8e9249cbbcd30480a036e34d : SUCCESS in 22m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2a447bd1344c48169528675c68e61006 : SUCCESS in 20m 29s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5f5b8f6645e34aea9fff3ee5d0736bd7 : SUCCESS in 25m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/10110c03d8a64e7b80c0669da5eb9b9f : SUCCESS in 8m 55s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a9ba87da79d1480f816efc7d10319d9a : SUCCESS in 55m 34s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/61b8aa7285ef47318c28b96b1022b9dd : SUCCESS in 52m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3d463526de6a40e28680e672479823d5 : SUCCESS in 38m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5156353ef4c74918a6f9228c6cf68b49 : SUCCESS in 46m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35617732af3c4dc1845aacb4b3edb299 : SUCCESS in 18m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a78af08054140cbb7df2d0be74c4f9b : SUCCESS in 8m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cfd30b06500f4b408fe907b0a34669da : SUCCESS in 24m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da07039225bd4aa99312803583a51d97 : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":1},{"id":"87178ae73db38917367ba40fa0dad4a922de5b4e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 18:31:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"96a379f47ddbe470ebe26ec2cd6d2cebc8b1d410","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 18:37:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8b44d0a67b9449fb54fb09568d4c998 : SUCCESS in 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d79727d55b3a42b08ea22179139306b3 : SUCCESS in 38s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/cdcb867cad034aec921a5fcf65d977da : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d55bfa7fb1cc15cc5c6ece3e2cfd9302a37e0eec","revisions":{"d55bfa7fb1cc15cc5c6ece3e2cfd9302a37e0eec":{"kind":"REWORK","_number":1,"created":"2021-03-30 18:28:33.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/47/783947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/783947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/783947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/783947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/783947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/783947/1"}}},"commit":{"parents":[{"commit":"fc7c0afbe4c151348af32a65ae47df8a3088ed7a","subject":"swift-manage-shard-ranges analyze: accept incomplete shard data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc7c0afbe4c151348af32a65ae47df8a3088ed7a"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-03-30 18:16:07.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-03-30 18:16:07.000000000","tz":-300},"subject":"s-m-s-r: read shard_ranges from stdin","message":"s-m-s-r: read shard_ranges from stdin\n\nChange-Id: I874e9c65e6209d727fa089be9d31462faf0d0c0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d55bfa7fb1cc15cc5c6ece3e2cfd9302a37e0eec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d55bfa7fb1cc15cc5c6ece3e2cfd9302a37e0eec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
