)]}'
{"id":"openstack%2Fswift~783968","triplet_id":"openstack%2Fswift~master~I309767b2c6352e097d9572053c08d8377806809c","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I309767b2c6352e097d9572053c08d8377806809c","subject":"Quiet more deprecation warnings on py2","status":"MERGED","created":"2021-03-30 21:02:50.000000000","updated":"2021-08-05 05:48:05.000000000","submitted":"2021-08-05 05:46:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783968","meta_rev_id":"2829e194b3bb49ed205b4488f927e3ff2ee25fdb","_number":783968,"virtual_id_number":783968,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-05 05:46:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-05 04:42:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-05 04:42:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-31 01:42:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aba4a7f3fced7359f55b729ecd883f83b16e9a9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-30 21:02:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9fb8f9604f496c8f0d3a324c2a3f0baf7d4a2d15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 01:42:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd238bbe877c48ac8bbd7b50e53402f8 : SUCCESS in 9m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/445f51c3e5f747c88b03749b7cee8b3d : SUCCESS in 58m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/09cc14d4a1fa4271b8bab18df6cd4cfe : SUCCESS in 42m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/27945ea7f1414a179aabfe28a984023d : SUCCESS in 9m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/382048fbfa9b4720a42f6a53ed8ae963 : SUCCESS in 6m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6b11310f2a6b42b2947f19e7bb82f796 : SUCCESS in 22m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/803307bcb7184b4e94e22e46dd936e04 : SUCCESS in 18m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b20e68e97bf54b228e6f0cfd15a9b43a : SUCCESS in 18m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a3122fe496db46caa53d974ab3e9b0c0 : SUCCESS in 17m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/de4518fe1dcf4446ba4ba7a6b8f3f10b : SUCCESS in 18m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/15c4b6d93d4a475ebe46f3b574dd8b71 : SUCCESS in 19m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/08a11d78455e4c378fc51b0ae4c11176 : SUCCESS in 21m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3d5a6f4e2d1f4c82be878d947963d4b6 : SUCCESS in 24m 23s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2bd4807e696a4fe0ab4f14186e3d913f : SUCCESS in 20m 51s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/e3c662bc97ed4fd39375713222b57c6c : SUCCESS in 20m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5590da9035b8482b8a6b27bc08ff783e : SUCCESS in 21m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bead7e30617b44cfbdc1b8cdfc31b175 : SUCCESS in 7m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/26e47c2f3d7445109fe645ec6453ff2d : SUCCESS in 29m 34s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6a3924f8c46a4f3a8f5c1fc3d8a7a1ca : SUCCESS in 52m 18s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/9761c6b1a59d442bb0928cacc1439c68 : SUCCESS in 1h 00m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bd796d33c419439ea67b9bc315cdeaf0 : SUCCESS in 44m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9513de00692b47a58944fae53dcdfb27 : SUCCESS in 47m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6169fccc945f4cc6a39aaf68b626af5b : SUCCESS in 19m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfecdf5177754d89af083a5f5a228e60 : SUCCESS in 6m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2dc53e62f8b049fc826d6f98a34e5837 : SUCCESS in 34m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1b0ea7daf3a4a9aa3ce2f7e92596d38 : SUCCESS in 1h 01m 24s","accounts_in_message":[],"_revision_number":1},{"id":"2a5df421c1b1b925a105df9e92622cc710353e8b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-08-05 04:42:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b0b43928f01d2a1ef1887752196094e7f2d0c7fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 04:42:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d44554f04b5b3820e954fc9c9a9c1ffa6ee10a76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 05:46:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dedf6f8689484e91a023c08d02dfcc33 : SUCCESS in 12m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/028723f66cdb4c95b7a5b59a094afea9 : SUCCESS in 49m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3c7a01e0070e47ef88609ed3b8af537c : SUCCESS in 54m 19s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1df4a942159c46668badf2d4b2944458 : SUCCESS in 7m 51s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/8ce3cd8d7ac74c41b22e8dfc4f709e4d : SUCCESS in 9m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d62c3799134542649891a2d303c05915 : SUCCESS in 14m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e3a67fc327a44325a32a4820063479ed : SUCCESS in 14m 12s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6202c1bb42994640bede46f5f98ab0fe : SUCCESS in 16m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a12d1bda8b13461489cdd9812b8533df : SUCCESS in 15m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/91bb06aaf9d546be9ed9e15d833033a5 : SUCCESS in 16m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8b84d148ccaa492da89b61f3227f8c71 : SUCCESS in 17m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4ab95adc164c49a6adc120eda42fc6fd : SUCCESS in 15m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1fd10b229ffc4c35b40d3e315603966c : SUCCESS in 17m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ad86cad7718c4fc88cc65dd3f3a49bfe : SUCCESS in 31m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2b862afd14354e91b771b557d5ed4d0c : SUCCESS in 30m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/dcbc2101d6754cb58468eda2aad5364d : SUCCESS in 16m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b329366878454352a40765d29b446684 : SUCCESS in 8m 21s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3f9efd55daf74eacb4b280f3f3a26b4c : SUCCESS in 43m 21s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/992ef0d9f3bd4213b5052809ca6b0c9e : SUCCESS in 44m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5e2963bb98f84719bb0246254ad489e0 : SUCCESS in 51m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3cd6f99d0ebc4888ac7a4a549d224022 : SUCCESS in 42m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54cf225fdaa2430aadacca825d7cc4a8 : SUCCESS in 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b49a1db36bd4fcb970a856f3b567415 : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9ab69e79117c4e1bb84ed25dead1edac : SUCCESS in 34m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c156a7010b834d688f0df0d14f0bd60f : SUCCESS in 59m 05s","accounts_in_message":[],"_revision_number":1},{"id":"6b85b3c08496189fd77a18ccd1ba638f70ec8dc4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 05:46:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2829e194b3bb49ed205b4488f927e3ff2ee25fdb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 05:48:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b4c0cfd40914969b3ba6fda42dcced8 : SUCCESS in 1m 06s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b480c6a8160541018ed1c0bef056e455 : SUCCESS in 44s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/9506289e113247ccab86b6ceb15b9d7b : SUCCESS in 45s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d44328c7500d688f7b398181ce7e6625d6fd178","revisions":{"0d44328c7500d688f7b398181ce7e6625d6fd178":{"kind":"REWORK","_number":1,"created":"2021-03-30 21:02:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/783968/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/783968/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/783968/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/783968/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/783968/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/783968/1"}}},"commit":{"parents":[{"commit":"6dd3c7cd153bea7d579d480763e7c2e39f4b8caf","subject":"Merge \"Update docs to discourage policy names being numbers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6dd3c7cd153bea7d579d480763e7c2e39f4b8caf"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-30 21:02:36.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-03-30 21:02:36.000000000","tz":-420},"subject":"Quiet more deprecation warnings on py2","message":"Quiet more deprecation warnings on py2\n\nChange-Id: I309767b2c6352e097d9572053c08d8377806809c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d44328c7500d688f7b398181ce7e6625d6fd178"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d44328c7500d688f7b398181ce7e6625d6fd178"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
