)]}'
{"id":"openstack%2Fswift~783997","triplet_id":"openstack%2Fswift~master~I0a0766cfb6672377de0f152ce179c874c327ec54","project":"openstack/swift","branch":"master","topic":"recon_refactor","hashtags":[],"change_id":"I0a0766cfb6672377de0f152ce179c874c327ec54","subject":"recon: refactor common recon names into a common location","status":"MERGED","created":"2021-03-31 04:48:09.000000000","updated":"2021-07-01 02:11:42.000000000","submitted":"2021-07-01 02:10:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":7,"has_review_started":true,"submission_id":"783997-recon_refactor","meta_rev_id":"5a65f34258c2e9d7996598ecbfa18de4fb34d226","_number":783997,"virtual_id_number":783997,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-01 02:10:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-29 22:23:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-29 22:23:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-31 05:56:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-31 06:10:32.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-06-07 10:05:24.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"ce5c19c7723ce32165f88338cb866a58fe0bd19b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-03-31 04:48:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75cee05323d8e8db4bc3d113bb8576e0b268c1f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 05:56:56.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/136d6e297adb4bea938fff9550beac12 : SUCCESS in 9m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/973e925966a346d1a97c72b6356cc1a3 : SUCCESS in 1h 03m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b9b4e68d8f85483fb4fcfe47c8f81934 : SUCCESS in 53m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/90f6b037bda14c2cb6e3896e88fd6f0a : SUCCESS in 9m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0f8666b0b09a4bebab6a0b6d502921ea : SUCCESS in 7m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5dbb9471f1d4497896001b87ac6695b0 : SUCCESS in 15m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d252832123624fcc9a63bf49061057a1 : SUCCESS in 15m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/92146eb1f1ae4537957b5e3ad1cb6b00 : SUCCESS in 16m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3a8b481b4b22403ca15a73769bcbb83e : SUCCESS in 15m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/70e4b8e8167c45f3971c1351a8bdef2f : SUCCESS in 15m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/36c5252107064f6bacddde086eaa54e0 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ebe4453e2ee8409098df3981d8b7824a : SUCCESS in 18m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/50f0f6e51da34e2789d633911c02cb13 : SUCCESS in 20m 12s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e7a2e2613d0544a98cfcde9058afa74d : SUCCESS in 19m 35s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8f89a6c4744e4380a9efca32563092dc : SUCCESS in 16m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2a6161e6852e452d87bb752c4077aa89 : SUCCESS in 18m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/efd72f3ffb4e4c878ae57241ff20edb3 : SUCCESS in 6m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6fba9628a3fa436f92b511a5b717b5c3 : SUCCESS in 21m 13s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7218e21569844c0f8ea10d470d75f3fc : SUCCESS in 57m 20s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/03347460cf9f4b80a9308d6439329a89 : SUCCESS in 51m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/36e2efdd96124eb2868655eb29d0bcab : SUCCESS in 42m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bfaf151831594816985b7d359af66ccf : SUCCESS in 43m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d958dac431a74a3ba3fa5b8c1b6a7bfd : SUCCESS in 17m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/783ae398dc884da9a864cca84bce925f : FAILURE in 5m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/92d7accdb68b421e805cc8eafe1ddd06 : SUCCESS in 22m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76f2904689fe49ecb64877c458bf2403 : SUCCESS in 1h 02m 39s","accounts_in_message":[],"_revision_number":1},{"id":"511ed0a4b5e9b68177328b23841dd6184d7c5273","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-03-31 06:10:32.000000000","message":"Patch Set 1: Code-Review+1\n\n(5 comments)\n\nThis touched more places than I expected, but that\u0027s probably just a sign of how overdue it is.","accounts_in_message":[],"_revision_number":1},{"id":"84361f145232d924ff537764f039b318c50be645","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-01 03:07:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e1c90afd967ad9a44b19e6907ce44815c1d4ff70","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-01 04:16:22.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6b56c3cc30727d3b18d4c9bb6797eee4449de6e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-06 09:53:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"89b492fc11c53f08c74b9d6cf6b9f726a754728c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 11:13:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c57ddcb80a14b27879663e480097ae7 : SUCCESS in 9m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/b510a53d79e8451aa31f279a6219f2d0 : SUCCESS in 1h 11m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6c5a7e7a0e7a402e84c16634122af0a7 : SUCCESS in 49m 08s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/457c15a8889c4db48aa282aa2c5908d1 : SUCCESS in 9m 20s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d7152160a9294e898e8f2501e8cce604 : SUCCESS in 7m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/70300a2342974809a0612ebc303d3106 : FAILURE in 8m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/51e7bef68cd640209453626253289566 : SUCCESS in 14m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b494cd2480f94515bcdfe32150f0a50f : SUCCESS in 16m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a882bf2b448b4b06bc05636742bd80ea : SUCCESS in 16m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7fe0b1c356e34c8a8ecc2c63eb5b1a7c : SUCCESS in 27m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/095168822719468aae17c96916fbc79b : FAILURE in 4m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9d5862f8144a4613beba26fd9c85a935 : SUCCESS in 20m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8ee940d44b754869a2de44ee27e6d5c4 : FAILURE in 4m 31s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f3c38838a86347e596237e50db734801 : SUCCESS in 19m 23s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f2e49b853678444d8360bfdcdbf9568e : SUCCESS in 21m 54s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/9b64c1aa22284536b6082a93a4a5308f : SUCCESS in 17m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e86ec99fc6ac44669e6292b5375bbc16 : SUCCESS in 6m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d237a1231ad04b7fb7bcba4272b326d5 : SUCCESS in 21m 10s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6f92bc31f42949568497268205f55cdb : RETRY_LIMIT in 3m 30s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/af354e1737bc4b1fbcc5492c45c33687 : SUCCESS in 52m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b280a68af4f34012a4fbb7029d747f86 : SUCCESS in 44m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9b113e982d6349f8ac6d99139f50cd60 : SUCCESS in 43m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/034cdec6a364429dadb11688996adc23 : FAILURE in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb99cd9984f346e1823147588706debe : SUCCESS in 6m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7aad2fdf69c84cc7a97cd7afceca3dc3 : SUCCESS in 22m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/263378a54ab649d5a8bfe3e99f537fb2 : SUCCESS in 56m 16s","accounts_in_message":[],"_revision_number":3},{"id":"fcfabbdde48ea5336d832a8ea18c0b88bca4b538","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-09 05:28:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ff3ad07c844fc7c6cbe32a7c4e4cff57401368bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 06:32:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60a78e3ec76c417e82ca0d3cc6cc4986 : SUCCESS in 9m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/c56fa0af2a124308a430ae91724b3586 : SUCCESS in 49m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1da3ac4a54e8457ba122e450e579cd6c : SUCCESS in 38m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3ef606d2f2394031b288282e84c4f82b : SUCCESS in 10m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4a6a34fa3d544cf2a95d82d08031edd2 : SUCCESS in 8m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a5e0dcd4a02d4139933cd1f79a883048 : SUCCESS in 15m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f49152700cfc45b1bd3dc8c92474b655 : SUCCESS in 14m 55s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5675640d3c5f4cfbbadc92591fc16f01 : SUCCESS in 18m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2d3c57e43d794fb89f5edc37afeae9d7 : SUCCESS in 14m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7ae5cf4a2d944efbb45a85ca4359d3dc : FAILURE in 13m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7e2d5cbf56ca41c39a0137de0f499588 : SUCCESS in 27m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c5c01456d7fb419f9566a9c298e26972 : SUCCESS in 18m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dd5cbdda6ecc4dd7af0438a4a9d21ac2 : SUCCESS in 18m 07s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/60cc0b636e5143c7bb5c251642b54354 : SUCCESS in 26m 09s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/942377402e804ad7a284ea5f8b5cc3ce : SUCCESS in 17m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/99f71276e21944cb8694be7c14d81c03 : SUCCESS in 18m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/689b257d635645c7b18bfa4c43696f20 : SUCCESS in 7m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5b2f3dfa1d1549bb987c46fd882dd3e5 : SUCCESS in 23m 18s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1f71776619ae40ecb1f4c962db8786a3 : SUCCESS in 57m 13s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/ec84e9ded5654500b3e10e1baf88f13a : SUCCESS in 52m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d311f2d8f06d48ca813c696faef8d6d4 : SUCCESS in 1h 01m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9b3d4187a0bc4850a66efb0a72dc2ccd : SUCCESS in 44m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/daf01fd747674537bcfa06ad63c4c8eb : SUCCESS in 23m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f07319c3a7114ad3aa1fadfd4a528a9a : SUCCESS in 6m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c9ccbb82556c495ba469a671352e8ddc : SUCCESS in 25m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df737cb7986449ed8e617a3ac6e51d44 : SUCCESS in 43m 52s","accounts_in_message":[],"_revision_number":4},{"id":"6a749c2b4866a0c10fb64d02243370a124e79cac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-12 00:32:51.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"131f3889fdb04bb4b795e9a7fc7488b65e24588d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 01:24:01.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07a6209341e847fe9018c4dd6160864e : SUCCESS in 9m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/5bbf6eaba74a49e492be407070ac1925 : SUCCESS in 47m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/14ab318c93024b54932f6081cacf5688 : FAILURE in 36m 11s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/670250e76c2949b1b5bc566b6021a9b6 : SUCCESS in 9m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d7e042071c71413fa00f5eab0717ed66 : SUCCESS in 7m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2d34acfa51154a7ea8d7f5a8e7c66e0f : FAILURE in 14m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e39244106d31415e97f79f5207c30c1b : SUCCESS in 19m 42s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2ffdefcffe61462f9fd3aa1186c63939 : SUCCESS in 15m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7cef72968ea74da18c8ed2da0811c557 : SUCCESS in 19m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/84cc0a7950c34b5f8cdf1b56ba7ae57b : SUCCESS in 15m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/09dd7c61a79f45a981cd3724687e4216 : SUCCESS in 22m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fbf13422e9db4f0c9d900dc6dac1312c : SUCCESS in 17m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6904aab3ff8d4a90be50881a84aec1fc : SUCCESS in 14m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2fd40aa95cd94f278dbd4984d7fa5513 : SUCCESS in 17m 11s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/989d0c73624b47269d67d22aaa16b90e : SUCCESS in 22m 33s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c430c8a9138a4c968f5944a4cde33ee3 : SUCCESS in 18m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/42d697e4e17447ad9bebc48a5f224e63 : SUCCESS in 6m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f8cc087bf50b4752a362ea12b4c57c93 : SUCCESS in 22m 14s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f852d7be7d624ee0919f8e69e99a7980 : SUCCESS in 43m 24s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/9f92b291407d41bea23849749e4ff252 : SUCCESS in 46m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/433387f818364dfbb501a7b26d1cae27 : SUCCESS in 39m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/293ec352978f4ecf8af6935eb7d5d7f5 : SUCCESS in 36m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/831ab7d6996646799a6d9c06c59b174b : SUCCESS in 18m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abe8f359e5f64b8db2965cbbc1d03123 : SUCCESS in 6m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d0ebd3973f21473896cc062e9a5e9fbc : SUCCESS in 24m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81bdcfdd79c54af0908f09c574af5a2c : SUCCESS in 46m 48s","accounts_in_message":[],"_revision_number":5},{"id":"de50b720a685dd5549bb319e3d9f451da1272ee0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-13 07:28:08.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"15057f66800edf9cbf7ce578e5b1a12a57e8af9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 09:05:00.000000000","message":"Patch Set 6: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/900969e87ced453a91b9a34e6ad8cb0f : SUCCESS in 7m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/3f1dc1d969ba405b9baaedd7faa42240 : SUCCESS in 53m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/45b2c96b9a284bff95390441338421ad : SUCCESS in 43m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/479cbc800c644a4da0df2b8af69ce384 : SUCCESS in 8m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f30a1575ec2444f090fdf76a24a5b086 : SUCCESS in 8m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/363be09f6a9a492ca765fd7f39a17078 : TIMED_OUT in 1h 31m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0eb7e4c28b4d439d80d53a572f06f34a : TIMED_OUT in 1h 31m 40s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/508cbed21b0f44d2bbb3332feba83bb7 : TIMED_OUT in 1h 31m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6b43745aad98441fb9b58c83d7afd996 : TIMED_OUT in 1h 31m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4db57e82e8204bc48b4241f746a29c0e : TIMED_OUT in 1h 30m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fc4315ea5ce84fb194d5a9c54cab9c0a : SUCCESS in 23m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/48a39a0b93884322a47a041d4fa521fd : SUCCESS in 17m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bfcd255cd6334e2d8e692432e9954bfa : SUCCESS in 17m 35s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ff06db6dafef4645836df8940c3500a7 : SUCCESS in 17m 56s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4dccdf7463f6439f97b23f71721798c1 : SUCCESS in 17m 45s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3d94f82db74b4468b8638971ac20da51 : SUCCESS in 20m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/219ee3f6de894e5fb84a8ebd0b1e9125 : SUCCESS in 6m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/12de79dc198446a6b0e77a7726bc792a : FAILURE in 23m 32s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/52c42934f4154996b383aa983e693258 : FAILURE in 48m 25s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/c872a9300d5b4c2187b2bfc517c1194e : FAILURE in 44m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a6c810be2a304d30961c090c93abd977 : SUCCESS in 41m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a47e506cf530498897106c23d2d9497e : SUCCESS in 46m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd90a3bd2ece4d01b4f0196f3dca0a53 : TIMED_OUT in 40m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a4f5892c3cd4607ad21a430b382f46b : FAILURE in 5m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2fb3817cbbb34dc98a44e1f526fbb618 : SUCCESS in 23m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11799f61d28c400597f18d1560c59015 : SUCCESS in 1h 00m 11s\n\nWarning:\n  Comments left for invalid file test/unit/cli/test_relinker.py","accounts_in_message":[],"_revision_number":6},{"id":"52d4fdbe103de61e42284bda470637b2fb5a1676","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-14 06:44:49.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"c05b74bede483abfe68f4406566beab139a0b5b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 08:13:33.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cf62570a89d47fea6eb8ad05269e42f : SUCCESS in 9m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/a62445735e9443a7b837637f58e22ba3 : SUCCESS in 1h 01m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dafc5b00e6144541b45011d222062740 : SUCCESS in 41m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/18c05db79b47467785f3f1589bb4bf78 : SUCCESS in 9m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a2db0c4a6c7e4df3853dca0660655ac6 : SUCCESS in 7m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9bb96dc295474df1b3d704cb6bdc60ad : FAILURE in 16m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b62c639b64b6474da94fc935876d3436 : FAILURE in 15m 12s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f0a624d6bb6044ae8ee68de205464b01 : FAILURE in 16m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/417916d6566c4767a8cef10b5e0d0336 : FAILURE in 15m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2f3b4f495e17439ca9a74cedb643b66d : FAILURE in 15m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a9160e1dbf3b41618a970583a5a34db4 : SUCCESS in 18m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/952a096984b243679e4efe98f2742f5c : SUCCESS in 19m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a3eb994e4f8a4e43ba75aff9ae714783 : SUCCESS in 18m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/60f6754c3c3f4c87a116edb01be475da : SUCCESS in 18m 56s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a7e908871802419d836e15cfa82d3601 : SUCCESS in 18m 19s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3e89486ae183463db258a4ce98007d46 : SUCCESS in 18m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/45dfb6f254cf432585a7838e6d22739d : SUCCESS in 7m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3ebfb123a0c54f4b907968366fa47fbe : SUCCESS in 20m 05s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6d153a65195c46749f98ac9bd67b2a50 : FAILURE in 54m 51s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/4a03af4fa662495fa37ecf646228aa6e : FAILURE in 47m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b805c09c770343d9862d83a94182376c : SUCCESS in 44m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b551a74705e547208775a3b9201362c9 : SUCCESS in 37m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09f9de07a4fe43f69f326dd4cae056e5 : FAILURE in 15m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea0eda9003344bfd900984ac138c913a : FAILURE in 6m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/70673ddb0b5a4986ba61c020b1a9ec8e : SUCCESS in 30m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9eab704deaba411da8bf9dd40d60650f : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":7},{"id":"5ecbc8490801f3f07117b0f812f6b80e6c20cafa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-06-01 01:29:24.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"2a87706415c422128614a0ba28c65f627efe28df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 02:46:29.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac58429e5caa4976b8a5be64c454731c : SUCCESS in 9m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/a081a054c3564e8d89a344ed8818cbdb : SUCCESS in 1h 10m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/34f46ed5811f47ceaeae2a00c9bc0af8 : SUCCESS in 45m 00s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/85ee3844b1e7471b8d10f33b0b4c9369 : SUCCESS in 10m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b991cde3301741fea29e37c99a14233d : SUCCESS in 7m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/436fd6da56d0408ca5f4532129a44cf9 : FAILURE in 15m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/230ec62c92e14b83811326d4122418df : FAILURE in 18m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/35c15149772643ddb9db9c641dd4ff3f : FAILURE in 16m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/94d5f9002ac545ad9301befab1154ea8 : FAILURE in 16m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6c342b89d7544f239d60a0f320b6a1b9 : FAILURE in 17m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/937c19bb9ddb4e36a3695cf45af59dbf : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b050e7d16d0048bf86040485f7f511b4 : SUCCESS in 17m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/74829cfc68544eac8e0393f0c7f8eaf5 : SUCCESS in 18m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e25f46aa357e4157aafaf28acb3dbba8 : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/c70d031ae71340339386169b15e1d995 : SUCCESS in 17m 15s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f3137398c26b43c1b7c35c60cfddf237 : SUCCESS in 17m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/efa4c1e1333248db84799b7f687483cb : SUCCESS in 7m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bdd4aeb6670a4c77ba0d784aa4e9d33e : FAILURE in 21m 30s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/eb7ffd30667143d5a4e63ecaa4611d02 : SUCCESS in 46m 52s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/8a39323050e0417490a3c6c1349cfd44 : SUCCESS in 44m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4d70a23860fc4efb8e0bdfe5379ad014 : SUCCESS in 43m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4c024598813f420f86f362c695266fa4 : SUCCESS in 37m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be87b8024dda4251a7e1062ecaf390de : FAILURE in 15m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b4e9c823c2e4d3da32177054b3ec317 : FAILURE in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/41c642567a9445c595074a1d82ef64c1 : SUCCESS in 22m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd0a4bc810734a80a11e864fe4a9388c : SUCCESS in 56m 38s","accounts_in_message":[],"_revision_number":8},{"id":"045d9d3a307f232cc94b3e3bad224cf7fcccfb6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-06-04 05:59:26.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f407534c5e0f05f548ecc8c6c07f2a19cb0a936a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 07:14:41.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da2b0a4f1d2b42ff89d174e41e6ffd7d : SUCCESS in 10m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/2179840d2ea24d4ca4a5ed1272babb49 : SUCCESS in 1h 08m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/54336788a5764a7bb52377eb15b350eb : SUCCESS in 40m 40s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/de421e35331c43e2b848e0532aa5d243 : SUCCESS in 10m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a4eb0d8d5ef24ae2b2222c1915893937 : SUCCESS in 9m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4dd34ea663fa49e9b81be6789f6ca28d : SUCCESS in 18m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/957b1c8d8a90435580172db78ed43ea7 : SUCCESS in 15m 56s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/628e092a25c746fda50a606cf727dc00 : SUCCESS in 16m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/87f588e379694192848c2288ede3c873 : SUCCESS in 18m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/17fa95640f464610ba3ea39ac2466a73 : SUCCESS in 17m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bbc6912df8544cdd9d4d1f236f4cbbe2 : SUCCESS in 21m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e548afaa03aa4708a45e0d3942e724a0 : SUCCESS in 18m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4dbaaf806cce45668725ed4c92266b20 : SUCCESS in 25m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e98c984ea2ae40eb86b7fefd463082ce : SUCCESS in 19m 05s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/2bfb0561becd4f3789602906cd31f9b2 : SUCCESS in 20m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8edd4326f47146f2a36e88ce68610ec2 : SUCCESS in 19m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/92d43e754f6b4f799f610daccddb884a : SUCCESS in 8m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/74b6357b9ce84cf089701fe38904ea89 : FAILURE in 21m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4ab7734b3c114844a421146f9c89e5a7 : SUCCESS in 57m 55s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/9aa25977e91e4d3e8543970a331de7f1 : SUCCESS in 48m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e1a70b1316cc439eb4001908ebd1cc78 : SUCCESS in 45m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cc82b6d800a0459d9938cbfa88ac9a36 : SUCCESS in 40m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d88b61dca608410baf451136a1977d77 : SUCCESS in 18m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b2e177d24e84e98b4d95d7af4360c61 : SUCCESS in 6m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7502531ad1e34c338967e50b01e457f1 : SUCCESS in 25m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6f748490e3848cba803d7974e4b2d73 : SUCCESS in 1h 05m 24s","accounts_in_message":[],"_revision_number":9},{"id":"42c3e40a5b64e3998ad75137a5a058559a2a5b90","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-05 00:04:12.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"875561bac5f2a89d531d584e736a8c6ad1ad4ddd","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-06-06 23:29:04.000000000","message":"Patch Set 9:\n\n(2 comments)\n\nThanks Tim, you always have awesome suggestions! New patch incoming.","accounts_in_message":[],"_revision_number":9},{"id":"55f66d43c3396c8be4c91b35165e0b6d9987dfe2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-06-07 00:23:47.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"10eb7e4b53863f93a46a77e46b202a621e6f306e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 01:34:27.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a708b449ddc6494c90adde71a0b75219 : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/96d6fbfb2c73458db153668c3680622f : SUCCESS in 1h 05m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7798e1cdbe5c4a78b3cf06c9942ae0d8 : SUCCESS in 44m 21s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2874cffde9134813b195b17c4063cc9f : SUCCESS in 7m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b97a857c20b844de900cf6b4decd2451 : SUCCESS in 7m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/81a3ec64d7f342a68408a693aad4af12 : SUCCESS in 16m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/58b37d5efb7344c38e2200f37f3c4603 : SUCCESS in 14m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/83e0d2b26d924d1486dfcfb9f4ee2e7e : SUCCESS in 14m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/264b530cb248462da51e02ecb06fbad1 : SUCCESS in 16m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3f73da59a98f46729f8c11118654581d : SUCCESS in 11m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/37cf8687f9cd45858d2c3354e1a802c9 : SUCCESS in 15m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4091707637ed43f581ebd49168ef494f : SUCCESS in 19m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/76104dddf8bc4f55a101cf5ef7a0a18b : SUCCESS in 17m 36s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f76e466dbb9f453ead8b81805bd101ad : SUCCESS in 14m 09s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/5aeb8076368944b7a85a16cd662097d4 : SUCCESS in 17m 02s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/5e81711c0e91403397b2531dbc166ae4 : SUCCESS in 19m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ce988a99e90f47d5a6c8e9be8e80d1dd : SUCCESS in 6m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ea676b3cb2cc4dd6906f1759c577331a : SUCCESS in 26m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/293279b3b8964e9c86e7061289688f41 : SUCCESS in 45m 00s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/d19184f574d8420687ae98d01375648e : SUCCESS in 51m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8afab0d19c754396a02f95b4bf72ec4a : SUCCESS in 27m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cef0de53767a4bf8a5ab919a54817c1c : SUCCESS in 36m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/179b363e372f4beaacecc84f5220a10b : SUCCESS in 16m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/053cf309b15846e0b569cdb3aac0ed12 : SUCCESS in 6m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/77d2c365e3dd4e2092a99c5724feb59f : SUCCESS in 23m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da1452fec11b4b778af7905d2c4b2b2c : SUCCESS in 46m 22s","accounts_in_message":[],"_revision_number":10},{"id":"ebfe7a492586806ecb07c2c31ddf64a1b11a343a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-06-07 10:05:24.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)\n\nGreat clean up! Just one comment on the test changes - I\u0027m interested in your/other opinions.","accounts_in_message":[],"_revision_number":10},{"id":"60f1d98792ffaeeb73710fbd31ede03064eaf2a0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-17 19:05:43.000000000","message":"Patch Set 10: Code-Review+2\n\nI see Alistair\u0027s point -- minimizing test churn during a refactor offers a pretty good signal that nothing has materially changed. The changes in middleware/test_recon.py make it less obvious that we\u0027re still doing the same thing that we were before. At the same time, I\u0027m not nearly as opposed to the use of the constant in obj/test_replicator.py.\n\n*shrug* Code in hand tends to win out.","accounts_in_message":[],"_revision_number":10},{"id":"734a7bb45da7e15a52ee3c7b2dba5055e8d761fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-29 22:23:06.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"d2a0f95be7e4263e358fd7950b7f627fdc2e51de","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-29 22:23:49.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\nMerging before it ends up in conflict again ;-)","accounts_in_message":[],"_revision_number":11},{"id":"484bf3054309ad2acecf66c822d94f5603cee254","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 23:08:15.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/63897efc4cd6451c8a6797c35209e794 : SUCCESS in 41m 11s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/439a1695028749488cd781cb2dfd2250 : SUCCESS in 43m 13s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"6b15df960f171d1bdaa6e2e31b2074f660d87b8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 23:35:20.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c008ec11ea94833ad528abe07814dc8 : SUCCESS in 10m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/f921bf13cc31437f8956c75ded57ca34 : SUCCESS in 57m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/60c622cb4be04ded9fe80fee75a2dc88 : SUCCESS in 43m 27s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5eca450d63cb4a12b1e3d9d33e9f78bf : SUCCESS in 10m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dae746ae08d3468b808dc4875a6709c1 : SUCCESS in 9m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/aa90493ae0e944d9abeb7ddc2a1ae4ef : SUCCESS in 17m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5b5a35b3d08e442d840f5f4aa43e9b6d : SUCCESS in 17m 24s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/412c1b6ec06f4a3a9f2fb73ea71aeece : SUCCESS in 17m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4b2daff0c3484f3eb1ef5c4246238741 : SUCCESS in 22m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1da44b62dcf64bb4b7f1d341e6bf5536 : SUCCESS in 19m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/98ba40612b74484ca0a2b8d9aa0fd81d : SUCCESS in 24m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9c6ff62db45343e1b92fb23bd7c999ef : SUCCESS in 18m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/62153348bb114335bd069004f0b1e24d : SUCCESS in 19m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1a966fa0e7fe4ad18d2a1417326cba16 : SUCCESS in 19m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/67a93d4498c54e60b2f773585cc66645 : SUCCESS in 20m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c72e9f8a23bd4058ad41e227459a8824 : SUCCESS in 20m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/357e65adeb1a486fb334df8acc87ba92 : SUCCESS in 7m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/83306a1151ce46feb91eda30b082925a : SUCCESS in 23m 20s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7285bbe47f164678aef08b374807a578 : FAILURE in 53m 11s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/39213bf0a900487abea019fedebdc633 : SUCCESS in 50m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/25d32fb5575a4a4fa66a90296c3095b0 : SUCCESS in 39m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d56ddce7c13d4e8d86d24f319ac50736 : SUCCESS in 42m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3691d202aac47db804d43a45b00cb9a : SUCCESS in 18m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd2d57fdf12c42679a2889ea8f5df4c4 : SUCCESS in 7m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/88f367ff0fdf45498696f30d27f24062 : POST_FAILURE in 1h 06m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa89c5c95b164f3aa8c18246bb84df2d : SUCCESS in 55m 11s","accounts_in_message":[],"_revision_number":11},{"id":"f7ae3a653bcd8b6df2d5202d955bd2b413e81bf4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-30 19:15:17.000000000","message":"Patch Set 11:\n\nrecheck","accounts_in_message":[],"_revision_number":11},{"id":"3aeb8c5feb00357ed17e4a9fb7b178c4d54fa453","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 20:05:47.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/65c7101526844387846ce4ffa6716da9 : SUCCESS in 39m 11s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f21776fb6a5c4498a4b809a24a26fb95 : SUCCESS in 48m 44s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"3723451c077fc6a3faa6094ed052e3c8e04be594","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 20:33:10.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62593471f089430ca72063eb2b268e42 : SUCCESS in 9m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/20af1cfe1f5f448b8bc96db98341450c : SUCCESS in 1h 13m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c985d43fbaa845e4b8d259cc5cd851be : SUCCESS in 48m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/97a9eba825b44f3dadb5e1bda31cf294 : SUCCESS in 9m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/38daa43cc618450dbb12421f3819c888 : SUCCESS in 10m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/077016059c2c483cb7c6b44fad1e22a3 : SUCCESS in 17m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0646aea5985f425b8560ad42dba45857 : SUCCESS in 16m 47s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d075cb3856814814863d75f638d3fe4f : SUCCESS in 17m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1159adb6ce754f398b56bf88ab632004 : SUCCESS in 16m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9851b7f67b6b4019be8f29ac13c95c7d : SUCCESS in 16m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/db4340788edd44cf8e09556ab44ae000 : SUCCESS in 16m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/306f9ddd1ca645208c003f71224e50b9 : SUCCESS in 16m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/75ffac08898e4daa975ee61293844dac : SUCCESS in 19m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1cc52dbc79c340a897f07384c4e52e1b : SUCCESS in 17m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/aa61e8c586244c0198e42029c375dc73 : SUCCESS in 16m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0c5cd2cdf3724b0eb354bca67b7e5098 : SUCCESS in 16m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/576cf58fa0c84c3ab2da6280d525e9f3 : SUCCESS in 7m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b3094b935a864e5bb2ab34c1baae30cb : SUCCESS in 24m 55s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/07c067d2574b4ab7a87c5442fdd8dfb1 : SUCCESS in 46m 07s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/1ae7f825412548c5a00a31692de081a4 : SUCCESS in 49m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/22616186f68d4465a649ea3a09ae609e : SUCCESS in 40m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cba9431198ce477aba53dbaec4160722 : SUCCESS in 36m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34bd4ec19468449190ad0853da7e9651 : SUCCESS in 18m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b70de71ea4a4e688d0fa0fcc4b9c43b : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/49b13c10c14642cd8de11a71a2ac5493 : SUCCESS in 21m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b26ad6ef5be3453cbac15c75d541a36f : SUCCESS in 48m 57s","accounts_in_message":[],"_revision_number":11},{"id":"73507ddaf972825442ef4fffb78d603c547af6a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 20:33:34.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"d8fcb5fc410d21a0b5b3696a9f7563862d1183ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 02:10:08.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/181daab2b0e14a359c50b86167847f14 : SUCCESS in 9m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/59116afa01b045cb83d850e5f7f37759 : SUCCESS in 58m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7e119cbfec64424dbec9c9d4437778d5 : SUCCESS in 42m 16s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/896f9afbe69443e6bb13b3fe6f00521c : SUCCESS in 10m 57s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ca20241fe9e64ba7a16adb49809c9b30 : SUCCESS in 9m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/25834dafb8094a109673ce68ad56b56c : SUCCESS in 15m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0fd29e87e3764af7803f4e97fa96c4eb : SUCCESS in 16m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0f8b511b26eb4b0c816bbfd780428e89 : SUCCESS in 15m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/55bbea8c009a452c96b86e3051a962c6 : SUCCESS in 19m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6bf713d6b0d74fd39026230134ef59df : SUCCESS in 22m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4c9d3c91ab6243ffa534e9a68d069fa6 : SUCCESS in 19m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fde6cd994b3d4096b1b78d84f1431ef5 : SUCCESS in 21m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8013a35022f640748a5c3efc5dc1bb12 : SUCCESS in 18m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1175e9f36ccd403d98c6661563f19d36 : SUCCESS in 19m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/79e039d9034a4603b51153985e140979 : SUCCESS in 17m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5b52d877efd34ffda195c7fe3b485e57 : SUCCESS in 22m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/10ca31d840494c01ae78494bd4b0d1eb : SUCCESS in 7m 31s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1ac50106583b4b3baf6cd19379fcf836 : SUCCESS in 50m 36s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/922e3d38cb3a44a790d54088d562ef44 : SUCCESS in 45m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4f6b2066a76d4921b817ba9f68ef5296 : SUCCESS in 41m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/592cf1a0d47e409e9051b8c78695c03f : SUCCESS in 38m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08f44f04cb7c4dc8b6ee3b10dfc7339d : SUCCESS in 17m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd663af9d5f542beaec94d3332d7c4f0 : SUCCESS in 6m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e1264bb4d05e495ea7bb73d095581759 : SUCCESS in 30m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0bd2f9711704acfbdcc15d435d79e22 : SUCCESS in 41m 08s","accounts_in_message":[],"_revision_number":11},{"id":"17cfd69fa4aa5266bf56ec0dcb57f09de6f78250","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 02:10:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"5a65f34258c2e9d7996598ecbfa18de4fb34d226","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 02:11:42.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a87685d39674c56b8b8d2f525585296 : SUCCESS in 1m 08s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/434d8b1a4a084205b6f04c2c05129c5f : SUCCESS in 48s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/75d4a144dc2946e18829620d9c8ba346 : SUCCESS in 48s (non-voting)","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"85e36f7122b83d53b5ce931f72ebf6c6f6fdfaf4","revisions":{"750e7632e772c6872c95f079e2fe06294cdfc5e5":{"kind":"REWORK","_number":1,"created":"2021-03-31 04:48:09.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/97/783997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/783997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/783997/1"}}},"commit":{"parents":[{"commit":"b9b6fc8ca26dd78cdb2c81583dd39465b99b6121","subject":"relinker: Add /recon/relinker endpoint and drop progress stats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9b6fc8ca26dd78cdb2c81583dd39465b99b6121"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-03-31 04:07:39.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-03-31 04:47:19.000000000","tz":660},"subject":"recon: refactor common recon names into a common location","message":"recon: refactor common recon names into a common location\n\nChange-Id: I0a0766cfb6672377de0f152ce179c874c327ec54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/750e7632e772c6872c95f079e2fe06294cdfc5e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/750e7632e772c6872c95f079e2fe06294cdfc5e5"}]},"branch":"refs/heads/master"},"24392d46d4529582833149a0c13c9b30449fc11d":{"kind":"REWORK","_number":2,"created":"2021-04-01 03:07:05.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/97/783997/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/783997/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/783997/2"}}},"commit":{"parents":[{"commit":"b9b6fc8ca26dd78cdb2c81583dd39465b99b6121","subject":"relinker: Add /recon/relinker endpoint and drop progress stats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b9b6fc8ca26dd78cdb2c81583dd39465b99b6121"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-03-31 04:07:39.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-04-01 03:06:39.000000000","tz":660},"subject":"recon: refactor common recon names into a common location","message":"recon: refactor common recon names into a common location\n\nChange-Id: I0a0766cfb6672377de0f152ce179c874c327ec54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24392d46d4529582833149a0c13c9b30449fc11d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24392d46d4529582833149a0c13c9b30449fc11d"}]},"branch":"refs/heads/master"},"ab86685897987ee6b04406f1547e3075492b5717":{"kind":"REWORK","_number":3,"created":"2021-04-06 09:53:27.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/97/783997/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/783997/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/783997/3"}}},"commit":{"parents":[{"commit":"421022a103f5bf749dc4e545bfc9f2628d9e0f9d","subject":"relinker: Add /recon/relinker endpoint and drop progress stats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/421022a103f5bf749dc4e545bfc9f2628d9e0f9d"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-03-31 04:07:39.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-04-06 09:31:20.000000000","tz":600},"subject":"recon: refactor common recon names into a common location","message":"recon: refactor common recon names into a common location\n\nChange-Id: I0a0766cfb6672377de0f152ce179c874c327ec54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab86685897987ee6b04406f1547e3075492b5717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab86685897987ee6b04406f1547e3075492b5717"}]},"branch":"refs/heads/master"},"054a93fd13e6511bc6f55297d162533aabee29d9":{"kind":"REWORK","_number":4,"created":"2021-04-09 05:28:22.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/97/783997/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/783997/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/783997/4"}}},"commit":{"parents":[{"commit":"ee5f8a0105ce5f4391209dcd42f25ad6db9517c8","subject":"relinker: Add /recon/relinker endpoint and drop progress stats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee5f8a0105ce5f4391209dcd42f25ad6db9517c8"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-03-31 04:07:39.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-04-09 05:27:49.000000000","tz":600},"subject":"recon: refactor common recon names into a common location","message":"recon: refactor common recon names into a common location\n\nChange-Id: I0a0766cfb6672377de0f152ce179c874c327ec54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/054a93fd13e6511bc6f55297d162533aabee29d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/054a93fd13e6511bc6f55297d162533aabee29d9"}]},"branch":"refs/heads/master"},"fe50ea9ed46689fb19bbe11a54247a59f12a4c96":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-04-12 00:32:51.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/97/783997/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/783997/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/783997/5"}}},"commit":{"parents":[{"commit":"b59449401ef30240b76f2549d4d1758fd19de10c","subject":"relinker: Add /recon/relinker endpoint and drop progress stats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b59449401ef30240b76f2549d4d1758fd19de10c"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-03-31 04:07:39.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-04-12 00:32:32.000000000","tz":600},"subject":"recon: refactor common recon names into a common location","message":"recon: refactor common recon names into a common location\n\nChange-Id: I0a0766cfb6672377de0f152ce179c874c327ec54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe50ea9ed46689fb19bbe11a54247a59f12a4c96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe50ea9ed46689fb19bbe11a54247a59f12a4c96"}]},"branch":"refs/heads/master"},"aa62dbd3e75ea0da68d76957448a3199c7338964":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-04-13 07:28:08.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/97/783997/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/783997/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/783997/6"}}},"commit":{"parents":[{"commit":"9835e54c5a5edfe9295544f33ffe3f04c5bfb970","subject":"relinker: Add /recon/relinker endpoint and drop progress stats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9835e54c5a5edfe9295544f33ffe3f04c5bfb970"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-03-31 04:07:39.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-04-13 07:27:21.000000000","tz":600},"subject":"recon: refactor common recon names into a common location","message":"recon: refactor common recon names into a common location\n\nChange-Id: I0a0766cfb6672377de0f152ce179c874c327ec54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa62dbd3e75ea0da68d76957448a3199c7338964"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa62dbd3e75ea0da68d76957448a3199c7338964"}]},"branch":"refs/heads/master"},"ce5b8896c8656594d93da3cfe649f15fa36dd983":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-04-14 06:44:49.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/97/783997/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/783997/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/783997/7"}}},"commit":{"parents":[{"commit":"5b355b34ab8c457b2f09734e3a5a6a17330932c9","subject":"relinker: Add /recon/relinker endpoint and drop progress stats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5b355b34ab8c457b2f09734e3a5a6a17330932c9"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-03-31 04:07:39.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-04-14 06:42:01.000000000","tz":600},"subject":"recon: refactor common recon names into a common location","message":"recon: refactor common recon names into a common location\n\nChange-Id: I0a0766cfb6672377de0f152ce179c874c327ec54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ce5b8896c8656594d93da3cfe649f15fa36dd983"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ce5b8896c8656594d93da3cfe649f15fa36dd983"}]},"branch":"refs/heads/master"},"534afc0f5b924f141f9378387522e9c03b2f4268":{"kind":"REWORK","_number":8,"created":"2021-06-01 01:29:24.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/97/783997/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/783997/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/783997/8"}}},"commit":{"parents":[{"commit":"2c71e7416c4fa3cc5886658bd5eaac77f30416a4","subject":"Merge \"Quote paths before sending them to swob.Request.blank\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c71e7416c4fa3cc5886658bd5eaac77f30416a4"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-03-31 04:07:39.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-06-01 01:27:08.000000000","tz":600},"subject":"recon: refactor common recon names into a common location","message":"recon: refactor common recon names into a common location\n\nChange-Id: I0a0766cfb6672377de0f152ce179c874c327ec54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/534afc0f5b924f141f9378387522e9c03b2f4268"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/534afc0f5b924f141f9378387522e9c03b2f4268"}]},"branch":"refs/heads/master"},"881f2da11b8d4aa7f32c0c332d4920168afd06f2":{"kind":"REWORK","_number":9,"created":"2021-06-04 05:59:26.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/97/783997/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/783997/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/783997/9"}}},"commit":{"parents":[{"commit":"2c71e7416c4fa3cc5886658bd5eaac77f30416a4","subject":"Merge \"Quote paths before sending them to swob.Request.blank\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c71e7416c4fa3cc5886658bd5eaac77f30416a4"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-03-31 04:07:39.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-06-04 05:59:09.000000000","tz":600},"subject":"recon: refactor common recon names into a common location","message":"recon: refactor common recon names into a common location\n\nChange-Id: I0a0766cfb6672377de0f152ce179c874c327ec54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/881f2da11b8d4aa7f32c0c332d4920168afd06f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/881f2da11b8d4aa7f32c0c332d4920168afd06f2"}]},"branch":"refs/heads/master"},"69e200f8b90eb36a39bd5fd6038e7f3f6a36c7f7":{"kind":"REWORK","_number":10,"created":"2021-06-07 00:23:47.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/97/783997/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/783997/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/783997/10"}}},"commit":{"parents":[{"commit":"2c71e7416c4fa3cc5886658bd5eaac77f30416a4","subject":"Merge \"Quote paths before sending them to swob.Request.blank\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c71e7416c4fa3cc5886658bd5eaac77f30416a4"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-03-31 04:07:39.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-06-07 00:22:35.000000000","tz":600},"subject":"recon: refactor common recon names into a common location","message":"recon: refactor common recon names into a common location\n\nChange-Id: I0a0766cfb6672377de0f152ce179c874c327ec54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69e200f8b90eb36a39bd5fd6038e7f3f6a36c7f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69e200f8b90eb36a39bd5fd6038e7f3f6a36c7f7"}]},"branch":"refs/heads/master"},"85e36f7122b83d53b5ce931f72ebf6c6f6fdfaf4":{"kind":"REWORK","_number":11,"created":"2021-06-29 22:23:06.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/97/783997/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/783997/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/783997/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/783997/11"}}},"commit":{"parents":[{"commit":"8e1209d2b1b41193ddb3f616036d3a1475ed1210","subject":"Merge \"Capture logs when running custom daemons in probe tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e1209d2b1b41193ddb3f616036d3a1475ed1210"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-03-31 04:07:39.000000000","tz":660},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-06-29 22:22:57.000000000","tz":-420},"subject":"recon: refactor common recon names into a common location","message":"recon: refactor common recon names into a common location\n\nChange-Id: I0a0766cfb6672377de0f152ce179c874c327ec54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85e36f7122b83d53b5ce931f72ebf6c6f6fdfaf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85e36f7122b83d53b5ce931f72ebf6c6f6fdfaf4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
