)]}'
{"id":"openstack%2Fswift~784506","triplet_id":"openstack%2Fswift~master~I5a73bce70a97bb6df724a60e469f7b53764c7897","project":"openstack/swift","branch":"master","topic":"p-relinker-eexist-tombstone-v2","hashtags":[],"change_id":"I5a73bce70a97bb6df724a60e469f7b53764c7897","subject":"Use debug_logger instead of FakeLogger in relinker tests","status":"MERGED","created":"2021-04-01 21:33:50.000000000","updated":"2021-04-07 01:20:49.000000000","submitted":"2021-04-07 01:19:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784506-p-relinker-eexist-tombstone-v2","meta_rev_id":"44aab1eeb4b6c8614f7943b2eac64b9f295c8fb9","_number":784506,"virtual_id_number":784506,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-07 01:19:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2021-04-06 08:55:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2021-04-06 08:55:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-01 22:28:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-06 08:55:23.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"abd7318ee731ea1677f174fb5d70654f9ad70133","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-01 21:33:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"362491d6f90827b44cd973bf12b0954594bee049","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 22:28:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e74cecfe712c4ba5a8a0d50864a62056 : SUCCESS in 9m 30s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/37b4bd72c4264a679700acbef182f447 : SUCCESS in 9m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/147de5fc9bbc4042ab80f743bd4b9d3a : SUCCESS in 7m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4cdb926ab5cd42788d4cd9a59c52cc7f : SUCCESS in 18m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e1b003d1dcc0413c9ac839ef3f68a6b7 : SUCCESS in 23m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a60693f00b854118ab55f51d9223c7e9 : SUCCESS in 19m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/47a583258d87412bb4d1fa2637e242d3 : SUCCESS in 16m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d307b92c6f01462c9021c6bbf32848d0 : SUCCESS in 18m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ab4cf43402384a7e8fc1d0147a4e203d : SUCCESS in 18m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/52baa0a775db4013946a70e558c65bad : SUCCESS in 29m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2eab55096bee46a3b8dc92a04e14acdf : SUCCESS in 23m 02s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1d9b8880e8de4eaeaeff1a5c6a14d620 : SUCCESS in 21m 25s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/517d92897f104fe98cdee7e998aea37b : SUCCESS in 30m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0024ec0bbdde401d8fb1494ba1e3eb01 : SUCCESS in 14m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/12883f8ef802492e80ffd2289c1a275c : SUCCESS in 52m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8ce6e49a7e304b99abfe359aab2cfe2e : SUCCESS in 43m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/253c62f8c4594db4b4a7756c79bbb059 : SUCCESS in 20m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6234f20431ad44cabb801e88e39becb7 : SUCCESS in 6m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8856f1c1f8b44cdfa606539995c55320 : SUCCESS in 31m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5d16563d052130e2010875a3cf7343d8472cc54c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-06 08:55:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"19880f1db1aa14480f098381954a98a315be9c36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 08:55:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ae05f1add87891056a2d8724a91ae9098aedfa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 13:14:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53e9b0a425da4be19d5341f9b483c72f : SUCCESS in 10m 43s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d889a10ad0be4bb68bd0d34179ed30fb : SUCCESS in 10m 55s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/ad5f9b22b7aa457c9400e97678065258 : SUCCESS in 8m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d40277b3b1f8468f9e2f74c03968d868 : FAILURE in 5m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8b8b9fa7cb624340bbc45251352d145a : SUCCESS in 15m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/66f0ba9f6ca84ee6929143ebfee4b59b : SUCCESS in 44m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4d9d28edc0734267a0bee50eb49c10c5 : SUCCESS in 18m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d0d7e4d6c5f74b34ba3e86bc15c837c1 : SUCCESS in 16m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/68c2c35dd3a742c9b9990acbb5084e55 : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/96bfdf024ace414dba5a9a2d37108c85 : FAILURE in 6m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b58991b847404ab5ae1356d9c187b3b0 : SUCCESS in 20m 04s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/afd497a2d40e4bdbb339641c532f5b91 : SUCCESS in 16m 37s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7f6bfb34a6124f1ba5797bd2ef5def88 : SUCCESS in 16m 22s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d01594e6088b422f86d799a495c779c5 : SUCCESS in 18m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7570c35e2cf74dfaa4c564aa41b67fe0 : RETRY_LIMIT in 4m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a9ada6a21f5a4a46a37d5667859c11bb : SUCCESS in 47m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/72031ec21c764ad09ade6968f32fa9e8 : SUCCESS in 49m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b7b9def606d480bace7b0e9c651a30d : FAILURE in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d04d290f1944ad48d94304b42539b90 : SUCCESS in 5m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c4ee1105d8e64dcda4b2a64e3ca4f2b1 : SUCCESS in 24m 03s","accounts_in_message":[],"_revision_number":1},{"id":"62ee2d5a8e115a0f41e255358c8bd72fc49a8d20","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-06 21:24:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ddd4f8802c3468bc4c4d820a9ec8cd5feb580fe6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 22:22:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14bcb51d00a34ac288872646ac485221 : SUCCESS in 9m 39s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ea9e4b2ef148456eb29f1ab6b2fa9863 : SUCCESS in 9m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a2ef47d28eb844188c488b79a30884b5 : SUCCESS in 9m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/84a9aa5177f84769a949932e9a81512d : SUCCESS in 24m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4df24d44f0654861b2eea9937ed07935 : SUCCESS in 20m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d8386e42b7874f8e99afe3b52aef1a8e : SUCCESS in 20m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b3496488e5c940ea9fc0430fd5035f49 : SUCCESS in 20m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/05a49006c9bb467596055608a52f8bbe : SUCCESS in 17m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9af0a702b5d74dbd8b00f57bd0624c68 : SUCCESS in 18m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fcfb9cbeda70481ea8cd92a0a8baf26f : SUCCESS in 18m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/15af94e9656d489e99063147f3453a7e : SUCCESS in 17m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/78bd535492e94551a82029f7766be2ee : SUCCESS in 19m 08s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f19ee878ad024081b223c4db0c9fd40d : SUCCESS in 24m 49s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/672240e8654347e7b606b260abceb1c6 : SUCCESS in 26m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3dcee0e76f8647a88eb160c9b6c1f69e : SUCCESS in 46m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/492a51e5bad840af8667870fc7092433 : SUCCESS in 52m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c18fdccf2d047a2844b36ee15356b66 : SUCCESS in 15m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d01ca6f988f4ee8b37ea3dc114f84b5 : SUCCESS in 7m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4e86af479cea48a3aeb893e97192b711 : FAILURE in 30m 10s","accounts_in_message":[],"_revision_number":1},{"id":"5a9744e367a7486315717860ad2c1f1420ae61e4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-06 22:23:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d4cdc3b73e63561e1a8b228170f170c5e5005043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 23:05:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93b896400f044d81a28b410c2b4f1d5a : SUCCESS in 8m 51s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d40a574c8c384d50bf3247628d25ab21 : SUCCESS in 8m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bc1419d175854c37984f631ee6dd8e0c : SUCCESS in 6m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/84c7e0fa5d114de890372f9344b21325 : SUCCESS in 18m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/90ac7243b5ae4d4c9cac8d17554af9c0 : SUCCESS in 18m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/84bb154936f841968f695b76f71a6587 : SUCCESS in 18m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a0b836ac5b0e41a295648926b983be88 : SUCCESS in 18m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/af2326cef118437185e18696788413dd : SUCCESS in 18m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7de8b79b3f6243bfb34980d74205891d : SUCCESS in 19m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/526b0496394a467da664698e7c79377e : SUCCESS in 21m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9b0b3eb772f642ba97a5232d0b2c5d5a : SUCCESS in 20m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9ac5b2072d72478bac936601bd2bf5ed : SUCCESS in 20m 55s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0220dc552f3b4379b9f9a70b5d652705 : SUCCESS in 21m 11s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/dab218f1ddf44958ba236a955ddb73f4 : SUCCESS in 20m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/79d5850a29644d1b9b613ecb4ebad586 : SUCCESS in 38m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d8c4dcd1ae7e4e45b0e330cb1d40c572 : SUCCESS in 40m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee3d1c2bc08b4cc7bd7be07cfb91fa9c : SUCCESS in 19m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef313a0106bb46a5bd3478ede1204ef5 : SUCCESS in 6m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a00a6d019d5d4823b31d5404fecc850b : SUCCESS in 24m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f89fe543fb7e1accb00dbe3164a05bd876c38f25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 23:06:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d215a4e3127bd9df4d1f50a0155a143ee3dda90f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 01:19:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3297ac004fe5435886ac02efdb50918b : SUCCESS in 10m 38s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/bfe4266c6a614ab4b4e9e806b15e6e0f : SUCCESS in 8m 38s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/503e6ec4e09a4c65b510be323ea2292f : SUCCESS in 7m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e3a61bb7a7794c93bfaad588f1ce5534 : SUCCESS in 18m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/83d6d3b1f4c9439090f4bd6ab736410b : SUCCESS in 18m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fcf06b23f6a44ff4b2930506b3b69caa : SUCCESS in 19m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3a2403b2a44b4a68bc283e74567ddf46 : SUCCESS in 18m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/35a86cd91dd84d68a881c4f6048bc37d : SUCCESS in 18m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f295b1fe67b54ece879bcb7f7d09b544 : SUCCESS in 20m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4f04bf11183c40728f3b01d8f2d0e83e : SUCCESS in 20m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4c6f07b6a3e44655820b51a9e74328a1 : SUCCESS in 21m 24s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/536cbbc1ab42423ca6fec957bb7cd3c4 : SUCCESS in 20m 45s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ea5471ae56bb4126960767e58fff34fe : SUCCESS in 21m 09s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b40b4d1220144fb3abe8557a5bdb0f78 : SUCCESS in 22m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/23ce781972fb472183da0f36305f7c59 : SUCCESS in 6m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6d6893b91a424cf0bf89e2a5201ee7eb : SUCCESS in 40m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/10acd23aaf2949978f724e93e22433b2 : SUCCESS in 40m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d79ab04e4cf4a59babfc5e4ab74ce2d : SUCCESS in 16m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6662aa3f4ed64b9c981fa7146ea5b8ef : SUCCESS in 6m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a9e8401d898844aebbc2e97e0378301c : SUCCESS in 30m 33s","accounts_in_message":[],"_revision_number":1},{"id":"07a51cb42eb808b2bc153a27458f8b73bf06ef35","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 01:19:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"44aab1eeb4b6c8614f7943b2eac64b9f295c8fb9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 01:20:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a8010372d7646f59988e15957f98fd2 : SUCCESS in 1m 12s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/687fb61952c74672a785c39cc36283b1 : SUCCESS in 35s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/cbc44fb6642b42c49a5c7d6e6f3125ec : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"20336f177631c56f4c62e9213eea47d71467e5aa","revisions":{"20336f177631c56f4c62e9213eea47d71467e5aa":{"kind":"REWORK","_number":1,"created":"2021-04-01 21:33:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/784506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/784506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/784506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/784506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/784506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/784506/1"}}},"commit":{"parents":[{"commit":"3bdd01cf4a1c8ee9ac953850106da505d381aa44","subject":"relinker: retry links from older part powers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3bdd01cf4a1c8ee9ac953850106da505d381aa44"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-01 21:33:39.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-01 21:33:39.000000000","tz":-420},"subject":"Use debug_logger instead of FakeLogger in relinker tests","message":"Use debug_logger instead of FakeLogger in relinker tests\n\nChange-Id: I5a73bce70a97bb6df724a60e469f7b53764c7897\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20336f177631c56f4c62e9213eea47d71467e5aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20336f177631c56f4c62e9213eea47d71467e5aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
