)]}'
{"id":"openstack%2Fswift~784617","triplet_id":"openstack%2Fswift~master~Iac71f68f6633764e0c926ca60990be3b16ef6855","project":"openstack/swift","branch":"master","topic":"sharding/reconcli","hashtags":[],"change_id":"Iac71f68f6633764e0c926ca60990be3b16ef6855","subject":"Add sharding to swift-recon","status":"MERGED","created":"2021-04-02 21:49:33.000000000","updated":"2021-04-28 21:33:12.000000000","submitted":"2021-04-28 21:29:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"784617-sharding/reconcli","meta_rev_id":"05b1ed69b4b36b406ef31eb35091934d55f88ba9","_number":784617,"virtual_id_number":784617,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-28 21:29:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-28 18:08:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-28 18:08:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-02 22:55:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-28 18:05:52.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"3c8fb1ad029ce22f8e01eb8be9b7461621af3407","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-04-02 21:49:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f74c201cd122755fe2c51c8912f59347cfdb7c21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 22:55:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d705adbd12f94b3d8a255d2ac401fc3f : SUCCESS in 9m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/fe73ba6a81ee40ea8ddb1eff541fd08b : SUCCESS in 1h 04m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/da977132aac1499bbfc8a53eb23eb632 : SUCCESS in 42m 42s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b6f442d5b0c043c6979c06685a93f027 : SUCCESS in 8m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ea38dd878d6d4ef197bf0f91239b7644 : SUCCESS in 6m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ad1a21f7e92b49bca31017d91fa92bca : SUCCESS in 15m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e5af8b4504c74b0497ea9992d1b4ce8f : SUCCESS in 18m 44s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/35868c1eeab54596b5ebdc1045d0e5fe : SUCCESS in 18m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c70a942c44e5446ebb59ede0b2609c80 : SUCCESS in 18m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/485e8594d9884aeca9a006ab2b521948 : SUCCESS in 18m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/13c993dc90ab49f09236b95c783a7e12 : SUCCESS in 20m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/325c526764b84df09f95fc9de844432b : SUCCESS in 16m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d0775d8918f34585934c941e9f677d31 : SUCCESS in 23m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/45765424cc6144f2838717c8308e6b54 : SUCCESS in 18m 59s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/144223deaf214c5dad3631e50ce005ad : SUCCESS in 21m 23s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4b23d2fcb3844d73a569d4b19b649214 : SUCCESS in 21m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b6c4c0dab9ea46ef859a723e370cb52c : SUCCESS in 7m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e476ecb6192a4a06aaddfdfe961ff064 : SUCCESS in 23m 16s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/bf22bef4a72740dd9ed13f7daf8f7e99 : SUCCESS in 44m 50s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/25dc4434a9af4182a322010da5d12a86 : SUCCESS in 49m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4df870c9efdd475c85cef78963a83f30 : SUCCESS in 40m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ec6e0dc9120a4c88bc2f421d5fccd614 : SUCCESS in 41m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c086da3cf1ce460091d888debbed0f45 : SUCCESS in 19m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45fff468c3bc46c585e2bd88702f05bf : SUCCESS in 6m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/812eee21b9f14920b8be38716486614f : SUCCESS in 25m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d6a002c4e0149e685557515ba7453d1 : SUCCESS in 53m 58s","accounts_in_message":[],"_revision_number":1},{"id":"20a5e1b3e067f605ef9d2ef204f10361313a89d8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-28 18:05:52.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nLooking pretty good! One little change I\u0027ll push up shortly.","accounts_in_message":[],"_revision_number":1},{"id":"9a8f3e1ba24baeda70bd3eee0fb34948932e96fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-28 18:06:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"759275924b36b621e5955a1d0b9a6d1d37d0fac7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-28 18:08:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"923877cd640c96af4bf962a9c7fa9a2c35ac6093","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 19:28:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c305f0e9b4a44d16b395994679f6d9f5 : SUCCESS in 9m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/06e7750829614526a0ebfbf815cdad37 : SUCCESS in 1h 15m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1e7f327daa154d598d3f92a73dc81c4b : SUCCESS in 49m 50s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d65826bd105c4f01b0dc3b43f6a71682 : SUCCESS in 10m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/174c3e5dd0db4eb8b893b11ef3f80696 : SUCCESS in 6m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b7058ae73b7948e49c5d6f698399e613 : SUCCESS in 15m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/686c82d95c4d4aa692d6c2a67b67cd79 : SUCCESS in 12m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/89ae238cabbe42dcaaf243bdcf66f778 : SUCCESS in 18m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/19904e82d1f346a688411e20631fa5a2 : SUCCESS in 19m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4cfff544187744c4a2c2c649179fed84 : SUCCESS in 19m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b0dc08db97b247e6918853b5b1849beb : SUCCESS in 18m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/79e687e788354fe0a5797872a19a29be : SUCCESS in 14m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dba5393d87074b4b8fa5894ce2098979 : SUCCESS in 14m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a8594f0062904ffbabe572949e33d035 : SUCCESS in 19m 19s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ecc3923383244762874b2166ad517437 : SUCCESS in 15m 26s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d05c7d70a3254102a20367db375d61da : SUCCESS in 31m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3253d46e45934fb4ac594a0e89552bbe : SUCCESS in 8m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/012a9521b361406dbc64b210d1f2770a : SUCCESS in 25m 15s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/657d7ce2527f41429a33a2a8a232cf28 : SUCCESS in 49m 16s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/36c4f87fa53c4792b332b5c82578f354 : SUCCESS in 46m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3b0a8c054fd241ec9dbdd9b8d7e8b6fc : SUCCESS in 42m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/08f3b96b07424432ab3c2a9cf6413398 : SUCCESS in 40m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78cbafc0fce54951b1d3acf81e0c2c2d : SUCCESS in 24m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7678dfac66c84c3391ebcbd7497dccc1 : SUCCESS in 6m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b851b69530fc4c1baea2f16d8c5ec126 : SUCCESS in 38m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd88988be8fb48379c3c1262327e4164 : SUCCESS in 48m 36s","accounts_in_message":[],"_revision_number":2},{"id":"06fdd30474c843d6f2e4de81648c02b899dedfcb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 19:28:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5bef7994f1d6ee4fcf972dbbf74855e4f918a946","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 21:29:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a5fa59543c0450aa36c9637e632b2f7 : SUCCESS in 9m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/30403170bdb3470fb6af27a82e9ae74e : SUCCESS in 1h 16m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5d2422a4cac240faa4ba39f6ebe81885 : SUCCESS in 45m 46s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8c1c1dc58c234273a0b2e1a45288b0cc : SUCCESS in 11m 26s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/8d053ab2f09743ca99d31fed8692b20b : SUCCESS in 11m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6741f9804a8c49bfabd2bbcf822bc6d8 : SUCCESS in 13m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/de1bac90f6514667996e7e9e4d947421 : SUCCESS in 16m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b200f2328e1640fa99c96bf5b5d36a4c : SUCCESS in 25m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e02320f3ebe04378b4a713cdb612be72 : SUCCESS in 16m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d6168a7a43c549ba88b22f99c8a95354 : SUCCESS in 40m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/55031fa7daa649afac195bdfa28ea974 : SUCCESS in 48m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6682a3b3e154482484d263dc108a429b : SUCCESS in 18m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/08ee29635f9e4578bb308a04198530d4 : SUCCESS in 36m 28s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/461eb6fb23cf4e94ae157a71d5d0a6c8 : SUCCESS in 21m 00s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/043af34326504c83933bbf589020401d : SUCCESS in 50m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7007973940e34db3a0b01c171c850ea6 : SUCCESS in 20m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/501dd174cc35481994d243e803d77ca9 : SUCCESS in 7m 56s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e146f60994824f6d976131f5916b9a7d : SUCCESS in 1h 03m 41s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/18f54f73761146f0a431f8eabc0f04c0 : SUCCESS in 53m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2ad4aaeeab104aa5adc33f1995469eb4 : SUCCESS in 50m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/53a81c85424d4a2da225a12dfd2164dd : SUCCESS in 39m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e90c3363a6da40aab3fe6b96ced88eb1 : SUCCESS in 16m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fd79cc0f7974cfaa40d4d43ed841ab0 : SUCCESS in 7m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/26df2e0fa28d41389ff8d956de4d9624 : SUCCESS in 27m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef3f8197d20d4c96abd9b3a74a087238 : SUCCESS in 1h 00m 52s","accounts_in_message":[],"_revision_number":2},{"id":"0619ad874fe632096f8634d282e7e98bb5f72c78","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 21:29:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"05b1ed69b4b36b406ef31eb35091934d55f88ba9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 21:33:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1649b2a911ea4fd3b9604fc19d906098 : SUCCESS in 1m 04s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/045905642af74aecba8de91d26b16bd6 : SUCCESS in 50s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/e1b27d8bcd2741eeb21124c1aece200e : SUCCESS in 44s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ba00ff43764bcca0e710eb03fb3fa6583473988c","revisions":{"99e4ca14a51c8ebfcf6ab7f11546cd1d562fbf00":{"kind":"REWORK","_number":1,"created":"2021-04-02 21:49:33.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/17/784617/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/784617/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/784617/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/784617/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/784617/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/784617/1"}}},"commit":{"parents":[{"commit":"8cc1ef92558bd3f7c345c955f386498b936bace1","subject":"Merge \"relinker: Add policy to relinker progress output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cc1ef92558bd3f7c345c955f386498b936bace1"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-04-01 20:43:24.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-04-02 21:49:26.000000000","tz":-300},"subject":"Add sharding to swift-recon","message":"Add sharding to swift-recon\n\nNote that this does not modify the recon middleware: we already\nsupport the /sharding endpoint in it. But inexplicably there\u0027s\nno CLI that interrogates and parses the sharding information.\n\nThe overview_container_sharding.rst document tells operators\nto read raw JSON. It has the advantage of always having the\nfull picture, but these days we deserve a digest by CLI.\n\nChange-Id: Iac71f68f6633764e0c926ca60990be3b16ef6855\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99e4ca14a51c8ebfcf6ab7f11546cd1d562fbf00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99e4ca14a51c8ebfcf6ab7f11546cd1d562fbf00"}]},"branch":"refs/heads/master"},"ba00ff43764bcca0e710eb03fb3fa6583473988c":{"kind":"REWORK","_number":2,"created":"2021-04-28 18:06:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/17/784617/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/784617/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/784617/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/784617/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/784617/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/784617/2"}}},"commit":{"parents":[{"commit":"8cc1ef92558bd3f7c345c955f386498b936bace1","subject":"Merge \"relinker: Add policy to relinker progress output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8cc1ef92558bd3f7c345c955f386498b936bace1"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-04-01 20:43:24.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-28 18:05:15.000000000","tz":-420},"subject":"Add sharding to swift-recon","message":"Add sharding to swift-recon\n\nNote that this does not modify the recon middleware: we already\nsupport the /sharding endpoint in it. But inexplicably there\u0027s\nno CLI that interrogates and parses the sharding information.\n\nThe overview_container_sharding.rst document tells operators\nto read raw JSON. It has the advantage of always having the\nfull picture, but these days we deserve a digest by CLI.\n\nChange-Id: Iac71f68f6633764e0c926ca60990be3b16ef6855\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba00ff43764bcca0e710eb03fb3fa6583473988c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba00ff43764bcca0e710eb03fb3fa6583473988c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
