)]}'
{"id":"openstack%2Fswift~784808","triplet_id":"openstack%2Fswift~master~I9b5cc6d5fb69a2957b8c4846ce1feed8c115e6b6","project":"openstack/swift","branch":"master","topic":"s3api-429","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-01-25 21:00:56.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-01-25 19:28:41.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2024-01-25 21:00:56.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-01-25 21:00:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9b5cc6d5fb69a2957b8c4846ce1feed8c115e6b6","subject":"Support swift.proxy_logging_status in request env","status":"MERGED","created":"2021-04-05 18:00:34.000000000","updated":"2024-01-25 21:03:07.000000000","submitted":"2024-01-25 21:00:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":67,"unresolved_comment_count":10,"has_review_started":true,"submission_id":"784808-s3api-429","meta_rev_id":"7bd13f5cf055d64b61abbe5d7ae31e1b4c873786","_number":784808,"virtual_id_number":784808,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-25 21:00:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2024-01-25 19:28:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2024-01-25 19:28:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-05 19:18:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-06 19:16:20.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2023-03-15 18:04:00.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-15 19:12:54.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"e0a1c976249c4f08534ff40f7a474d8da7df7554","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-04-05 18:00:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6634dab9da73f375365316903103288126813935","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-04-05 18:01:23.000000000","message":"Patch Set 1:\n\nLooks like a 429 resp to logging (but notes the 503 was sent to the client as \"wire_status_int\" on the end)\n\n    Apr  5 17:58:03 saio proxy-server: 127.0.0.1 127.0.0.1 05/Apr/2021/17/58/03 GET /foo%3Fencoding-type%3Durl HTTP/1.0 429 - aws-cli/1.19.13%20Python/2.7.18%20Linux/5.4.0-67-generic%20botocore/1.20.13 - - 183 - tx9937e976237641ca9688e-00606b4fab - 0.0241 - - 1617645483.195113897 1617645483.219254017 - 503","accounts_in_message":[],"_revision_number":1},{"id":"6a9453950155fd14079673d8517c4139d4884790","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 19:18:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/627f1f4529c64d578770c1184637af5f : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/457a0ca624ab49b7b169cbc5dc86f799 : SUCCESS in 1h 12m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3973259ac9d64e778f1c3d45aa21c98f : SUCCESS in 51m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/40304f9b74f341388e186bb839dcdc8d : SUCCESS in 8m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/edeb5ff103bc4013bce26ebef869fa85 : SUCCESS in 5m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/640d934b7cd447a4b29b6aeb5d68eb8a : FAILURE in 3m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c0b3ca11a0c5477aa67d9fca85bb1087 : SUCCESS in 15m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7a9fc60159c646c5b47c38fbd72adccf : SUCCESS in 14m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/94be2658627e4202985cdc2399932a39 : SUCCESS in 18m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/97aae997dfdb41d58e0e14bf8748615c : SUCCESS in 29m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d1f06329ca544bd2868b3e378cf0c2ad : FAILURE in 3m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1705ac35fb8649b8a284b04715f9d2b0 : FAILURE in 4m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/347a76f06cc9452f861a89f9e19d02b0 : FAILURE in 3m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7252833ff4864bac9d2b1612d1a24567 : SUCCESS in 23m 40s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/eff000cebb32426b9b123478a059ef15 : SUCCESS in 32m 48s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/39a28daf09344eeb886402b3d1ef8de3 : SUCCESS in 18m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1d823e06e5d244179675df9760367bc8 : RETRY_LIMIT in 3m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5eb05cafee4b4829a552220e61cdaea7 : RETRY_LIMIT in 3m 36s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/efd3473fe2774c65ad7a90c3f39a23f1 : RETRY_LIMIT in 3m 27s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/040377e66ce04dff9d5e1dca08fec342 : SUCCESS in 50m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ccaf7ce3d7c146beba8729c1d6e10e57 : SUCCESS in 39m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3d1ab1a43ef44dd7b19bcf201513dd69 : SUCCESS in 33m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9b303a6db3574c39a7012bce68aea620 : FAILURE in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f6d8213df7348fd8a1eba266678244c : SUCCESS in 6m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e76b7821507d4e009bb05b6339a6abd2 : SUCCESS in 23m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/633f473630f74a5b80a096b018e37dad : SUCCESS in 56m 10s","accounts_in_message":[],"_revision_number":1},{"id":"465c08b5df2858bcdb8e6b3a9005103d4ab0cb6a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-06 19:16:20.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nI\u0027m a little worried about the upgrade impact for ops that haven\u0027t added wire_status_int to their logging...","accounts_in_message":[],"_revision_number":1},{"id":"9e1426f6b9ce4385135dfeb09d1e4221a772046e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-13 21:25:09.000000000","message":"Abandoned\n\nI think this was obviated by https://review.opendev.org/c/openstack/swift/+/729093/ and https://review.opendev.org/c/openstack/swift/+/755639","accounts_in_message":[],"_revision_number":1},{"id":"2ba98878a30d45756e8109889a685cca721e07cf","tag":"autogenerated:gerrit:restore","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-03-15 18:03:38.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"5f5d87a22b0906b768585e7406749b0f1aef13ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 18:03:53.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 784808,1","accounts_in_message":[],"_revision_number":1},{"id":"c372c6b38daf743f90e202ec85103f62e04d1326","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 18:03:53.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 784808,1","accounts_in_message":[],"_revision_number":1},{"id":"09748be83bf326df749b390a2fa70cb7c833d8a2","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-03-15 18:04:00.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/swift for 784808,1","accounts_in_message":[],"_revision_number":1},{"id":"201eb3d1075e9c422246ee0e5281c6fdfda5a1fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-03-15 19:09:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"496d9334788b5db7ecbdff5108617ab95ee0fe03","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-03-15 19:12:54.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cc0036b62facf3986f9a196ec11377bfd3aeb628","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 19:29:34.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64b7ba8144a94fc4bac78bac91209a85\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f83a81ec4bd44478cef572ddebc4ea6 : FAILURE in 13m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c943b06ae7144c48893f632a88c590d9 : FAILURE in 17m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3babafbf7cae4369bbf6573c33c1b261 : SUCCESS in 17m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/470ae78290db48c79b0667e7eac337f3 : SUCCESS in 18m 53s","accounts_in_message":[],"_revision_number":2},{"id":"f7372b050cf64c44fc25e5bd22f83de7df18fbaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-03-15 19:38:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3f7a5d91974b8122fb8fad338959afa6d4e50a60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 20:00:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96adeb99939442bfb3b9572b1b5152ec\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7bf8cb5c28be49e88ea6c9cf7ecc9c13 : SUCCESS in 15m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c402c13cebd34429918f122d37972da6 : SUCCESS in 14m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/34af562033774cbdb4bee84444aa1617 : SUCCESS in 20m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ddb136c67a9b48e9ad78bcb3f2dd2ede : SUCCESS in 17m 33s","accounts_in_message":[],"_revision_number":3},{"id":"760c6237b49be685abb16ca06f540cacd337c15b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 21:09:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a91d06f8b4b0472e9381d6582ee719e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9605438a5ced4e4381aee08a434d5c29 : SUCCESS in 8m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/68f265c73f4347b2b636956b4aa34f29 : SUCCESS in 1h 13m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fab65f741e1a4040928dd87c197e3653 : SUCCESS in 46m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb3b2972466a48f086dcbfd0ba4066d8 : SUCCESS in 1h 02m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/080a062f53dc4877a46c968620d3a60e : SUCCESS in 17m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d522c030ff3474ca90528a1430aa5c8 : SUCCESS in 19m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b201c0dd8324822be02a52fbdd9bcc4 : SUCCESS in 17m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/602315530f9e40eaa99e86da392c01f9 : SUCCESS in 8m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c22477cf54fb4c37bba2a392d4e2ea9a : SUCCESS in 6m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e205175c32084314b57d641df3a1ae34 : SUCCESS in 14m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4bf8c5d4082e46b1aedc35fdcc538085 : SUCCESS in 18m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e252b375788047a6a3fc1506ecee2d67 : SUCCESS in 16m 37s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/1156a38257764f0496d861cc32a9caad : SUCCESS in 15m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/84929ad714124d8681b6ed282f1c6bc0 : SUCCESS in 15m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e94cac92c13e4765b4802b3cd67ce442 : SUCCESS in 18m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/07ce15f44baa4426843466f2d93bc0ac : SUCCESS in 18m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d9fffa54d15d48bd8f91387441089b68 : SUCCESS in 14m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9669a03267ee475c8a26f6a57587c84f : SUCCESS in 15m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c912b97310a54a1995ddb5bfffaf0778 : SUCCESS in 16m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ec06ed90ec0c454ab10c5790a594276e : SUCCESS in 6m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/10716e3cf048455481a2f94e29a4b57e : SUCCESS in 19m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f41500ca644f44eaaded1267b047350d : SUCCESS in 7m 04s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3dfa5ef62d6c4157ac65d0c9ad83db49 : SUCCESS in 1h 01m 37s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/46e15c38023d411993ccbe395e803e9a : SUCCESS in 1h 11m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/44f57c4747934181b6d9083b12bab7f5 : SUCCESS in 57m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/99ff23280fb74c89938b37cfb6e7ff36 : SUCCESS in 38m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e1430623989469a84636e91baa848aa : SUCCESS in 17m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c48769b7ba043c7a459576f2db88dbf : SUCCESS in 5m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c676cf555c4e4d5bb462a13b1d546970 : SUCCESS in 36m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bae3f567f59247328e7befb03e2f9bb8 : SUCCESS in 56m 46s","accounts_in_message":[],"_revision_number":3},{"id":"050f705d94162bc34801f8533cb6058ec4df5152","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-03-20 15:35:40.000000000","message":"Patch Set 3: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9e3edfe44f0fcb17a2f7aeb1f5d172af0eec1904","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-03-22 17:30:50.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c18e28a60320ca9ce90278c69fda9e624b86b017","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-11 11:55:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"848fb00371bf438ca5aca940b570af709e597512","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-11 17:01:03.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"8faf5ef98048e927cd2022dc30de68247e947654","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 17:22:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e3384c495ec4b81b343de745b7981fd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/90983c94675e42feabf65fc7fcece016 : SUCCESS in 15m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/517341860a4048d1995d49b4912b949f : SUCCESS in 15m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/33df4425492a42f09d19e35b2d2ea41c : SUCCESS in 20m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7930d6f054354911bd376a57cf303106 : SUCCESS in 17m 17s","accounts_in_message":[],"_revision_number":4},{"id":"65f71080b8af8c7eee711d8e260f06e42c3cf466","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 18:25:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10b640f223c74069a3c81cd1323231c2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a5edaa4ebd941da9d8d36e4c930f1da : SUCCESS in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/10cc65ebc3174f809c46537cf3c1b9fb : SUCCESS in 56m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9456182a174643a6be50f3c08e1656ec : SUCCESS in 37m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7312bff805894cb19a1374ad895fcf60 : SUCCESS in 49m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d0e34153e0624709bdd4e929379b42d6 : SUCCESS in 17m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e2ec5d77971945db8184a913d1aa9774 : SUCCESS in 17m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/093546d4dd1544e19127f589c7a66c47 : SUCCESS in 18m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bf8a9ed427914fa193ea252b1ec787f6 : SUCCESS in 10m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7f6ce5f10c0d4c069a993f7a1011cbf8 : SUCCESS in 7m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/64f13c83cc9d4c0488e312ee72a7dde4 : SUCCESS in 21m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1eaef5cce5da4b7cbaaa6cba33bf7da0 : SUCCESS in 19m 48s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8922d87de3b34db6874eee2166e4ce78 : SUCCESS in 18m 59s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/af6f48c0f4ea48fdb0b5402b11423f00 : SUCCESS in 18m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/59f32ce20eb64634ab8f4ccb08e9feff : SUCCESS in 18m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/94c8e738856c4faab71e9e87dab9c4d0 : SUCCESS in 20m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/513394c2f6ab45b585b4671e650e36d2 : SUCCESS in 21m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9bf7f6fef1ed4ec49772f1f8850d9426 : SUCCESS in 17m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6c7d83b1f8114f9f93cd1bc03c07f05d : SUCCESS in 15m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/156866fdcd1d496283beae7fc1f20b46 : SUCCESS in 20m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/92dcc9d5f0be4bd19de8c96625560bba : SUCCESS in 7m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b3a259f4c6bd44fd93e2b809d66f3439 : SUCCESS in 20m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e2ba8e866ec4496fbb79d5ff541445f9 : SUCCESS in 7m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/451859872c444fb2b7533eb96bae4b0c : SUCCESS in 51m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0c94f0d11bd475380d16a3a9a5723c2 : SUCCESS in 1h 04m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/85bed0c654664b5cbee3bd29931824d7 : SUCCESS in 47m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/698836461a274fdab694809c2357562c : SUCCESS in 41m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d6ad705ecdd4b45944d1d29d30cbd08 : SUCCESS in 18m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3b1aeb6450149a68a2d86f53d2cece4 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a5f9df0bf12d446fb161f2d5eb802ac7 : SUCCESS in 23m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b323978722754f339e5d9d270d448315 : SUCCESS in 40m 15s","accounts_in_message":[],"_revision_number":4},{"id":"e200f3553148affe8442c7136b74bb40d0ab09e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-17 16:04:02.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"bf8f9b4867a8226fcb1940171ae1987d0116dde8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-05-17 16:06:34.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"1b8b96de0c5bb4b5dd78e3813ae849db7d1fbf21","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 17:12:48.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b17380aac1cf477ebc26facedecf66af\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c50d605e6a924f4897297ef6337f24ea : SUCCESS in 15m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3edfa8145f8140db97074b05b62db42b : SUCCESS in 1h 06m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ce81a60de0242739be60197ecf8e48e : SUCCESS in 18m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/17c9416bd4ed43e3b58008bcb019e5ed : SUCCESS in 18m 00s","accounts_in_message":[],"_revision_number":5},{"id":"b4206ebe8a6c35257e1b98f26f3b7bc513bf06a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 17:28:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/240694b6504f48aebe104df54e75f931\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a88c4c16f864ef1b71bf4d943dda530 : SUCCESS in 7m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/2500349b8bbf468fbc73f54378d8e9a3 : SUCCESS in 42m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ce8480ce379b451db06ce361841814a3 : SUCCESS in 47m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/945f21639ca142a192604c40d552826d : SUCCESS in 51m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83ab28ed717846329d2746c3164757d0 : SUCCESS in 22m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2910e1daaf0e485ab8ca1ebd9a151296 : SUCCESS in 18m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/22f830d2cda94354833f4cc580dd6fc9 : SUCCESS in 22m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/47ece82fd35d471cb65460b58f61c666 : SUCCESS in 9m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/33faf650cfe6497bbb3fd5ae958cc5c5 : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/06cd4cb446d04582b68836fc1734d022 : SUCCESS in 19m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e084e4a0bc06420ca218aee75a15a256 : SUCCESS in 13m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d8065171e66a42699a9fb652882bf184 : SUCCESS in 24m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c4ba3d040e2b4148a8a3b394dcf1a9a2 : SUCCESS in 15m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7548bebd33af4696914aa3bd2f32f2db : SUCCESS in 18m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f12dc1dce9724898a1d64d7e766212b2 : SUCCESS in 22m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/88a41b8e6d234921abfe20cdc95a0d24 : SUCCESS in 17m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/94bddb8049d74dc4bbc829af4a09c798 : SUCCESS in 15m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b42dc400a3c94764a64932ed1decec48 : SUCCESS in 13m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ab2d53e4ba5d4bef83f27c8a7d4d9fc6 : SUCCESS in 16m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/53ba4087c5f84a13af5813ee7c1b5b52 : SUCCESS in 6m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/43070c48bf314715b96ee3e21e1d8d43 : FAILURE in 15m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/13f065935ca54046aaf2e30348bd08ad : SUCCESS in 7m 59s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e13cc56a99964ab19e55d1762784b441 : SUCCESS in 1h 00m 13s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/817747d817ae496f800546a927c1130c : SUCCESS in 1h 04m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5076ccdf3ba0462bb1b95decb05818ad : SUCCESS in 43m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0635b13f3bec4115bb1035bb445fe6b0 : SUCCESS in 51m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a35add086d7945c8a35c9bb18f2aab26 : SUCCESS in 25m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a0a0df099b647679989ccd3690ee05a : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/881748f17f604c1e8f56eaf1eb788b36 : SUCCESS in 18m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86142a593ba64a3aabc6cbc74ffad5d9 : SUCCESS in 40m 39s","accounts_in_message":[],"_revision_number":5},{"id":"aa9f6bb082d9078511a15b9ca2fd1cd98b514de5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-24 18:56:21.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"d2f8d68e565a0f09b2287992d5a04692b660d414","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-06-28 12:35:02.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"ab0a10d70d804bb843227d8bcd559e1770ef91bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 14:08:49.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1883236e156b4edf9888d154e35f77fd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a43cd891c21240988ce946ecd9279771 : SUCCESS in 15m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7c9bf3d3b11346ee9a79dc4a81599b16 : SUCCESS in 17m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4bf93d75becf4df8af7e9420dbd98800 : SUCCESS in 25m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c319a4b1f76431cad3a653ea950dbec : TIMED_OUT in 1h 31m 06s","accounts_in_message":[],"_revision_number":6},{"id":"d3c525e89eec7e8bffc0061615a43d416fa1324b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 14:57:14.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3770730014b94fa38c97b1cbe3745286\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11aca40009274bbbb58ca439ea70380c : SUCCESS in 8m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/e02b94b615334092b426269b21284326 : SUCCESS in 1h 00m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3462d487c8de4ceb9fcb0182384d6b57 : SUCCESS in 1h 00m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/30cab2b6ee604644bdf7077491303380 : SUCCESS in 1h 11m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05cdf30c846f4ca9bc1f031a8d800f57 : SUCCESS in 17m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/031e80b3e8bb4ed2bca8a45896d0d0ba : SUCCESS in 15m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b370ef7373a24f818c0acebb78fcb5d0 : SUCCESS in 17m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a71db48e6b9e47eb883c7a7d13086af1 : SUCCESS in 8m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ea8d0464140e409494ab4d47d229dacf : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b1e73d98df124597b6ffa04371af7ea6 : SUCCESS in 12m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/36da57f2953d45ff91cda2423e34676c : SUCCESS in 15m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/72b5c2940f244617bd2fdf1d97e3b817 : SUCCESS in 14m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3cc487461748497882c1a9b2b1a12eec : SUCCESS in 14m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/879fa2155fce4078a74c6a315c36e9b3 : SUCCESS in 12m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/210cf42754e348dca3ff98c50c6a7126 : SUCCESS in 15m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/df60dbc9dca940d9a201a195a965ef17 : SUCCESS in 15m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ecad7a2625c747b29e35bd070f461245 : SUCCESS in 13m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6ec8dd6c18d4443a9417859daab2dbb2 : SUCCESS in 13m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0e14e70274e249b58d3c4ad235bd91ce : SUCCESS in 15m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/98f3144aa89345978bdd3a63481ffd8f : SUCCESS in 5m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e937e7063af944f3b7079490b9e0ee6e : SUCCESS in 12m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5d32044921c64eaea46d67f24a0aa918 : SUCCESS in 6m 23s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1241bfdb473c4692ba6b95812c3b3994 : SUCCESS in 59m 27s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/198fea5d339f4f8394d5cc8610615073 : SUCCESS in 48m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bd693e29bb384a8ba3bdaee5364f80c0 : SUCCESS in 49m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/18de94fe594b4e61b3383b7b4aceaa04 : TIMED_OUT in 1h 34m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13272416b416497c94f308e8408c7c12 : SUCCESS in 17m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/733d156ab56544e8adb64ec268625f2d : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d434eae3d3e74785a6e2c0ab1a503e1d : SUCCESS in 25m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c50495fa1b5f431fbf0f1c22a28c7511 : SUCCESS in 1h 05m 03s","accounts_in_message":[],"_revision_number":6},{"id":"a63f284990f708b8587fe9705e0c3d2b203d9c58","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-28 21:31:56.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"bdce92731ffdf3a414d099918e5fed4d6f8dbaf0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 22:07:32.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bafd0a1323f14c4d8db7712785145b54\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c57e5cd291f5493993cb14e6ce9ab676 : NODE_FAILURE Node request 300-0021569468 failed in 0s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd99561c251d4743a9e3a4c7fc4de4ec : SUCCESS in 28m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b30ff7ba13f44268bc3766f0318fadf : NODE_FAILURE Node request 300-0021569470 failed in 0s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b73b76007a5a4ab4aa35d9da8ec84a91 : SUCCESS in 31m 29s","accounts_in_message":[],"_revision_number":6},{"id":"a3f84bbd993a6380a960def15503df9058a4262c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 22:42:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f0071be28e84f86b9eb44f3d418e4fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/316941824c5d477582174d46b7481ba3 : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/fcf70c7665fe47b39f4444f77e27196e : SUCCESS in 1h 01m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d692b4055d5b4d0f82e92c5d7dc1e15f : SUCCESS in 52m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/70173d3fb8204dc08b9d50ae823a5a31 : SUCCESS in 50m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51d65242f45748a7b3375506f46612a0 : SUCCESS in 15m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/268e24de50b64196879ae773286e56a4 : SUCCESS in 17m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/850860b5912c49c991bf6418450b3533 : SUCCESS in 19m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8443492ecb79426a9866828c9324ac73 : SUCCESS in 9m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a9222171d56e4461bbe94ea70acf98b6 : SUCCESS in 6m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/68bcca5df90944a2ae7aeba5f8a2356e : SUCCESS in 16m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1248d838493d4ec2999983a6334293f0 : SUCCESS in 18m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9e5970eace2f45a98a210f8a763d5fca : SUCCESS in 17m 43s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5f6c8fe1baf54700a2d96483b2cf1a30 : SUCCESS in 15m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3e6ba3c77d9147dea896ce24df4a071c : SUCCESS in 19m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6f06203d07d348e09ee7e3edb2bc769a : SUCCESS in 16m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e827f04327ed47bcaa5087b068110998 : SUCCESS in 18m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/dd432bfe09a544c4a22141b5b9578400 : SUCCESS in 15m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ea283e112ed3427e93779b8ad99962ac : SUCCESS in 15m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1ded6cb1b5ea433cbfcaea5522ac6320 : SUCCESS in 18m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/064fc60f9833446a9c1822ad2ca5fcd4 : SUCCESS in 6m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e7b8ad0c4d1143fba86265ace30c5f4c : SUCCESS in 42m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fe3f5e718646483fb7a4453871b69adc : SUCCESS in 7m 16s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1ec0e06186584ca3a0cee5a2ee865a7b : SUCCESS in 55m 44s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7a5a2b76c864b8aaba71f09bf3c3439 : SUCCESS in 1h 00m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5d522442af724da48e19250b497c0190 : SUCCESS in 47m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3d625001d2674f11861e63026a218dc7 : SUCCESS in 52m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39153158b50848668cf8e62a498e1f56 : SUCCESS in 16m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65ad1eb560d54f1aa45531aa2304e5e9 : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/39c8c027a64a481a899fd35e8abd6a16 : SUCCESS in 18m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84a71aa81b744aa0a42187919ba2ef26 : SUCCESS in 1h 01m 30s","accounts_in_message":[],"_revision_number":6},{"id":"f63f04a6a9e3d7ddb8252f04f75679f2ac85cc48","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-10-27 03:30:44.000000000","message":"Patch Set 6:\n\n(11 comments)","accounts_in_message":[],"_revision_number":6},{"id":"bf041d173a3d263aad4ef3a4b227ade1932ccd3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-10-27 13:05:39.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"3905a34d2969d1b4b5a0fd8528021336302eda57","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2023-10-27 13:10:42.000000000","message":"Patch Set 7:\n\n(11 comments)","accounts_in_message":[],"_revision_number":7},{"id":"45858d15c1586a5a1831e8d9d6a747f7dcd6f618","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 14:16:57.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e95c7f435c64cd894c4d97ba4ea8b7b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ad2eb9f8d46c4891b8d284a3487661eb : SUCCESS in 17m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/96a6050afb124ffc87e4aadf25cbadc0 : SUCCESS in 1h 08m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d59ee3b767f743ea87aa056df4e8fcfd : SUCCESS in 1h 08m 35s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2b7d60c68be4e3885cf1834593d5d97 : SUCCESS in 1h 04m 50s","accounts_in_message":[],"_revision_number":7},{"id":"77dd4fcbe8e53f1e3d636b8ebda85043c4a24a5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 14:32:40.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73dd937d82fd4ea08cddbba63fb1b613\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d7f73f3ff8d414caf5ae982ac3941f4 : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/89f1ba0fa63f44619d1d436b23aff481 : SUCCESS in 54m 49s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/edfa4a71f01f41968eea35a14d3643e1 : SUCCESS in 51m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/85f350af00c8439082ec8b438f1c09d5 : SUCCESS in 39m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01000904993b45c0bd952ad579ab0952 : SUCCESS in 1h 00m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ad1c62a814a43398ffe6398d3a09a20 : SUCCESS in 18m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/53894c31b7b841149499e4572400aac5 : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d9de1cf122b405da5c2e82e1ea6f8b3 : SUCCESS in 20m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d2329ada59d34e618568f974d57ace9e : SUCCESS in 8m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3eb4d64beecc40b2a6208677c73a9a24 : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6bcd990b9f3d4e548a6e463294a2f646 : SUCCESS in 15m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1ea42c1ea72a4563ad28d22e8944e524 : SUCCESS in 14m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2c6077dd2cdb4009a992b732e64a974a : SUCCESS in 16m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/de0b4841e32b4f72934bd174e878b377 : SUCCESS in 14m 56s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d97ad53b89204edda7b3485db4c0ea9d : SUCCESS in 15m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1596dcd616e048e9b8b2f848f9a9e7f6 : SUCCESS in 29m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/22e49f4e29e84ad19f362ede4ae101bd : SUCCESS in 32m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6f83631ce7284d298f14c958cefcf5a7 : SUCCESS in 25m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1cb1f910c13b435da6cc8fc296603db1 : SUCCESS in 16m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/314b9c0e35ef4bf9a2bbc449866a674a : SUCCESS in 16m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/877f28ef91924fedb14a2415dd451ae4 : SUCCESS in 17m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e7f9ea8f76aa4a57b4c5006bec22f1c2 : SUCCESS in 7m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bba70fc3e02f45c49484a35f33d7d469 : SUCCESS in 16m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f0d691b6a785486ab8c8611556535aab : SUCCESS in 7m 32s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/75100a6e73b640aeba57d5da77b5b715 : SUCCESS in 54m 12s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/53bcc03c9043436bb73837f8b74685ea : SUCCESS in 1h 20m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f09e8b3bf4a84646bb96498fa02ed992 : SUCCESS in 46m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8c3df5c7c27f42e79dc361c53b712e02 : SUCCESS in 47m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c26ef32b62a44d0aaf773611e2ad3120 : SUCCESS in 17m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d903f140e2fb4d769a0158127277ea54 : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d73657cba0a049d8832c93ffc6d8a451 : SUCCESS in 26m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d14e61b90ae4a6482eae894198fde53 : SUCCESS in 53m 40s","accounts_in_message":[],"_revision_number":7},{"id":"ab1b5781ca4ab4cded523e5a39e285ee19c03415","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-11-06 17:44:48.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"5b60c2319768f3e263bfb9f150482c610e4021b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 18:45:01.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e4184b54c4d4296800602920ebd6991\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/878fa41fee5840269d81025596fad0ed : SUCCESS in 14m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/995f2a2a997f4baabd4e7f3c9bb50d67 : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/296a0fd36bd74c9d989bad98e7121344 : SUCCESS in 58m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0099bd080b8246869347f42ffbe66d56 : SUCCESS in 22m 30s","accounts_in_message":[],"_revision_number":8},{"id":"50c76eb0cf98a2b8be2bd52013381e890fc5a530","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 19:03:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f18ac742cb4341aa89ae49cf92651d8a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eb3efae7faf4af79c0fa824b9f955be : SUCCESS in 8m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/41135f5533394f888377b4358d385ef1 : SUCCESS in 42m 41s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c71b8269fca143d7ab59139536ede72c : SUCCESS in 43m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/defb03de6efc429f8e7815657cf3b6a1 : SUCCESS in 41m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74b36a37b4eb40bdbb8457cb44cefdf2 : SUCCESS in 1h 11m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9b1f916836a431ba59972392c3e02ab : SUCCESS in 22m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a4a6826f93c4436f951f411699ff0b2c : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b8337826b11540e2adbae172be72a071 : SUCCESS in 23m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7e99b814008e4454b2e611151f9b0036 : SUCCESS in 10m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/20e5cc7471154ceb9682f04c06da4625 : SUCCESS in 5m 47s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/377b23d5a222445cb45dd6c7f9e1a96f : SUCCESS in 21m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/da381a422f39476ebda86f7558c2f8a2 : SUCCESS in 19m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cbb7643571a344b3aaebb986bdf447b5 : SUCCESS in 17m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d68a2c463be44b9aa8ef03b39396fdf0 : SUCCESS in 17m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8ae078bd5c9043939b79619fab9cc30e : SUCCESS in 19m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/62aa289413ce4400b238e06fe6c3a733 : SUCCESS in 17m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/74c216464ca840dc8eecbe394f4dae19 : SUCCESS in 19m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5561a13114e642dead6721f1f3ea793f : SUCCESS in 19m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/95f2cf4a985340d3b236b0d0c430bfa5 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/695c262a29e848e497d91b0b5d0a820c : SUCCESS in 16m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/73784cbd005a47a2be280c50d6fb50d2 : SUCCESS in 14m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fda6deed59f6474598869a657af3b421 : SUCCESS in 6m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/234147e464904a75919416051134c6e6 : SUCCESS in 14m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4fbb012a2e1d49c3ad213af952305630 : SUCCESS in 7m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/294d5a4f3c0f4f93bd1d21ddbf76df15 : SUCCESS in 53m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2f9ac4c656fa4fb3938fdf3e3932db75 : SUCCESS in 59m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e61b421f59ef4054b15d2abb46972d3a : SUCCESS in 44m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4ecc165bc9dc4ebd8417a1ab3a8dc07b : SUCCESS in 58m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bec514985bda4367b1343180bf486f76 : SUCCESS in 21m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f43d21b21a3a4f909df7cab7652dc2cf : SUCCESS in 4m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9b976e3a28f944f2b870c87dd9eedaa1 : SUCCESS in 21m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad6a214646ec4ad4aa2c4aff5eb6a67d : SUCCESS in 45m 39s","accounts_in_message":[],"_revision_number":8},{"id":"c8db258ec0db5e45a4d04b4582e5612369748007","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2023-12-20 17:31:06.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"7063b01b677291ac062fcb9d73d2a8c9239e3a5c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 19:15:32.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d9a124664704ce090410ab714c665e3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce3cd4825ba84234a70997ab8d15163f : SUCCESS in 19m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e598a790108745daa778294f169d6ae1 : SUCCESS in 21m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/fdff3d10ca2647bc85471518d51a61ca : SUCCESS in 51m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/554c18d39a6a4b5a87e24573b67a7b9d : SUCCESS in 22m 19s","accounts_in_message":[],"_revision_number":9},{"id":"bbbae981d53e50a38182231d6f38fe437b4d6e1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 19:39:40.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a005b9870be34c2aaf7e7308d289fb6e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbc0fdc1a1a443ca908d1e8bf84af91d : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/8c9e3db4d5094c7d99a92fac9264f913 : SUCCESS in 59m 15s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b6d7bacae7cc492c980b5cd9d4d511c1 : SUCCESS in 1h 09m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e0f9d03185d642538ba4dc05058e926f : SUCCESS in 50m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/43eaee4f0f2a46b8bf489a020dced6c7 : SUCCESS in 1h 10m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/443488bd6f2446b297936d00cff05e8e : SUCCESS in 19m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dbfcb6da16504c98a5003257713bcabf : SUCCESS in 20m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6bbf60bb0c314fd087267988303c7abc : SUCCESS in 21m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ec0ac1c43e3345b1a09cd8ccd4363e17 : SUCCESS in 8m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e82c0d07f8ce4266818018aa9535e3c4 : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/aa82f439ca29417d8f2c297747fc71f5 : SUCCESS in 17m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b52d45748d2e4b3daaf2b16ca9764af5 : SUCCESS in 16m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b860992dd7c6429b958c2aef985a1672 : SUCCESS in 15m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/10ed44c662e242d38761cfa0d92f0050 : SUCCESS in 16m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b21aa7dac89a48659915edb9773d1a87 : SUCCESS in 17m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f29f13af718a48bf912ef8c936c7a4d0 : SUCCESS in 17m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2de1d8d44f1848b3937fdef1c77b4e8f : SUCCESS in 18m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2966546fc28b49faab3f206e5734df99 : SUCCESS in 17m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7b0f3a4d21f4401e9c06486325610198 : SUCCESS in 15m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/96e31c8fb74a4a5ebcb32836eb050a42 : SUCCESS in 16m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d518fe722e1f4ca69ea58c4c62fbce89 : SUCCESS in 17m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/da61549b51ff4099b4d860421e3f7762 : SUCCESS in 7m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a1fba8a282d948a68b0474c9e894faab : SUCCESS in 16m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/96678237880041499e754f9ad11ae5ab : SUCCESS in 8m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/adfba7096aa24a6f898514f10015d3ae : SUCCESS in 1h 09m 39s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e4207d62d8014ffc9b12ac32046cb3fb : SUCCESS in 1h 59m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9221cf4faa37440396301515d73aafee : SUCCESS in 57m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eb45dab3780e4ff8a40521f76232c313 : SUCCESS in 1h 03m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afc0c0f96eab410793e54c48a29c6c8f : SUCCESS in 20m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0a45b17ee694fd0b889a4b76baad711 : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/323ce92008ef414d8f9321a8371dbc6c : SUCCESS in 21m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca2b58a22c95413b90884be789a7844c : SUCCESS in 1h 03m 21s","accounts_in_message":[],"_revision_number":9},{"id":"d4fb411477e58783a4252f73ef4ece530c9272a3","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-01-25 19:28:41.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(22 comments)","accounts_in_message":[],"_revision_number":9},{"id":"4edc9955b25ca74747969d7b955327b2369f3cad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 19:29:16.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"c4a30de1e70ede249c52f9cc9480c855c70d732d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-01-25 19:31:29.000000000","message":"Patch Set 9:\n\n(8 comments)","accounts_in_message":[],"_revision_number":9},{"id":"3c8edfa66fc575343c0942c1a7300b477a055be4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 21:00:56.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb5456d2cd494ee5a47b80868efaa0df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02344642ff034a99b99cbde8fac668af : SUCCESS in 10m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/70011fc4aaf9477faf04c41daa3770db : SUCCESS in 56m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c57deaa7056c4504b0f0be5403f8645b : SUCCESS in 53m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/95cd28d6b8294925a20c2783885e888c : SUCCESS in 30m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79d7ee57fafa426b89a324e2339d122f : SUCCESS in 24m 51s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c116db1215ff497ebcde88733cfbe193 : SUCCESS in 12m 27s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/888ae2cebbb14f2abb02c4c088db010e : SUCCESS in 6m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/27486a4bab604f57935892ab267bf891 : SUCCESS in 17m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/93775d0e6e42418cac931ff4b81ab627 : SUCCESS in 16m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d5b999d7a0e24123bc8e4cf389913c50 : SUCCESS in 15m 28s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/865b2026b3df4ebb940aaeac3f9ee496 : SUCCESS in 17m 08s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/973eae848d364e04ac7ab0ea72a0c025 : SUCCESS in 17m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1ebfa713627549e39641775595d385f5 : SUCCESS in 17m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/33ba7a91869e4bccbb67d7fd3e26aa01 : SUCCESS in 19m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0b3ae4f96d2440b9bf9882d84272bf50 : SUCCESS in 19m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7906aa7b0da04e12832c21767c3fe663 : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/27a07959d0dd4e94af54b7e1a5f9fb5f : SUCCESS in 15m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/222bd531774d40659300c9b9637500fb : SUCCESS in 16m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5bb5139af6c7459d969daddd4f9e9e1d : SUCCESS in 7m 59s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/31eef871474d4e3eaec30627e3396bd4 : SUCCESS in 7m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ce56bf8c263b46c798bb8eb36070be72 : SUCCESS in 54m 19s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d8af97bfaf2344768c2b9b2ca251e127 : SUCCESS in 1h 07m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2e727862d0c14cd1a0114171a05e63f5 : SUCCESS in 45m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3657c519d31549e6a2515be85a1a9237 : SUCCESS in 54m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f372df093384b0984c28c9d2bbe4ea2 : SUCCESS in 16m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/252ea73c81644878983dd88402131bff : SUCCESS in 5m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3703c2d1106448f97a443e77e81299d : SUCCESS in 50m 27s","accounts_in_message":[],"_revision_number":9},{"id":"1891207f1d49d056a0164c0cca5cdcdbde327afe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 21:00:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"7bd13f5cf055d64b61abbe5d7ae31e1b4c873786","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 21:03:07.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76b7df4235a84baea7af77d4c72e105c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7e851b71b2846d5bf2e9591c8cf03fb : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2cd28fdf08994639a376d4a901c48f36 : SUCCESS in 35s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/c133242630f7461fa3fc616eefdee6dc : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"5af7719ef346e1594263536ccd9a952817cb2f7c","revisions":{"261659aa85e8847251287a2dce8dd9fd31992230":{"kind":"REWORK","_number":1,"created":"2021-04-05 18:00:34.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/08/784808/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/784808/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/784808/1"}}},"commit":{"parents":[{"commit":"b79ca57fa98323b17f896af86c617d465066d544","subject":"Merge \"Create a separate doc for audit watcher\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b79ca57fa98323b17f896af86c617d465066d544"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-04-05 18:00:24.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-04-05 18:00:24.000000000","tz":-300},"subject":"log s3api rate-limit resp as 429","message":"log s3api rate-limit resp as 429\n\nChange-Id: I9b5cc6d5fb69a2957b8c4846ce1feed8c115e6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/261659aa85e8847251287a2dce8dd9fd31992230"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/261659aa85e8847251287a2dce8dd9fd31992230"}]},"branch":"refs/heads/master"},"584926d0e924557ccb1d9164c62a51e9e20f9d81":{"kind":"REWORK","_number":2,"created":"2023-03-15 19:09:06.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/08/784808/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/784808/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/784808/2"}}},"commit":{"parents":[{"commit":"c0483c5b940d559b4a9efd2b5c64bf5b5528fa11","subject":"Merge \"sharder: make misplaced objects lookup faster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0483c5b940d559b4a9efd2b5c64bf5b5528fa11"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-04-05 18:00:24.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-03-15 19:07:10.000000000","tz":0},"subject":"Support swift.proxy_logging_status in request env","message":"Support swift.proxy_logging_status in request env\n\nWhen logging a request, if the request environ has a\nswift.proxy_logging_status item then use its value for the log\nmessage status int.\n\nThe swift.proxy_logging_status hint may be used by other middlewares\nwhen the desired logged status is different from the wire_status_int.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I9b5cc6d5fb69a2957b8c4846ce1feed8c115e6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/584926d0e924557ccb1d9164c62a51e9e20f9d81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/584926d0e924557ccb1d9164c62a51e9e20f9d81"}]},"branch":"refs/heads/master"},"ac4b91b79766d5a9c3eef1331045e481aaf0d95b":{"kind":"REWORK","_number":3,"created":"2023-03-15 19:38:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/08/784808/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/784808/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/784808/3"}}},"commit":{"parents":[{"commit":"c0483c5b940d559b4a9efd2b5c64bf5b5528fa11","subject":"Merge \"sharder: make misplaced objects lookup faster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0483c5b940d559b4a9efd2b5c64bf5b5528fa11"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-04-05 18:00:24.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-03-15 19:38:42.000000000","tz":0},"subject":"Support swift.proxy_logging_status in request env","message":"Support swift.proxy_logging_status in request env\n\nWhen logging a request, if the request environ has a\nswift.proxy_logging_status item then use its value for the log\nmessage status int.\n\nThe swift.proxy_logging_status hint may be used by other middlewares\nwhen the desired logged status is different from the wire_status_int.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I9b5cc6d5fb69a2957b8c4846ce1feed8c115e6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac4b91b79766d5a9c3eef1331045e481aaf0d95b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac4b91b79766d5a9c3eef1331045e481aaf0d95b"}]},"branch":"refs/heads/master"},"795193f406050f4f188ebae4546e2d58af2ab3c1":{"kind":"REWORK","_number":4,"created":"2023-05-11 17:01:03.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/08/784808/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/784808/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/784808/4"}}},"commit":{"parents":[{"commit":"c0483c5b940d559b4a9efd2b5c64bf5b5528fa11","subject":"Merge \"sharder: make misplaced objects lookup faster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0483c5b940d559b4a9efd2b5c64bf5b5528fa11"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-04-05 18:00:24.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-11 16:53:27.000000000","tz":60},"subject":"Support swift.proxy_logging_status in request env","message":"Support swift.proxy_logging_status in request env\n\nWhen logging a request, if the request environ has a\nswift.proxy_logging_status item then use its value for the log\nmessage status int.\n\nThe swift.proxy_logging_status hint may be used by other middlewares\nwhen the desired logged status is different from the wire_status_int.\n\nIf the proxy_logging middleware detects a client disconnect then any\nswift.proxy_logging_status item is ignored and a 499 status int is\nlogged, as per current behaviour. i.e.:\n\n  * client disconnect overrides swift.proxy_logging_status and the\n    response status\n  * swift.proxy_logging_status overrides the response status\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I9b5cc6d5fb69a2957b8c4846ce1feed8c115e6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/795193f406050f4f188ebae4546e2d58af2ab3c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/795193f406050f4f188ebae4546e2d58af2ab3c1"}]},"branch":"refs/heads/master"},"7b46fe05d8e42a5c8ad477c809ba3e488a368ff9":{"kind":"REWORK","_number":5,"created":"2023-05-17 16:04:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/08/784808/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/784808/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/784808/5"}}},"commit":{"parents":[{"commit":"c0483c5b940d559b4a9efd2b5c64bf5b5528fa11","subject":"Merge \"sharder: make misplaced objects lookup faster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0483c5b940d559b4a9efd2b5c64bf5b5528fa11"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-04-05 18:00:24.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-05-17 15:56:05.000000000","tz":60},"subject":"Support swift.proxy_logging_status in request env","message":"Support swift.proxy_logging_status in request env\n\nWhen logging a request, if the request environ has a\nswift.proxy_logging_status item then use its value for the log\nmessage status int.\n\nThe swift.proxy_logging_status hint may be used by other middlewares\nwhen the desired logged status is different from the wire_status_int.\n\nIf the proxy_logging middleware detects a client disconnect then any\nswift.proxy_logging_status item is ignored and a 499 status int is\nlogged, as per current behaviour. i.e.:\n\n  * client disconnect overrides swift.proxy_logging_status and the\n    response status\n  * swift.proxy_logging_status overrides the response status\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I9b5cc6d5fb69a2957b8c4846ce1feed8c115e6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b46fe05d8e42a5c8ad477c809ba3e488a368ff9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b46fe05d8e42a5c8ad477c809ba3e488a368ff9"}]},"branch":"refs/heads/master"},"2dc7047bbdde8402ca014627101b5da62dd0d0b9":{"kind":"REWORK","_number":6,"created":"2023-06-28 12:35:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/08/784808/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/784808/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/784808/6"}}},"commit":{"parents":[{"commit":"39127fee5be6eb7733938d31e82bb6553ea9cbeb","subject":"Merge \"sharder: Fix completed metric emission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39127fee5be6eb7733938d31e82bb6553ea9cbeb"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-04-05 18:00:24.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-06-27 16:54:15.000000000","tz":60},"subject":"Support swift.proxy_logging_status in request env","message":"Support swift.proxy_logging_status in request env\n\nWhen logging a request, if the request environ has a\nswift.proxy_logging_status item then use its value for the log\nmessage status int.\n\nThe swift.proxy_logging_status hint may be used by other middlewares\nwhen the desired logged status is different from the wire_status_int.\n\nIf the proxy_logging middleware detects a client disconnect then any\nswift.proxy_logging_status item is ignored and a 499 status int is\nlogged, as per current behaviour. i.e.:\n\n  * client disconnect overrides swift.proxy_logging_status and the\n    response status\n  * swift.proxy_logging_status overrides the response status\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I9b5cc6d5fb69a2957b8c4846ce1feed8c115e6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2dc7047bbdde8402ca014627101b5da62dd0d0b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2dc7047bbdde8402ca014627101b5da62dd0d0b9"}]},"branch":"refs/heads/master"},"8a3c8a33037aa1deabbf064516bfe694872ec8c8":{"kind":"REWORK","_number":7,"created":"2023-10-27 13:05:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/08/784808/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/784808/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/784808/7"}}},"commit":{"parents":[{"commit":"39127fee5be6eb7733938d31e82bb6553ea9cbeb","subject":"Merge \"sharder: Fix completed metric emission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39127fee5be6eb7733938d31e82bb6553ea9cbeb"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-04-05 18:00:24.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2023-10-27 11:52:57.000000000","tz":60},"subject":"Support swift.proxy_logging_status in request env","message":"Support swift.proxy_logging_status in request env\n\nWhen logging a request, if the request environ has a\nswift.proxy_logging_status item then use its value for the log\nmessage status int.\n\nThe swift.proxy_logging_status hint may be used by other middlewares\nwhen the desired logged status is different from the wire_status_int.\n\nIf the proxy_logging middleware detects a client disconnect then any\nswift.proxy_logging_status item is ignored and a 499 status int is\nlogged, as per current behaviour. i.e.:\n\n  * client disconnect overrides swift.proxy_logging_status and the\n    response status\n  * swift.proxy_logging_status overrides the response status\n\nIf the proxy_logging middleware catches an exception then the logged\nstatus int will be 500 regardless of any swift.proxy_logging_status\nitem.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I9b5cc6d5fb69a2957b8c4846ce1feed8c115e6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a3c8a33037aa1deabbf064516bfe694872ec8c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a3c8a33037aa1deabbf064516bfe694872ec8c8"}]},"branch":"refs/heads/master"},"3268c5bb12c85955f5b1bd6b5d2e63d6c6dd2cd0":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-11-06 17:44:48.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/08/784808/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/784808/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/784808/8"}}},"commit":{"parents":[{"commit":"ed15681349ab1394b285a43948682b36613481ce","subject":"Merge \"tests: refactor SLO size/etag sysmeta tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed15681349ab1394b285a43948682b36613481ce"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-04-05 18:00:24.000000000","tz":-300},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-11-06 17:44:48.000000000","tz":0},"subject":"Support swift.proxy_logging_status in request env","message":"Support swift.proxy_logging_status in request env\n\nWhen logging a request, if the request environ has a\nswift.proxy_logging_status item then use its value for the log\nmessage status int.\n\nThe swift.proxy_logging_status hint may be used by other middlewares\nwhen the desired logged status is different from the wire_status_int.\n\nIf the proxy_logging middleware detects a client disconnect then any\nswift.proxy_logging_status item is ignored and a 499 status int is\nlogged, as per current behaviour. i.e.:\n\n  * client disconnect overrides swift.proxy_logging_status and the\n    response status\n  * swift.proxy_logging_status overrides the response status\n\nIf the proxy_logging middleware catches an exception then the logged\nstatus int will be 500 regardless of any swift.proxy_logging_status\nitem.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I9b5cc6d5fb69a2957b8c4846ce1feed8c115e6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3268c5bb12c85955f5b1bd6b5d2e63d6c6dd2cd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3268c5bb12c85955f5b1bd6b5d2e63d6c6dd2cd0"}]},"branch":"refs/heads/master","description":"Rebase"},"5af7719ef346e1594263536ccd9a952817cb2f7c":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-12-20 17:31:06.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/08/784808/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/784808/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/784808/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/784808/9"}}},"commit":{"parents":[{"commit":"365db20275d7798c30f10ee03221561a28bb7a79","subject":"FakeSwift: use HTTPMethodNotAllowed not HTTPNotImplemented","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/365db20275d7798c30f10ee03221561a28bb7a79"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-04-05 18:00:24.000000000","tz":-300},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2023-12-20 17:31:06.000000000","tz":0},"subject":"Support swift.proxy_logging_status in request env","message":"Support swift.proxy_logging_status in request env\n\nWhen logging a request, if the request environ has a\nswift.proxy_logging_status item then use its value for the log\nmessage status int.\n\nThe swift.proxy_logging_status hint may be used by other middlewares\nwhen the desired logged status is different from the wire_status_int.\n\nIf the proxy_logging middleware detects a client disconnect then any\nswift.proxy_logging_status item is ignored and a 499 status int is\nlogged, as per current behaviour. i.e.:\n\n  * client disconnect overrides swift.proxy_logging_status and the\n    response status\n  * swift.proxy_logging_status overrides the response status\n\nIf the proxy_logging middleware catches an exception then the logged\nstatus int will be 500 regardless of any swift.proxy_logging_status\nitem.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I9b5cc6d5fb69a2957b8c4846ce1feed8c115e6b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5af7719ef346e1594263536ccd9a952817cb2f7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5af7719ef346e1594263536ccd9a952817cb2f7c"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
