)]}'
{"id":"openstack%2Fswift~784868","triplet_id":"openstack%2Fswift~master~I8bfb62427438a41fe2df52d36d9b36b978c3d4dd","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I8bfb62427438a41fe2df52d36d9b36b978c3d4dd","subject":"reconstructor: Ignore already-deleted files indelete_reverted_objs","status":"NEW","created":"2021-04-06 01:01:13.000000000","updated":"2021-04-06 18:06:51.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"71498ed70869e58a86a1312b82132d94d44a6af8","_number":784868,"virtual_id_number":784868,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-04-06 02:13:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2021-04-06 15:42:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-06 02:13:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-06 15:42:16.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"8e2aa74b12edf021d71ecfe78b649869122445a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-06 01:01:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b2d0ed6fa3bf5bdc5494e79c5803ab15cec71ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 02:13:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5122a464c0a439fa536a7f629f68eca : SUCCESS in 9m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/8aa73f61770c41ecbd9e2c7d7fd19ff0 : SUCCESS in 1h 02m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6314bc94da9a4896aa9476c5cd4c9302 : SUCCESS in 43m 02s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e5065fb997874be090a8ad2274be73a2 : SUCCESS in 8m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9bd2b4ea736f46f886537d62cd54d274 : SUCCESS in 5m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b3ac6d0e849f416bb031c9db11b3893a : SUCCESS in 16m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/03c1976ddf3f4d549d38173e12620974 : SUCCESS in 20m 03s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/375d5d90e302404dbb00a1eb5a0cfc49 : SUCCESS in 14m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e65e26cd261a4b5ab6e5e1682514d19b : SUCCESS in 16m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9dfd9f48c10d457aadc5ad1d078cb8b7 : SUCCESS in 16m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/14d1c72e316445cdad4626b5d97a463e : FAILURE in 4m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/732f70a6be344715bc87597c8c0d4049 : FAILURE in 4m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c7e7079a1fac40a8ac7a2fc7b47f70b9 : SUCCESS in 18m 06s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/87e7d4915b554d0db3f32ed62b03e725 : SUCCESS in 18m 01s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9a723ca108b54288af75ace7262b8ac8 : SUCCESS in 21m 32s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/284cbdb7352846afb278fa21c846af90 : SUCCESS in 22m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/35a9333f44674280aa85f0f7e3f1b781 : RETRY_LIMIT in 4m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8c83014c0bb34a2b944bc336610cd23d : SUCCESS in 20m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/24b3d3d698f84602a2dc65badf808d10 : SUCCESS in 44m 28s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/6b55078d965e4db7947fd88acb4d5f17 : SUCCESS in 48m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b3043f4b3b824a5f91fa275f26915eaa : SUCCESS in 38m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f0cfba2a8e934155b5862393482ca41e : SUCCESS in 42m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f80c7f48c6944fdafaf72dc5583f6ea : FAILURE in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/086aaf7a9ccf4870a741fb7de1866df1 : SUCCESS in 5m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/39e47e988d1e4427a34891fc75e917e4 : SUCCESS in 26m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bea366f9288e404586700d2a438befda : SUCCESS in 54m 30s","accounts_in_message":[],"_revision_number":1},{"id":"1ba07c6c88752c82fac181eb4d2bdd749a8ad84f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-06 15:42:16.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\ndoes look like some error handling is needed, query inline","accounts_in_message":[],"_revision_number":1},{"id":"71498ed70869e58a86a1312b82132d94d44a6af8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-06 18:06:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d705af54297e105da18f9ee74eb3f4dc601cb68","revisions":{"4d705af54297e105da18f9ee74eb3f4dc601cb68":{"kind":"REWORK","_number":1,"created":"2021-04-06 01:01:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/784868/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/784868/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/784868/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/784868/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/784868/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/784868/1"}}},"commit":{"parents":[{"commit":"b79ca57fa98323b17f896af86c617d465066d544","subject":"Merge \"Create a separate doc for audit watcher\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b79ca57fa98323b17f896af86c617d465066d544"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-06 00:46:31.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-06 01:01:07.000000000","tz":-420},"subject":"reconstructor: Ignore already-deleted files indelete_reverted_objs","message":"reconstructor: Ignore already-deleted files indelete_reverted_objs\n\nChange-Id: I8bfb62427438a41fe2df52d36d9b36b978c3d4dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d705af54297e105da18f9ee74eb3f4dc601cb68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d705af54297e105da18f9ee74eb3f4dc601cb68"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
