)]}'
{"id":"openstack%2Fswift~784893","triplet_id":"openstack%2Fswift~master~Ie7d8277fc2626c20bd7bcf108959ac4d3e6c64ac","project":"openstack/swift","branch":"master","topic":"p-relinker-test-print","hashtags":[],"change_id":"Ie7d8277fc2626c20bd7bcf108959ac4d3e6c64ac","subject":"relinker: trivial comment and test fixes","status":"MERGED","created":"2021-04-06 08:52:17.000000000","updated":"2021-04-06 23:46:43.000000000","submitted":"2021-04-06 23:45:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784893-p-relinker-test-print","meta_rev_id":"6c5e8f53bb0ee26ca0bc69603b7bf335c1a200df","_number":784893,"virtual_id_number":784893,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-06 23:45:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2021-04-06 08:52:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2021-04-06 08:52:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-06 10:33:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-06 21:24:39.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"f6c401d852fc5f99d591b01eb0b0f38fa8d1c917","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-06 08:52:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71b93fb39ac6ae015fdd3db5ef672c9728581a97","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-06 08:52:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"05194d3aa5da5402b285bd1f0e4254a3df641b2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 10:33:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a893571e891344f7a39a9aac90f0c8a4 : SUCCESS in 10m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/75c0734fb32849bb8390fc44fc50c68c : SUCCESS in 1h 10m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ba9af586c2774ff0a03a2598c1c71b63 : SUCCESS in 45m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e29f1992d1de40f69f6d2512ad5c7db4 : SUCCESS in 11m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/745b2de644974c9994f812e1c75ee795 : SUCCESS in 6m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8dc2fe1502ef49a8b982e9628712c1e0 : FAILURE in 4m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f5eecc5c4cf84c9786f943acda599821 : SUCCESS in 16m 17s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/069901fbd4e94ae6b6437806adece33e : SUCCESS in 17m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/847b565524934e58a120e80298a9b7e2 : SUCCESS in 16m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ad67c5b9d3de49659d5d03d254636a28 : SUCCESS in 24m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a0cf153b86154bc6a6f847042c3db182 : FAILURE in 4m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4c88543806c54c12bfc0e999ebd4cee7 : FAILURE in 4m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cd545b8eeb8b490ab14ff37947955ee9 : FAILURE in 4m 20s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6f87efa15bf448f89519b8a2c761aebb : SUCCESS in 17m 26s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d1a36d9e36d54d71a492af3cef8fc95e : SUCCESS in 1h 25m 10s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0316b225a2394957bbf90698713b129f : SUCCESS in 22m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cb394eb218dd4812aca33e7b1d887bbf : RETRY_LIMIT in 3m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/90bc835e3a384107b9d8aafe326565aa : RETRY_LIMIT in 4m 21s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7b469daea2aa4ba1b8dcaa79272f28dc : SUCCESS in 49m 39s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/d81396f826144ab699264b3a095ca029 : SUCCESS in 55m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/99a7930a569c431b8a9b32258de96277 : SUCCESS in 42m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/be490c304dc6450f94d5fa27a0099ec5 : SUCCESS in 45m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18e6f3759f974978aacf58e64a38da11 : FAILURE in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c172e776b38432cb35ff8894e88c5ad : SUCCESS in 7m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cb0a063027574dcd98eb633a877e9931 : RETRY_LIMIT in 16m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d432cf8f24c34133b0c785eb0f336a67 : SUCCESS in 46m 56s","accounts_in_message":[],"_revision_number":1},{"id":"b755200a327cd172449164a659aa7dbb702b5a82","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-06 11:39:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1c3cd35aa3724dad9cd048c668480950ed266218","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 13:19:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59e77d6836eb4b64985c473fee688a71 : SUCCESS in 9m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/ba9a3b94a7804a81bc99b481f7301500 : SUCCESS in 54m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/db2ce9af418f42edb3813485a0a7eb43 : SUCCESS in 1h 02m 15s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/72ac6cc62f074e54b67e54979d53a8eb : SUCCESS in 9m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ccd705cd879d40718fba6bcf8cbfbeef : SUCCESS in 7m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c728be5442644429b75cc0a825e9a10b : FAILURE in 5m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/329e1015ffd74489984d58d275a92459 : SUCCESS in 41m 02s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2d612b5b938d44f697d06deffe05e59c : SUCCESS in 26m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/42e8d6d75486455bad16828728692f6c : SUCCESS in 23m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ea5267612150495381c3b07d6f57c381 : SUCCESS in 23m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4d382327ff82440280f0020914a1824e : FAILURE in 5m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/668d9c5aeadc4306b19ca6d048bd6d7f : FAILURE in 4m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/20c32b1117ea41f58d305e598aa2705c : FAILURE in 4m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/7b158f9040ce4b9f883c2d5ca39b6142 : SUCCESS in 19m 23s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cfc6c1c4ebee4459b89871a2f3d0eb2f : SUCCESS in 21m 52s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/061d119a93b440de9c6d60eb7e142864 : SUCCESS in 24m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ee1c38ec8b2f4048b814fd05ba27b0c1 : RETRY_LIMIT in 5m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/17bb0be8911e49509c02e1be1dbdca0b : SUCCESS in 21m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/7d4091ee73ab4835a321bee8dd07127e : SUCCESS in 49m 20s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/7596a05655bc4759b6429899cb7b3e81 : SUCCESS in 50m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d28be3bff2a04146a41056e67c3203d6 : SUCCESS in 43m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8668b065b254446db2d902fffbfcd1ec : SUCCESS in 41m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/903b5107777b42e6b2a5bf19acb2fb0b : FAILURE in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc3fefec67d348769454caaa9adea01c : SUCCESS in 7m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1d7cc2de1d7c46068282f1552d7ba76d : SUCCESS in 23m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/702d3e014f894b2b9ab9ae10d9169231 : SUCCESS in 1h 01m 39s","accounts_in_message":[],"_revision_number":1},{"id":"18c4ed61661b17fd4156db52f11c11e0f0d25337","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-06 21:24:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f53359ecd50f5275381a3dbee2e231b47a5719e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 22:41:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/917fd64f9dfb4fa8af227add63a940dc : SUCCESS in 7m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/ecade2a1d99e4a7483d76e756d68125c : SUCCESS in 1h 12m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/30fd0a2637a24cf59eeb98086198fa09 : SUCCESS in 51m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e17f5662428841df946087bf1b9715cf : SUCCESS in 9m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9c0df0bee2c248fc90270f80bc249065 : SUCCESS in 8m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dadc6d92bfd54275a993dd4cfd16ba60 : SUCCESS in 12m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/119df9dd386c4039b95a83b83e759d33 : SUCCESS in 13m 00s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/846d25820a8042a5afd84828504f9d1c : SUCCESS in 12m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c4c2f4d4c2854b3c95de6552098a2370 : SUCCESS in 11m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a3134ec7a5304661a5a1fd0697484224 : SUCCESS in 18m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b871e3ebbfc54623b89a2b61a4e6633f : SUCCESS in 19m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c3cabcbd2fb64521885c17a1df28a0ff : SUCCESS in 20m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6a01b1c719ee49b4b62ed6300c07e788 : SUCCESS in 21m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/43bb9500a081405f95a2d30411ac2cca : SUCCESS in 20m 46s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/31858f9a2c3f4ce1ac094222e1635c40 : SUCCESS in 15m 02s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b04b2c16a0a544aca689750676e6c884 : SUCCESS in 16m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/887533ab9f284db0992be613a7407cc6 : SUCCESS in 5m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ee973c4f8d0a47459dedb7f39ffef948 : SUCCESS in 17m 58s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/08aa88570bf64eb1920e69e080e1df11 : SUCCESS in 39m 08s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/07f5744e1e3d48f4b5e3b26e94a5ee36 : SUCCESS in 40m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a981cb0bffae4aa0908801d273e4d5c0 : SUCCESS in 30m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1d8bcbb567b04103bdc437a8a0f10deb : SUCCESS in 40m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4bdd2eab6a948429dabbf078e504d58 : SUCCESS in 18m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bcf5683e10849a1a4fccc42f0fdfea5 : SUCCESS in 5m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fa80c95393da41cb82c0bf9367abeae3 : SUCCESS in 18m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/03962da4b4c44ec592fdcfcdcbe0fe6d : SUCCESS in 56m 54s","accounts_in_message":[],"_revision_number":1},{"id":"4231ba34f66e1d18d52a36d1075d6033ca2eef1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 22:41:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"797e456d1a1b7ace41f896064366ce2b68eae716","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 23:45:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e688fc984864f58b47e55ef79dd316a : SUCCESS in 8m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/2d0585a909124a48a3541a9d144013b7 : SUCCESS in 1h 02m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6995996a1100456989038d0ca960e9da : SUCCESS in 39m 01s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/cccc64b67b8e491ca9721349cfd27551 : SUCCESS in 10m 22s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/b7287a752e204c339f5993de514210b8 : SUCCESS in 6m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/abd0fff78dc14481bf58e1e0fda683d9 : SUCCESS in 13m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4522a2dd36d946bfbb563956c526d26e : SUCCESS in 12m 46s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/008ab26c11df4b9cbf114b7189bd6ef5 : SUCCESS in 14m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ba191359905a42768c744ba35318735d : SUCCESS in 11m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1355d0f107f649a1af1ffa59f2f02688 : SUCCESS in 14m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/054cfa6cf4354d078520f7c18d385b56 : SUCCESS in 13m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8ec6d3a41c2d49eca9ca9dc2fd9c4cd8 : SUCCESS in 16m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/73bb837390b5463bbbeebf5d5df2c564 : SUCCESS in 17m 12s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/936fee8db2c3471db3c827150dbb6a3d : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/476a4a21536749c2b8a20e73b801fa35 : SUCCESS in 18m 06s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/650006e270ee4ecfa23bf131d8525e0f : SUCCESS in 17m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/43c3c9f318c5455d96ca4bc212918e37 : SUCCESS in 7m 19s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6a596dcf584a40949a1931e1f912fc0c : SUCCESS in 46m 42s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/000a0996f01a4082af4f8c40576ea894 : SUCCESS in 42m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4ab03fbf85574e08bca0bf8ae68c4995 : SUCCESS in 44m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2d1eb616c8e54ded9b1459a695025c07 : SUCCESS in 51m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/760cd05c5e0142bab947d6bebae031d0 : SUCCESS in 18m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/833202ea66704eaa80c594b0da925f9e : SUCCESS in 7m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/79d94d255b7d411b9bab2ead2522a952 : SUCCESS in 18m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a37ad9e20cb24d5a96ff3a9d4c84adb8 : SUCCESS in 54m 43s","accounts_in_message":[],"_revision_number":1},{"id":"239cdf60951ff9a29350687e91f60e77092eacf7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 23:45:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6c5e8f53bb0ee26ca0bc69603b7bf335c1a200df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 23:46:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fe2d6cf29754580852146b07162f6b3 : SUCCESS in 1m 07s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/691c0267e3b54fb0848227650f17038c : SUCCESS in 37s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/67ef1df4c72e4c98bfc5580f22bc4ad1 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ecb5aa120f9977b69e01d8dc4140df5437222c26","revisions":{"ecb5aa120f9977b69e01d8dc4140df5437222c26":{"kind":"REWORK","_number":1,"created":"2021-04-06 08:52:17.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/93/784893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/784893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/784893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/784893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/784893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/784893/1"}}},"commit":{"parents":[{"commit":"b79ca57fa98323b17f896af86c617d465066d544","subject":"Merge \"Create a separate doc for audit watcher\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b79ca57fa98323b17f896af86c617d465066d544"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-06 08:51:34.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-06 08:51:34.000000000","tz":60},"subject":"relinker: trivial comment and test fixes","message":"relinker: trivial comment and test fixes\n\nChange-Id: Ie7d8277fc2626c20bd7bcf108959ac4d3e6c64ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ecb5aa120f9977b69e01d8dc4140df5437222c26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ecb5aa120f9977b69e01d8dc4140df5437222c26"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
