)]}'
{"id":"openstack%2Fswift~784929","triplet_id":"openstack%2Fswift~master~I16d91a5a3e5c785244497bbc2882b305d2b3ca56","project":"openstack/swift","branch":"master","topic":"p-get-shard-ranges-fill","hashtags":[],"change_id":"I16d91a5a3e5c785244497bbc2882b305d2b3ca56","subject":"sharding: constrain fill_gaps to own shard range bounds","status":"MERGED","created":"2021-04-06 11:41:20.000000000","updated":"2021-04-07 00:02:22.000000000","submitted":"2021-04-07 00:00:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784929-p-get-shard-ranges-fill","meta_rev_id":"ebb8444e8df50608f3d0ce9c5f9459e6280097ba","_number":784929,"virtual_id_number":784929,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-07 00:00:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-06 18:10:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-06 21:27:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-06 13:44:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-06 18:09:53.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2021-04-06 18:10:06.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"4dd997e07ed88abd369d770fdefed76091002536","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-06 11:41:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fabda192c2f57063870625d3ecc61c7f4e7c8c25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 13:44:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0145f46aa92949699c42fd3c9418d359 : SUCCESS in 11m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/6f52db1850c74d5a809d734b961c7f7c : SUCCESS in 1h 08m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d5997a4983bb4e63bb7fdd443ed7ac7f : SUCCESS in 48m 48s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a2d623e9809043fbb15cb50bd4ec48ad : SUCCESS in 10m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/681d3e1285c14bd48ce6d2b12a6dcaa1 : SUCCESS in 8m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a158f83a66dc4e67851a6ed01bd481a8 : SUCCESS in 14m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c2a2aa93e52e468698ea58251d7779f3 : SUCCESS in 17m 24s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e25e5bb8d9c14c2abdcf9bf518268980 : SUCCESS in 15m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d2b7ad3446b84d1ea99747350fb364fa : SUCCESS in 15m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e7d35c4236724c1da1dceaf95881f3ba : SUCCESS in 17m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/de84a549149540c6a40fc2ca02f9ff18 : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3e67f572f366451b80f074dd3056f6c6 : SUCCESS in 19m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d4468000b3c640a181408bc1ba1eb49a : SUCCESS in 17m 22s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/58ec0dcfdf594891babf788942fb3e2f : SUCCESS in 18m 56s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4fa8241acb5d4baa95b9791252679f2d : SUCCESS in 17m 57s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a9543eaff0e54a98aeeb651a37763d9d : SUCCESS in 20m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a03a8aaa4fe24323ad4e50b98eba1d77 : SUCCESS in 7m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a0ff17e13f6149df8ddfd9bf4338f00e : RETRY_LIMIT in 4m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/81f6e1227bcb45c791a3a5ccd2011569 : SUCCESS in 50m 41s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/369842e8935c485889578401fac76eb2 : SUCCESS in 51m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5f5d9e1993474d85b2b8f3f0aa4d90a2 : SUCCESS in 41m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8bdedde682264e57aad1e12ff22270ea : SUCCESS in 42m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43f120f98edc4ba9a3f29ff4ad282fa9 : FAILURE in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f6f8d2952754a5ba3686d2eeafd2c0e : SUCCESS in 9m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/de5b6072d8c54f59bdc57dcd836f77b9 : SUCCESS in 24m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/180a70e0bbf3420c83175ac81826a208 : SUCCESS in 1h 15m 16s","accounts_in_message":[],"_revision_number":1},{"id":"fd9b80925708a352fa590b3c231013ec3fbfe9ac","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-06 18:09:53.000000000","message":"Patch Set 1:\n\nWay better than https://review.opendev.org/c/openstack/swift/+/784606 -- even has tests!","accounts_in_message":[],"_revision_number":1},{"id":"df6b583f297656b08e9e5698e2135d45bf8e4910","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-06 18:10:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2ebff1ac1a10bf0cb506f1211e53b438237bab66","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-06 21:27:39.000000000","message":"Patch Set 1: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d4bebffe0ee272038600df84baaa28de1b92cd03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 22:58:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b1a8fec31734849a7605b381b1d6c0c : SUCCESS in 9m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/954c960aa42e47f2ae93a6b86b154483 : SUCCESS in 1h 28m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/18dafafd0aa64b29997ea447661c11e9 : SUCCESS in 48m 09s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5e8eb9ae1a224ebf93b8b1599f92e425 : SUCCESS in 8m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/615307288ba547b0b0725d3226248117 : SUCCESS in 7m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/97f1b61230644b5584874801d0f3ad44 : SUCCESS in 19m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3d82d5c1d0a34beb9b591d8c7c4aeaf6 : SUCCESS in 17m 19s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8110deb1b82042bbb5eed58c57ccbb2c : SUCCESS in 19m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/69d89ac5c345421a87afabb062891965 : SUCCESS in 19m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3c96f10d9378414493e8d594515a9cd6 : SUCCESS in 19m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e317edf32cbd468d834a842fe8f9b5c3 : SUCCESS in 17m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/560f0cee681c4090b072a71a75ca6bcf : SUCCESS in 13m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a005943c62064834944a64a6bec467b2 : SUCCESS in 21m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6bfa28c27df7407fb2f1c1c3ba08ce63 : SUCCESS in 18m 18s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f356b56c056f4df1b8b4d04544b77417 : SUCCESS in 21m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b85afd7ff637438dad50fc3d15f8c411 : SUCCESS in 15m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/835aff697c134bd9ad446de6cbf5da78 : SUCCESS in 13m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/39b7e693a7ed47c98bc039381d24c61d : SUCCESS in 24m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3ae334297bec472aad09af5a44a2b376 : SUCCESS in 35m 19s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/2df55b6b53974f1bb8f63515e11f5ecc : SUCCESS in 35m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/93cd5a13984d40e9966a5d785a349bde : SUCCESS in 37m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/92d940217ea4475cbef66911edd7f689 : SUCCESS in 41m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b20d4ebd245426986dd31b619ae89c2 : SUCCESS in 11m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b06ae3331c943abab452a3956497131 : SUCCESS in 7m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/26a3409f2fd648fa8ef9234ea7e645ea : SUCCESS in 28m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ecd34cb4fa9348f8bbae56bfb27da12b : SUCCESS in 51m 45s","accounts_in_message":[],"_revision_number":1},{"id":"31f4e60bbb3df641724038d551b1f09cc6e51086","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 22:58:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ee60bd4e39ba92149d92ecbcf8a435743c35042","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 00:00:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d51a808c2e444350a8a4c7cca0c69ccf : SUCCESS in 11m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/7e77aa12ba0d4565b8c87254d063c8a1 : SUCCESS in 1h 00m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0e9e1a1d3bf742ab86fb2eb0bffbf30f : SUCCESS in 42m 24s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/944d7d8887c641df8cf38f6eeb0e03de : SUCCESS in 8m 29s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/652ffb19be5f43d79f25cd83a9ba1516 : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0a60a7ddebf94116994dd377c398de68 : SUCCESS in 13m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/12b05c75175144fbbad9a2b5c5114029 : SUCCESS in 20m 46s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c2b430d4c1e648ccba8652585633dc7a : SUCCESS in 15m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a8417e80aec747ba9d93afc27b4b5570 : SUCCESS in 13m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/98c077854e4a4c9d9b3588fa79e5e72e : SUCCESS in 16m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/637c93698a074978b4c9079bcdd43c6a : SUCCESS in 17m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aaa8682fef57413c99a3d9949cae1006 : SUCCESS in 23m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/45065e0dd4e246c0a1c6a1764880d22d : SUCCESS in 16m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/76db4bbe4e4c4ae79444c4c9cc89cf66 : SUCCESS in 20m 58s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0d12f29fef6a45e0be8d5418dee32b44 : SUCCESS in 18m 38s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b68cf72cfd36424ab9216edec44b1e94 : SUCCESS in 18m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/387398b4597643ba9e5dfae817fa413d : SUCCESS in 6m 42s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a635bc89b81642a3b05d3d49a544410c : SUCCESS in 47m 11s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/efd6cdd9fe56460fb04e8a27338a6a82 : SUCCESS in 54m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e32669bbb3654d6199a8158353d31d5e : SUCCESS in 34m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/030259dfeca7456ca81d07887f50a7a1 : SUCCESS in 38m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4649dac105b749a29603690a61f7974c : SUCCESS in 14m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b75227ca220248efbf12bd64b33f3c6e : SUCCESS in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/43526461cb8b4e5b82f06263bba6eb46 : SUCCESS in 25m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b0f6326f88e4a2c964181fa88705148 : SUCCESS in 43m 57s","accounts_in_message":[],"_revision_number":1},{"id":"7376c788cb2add9ff6d9bac2f5a35b68d9db8949","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 00:00:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ebb8444e8df50608f3d0ce9c5f9459e6280097ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 00:02:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37d83d4128e24fe0a2f8d816f04cdf62 : SUCCESS in 1m 08s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0c9c6b9e849344fbb7ea58c0b0e6bc8e : SUCCESS in 37s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/590ea633ca1348adbb26a8a92ef7d890 : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7bbc73a91c22e53c45d7097dfd320a9990580ce7","revisions":{"7bbc73a91c22e53c45d7097dfd320a9990580ce7":{"kind":"REWORK","_number":1,"created":"2021-04-06 11:41:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/29/784929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/29/784929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/784929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/784929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/29/784929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/29/784929/1"}}},"commit":{"parents":[{"commit":"b79ca57fa98323b17f896af86c617d465066d544","subject":"Merge \"Create a separate doc for audit watcher\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b79ca57fa98323b17f896af86c617d465066d544"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-06 11:12:32.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-06 11:27:12.000000000","tz":60},"subject":"sharding: constrain fill_gaps to own shard range bounds","message":"sharding: constrain fill_gaps to own shard range bounds\n\nRestrict the fill_gaps behaviour of ContainerBroker.get_shard_ranges()\nto only insert a filler range to the upper bound of its own shard\nrange.\n\nChange-Id: I16d91a5a3e5c785244497bbc2882b305d2b3ca56\nCloses-Bug: #1922386\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bbc73a91c22e53c45d7097dfd320a9990580ce7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bbc73a91c22e53c45d7097dfd320a9990580ce7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
