)]}'
{"id":"openstack%2Fswift~785038","triplet_id":"openstack%2Fswift~master~Ib99c9fb61125122d2e3cca633ce3504fd311dc6d","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ib99c9fb61125122d2e3cca633ce3504fd311dc6d","subject":"relinker: Add some logging when we want to clean up but can\u0027t","status":"ABANDONED","created":"2021-04-06 20:00:57.000000000","updated":"2021-06-23 23:51:08.000000000","total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"15066e1c26daec3b0ca2e8e4a12013a8c3c3bc18","_number":785038,"virtual_id_number":785038,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-06 21:17:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2021-04-07 10:42:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-04-07 19:09:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}],"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-06 21:17:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-07 01:15:27.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2021-04-07 10:42:49.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"3591bd2b947168f26babacc379a2a0bb94a42156","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-06 20:00:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dbada6be364e1660027fe48b4a0060cde14b285d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 21:17:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ba93ab3856e420a9fa1e537c9590afe : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/9d879859aea84588b0417dd959d53b20 : SUCCESS in 1h 11m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a01a5cacaa004bd7a659a485cf9dfae3 : SUCCESS in 31m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f718203b519441529056351247c9ad49 : SUCCESS in 8m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b0b56970849345c9803fcc4a22150685 : SUCCESS in 6m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dde9b1f8e78d4ad0b73dc3c7f2a2aa3c : SUCCESS in 17m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/990f12bc8d0047abb67fabd7b1d914b6 : SUCCESS in 18m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d4fd1db837cd45f9b01c063e5eb09a9f : SUCCESS in 14m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/205be179fae34c469ab5e499d1c76093 : SUCCESS in 13m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0aab045a7cbc49b39154db07b8878090 : SUCCESS in 18m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3dd97a9e011e486d861eeb2396cd576c : SUCCESS in 16m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1af28dfe6a4146e88850fe60309f7174 : SUCCESS in 16m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/96215e84419345d8894ae903603e1f9e : SUCCESS in 20m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3a8503ec2b734796affee1dd8c54e2c3 : SUCCESS in 16m 20s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/df7af2be38ab4b6c8a50646964c843df : SUCCESS in 17m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7ba3d4f501a548448604b15c01729e33 : SUCCESS in 20m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/24e73611667d467eb68c55feb8ad52c5 : SUCCESS in 5m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/78f0afec298e408eba12924e63276822 : FAILURE in 18m 54s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4d8ff41e47c049a493105d4674120e63 : SUCCESS in 49m 47s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/ce3562057b0a4c77b091619542deaebb : SUCCESS in 48m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b3c6296e06904dc98707ef965b82a45b : SUCCESS in 38m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3792550a93b34b99833b2ab17b717d93 : SUCCESS in 30m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d0211c367df444badd12b40e668f902 : SUCCESS in 15m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/deb25b3d367d4a5f88afaab585122723 : SUCCESS in 5m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b3672963bbd549b0a6a00aa98db0d235 : SUCCESS in 27m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c8ffd4d2bdc4bb88cd0cb797d09691e : SUCCESS in 51m 56s","accounts_in_message":[],"_revision_number":1},{"id":"dfa627b37913046825a8e8bff2b5757c77da26c5","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-07 01:15:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"afbcdaffd413f9845b2494633371e10db58692ff","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-07 10:42:49.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nspent a while trying to figure out how a .lock file could be left in an otherwise empty dir, but not got any ideas, so having some logging may help.\n\nholding off on +A pending response to Matt\u0027s question. My 2cents worth...info may be ok, we don\u0027t necessarily need to be generating alarm by adding warnings that we do anticipate firing","accounts_in_message":[],"_revision_number":1},{"id":"6a0f679fde2016274034f492d9f4a1174997f34f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-07 19:09:10.000000000","message":"Patch Set 1: Workflow-1\n\n(2 comments)\n\nReally, I kinda just want this on one or two of the hundreds of nodes the next time I\u0027m running a part-power increase -- maybe it\u0027s worth a manual hack instead of pushing it upstream?","accounts_in_message":[],"_revision_number":1},{"id":"15066e1c26daec3b0ca2e8e4a12013a8c3c3bc18","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-23 23:51:08.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e5c1f3bdae287f470e38ce86754954a1685f021f","revisions":{"e5c1f3bdae287f470e38ce86754954a1685f021f":{"kind":"REWORK","_number":1,"created":"2021-04-06 20:00:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/38/785038/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/785038/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/785038/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/785038/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/785038/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/785038/1"}}},"commit":{"parents":[{"commit":"1895213d2550e25bbc5313d67ee5b32f32adf970","subject":"Update some constraints for py2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1895213d2550e25bbc5313d67ee5b32f32adf970"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-06 20:00:10.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-06 20:00:10.000000000","tz":-420},"subject":"relinker: Add some logging when we want to clean up but can\u0027t","message":"relinker: Add some logging when we want to clean up but can\u0027t\n\nChange-Id: Ib99c9fb61125122d2e3cca633ce3504fd311dc6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5c1f3bdae287f470e38ce86754954a1685f021f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e5c1f3bdae287f470e38ce86754954a1685f021f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
