)]}'
{"id":"openstack%2Fswift~785276","triplet_id":"openstack%2Fswift~master~I1805fa2b471882c500755c6136b0d5a9ba7cd5b3","project":"openstack/swift","branch":"master","topic":"updateAuthors","hashtags":[],"change_id":"I1805fa2b471882c500755c6136b0d5a9ba7cd5b3","subject":"Update AUTHORS","status":"MERGED","created":"2021-04-07 20:54:48.000000000","updated":"2021-04-13 06:15:35.000000000","submitted":"2021-04-13 06:12:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"785276-updateAuthors","meta_rev_id":"4c4e960bf9a5b74ddb3bf8949f9ab8808c470522","_number":785276,"virtual_id_number":785276,"owner":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-13 06:12:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2021-04-08 08:36:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2021-04-07 22:36:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":1,"date":"2021-04-12 22:46:28.000000000","_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-07 21:25:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-07 22:36:08.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2021-04-08 08:36:53.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"4183e8fd057463455c596c7a3fee135c8c333c84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2021-04-07 20:54:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91e6952612559f333e7fa7ec49f3ae40235bbfac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 21:25:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a77e02aafe74cddb662178d9c8246d3 : SUCCESS in 10m 37s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a0d1099fcdf640b9af3cb75424005bab : SUCCESS in 9m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bff09b1546414ecd8deaf95759445f57 : SUCCESS in 7m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f0b10f9e7f8347d6baf7032261f3f020 : SUCCESS in 22m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/855ee159a79443ecbe1f1d1ee73bcff7 : SUCCESS in 17m 17s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/047437eea2104ff88d07384d1666892d : SUCCESS in 15m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2999d76701b14349ad4ab9ec29708fd0 : SUCCESS in 27m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/de3f442cb53b4fec936b7c2b4a47fce4 : SUCCESS in 15m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58a0359148944425842050600d328ebb : SUCCESS in 15m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04553489d92f4f1dbd81eeb6b927cd2f : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"30138b9f4f7aba85f51b4b3f4bcb4c4516bf3c21","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-04-07 22:36:08.000000000","message":"Patch Set 1: Code-Review+2\n\nRomain has been a staple of the Swift community for years.  I recognize his direct contributions as having board impact on Swift\u0027s operational excellence.  He\u0027s extremely technical and knowledgeable.  His experience has been a valued asset during the implementation of major features and critical bug fixes.  His good judgement and ability to articulate a vision have often thrust him into the position of leader and decision maker - and he never let us down.  Thank you for all the help Romain - I\u0027d welcome an oppertunity to work together again in the future!","accounts_in_message":[],"_revision_number":1},{"id":"72a93c3e4f07a53fc447dbbe36cfa39f52356856","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-08 08:36:53.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nThank you Romain for all your contributions. You are respected for your expertise as an operator of Swift and also as a contributor of significant code improvements. It has been a pleasure to work with you.","accounts_in_message":[],"_revision_number":1},{"id":"b806c8ef039180649571966269e5d45f3fdbbe9e","author":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"date":"2021-04-12 22:46:28.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f7add014354d832b92fbb749d13412d6d37c6f88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 01:43:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93e0c1481f6124688d46290afd03a5ddb11bc85a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 06:12:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c617f5335f504262a0614e63e198ffb4 : SUCCESS in 10m 25s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b1e691fff6854c76aa897675234c00d3 : SUCCESS in 11m 19s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/5792d2eb07b646059b5638657c93f9ff : SUCCESS in 8m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/59f3e3cf7f2446c39562699f4b914156 : SUCCESS in 15m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7ee7a96ceba34862b9fb3b70f05940b8 : SUCCESS in 17m 39s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/494d58fd8dc54f6aac87a2db6396c953 : SUCCESS in 15m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/073bc864d7944e1a885244111ce78c9e : SUCCESS in 19m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ef8d9093f76949cc9eae3acab8f43632 : SUCCESS in 15m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5a139e8878af4f619992fd2466249849 : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/02606da2e40948f4b692165dfdc946b5 : SUCCESS in 17m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ad33974980334f19b124541059e7de1f : SUCCESS in 17m 00s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/37669678f10d4042b786c0629b3041fb : SUCCESS in 18m 42s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cdf6ad044e7c47bdad05204d1539b55d : SUCCESS in 20m 00s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/2a925421a1734bfca9db500993793bba : SUCCESS in 18m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/381afd705a724d16b0fe1d35f22c91b4 : SUCCESS in 8m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cc89813b79f4f198e75788776614c40 : SUCCESS in 14m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/763211c5eaed40fdbe95cb4c85e76fb8 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":1},{"id":"186df3c36bcc15173df1bba518fb65b7d90e2ae3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 06:12:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4c4e960bf9a5b74ddb3bf8949f9ab8808c470522","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 06:15:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4aa4392a8a114737b6728492dbd94e2b : SUCCESS in 1m 14s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b57fb1e8295d4da2a93111785b3d9007 : SUCCESS in 53s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/0573fa170cdb40d1bced57c9b0b3ef54 : SUCCESS in 1m 24s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c13c9cc6754d8a47bd6a30439b7c883f3daa4ac7","revisions":{"c13c9cc6754d8a47bd6a30439b7c883f3daa4ac7":{"kind":"REWORK","_number":1,"created":"2021-04-07 20:54:48.000000000","uploader":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"},"ref":"refs/changes/76/785276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/785276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/785276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/785276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/785276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/785276/1"}}},"commit":{"parents":[{"commit":"d9a6fe4362852b07770c46b8ca549ea6f1475ae2","subject":"Merge \"sharder: Prevent ValueError when no cleaving contexts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9a6fe4362852b07770c46b8ca549ea6f1475ae2"}]}],"author":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2021-04-07 20:27:42.000000000","tz":-240},"committer":{"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","date":"2021-04-07 20:38:25.000000000","tz":-240},"subject":"Update AUTHORS","message":"Update AUTHORS\n\nIt\u0027s been a real pleasure to work on Swift all these years with you\nguys. You\u0027re doing an amazing job in the best mind. Don\u0027t change\nanything!\n\nChange-Id: I1805fa2b471882c500755c6136b0d5a9ba7cd5b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c13c9cc6754d8a47bd6a30439b7c883f3daa4ac7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c13c9cc6754d8a47bd6a30439b7c883f3daa4ac7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13852,"name":"Romain LE DISEZ","email":"romain.le-disez@corp.ovh.com","username":"rledisez"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
