)]}'
{"id":"openstack%2Fswift~785363","triplet_id":"openstack%2Fswift~master~I206eea1b0c36d941307f80447c4fd64fcf83b6f2","project":"openstack/swift","branch":"master","topic":"db_auditor_refactor","hashtags":[],"change_id":"I206eea1b0c36d941307f80447c4fd64fcf83b6f2","subject":"WIP: Account and Container Audit Watchers","status":"NEW","created":"2021-04-08 10:01:56.000000000","updated":"2021-04-22 02:07:24.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"e812fbf971d2412a617e1132caa95263b546fc3f","_number":785363,"virtual_id_number":785363,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-12 01:29:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-08 11:10:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-22 02:07:24.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"da3f4390d82235058867151f39131be87ec68f75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-08 10:01:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34172b393aaa6895fbafd8026d48316c3645c483","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-08 10:02:31.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"1d4655ea1882c4e66d4a1b4a194eb65e729a93a6","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-08 10:03:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2043b38bffcbc6ca84dcf49677cb37c40c4c0b57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 11:10:15.000000000","message":"Patch Set 1: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0046ac17ecd84712a836e1d5bf9768cc : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c50bc6ac75444c399e3d6efc998e2026 : SUCCESS in 10m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/a53c57ad0a904de0be307e15a29038ae : SUCCESS in 48m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8bf3d03aec784495a98d18ee1ad20701 : SUCCESS in 43m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/34861334faf44fc38050400d03c1c276 : SUCCESS in 9m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/55039d1a47734bfc917f6a28509eca4a : SUCCESS in 7m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fd3ebb7f322f4d6dae82ed5c4d7e382e : FAILURE in 14m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7b52a0cc2e5947f0aebc913e438f795a : FAILURE in 14m 58s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ad99b20f8cc04a16b351e8b44acd9067 : FAILURE in 17m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/44a3c18d81704ae099afde3a19098503 : FAILURE in 16m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5dad38a9f4fe4585b63cb90e5f8eb048 : FAILURE in 15m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/204a7b8805b54819a5fd6d473c6187da : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a213a888c6c04d8fadc7ceb4d6594db3 : SUCCESS in 17m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7bc51055cb284043b83380e7a775dadf : SUCCESS in 16m 58s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/6413bfca176f4db0aedafc6a0c6b0fdc : SUCCESS in 20m 46s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/10e201bbb7094f7b867a2d8358dab6e3 : SUCCESS in 17m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e3e3e0e97c964919bf33f8a95dceb38c : SUCCESS in 20m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/977ec5f035d54244a1c5b2cd69aa7d59 : SUCCESS in 9m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d56321424f444ba9a778f2410bce63bb : SUCCESS in 26m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5fd1342539c7494a84af4351d06b0b4f : SUCCESS in 45m 41s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/662a3b1f8888404ebdfcf1c457abfbf5 : SUCCESS in 50m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/786227c098dc42b38592306c796a0d89 : SUCCESS in 42m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/32d5a0ea49aa4a07a2226092da2e5fac : SUCCESS in 39m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c5c99c4cabfe46e3b6908f36b6e322d1 : FAILURE in 18m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4d564fdd96e4b06a13213b90fba82cc : FAILURE in 6m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7734f018ff3f4f399dfce723b05c2e02 : SUCCESS in 24m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bea3e06b5cc54055819a8c2f4901b705 : SUCCESS in 1h 03m 12s","accounts_in_message":[],"_revision_number":1},{"id":"424a59036a7b3238b65764027dd4a0160e40de51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-09 00:51:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"45a5efb3493cee6c4093d9ee682833309fb5949f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-09 00:56:53.000000000","message":"Patch Set 2: Workflow-1\n\nSLOs for sharded conatiners in the container walker bucket inventory now mostly work (once you add SLO to the interenal client), however SLO\u0027s will update their manifest with etag and bytes which may change behind the scenes meaning you can\u0027t actaully get the SLO .inventory object.\n\nIf we decide to use an approach like this we need to have a way of having a manifest where we don\u0027t care about the etag or bytes of a segment. We could use DLO, but as things shard and change then the point of the SLO is that the object will only have what the sharded container things is appropriate. So kinda need a less strict manifest version.","accounts_in_message":[],"_revision_number":2},{"id":"a5bcdf1bad37ab45d58abbbede05010372be88e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 01:58:40.000000000","message":"Patch Set 2: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0cb115a740e44ccab1c3a1bf136869e6 : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ff1a62ab0ea446d86aec508a3158f45 : SUCCESS in 10m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/40e5db13c49d4dbc9f5d33bfe952f8a2 : SUCCESS in 1h 01m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0b8cec9507d943d0a1439eeaf78f1b77 : SUCCESS in 48m 27s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/351ac07cb1e84a798088e0cb697963bb : SUCCESS in 9m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a0d32f82ef5c475a801bab36a04a53d0 : SUCCESS in 7m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1ed4e012ffcf463baccdb26607598706 : FAILURE in 14m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/403a60ddcbc94365ae7cb9d259c50c87 : FAILURE in 15m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/69f3201e854f4cdb90bd51a5503b8d29 : FAILURE in 14m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5b38b53c27b5489ab21cfa9ff9f92363 : FAILURE in 14m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1b60b99375984c1d974bc2190763d64e : FAILURE in 14m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/67ab04133a06455796ced80ccebc4ad1 : SUCCESS in 16m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/13c1c8e4e19b47138a9b6488bdaf69c8 : SUCCESS in 16m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/86968989b67e4a51a5095bf4f277ce35 : SUCCESS in 16m 38s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/978e519ee1914b14bc6467ef01ab4605 : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/318c667e518c422c97d1d82e8daebab4 : SUCCESS in 17m 27s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/c354d15a37a945c68b610d033b761cb0 : SUCCESS in 17m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4299c3fe469c460a8f21ee22eb037a2f : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a218a5a175bc428896c2b51f88ccabf7 : SUCCESS in 20m 05s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1e0949fcba2c49418fed5535c358ecba : SUCCESS in 46m 24s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/9ad6922cb32e4ac4ad55c10800e12abe : SUCCESS in 45m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/846f3b01b6954faeb002e85b1da4123a : SUCCESS in 37m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bd42d375687a4dd4bd936cebbe87cac0 : SUCCESS in 36m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62953de69dcf4aa9bb42b4ad2d6bc523 : FAILURE in 14m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9db4578ce270465e934c9f8b3388c613 : FAILURE in 5m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/45a0370743c947a196faf2cd5c9dec21 : SUCCESS in 22m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b4069a7a59944e08b9bf335748dff6e : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":2},{"id":"4855e2dee8b4612b8f724582c8f1ac318d66c07f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-09 02:02:20.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"9499590cff081148bf9bc30cd86e3c8e530030fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 03:16:14.000000000","message":"Patch Set 3: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a274b264bf134441afcd04e5c3aa3dc5 : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18346af220de42faab12a6dd037a045d : SUCCESS in 10m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/ec6883d94ce94cdca81dafd855f4c9fe : SUCCESS in 1h 07m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/98f4fa51ceaa45eca41af7bcd3328773 : SUCCESS in 1h 11m 43s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/69082f299e67413c9fa5403c74033119 : SUCCESS in 10m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c889a900add44a3284d36f043fce95e2 : SUCCESS in 9m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4711714e445240e4983ff12b1349c6db : SUCCESS in 39m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8f09935a4a074bf4acba25e157ec469c : SUCCESS in 31m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/863bba69e2a645c2b62f730b914a9259 : SUCCESS in 15m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9095dbfd89c9413da7f05d07d28a7685 : SUCCESS in 16m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0e28c981044c426e886a04f918355ab4 : SUCCESS in 27m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d27faa35163d4d1f8cf489367bf82bd5 : SUCCESS in 57m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/40776f2379fd4ab9977ab52fb21e812a : SUCCESS in 1h 00m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f983918ce90241e4ab84c3f8926e9e7f : SUCCESS in 17m 30s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/42aef86b6cc64bdd9fb43474c72a43f3 : SUCCESS in 17m 59s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/54bed63b2e9741a7b83f5b685c067e41 : SUCCESS in 1h 01m 44s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/10cb4262bf634cd284ad1801803956b5 : SUCCESS in 19m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5a1daa8e775e4b389309b04995aa77ba : SUCCESS in 6m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/58c2f19bad4b4a8ba72704e246a6bc32 : FAILURE in 19m 33s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/52f1af8e877d4b58be9d5a665060bf5c : SUCCESS in 54m 34s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/01f6c974d6f440e7a81243240f04ce5e : SUCCESS in 40m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e9176105b43a4a0d80be88a523c4a305 : SUCCESS in 38m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/35c2a3c6d2474c5f9a86c133a396efc3 : SUCCESS in 46m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5e15129bd5d4a66a6b08dd0b4f171d4 : SUCCESS in 14m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ed0bb0f454148b98e2beefd497430e8 : FAILURE in 5m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/497cb12250ef401ab6cccdf7db8eb6ff : SUCCESS in 22m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63f3ad98c3664b09b67cd6c767bc746e : SUCCESS in 51m 15s","accounts_in_message":[],"_revision_number":3},{"id":"2f1574f73b4c57f40f7f92517df441ded988fad2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-12 00:22:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"abab33c56e33de496b48c49922b787ade5edbf5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-12 00:35:28.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"f3d67f431f6974cd5752d92a0afafdd471cc650c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 01:29:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1b915cb9fbf4451fb1d5c73d5782796a : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d23f742457b424e8d1966043832354e : SUCCESS in 12m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/cc19b5d48cc147daa5741f6341a6d932 : SUCCESS in 38m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/04ad8dc1dbed4a279721786df72b0a88 : SUCCESS in 41m 27s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/22a8148f2425456ebc391b6b0c40fcb7 : SUCCESS in 7m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/181cd3ab5276457faa752bed3d7346d3 : SUCCESS in 10m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a58f3ee917984d1781b5e1c34f0e7bd7 : SUCCESS in 19m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/03aa7e6ea8f9418e99c6d18a8ea99c25 : SUCCESS in 19m 26s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/647f7dc8479d422f95ba02ffedb313d6 : SUCCESS in 13m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6f24543fa7f446089128e886d08def7f : SUCCESS in 19m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a3ba2a47e85d416c928e661099643b12 : SUCCESS in 19m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/abf15d985b6843a097de9b15086c1bd3 : SUCCESS in 19m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/44f73d2b904f456ea22f3524f0fd1f06 : SUCCESS in 19m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4a53b994214d4434ade5deeb969ef916 : SUCCESS in 22m 27s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/254afa779bbf431585b6c5d8ca259ac0 : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cd8552c630e14219af77f779d6e02c7e : SUCCESS in 22m 04s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/4fbfd89ee4ea450389747d4047fcc2fa : SUCCESS in 44m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b997b35f39684d6bb3ba78e63299b5d8 : SUCCESS in 6m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/64aa566ca93a42b5ab5001f24918c1c9 : SUCCESS in 51m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/2b03b8569ca64f44b5b5cf65591a8f48 : SUCCESS in 42m 17s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/ab1087fe2e48431d82fcaa35454adf63 : SUCCESS in 48m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/004da5e62f0947e3ac6e9df77f2fba8b : SUCCESS in 39m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a5b45fb55c9d4e0ca4efdecaef8f9367 : SUCCESS in 39m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2b70595cac94c7cb98cc2f56e8fee5b : SUCCESS in 37m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40b18f144de14efbb4544089f0d84e96 : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/91aa71f563c44c19a66f535370c5b007 : SUCCESS in 30m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e82e2f75fb7b4af98e5ff0814879a191 : SUCCESS in 46m 00s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d084842094910d293b9014c03a82c55c7ec12837","revisions":{"a92aa93fd9709f9fca65ecccd438ac3473c171c2":{"kind":"REWORK","_number":1,"created":"2021-04-08 10:01:56.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/63/785363/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/785363/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/785363/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/785363/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/785363/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/785363/1"}}},"commit":{"parents":[{"commit":"b7e5bb5c31b705dc24f2dc8c97a9006115196d6b","subject":"Refactor db auditors into a db_auditor base class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b7e5bb5c31b705dc24f2dc8c97a9006115196d6b"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-04-08 09:58:16.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-04-08 09:58:16.000000000","tz":600},"subject":"WIP: Container Audit Watchers","message":"WIP: Container Audit Watchers\n\nNow that we have object audit watchers, let\u0027s have the ability to create\ndatabase audit watcher. That is account and container watchers.\n\nInstead of a see_object, db watchers have a see_broker(broker).\n\nThere is also a very experimental bucket inventory container watcher\nthat by defaults puts the inventory in an object called .inventory in\nthe said container. On sharded containers this will be an SLO (once I\nget it working).\n\nChange-Id: I206eea1b0c36d941307f80447c4fd64fcf83b6f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a92aa93fd9709f9fca65ecccd438ac3473c171c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a92aa93fd9709f9fca65ecccd438ac3473c171c2"}]},"branch":"refs/heads/master"},"c79ff2a5dfede7fbbd7a9c4cee707bff2c6b9acf":{"kind":"REWORK","_number":2,"created":"2021-04-09 00:51:00.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/63/785363/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/785363/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/785363/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/785363/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/785363/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/785363/2"}}},"commit":{"parents":[{"commit":"b7e5bb5c31b705dc24f2dc8c97a9006115196d6b","subject":"Refactor db auditors into a db_auditor base class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b7e5bb5c31b705dc24f2dc8c97a9006115196d6b"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-04-08 09:58:16.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-04-09 00:50:29.000000000","tz":600},"subject":"WIP: Container Audit Watchers","message":"WIP: Container Audit Watchers\n\nNow that we have object audit watchers, let\u0027s have the ability to create\ndatabase audit watcher. That is account and container watchers.\n\nInstead of a see_object, db watchers have a see_broker(broker).\n\nThere is also a very experimental bucket inventory container watcher\nthat by defaults puts the inventory in an object called .inventory in\nthe said container. On sharded containers this will be an SLO (once I\nget it working).\n\nChange-Id: I206eea1b0c36d941307f80447c4fd64fcf83b6f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c79ff2a5dfede7fbbd7a9c4cee707bff2c6b9acf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c79ff2a5dfede7fbbd7a9c4cee707bff2c6b9acf"}]},"branch":"refs/heads/master"},"9a90e2d61aba93df9120b31fa0133495f3d5ba8d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-04-09 02:02:20.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/63/785363/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/785363/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/785363/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/785363/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/785363/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/785363/3"}}},"commit":{"parents":[{"commit":"4cb52b44dd490b6493798dc3c8188a1da7667559","subject":"Refactor db auditors into a db_auditor base class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4cb52b44dd490b6493798dc3c8188a1da7667559"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-04-08 09:58:16.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-04-09 02:01:51.000000000","tz":600},"subject":"WIP: Container Audit Watchers","message":"WIP: Container Audit Watchers\n\nNow that we have object audit watchers, let\u0027s have the ability to create\ndatabase audit watcher. That is account and container watchers.\n\nInstead of a see_object, db watchers have a see_broker(broker).\n\nThere is also a very experimental bucket inventory container watcher\nthat by defaults puts the inventory in an object called .inventory in\nthe said container. On sharded containers this will be an SLO (once I\nget it working).\n\nChange-Id: I206eea1b0c36d941307f80447c4fd64fcf83b6f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a90e2d61aba93df9120b31fa0133495f3d5ba8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a90e2d61aba93df9120b31fa0133495f3d5ba8d"}]},"branch":"refs/heads/master"},"36e18bcbd13494f79b5447e841028c0b1fb6184f":{"kind":"REWORK","_number":4,"created":"2021-04-12 00:22:18.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/63/785363/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/785363/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/785363/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/785363/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/785363/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/785363/4"}}},"commit":{"parents":[{"commit":"4cb52b44dd490b6493798dc3c8188a1da7667559","subject":"Refactor db auditors into a db_auditor base class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4cb52b44dd490b6493798dc3c8188a1da7667559"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-04-08 09:58:16.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-04-12 00:21:59.000000000","tz":600},"subject":"WIP: Container Audit Watchers","message":"WIP: Container Audit Watchers\n\nNow that we have object audit watchers, let\u0027s have the ability to create\ndatabase audit watcher. That is account and container watchers.\n\nInstead of a see_object, db watchers have a see_broker(broker).\n\nThere is also a very experimental bucket inventory container watcher\nthat by defaults puts the inventory in an object called .inventory in\nthe said container. On sharded containers this will be an SLO (once I\nget it working).\n\nChange-Id: I206eea1b0c36d941307f80447c4fd64fcf83b6f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/36e18bcbd13494f79b5447e841028c0b1fb6184f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/36e18bcbd13494f79b5447e841028c0b1fb6184f"}]},"branch":"refs/heads/master"},"d084842094910d293b9014c03a82c55c7ec12837":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-04-12 00:35:28.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/63/785363/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/785363/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/785363/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/785363/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/785363/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/785363/5"}}},"commit":{"parents":[{"commit":"4cb52b44dd490b6493798dc3c8188a1da7667559","subject":"Refactor db auditors into a db_auditor base class","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4cb52b44dd490b6493798dc3c8188a1da7667559"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-04-08 09:58:16.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-04-12 00:33:39.000000000","tz":600},"subject":"WIP: Account and Container Audit Watchers","message":"WIP: Account and Container Audit Watchers\n\nNow that we have object audit watchers, let\u0027s have the ability to create\ndatabase audit watcher. That is account and container watchers.\n\nInstead of a see_object, db watchers have a see_broker(broker).\n\nThere is also a very experimental bucket inventory container watcher\nthat by defaults puts the inventory in an object called .inventory in\nthe said container. On sharded containers this will be an SLO (once I\nget it working). This inventory is just a json-ifyed version of the\nobject table.\n\nChange-Id: I206eea1b0c36d941307f80447c4fd64fcf83b6f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d084842094910d293b9014c03a82c55c7ec12837"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d084842094910d293b9014c03a82c55c7ec12837"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
