)]}'
{"id":"openstack%2Fswift~785405","triplet_id":"openstack%2Fswift~master~Ic2bc4efeba5ae5bc8881f0deaf4fd9e10213d3b7","project":"openstack/swift","branch":"master","topic":"p-probe-use-assert-subprocess","hashtags":[],"change_id":"Ic2bc4efeba5ae5bc8881f0deaf4fd9e10213d3b7","subject":"probe test: use helper functions more widely","status":"MERGED","created":"2021-04-08 12:45:30.000000000","updated":"2021-04-08 22:23:55.000000000","submitted":"2021-04-08 22:22:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785405-p-probe-use-assert-subprocess","meta_rev_id":"ed8311b779201edf7c0e731d96036a570fef9228","_number":785405,"virtual_id_number":785405,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-08 22:22:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-08 21:27:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-08 21:27:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-08 13:42:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-08 21:27:51.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"1951d895f60a91ca00158472cb5545416da5e213","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-08 12:45:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42280fde9398329da2ee772ddd7975535e926998","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 13:42:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/920d9b76c76a4ee5ad8ea002b80e5ce1 : SUCCESS in 9m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/527fdeda8aca44a1a60d56c414c12723 : SUCCESS in 52m 18s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/33917bb07686411fbc982b3488c53fcc : SUCCESS in 51m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc62294f730848a0a69aa31c2fbf6236 : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":1},{"id":"800d11fac52349132cba9352dec733ab494f6253","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-08 21:27:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"549d62869b3e34b05b85df7f56becb3c78889a11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 21:28:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b3a192f454f74d39c79819d9bc60578262b3bf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 22:22:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cb7da48a7444dbaab3409e402584ef3 : SUCCESS in 8m 52s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5b08fbda6b134145a940dbbc53498e89 : SUCCESS in 19m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/57c908272f2c4273990fd122f110dcb3 : SUCCESS in 15m 02s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fbf5d14b3a044ad991da90a04a6b39ad : SUCCESS in 18m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/79c1e502e92b4533a8c703d29e122f58 : SUCCESS in 14m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/91e009ef0de2455ca02ebdb2f7516436 : SUCCESS in 16m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/41db6d47715f4839a870fc61bf466ec1 : SUCCESS in 20m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0d6e515bd8c949d0add9e3ae511808c7 : SUCCESS in 19m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f5cc0152bea449afb6a3aa2b644f8668 : SUCCESS in 19m 02s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b04287388a26429bb5ae58301206657e : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/19b80569b7724aa4ab5c8c2356911697 : SUCCESS in 19m 17s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d6a4758e23e04d329dab6be383ead257 : SUCCESS in 19m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ef5a741910df4177b3f36871f6d2363e : SUCCESS in 6m 49s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/701ce71c1b664390a1e6dc0ef2f90296 : SUCCESS in 50m 26s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/f9e5dc8f04544bd6ba7f86869346990e : SUCCESS in 48m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef8f4fc9994640f1b4413a69ee468412 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":1},{"id":"a6fd025cc0b1e521a6001745504a0d0a6001faa1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 22:22:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ed8311b779201edf7c0e731d96036a570fef9228","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 22:23:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d99b94b60d2429396daa10d5568e923 : SUCCESS in 1m 01s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/74edef3bb0514d8c80fd928431e23053 : FAILURE in 35s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/359a28679ae24f1093fafdaee74617dc : FAILURE in 37s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"122840cc04cd71f0f0ae6a1fbd14d4fe5bf5596c","revisions":{"122840cc04cd71f0f0ae6a1fbd14d4fe5bf5596c":{"kind":"REWORK","_number":1,"created":"2021-04-08 12:45:30.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/05/785405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/05/785405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/785405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/785405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/05/785405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/05/785405/1"}}},"commit":{"parents":[{"commit":"d9a6fe4362852b07770c46b8ca549ea6f1475ae2","subject":"Merge \"sharder: Prevent ValueError when no cleaving contexts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9a6fe4362852b07770c46b8ca549ea6f1475ae2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-08 08:51:40.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-08 11:18:40.000000000","tz":60},"subject":"probe test: use helper functions more widely","message":"probe test: use helper functions more widely\n\nUse the recently added assert_subprocess_success [1] helper function\nmore widely.\n\nAdd run_custom_sharder helper.\n\nAdd container-sharder key to the ProbeTest.configs dict.\n\n[1] Related-Change: I9ec411462e4aaf9f21aba6c5fd7698ff75a07de3\n\nChange-Id: Ic2bc4efeba5ae5bc8881f0deaf4fd9e10213d3b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/122840cc04cd71f0f0ae6a1fbd14d4fe5bf5596c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/122840cc04cd71f0f0ae6a1fbd14d4fe5bf5596c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
