)]}'
{"id":"openstack%2Fswift~785425","triplet_id":"openstack%2Fswift~master~I9e15762e6886b3f63f20da2452de55e699a32900","project":"openstack/swift","branch":"master","topic":"p-shard-shrinking-use-tombstones","hashtags":[],"change_id":"I9e15762e6886b3f63f20da2452de55e699a32900","subject":"Fix reclaim to use RECLAIM_PAGE_SIZE batches","status":"MERGED","created":"2021-04-08 14:05:16.000000000","updated":"2021-04-10 01:57:33.000000000","submitted":"2021-04-10 01:56:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"785425-p-shard-shrinking-use-tombstones","meta_rev_id":"e6ca69939e8fbd1691b5d345218656ee73678feb","_number":785425,"virtual_id_number":785425,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-10 01:56:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-09 17:31:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-09 17:31:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-08 14:39:30.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2021-04-08 16:36:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-08 21:21:59.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"5a077c6d23f8398cb2f062783575068ee19d2b9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-08 14:05:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7c93d62f025c86c47e6df85115beab7c2d88fe2","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-04-08 14:39:30.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nCan you elaborate on: \"caused me some confusion when first grokking it\"\n\nsqlite\u003e select * from object limit 10 offset 0;\n1|obj0083|1617892644.46749|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\n2|obj0028|1617892642.10669|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\n3|obj0309|1617892654.00094|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\n4|obj0298|1617892653.53964|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\n5|obj0095|1617892644.95953|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\n6|obj0038|1617892642.52715|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\n7|obj0105|1617892645.36678|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\n8|obj0037|1617892642.48733|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\n9|obj0180|1617892648.57317|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\n10|obj0133|1617892646.55379|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\nsqlite\u003e select * from object limit 10 offset 10;\n11|obj0287|1617892653.08703|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\n12|obj0159|1617892647.65888|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\n13|obj0232|1617892650.78772|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\n14|obj0116|1617892645.83171|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\n15|obj0306|1617892653.88072|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\n16|obj0210|1617892649.84578|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\n17|obj0208|1617892649.75629|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\n18|obj0065|1617892643.70308|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\n19|obj0055|1617892643.26560|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\n20|obj0145|1617892647.05545|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\n\nwouldn\u0027t offset 9 here be confusing?","accounts_in_message":[],"_revision_number":1},{"id":"8833f1babaebd236177eda4b292bb6f667e86cf9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-08 16:02:26.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nI found it confusing that RECLAIM_PAGE_SIZE is not the size of batches that _reclaim deals with\n\nlimit 1 offset N gives back the N+1th row:\n\n  sqlite\u003e select * from object;\n1|obj-0002|1617882917.44585|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\n2|obj-0000|1617882917.39881|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\n3|obj-0006|1617882917.50510|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\n4|obj-0004|1617882917.47870|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\nsqlite\u003e select * from object limit 1 offset 1;\n2|obj-0000|1617882917.39881|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0\nsqlite\u003e select * from object limit 1 offset 2;\n3|obj-0006|1617882917.50510|0|application/octet-stream|d41d8cd98f00b204e9800998ecf8427e|0|0","accounts_in_message":[],"_revision_number":1},{"id":"133651a74442f4d7afa4e64e2f396031c0d8b965","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 16:36:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1facd981decf4e718154ce0d36c45833 : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/43e0633a6ca3422cb9e492c524bc35de : SUCCESS in 52m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/20892cb434ec49f8b367909b4d3275cd : SUCCESS in 40m 46s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/58c9fabeb826414b993a75720f061c0b : SUCCESS in 9m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/15269ab4b10b41a0b1c6a8b76b5947d4 : SUCCESS in 7m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d63973f077e442d3b9aa081bfe196499 : SUCCESS in 17m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dd81c4c83b6c40e897077634d72978a2 : SUCCESS in 15m 33s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c6917ae8657444ab8d6880a60aa760f8 : SUCCESS in 15m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7904a248191a43ba87f13d407f5c7e96 : SUCCESS in 14m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bdc312b7d4a247a79f147db36aa8d60c : SUCCESS in 18m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c35c4810471c4da98c3758772a0754a9 : SUCCESS in 19m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9c6b49efdbbe4ae892d30159836b2bda : SUCCESS in 21m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ecc866760a40462db1e40e0ec9acb6ed : SUCCESS in 17m 50s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/099f4604469740f4a555d8affe43e59a : SUCCESS in 18m 49s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3197a00fb723469d83b9f1bd54a637e3 : SUCCESS in 14m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f0003c225f3e40d49699541e3ff457b9 : SUCCESS in 19m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f5aa5ebd411247f2a99a367a912a3f17 : SUCCESS in 7m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8065c5b3641a49ac993bde3ea98ab3d1 : SUCCESS in 24m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/36c9d4db660d4acda540fca8654c098f : SUCCESS in 52m 33s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b86648314280476d81d6ad21f813605b : SUCCESS in 1h 15m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cb867f667b4e478eaa2b0cc36d6163de : SUCCESS in 35m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/98afe8a2efa44232b0f4df5f3558d8a7 : SUCCESS in 50m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df58017f8fe14b8fa021f54b76bbc3e6 : SUCCESS in 15m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa80a68cf9874b2cb372205c0a69458f : SUCCESS in 5m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/41ee89d8f4404c7691692398b4e0a233 : SUCCESS in 29m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4733275f461a47dfb03b82d306fc7ba8 : SUCCESS in 1h 02m 13s","accounts_in_message":[],"_revision_number":1},{"id":"e02fea42d7659b64842248b60f9e91843e7b23a2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-08 21:21:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d2dd1c8ec01e02f07782d74c2094c5ddae002402","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-09 15:14:16.000000000","message":"Patch Set 1: Code-Review-2\n\nI think Tim\u0027s idea of changing the comparisons works out better - new version on the way","accounts_in_message":[],"_revision_number":1},{"id":"2980c8b209bed5efa66cd312f82f0c97b70b3645","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-09 15:42:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"21d0d4343359f69982f6832b058d241f6b341bd6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-09 15:45:17.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e92abaf42181943160ca638e2ae04939cc27b147","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-09 15:57:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1b7b68595212c260929f70b8079b92b81d82b60b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-09 17:31:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9a582b1b546e07e6c89850bf5200a2dcf13ca386","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 18:04:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67f3aab7029849aeafb499b2b1dea5c1 : SUCCESS in 10m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/4bb160d5c26047878046f603dd99345e : SUCCESS in 1h 08m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/993e73025b994311bbf83682d9ed953e : SUCCESS in 1h 20m 54s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1016ff7eb3a1410787d48a6692f1e019 : SUCCESS in 10m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a3ecbfac715f41ce94769568b9483d40 : SUCCESS in 11m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/59fa7ed1f99441b084df84088d897c0c : SUCCESS in 15m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3789f70cec404322847f1b7436683bb0 : SUCCESS in 1h 03m 43s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/af6b4a01d7f64410899f11f95fa589f2 : SUCCESS in 32m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1c84f50912df412bbacdb3a147085ff4 : SUCCESS in 17m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8844b3819a974ee9bc180783a74818e6 : SUCCESS in 16m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6570c29110494a99a1edb3c6be6686a4 : SUCCESS in 18m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/66a9fe747eda4704b995e281232bc1e2 : SUCCESS in 18m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a063d0290de54d8f97574cac8f16548c : SUCCESS in 19m 21s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c11d84e49a154fbcbcff82fa15d7db3d : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/442684517c654fc09998dc08cc6990b7 : SUCCESS in 17m 12s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/18a97af84329446b9765e9e529c0ec09 : SUCCESS in 18m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/187c8626215d4b559705657294d24fbd : SUCCESS in 7m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1048633e79ad401cb3110bc1741863a5 : SUCCESS in 22m 07s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5eb85c0f9d434e2b970d63208f4b2a3d : SUCCESS in 47m 04s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/3428adc0eccf4440bee50fee3abd2823 : SUCCESS in 45m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7405999815b34781be84e041ed1a96d9 : SUCCESS in 41m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/685ac7a9f1b847c7990ee242ff34e96e : SUCCESS in 40m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e44c0f178dc4f2fbe4fce06bb3033fa : SUCCESS in 15m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cd7616b5b6644728b343c18b3725dd4 : SUCCESS in 6m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/51e6437d60c541c180d4825a6ea8c46a : SUCCESS in 28m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48630380fe454fde99d1fe8e3bce037a : SUCCESS in 1h 05m 07s","accounts_in_message":[],"_revision_number":3},{"id":"9574b25c2aa7594e1ffe0d21f0bb07708b5527e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 18:05:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7070bcb48c0c778072be175edc1edbc55722f7a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-10 01:56:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0416a54a13c54dfe8d746e8513b3e8e5 : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/49165676792a4e9da6c4455d50f14176 : SUCCESS in 47m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/af03025cb15d47d8b12e2ee77c691bed : SUCCESS in 38m 22s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a9e1372077da4aa6802e292520033611 : SUCCESS in 9m 46s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/bbf5da76e434493b8fefe034b95a3353 : SUCCESS in 7m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0afe0b56e4b74ede877d62c0c9469415 : SUCCESS in 18m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9447d4b5e5ea4c4ca0edc52ca42a07c0 : SUCCESS in 20m 09s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f1cc3cd9b30b4d178be6ad3a783b2296 : SUCCESS in 20m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/be07b1a03fc544ae93d4dd06a6d24dfa : SUCCESS in 18m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/77297a29a283413a81964df207ba66ea : SUCCESS in 16m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/67a11e1cdaaf40f59ad1a73617dca9c7 : SUCCESS in 22m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f6ecf479ec6d488ba0f8749d818130b5 : SUCCESS in 20m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fc85dbcfc0104c2aa9cfa59bb428aff7 : SUCCESS in 21m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/79cc17d21a7d43e98cd2f1a41a5d4e16 : SUCCESS in 19m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/1286e47dc4484350a86ddd9324877a44 : SUCCESS in 37m 58s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f7c7589d7e0547478c0c2458f9abb7ba : SUCCESS in 25m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/44a37f3da1404514af1107da0726d270 : SUCCESS in 5m 36s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/3cc430ea2aa84f1ea5fe5074cd2ea945 : SUCCESS in 45m 17s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/7c45329fb4c249c0b77a7af4b24a9dec : SUCCESS in 48m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2e8db4576eed4cb38a401dcd4bbf39dc : SUCCESS in 41m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b7dfbf4632df4558ba1d096f220b1f92 : SUCCESS in 40m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a2d54dc6d9c402c92eeb28a9c3fdfae : SUCCESS in 12m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d904f293b54445928ebab286a2965497 : SUCCESS in 6m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/24e2ae9fa979487687eb401a78842963 : SUCCESS in 22m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c33824f7976046698eb8b61aa85b3756 : SUCCESS in 49m 02s","accounts_in_message":[],"_revision_number":3},{"id":"7c9b4731f9d5e02e38c4a8b3607557721d80316c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-10 01:56:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e6ca69939e8fbd1691b5d345218656ee73678feb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-10 01:57:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddc8faaefc0946e4b38570f60e1bb392 : SUCCESS in 1m 00s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/c57d5cc198e443599684c33752b8b02b : SUCCESS in 42s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/e016733834814cdf947e4c2f55697bd7 : SUCCESS in 43s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"751deb9881947eda4f5880a92cadbd0a13663c32","revisions":{"4f3ea9d6f882c862a33e1a0cd6b63bc0f77fee5b":{"kind":"REWORK","_number":1,"created":"2021-04-08 14:05:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/25/785425/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/785425/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/785425/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/785425/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/785425/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/785425/1"}}},"commit":{"parents":[{"commit":"122840cc04cd71f0f0ae6a1fbd14d4fe5bf5596c","subject":"probe test: use helper functions more widely","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/122840cc04cd71f0f0ae6a1fbd14d4fe5bf5596c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-08 13:57:00.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-08 13:57:00.000000000","tz":60},"subject":"Fix reclaim to use RECLAIM_PAGE_SIZE batches","message":"Fix reclaim to use RECLAIM_PAGE_SIZE batches\n\nThe batched reclaim would previously take batches of size\nRECLAIM_PAGE_SIZE + 1 from the database: fixed this to be\nRECLAIM_PAGE_SIZE.\n\nChange-Id: I9e15762e6886b3f63f20da2452de55e699a32900\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f3ea9d6f882c862a33e1a0cd6b63bc0f77fee5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f3ea9d6f882c862a33e1a0cd6b63bc0f77fee5b"}]},"branch":"refs/heads/master"},"66e48542dfcd3c19428b28f35bb1cda9e4385cf1":{"kind":"REWORK","_number":2,"created":"2021-04-09 15:42:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/25/785425/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/785425/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/785425/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/785425/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/785425/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/785425/2"}}},"commit":{"parents":[{"commit":"122840cc04cd71f0f0ae6a1fbd14d4fe5bf5596c","subject":"probe test: use helper functions more widely","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/122840cc04cd71f0f0ae6a1fbd14d4fe5bf5596c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-08 13:57:00.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-09 15:27:49.000000000","tz":60},"subject":"Fix reclaim to use RECLAIM_PAGE_SIZE batches","message":"Fix reclaim to use RECLAIM_PAGE_SIZE batches\n\nThe batched reclaim would previously take batches of size\nRECLAIM_PAGE_SIZE + 1 from the database: fixed this to be\nRECLAIM_PAGE_SIZE.\n\nChange-Id: I9e15762e6886b3f63f20da2452de55e699a32900\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66e48542dfcd3c19428b28f35bb1cda9e4385cf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/66e48542dfcd3c19428b28f35bb1cda9e4385cf1"}]},"branch":"refs/heads/master"},"751deb9881947eda4f5880a92cadbd0a13663c32":{"kind":"REWORK","_number":3,"created":"2021-04-09 15:57:27.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/25/785425/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/785425/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/785425/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/785425/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/785425/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/785425/3"}}},"commit":{"parents":[{"commit":"122840cc04cd71f0f0ae6a1fbd14d4fe5bf5596c","subject":"probe test: use helper functions more widely","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/122840cc04cd71f0f0ae6a1fbd14d4fe5bf5596c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-08 13:57:00.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-09 15:46:38.000000000","tz":60},"subject":"Fix reclaim to use RECLAIM_PAGE_SIZE batches","message":"Fix reclaim to use RECLAIM_PAGE_SIZE batches\n\nThe batched reclaim would previously take batches of size\nRECLAIM_PAGE_SIZE + 1 from the database: fixed this to be\nRECLAIM_PAGE_SIZE.\n\nChange-Id: I9e15762e6886b3f63f20da2452de55e699a32900\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/751deb9881947eda4f5880a92cadbd0a13663c32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/751deb9881947eda4f5880a92cadbd0a13663c32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
