)]}'
{"id":"openstack%2Fswift~785936","triplet_id":"openstack%2Fswift~master~I2a5021c4833ff713fb6d0d39f16a6d0e523e3697","project":"openstack/swift","branch":"master","topic":"p-relinker-log","hashtags":[],"change_id":"I2a5021c4833ff713fb6d0d39f16a6d0e523e3697","subject":"relinker: Add start/end logs to parallel_process","status":"MERGED","created":"2021-04-12 16:29:28.000000000","updated":"2021-04-26 18:33:49.000000000","submitted":"2021-04-26 18:32:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785936-p-relinker-log","meta_rev_id":"099fda9d34959672c1c2fe70ce5acc4dbce3b7fb","_number":785936,"virtual_id_number":785936,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-26 18:32:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2021-04-26 17:03:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2021-04-26 17:03:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-12 16:35:40.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2021-04-12 17:40:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"603a50f715b101bb427f578487d23062d69b641c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-12 16:29:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c6e53d506b6c889dbc263e645aacdd578a256ec","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-04-12 16:35:40.000000000","message":"Patch Set 1: Code-Review+1\n\nlooks like we should do this, at least until we get it all in /recon:\n\nhttps://review.opendev.org/c/openstack/swift/+/779655\n\ndidn\u0027t notice at first the tests use assertIn (so I can\u0027t tell excactly what the formatted date looks like unless I run it)  But I\u0027m sure this is going to be a big help!","accounts_in_message":[],"_revision_number":1},{"id":"d74b2d1f3672f34ff62f9c1fbd0098d8903a2c8f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-12 16:43:08.000000000","message":"Patch Set 1:\n\nfrom a probe test on my vsaio:\n\n  Apr 12 16:28:43 saio object-6030[2260]: Starting relinker (cleanup\u003dTrue) using 1 workers: 16:28:43 04/12/21 UTC\nApr 12 16:28:43 saio object-6030[5629]: [pid\u003d5629, devs\u003dsdb3,sdb7] Processing files for policy default under /srv/node3 (cleanup\u003dTrue)\nApr 12 16:28:43 saio object-6030[5629]: [pid\u003d5629, devs\u003dsdb3,sdb7] Removed /srv/node3/sdb3/objects/6733/548/d26939634cd0c03aff8927b13994e548/16182449\nApr 12 16:28:43 saio object-6030[5629]: [pid\u003d5629, devs\u003dsdb3,sdb7] Run listdir on /srv/node3/sdb3/objects/6733\nApr 12 16:28:43 saio object-6030[5629]: [pid\u003d5629, devs\u003dsdb3,sdb7] Step: cleanup Device: sdb3 Policy: default Partitions: 1/1\nApr 12 16:28:43 saio object-6030[5629]: [pid\u003d5629, devs\u003dsdb3,sdb7] 1 hash dirs processed (cleanup\u003dTrue) (0 files, 0 linked, 1 removed, 0 errors)\nApr 12 16:28:43 saio object-6030[2260]: Finished relinker (cleanup\u003dTrue): 16:28:43 04/12/21 UTC (0:00:00.090393 elapsed)\nApr 12 16:28:43 saio object-6040[2260]: Starting relinker (cleanup\u003dTrue) using 1 workers: 16:28:43 04/12/21 UTC\nApr 12 16:28:43 saio object-6040[5650]: [pid\u003d5650, devs\u003dsdb4,sdb8] Processing files for policy default under /srv/node4 (cleanup\u003dTrue)\nApr 12 16:28:43 saio object-6040[5650]: [pid\u003d5650, devs\u003dsdb4,sdb8] Run listdir on /srv/node4/sdb4/objects/3289\nApr 12 16:28:43 saio object-6040[5650]: [pid\u003d5650, devs\u003dsdb4,sdb8] Step: cleanup Device: sdb4 Policy: default Partitions: 1/2\nApr 12 16:28:43 saio object-6040[5650]: [pid\u003d5650, devs\u003dsdb4,sdb8] Removed /srv/node4/sdb4/objects/1644/859/33652bfb926ab90fafff3575c337c859/16182449\nApr 12 16:28:43 saio object-6040[5650]: [pid\u003d5650, devs\u003dsdb4,sdb8] Run listdir on /srv/node4/sdb4/objects/1644\nApr 12 16:28:43 saio object-6040[5650]: [pid\u003d5650, devs\u003dsdb4,sdb8] Step: cleanup Device: sdb4 Policy: default Partitions: 2/2\nApr 12 16:28:43 saio object-6040[5650]: [pid\u003d5650, devs\u003dsdb4,sdb8] 1 hash dirs processed (cleanup\u003dTrue) (1 files, 0 linked, 1 removed, 0 errors)\nApr 12 16:28:43 saio object-6040[2260]: Finished relinker (cleanup\u003dTrue): 16:28:43 04/12/21 UTC (0:00:00.089845 elapsed)","accounts_in_message":[],"_revision_number":1},{"id":"5ac736ea6f64dd0b00836a3e7f31a3a7a5aa8b08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 17:40:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4755283b879499da98cdf79571eb002 : SUCCESS in 11m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/8adf77e3e0b940e3a62415689ca80022 : SUCCESS in 58m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a1da4ddcec89470f8beae1715c02c72b : SUCCESS in 47m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4fab4cbd9cfb476fa6439a50bfeaf69a : SUCCESS in 10m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/299ad4375a1c412b8a38af9ff60c2eb0 : SUCCESS in 7m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/24fb8e0ea52b4c4cbb22be46c8fa7435 : SUCCESS in 27m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/52fa72176b3145f790835d9ef0ab674d : SUCCESS in 18m 10s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2ffe07dbe89740148303395abcd793f3 : SUCCESS in 15m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8d25f1034f9d4eb89330c28e24cc6715 : SUCCESS in 28m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1ecc96307aea4f0c9610cfe776d8e11a : SUCCESS in 18m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/062e33ded767451796600d2321700afa : SUCCESS in 17m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8ea0268d106e40578b55f99bdcf3df2d : SUCCESS in 30m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9eeb34c1e4354fc0a6cc508d8b4c4757 : SUCCESS in 21m 17s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5239693f98cb45f39676f18ba9265bf6 : SUCCESS in 19m 00s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/fa207fd1f35d4797b0a47e9e29acde0f : SUCCESS in 17m 08s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8ac511809573426a87ba0776eb5f766f : SUCCESS in 21m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2da7163e779a420e970722f1c66a8414 : SUCCESS in 6m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5b26c54f341c4134b0537294512c2078 : SUCCESS in 21m 53s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/57d47d21deb44f9eb7ddc0c33405a074 : SUCCESS in 51m 47s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/690b26d0b7c64c9ca65b23638ba9a756 : SUCCESS in 44m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ddf7bc06763244e1a19916eec6aa022e : SUCCESS in 44m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/be800015849747eeafc53691b01f5525 : SUCCESS in 39m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf1322fef11b498aaa479225eb4043e3 : SUCCESS in 17m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb9aa8f2c841404aa5f8d08b3fe6d5df : SUCCESS in 6m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/14a692238c9842ee8db8bda7e870a1c6 : SUCCESS in 24m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fa7b729501a48e9a9c284e0971bf769 : SUCCESS in 1h 08m 53s","accounts_in_message":[],"_revision_number":1},{"id":"82e7ed2f05c2b99d2884e5b1ddc254023f0ea2ab","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-04-26 17:03:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"01ddcdb351460a768105f3aacc54a2ba0578add7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 17:04:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d0b229d40e4775706b615e92e2e929e297579ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 18:32:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8130af8fe1a49dfadf1fbd5fa1a039f : SUCCESS in 10m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/757d473fb06148b38f1d2efd554767c6 : SUCCESS in 1h 21m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5fc10da3afef4808bb962fd390db75e2 : SUCCESS in 47m 00s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ed1ab9e8b49b4a4792a37d9bb146f163 : SUCCESS in 10m 23s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e8c444fd52b844e9b349bf4f9eb58e73 : SUCCESS in 7m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1b07806060414950bba8d97d4748fd1a : SUCCESS in 18m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fcc5276ea4cd47968ae0ca2592ee805f : SUCCESS in 16m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0e128a7418ec49ebb24050f35598fb7f : SUCCESS in 16m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a8f4dbcf744d4572b15be972f2538a65 : SUCCESS in 16m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c88fb08c5f654baa947ad4742f5734d1 : SUCCESS in 16m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9ed55bc51d60426ca0a114df59296910 : SUCCESS in 18m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/69add22b923e453d987b1811a8b59510 : SUCCESS in 18m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/723a1af0780d4b519c3c52a1749e6af1 : SUCCESS in 17m 29s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f8f25ae76d5b4ec0aff7be85beecff5e : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/a6fb5049819e4a15bddfaf38a4d97b4a : SUCCESS in 17m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/ec3c1aedb1544e5283bd77b03ba94011 : SUCCESS in 19m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/986f115f3d394dcfbe3740153c30c775 : SUCCESS in 8m 15s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/5c067c7582c549a28b98abc2b97a71bd : SUCCESS in 50m 02s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/6339ed74c7bc49268e88a21a202c2741 : SUCCESS in 51m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/48058a9d236349dcb63d529d7a1bf008 : SUCCESS in 42m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/471a9054ea8e42028d5de88cefc623ac : SUCCESS in 45m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3395f0344cc34b38a16756b1537347bf : SUCCESS in 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57e7a850c63e4e46968162bb23c2b33e : SUCCESS in 6m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/670d55240dbf4ead85066cbb284adb67 : SUCCESS in 30m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f63ba38d68ab4d41918813a271074943 : SUCCESS in 1h 02m 01s","accounts_in_message":[],"_revision_number":1},{"id":"8c44d14f2f7175376258bc006c825d1631cbc578","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 18:32:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"099fda9d34959672c1c2fe70ce5acc4dbce3b7fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 18:33:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f47ce6bc3324c778475146c1ef431b9 : SUCCESS in 1m 08s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d661f5575a3a450fa88ebbef76abe236 : FAILURE in 47s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/1e15af033f37444aaddc0b00faea051a : SUCCESS in 47s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e76ba21077a97543a11c2c0a1b65424d581bb265","revisions":{"e76ba21077a97543a11c2c0a1b65424d581bb265":{"kind":"REWORK","_number":1,"created":"2021-04-12 16:29:28.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/36/785936/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/785936/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/785936/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/785936/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/785936/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/785936/1"}}},"commit":{"parents":[{"commit":"c0fad7714e1c0221b15c8cf3d5f4aa74f18b2f33","subject":"Merge \"Fix reclaim to use RECLAIM_PAGE_SIZE batches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0fad7714e1c0221b15c8cf3d5f4aa74f18b2f33"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-12 16:08:43.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-12 16:26:39.000000000","tz":60},"subject":"relinker: Add start/end logs to parallel_process","message":"relinker: Add start/end logs to parallel_process\n\nChange-Id: I2a5021c4833ff713fb6d0d39f16a6d0e523e3697\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e76ba21077a97543a11c2c0a1b65424d581bb265"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e76ba21077a97543a11c2c0a1b65424d581bb265"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
