)]}'
{"id":"openstack%2Fswift~786084","triplet_id":"openstack%2Fswift~master~I552d23ab60954ac32fe31e7f85e6d9e58fa25da3","project":"openstack/swift","branch":"master","topic":"p-lonely-frag","hashtags":[],"change_id":"I552d23ab60954ac32fe31e7f85e6d9e58fa25da3","subject":"Quarantine stale EC fragments","status":"ABANDONED","created":"2021-04-13 17:43:41.000000000","updated":"2021-05-11 09:03:07.000000000","total_comment_count":41,"unresolved_comment_count":23,"has_review_started":true,"meta_rev_id":"21a9f629ae4facfbc10aa9ba7aa22454f72af7da","_number":786084,"virtual_id_number":786084,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-04-29 23:24:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":-2,"date":"2021-04-30 17:24:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-13 18:24:02.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2021-04-13 18:44:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-15 07:37:15.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2021-04-19 23:11:46.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"0368e68040b78827830905fdc8c5390a1e4c3695","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-13 17:43:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"416dda5fe6e158b6c04a1512d5b413c431861d72","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-13 17:44:45.000000000","message":"Patch Set 1: Code-Review-2\n\nWIP:\n- is this the correct approach?\n- more tests needed","accounts_in_message":[],"_revision_number":1},{"id":"15b615f1323f2eb9aa79e006b56b2906d98fe4c1","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-04-13 18:24:02.000000000","message":"Patch Set 1:\n\n(6 comments)\n\nI think it\u0027s probably the right direction!","accounts_in_message":[],"_revision_number":1},{"id":"f1c587326aecb2fe9b5589bb325e0d9df7b684ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 18:44:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beebcd0cedbc4b009de55dceb6460a61 : SUCCESS in 12m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/d77f5964856e4da29e72d187ec493b4f : SUCCESS in 58m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e178e3d503a247f8ad9d3b44ee094e03 : SUCCESS in 47m 49s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ebecc085344044af98578ea071ffecb9 : SUCCESS in 9m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7634bf53711f4abaa25e2b32bfaccfea : SUCCESS in 8m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/35d83d606b094033921bfe99bfecb3e2 : FAILURE in 18m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9de58a1a6fbd41819f213bff7636be9a : FAILURE in 18m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4198978dc46a44028f65bee5b5d889f0 : FAILURE in 16m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6f30905328944f18a9c1674cf8f6a50c : FAILURE in 18m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f684415d289b495b803724af2ad7cf5a : FAILURE in 17m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e73cad1bd9ad46659aeca09ea534649e : SUCCESS in 20m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/beca831c422846edb0f8be36256486a7 : SUCCESS in 19m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1d131fd3b177407f9956ed1b92ab7479 : SUCCESS in 21m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ad6d5f54ff0b40c595dceb96b197eb4b : SUCCESS in 19m 59s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cef35da9cde64419b14f5151b2a4cbeb : SUCCESS in 21m 35s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/88f2a27f98884d1aa226911a0d2c7170 : SUCCESS in 19m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b78ff017887c447c99484a4a034fd8c5 : SUCCESS in 6m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8785bed26ba948b0996b4cbd385ebb5d : SUCCESS in 23m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/288df6873747446685ee8ff140bcf359 : SUCCESS in 49m 07s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/e74e5f46cd4b4854ba1e8142138efa8f : SUCCESS in 49m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fe1e1e24cfa2474a809d602088809b64 : SUCCESS in 41m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/217042acc88547cbbebb478e6476cca2 : SUCCESS in 39m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ec88d2386b6496ca0a92f7b880e50a9 : FAILURE in 17m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/452d2e97d9724c5b9277d6a202097a7d : SUCCESS in 6m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d0d16a3837964f30981ad60a6de85899 : SUCCESS in 30m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12a4ac93d90c42989073207b0dab987d : SUCCESS in 53m 49s","accounts_in_message":[],"_revision_number":1},{"id":"4b4f67c0bc95675d4c6eeb6e326d839c2f8eba86","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-15 07:37:15.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nSorry got distracted talking to banks and then with more aggregate stat fun in the relinker recon drop. I\u0027ve had a quick review of this, it\u0027s looking pretty cool.\n\nWant to dig into it more tomorrow.","accounts_in_message":[],"_revision_number":1},{"id":"74bb6da7f7163658cfed51279e3ec7f91f0ca6ac","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-16 06:55:01.000000000","message":"Patch Set 1:\n\nI fixed some stuff up here: https://review.opendev.org/c/openstack/swift/+/786575\n\nMostly fixed up most other tests, added another to test lonely_frag\u0027s 2-9 and make the reconstructor changes py3 compatable. Well at least py39 which I\u0027m running\n\nIt\u0027s meant to be suqashed in.\n\nI really like how you go another unique frag count into the handoffs! great work Al.","accounts_in_message":[],"_revision_number":1},{"id":"33379fc778e0612ec380bb9521c895d9d55b11bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-16 15:52:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"39ee3f1c7385183166b880201924b1cb495e9f8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 17:08:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c77de3b70834a0a8bc627ac2bf06659 : FAILURE in 6m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/4b779c6b2be3404bb63c53d218b5f745 : SUCCESS in 1h 05m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9f8b4f093274413c99434e3f89da3005 : SUCCESS in 43m 31s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/77b21aa969454f53be940905bc3a60a6 : SUCCESS in 9m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a46ba9374bc74113a1bf16df73024b43 : SUCCESS in 8m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/32da067a281e4ab6b79337184aaf89fc : SUCCESS in 14m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b28cbf64b4b44827a2932f80c19c6d62 : SUCCESS in 15m 15s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/72fed2e7f3ef4f61826885d0d8c2c523 : SUCCESS in 15m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2c69c75e8abb49389f6ad5529448d279 : SUCCESS in 14m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/01e55e756f204aaaaccb6b5e220d0726 : SUCCESS in 15m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/656e29e111fb498c8bfd00f6ffdc402b : SUCCESS in 14m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/48090dea812b464a88665ba90b7eb048 : SUCCESS in 17m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6c6156027b8f4a07a7ef18224ba6d31a : SUCCESS in 16m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/5368d24e6b9d46c6bd3e1419d1029dd4 : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/949adf18deb044cf9da2065091842eb8 : SUCCESS in 17m 31s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8e39417539cf4224adc83d12eddbc95c : SUCCESS in 23m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e9b182e1e25a403c979a13f626547ec7 : SUCCESS in 6m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/93cf189fc74743d3b25ed8f5b6fda88d : SUCCESS in 19m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8cca2087db1f4299873462646efc4a2e : SUCCESS in 48m 19s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/dc586d07d2e4448ca97361db0a20a4d1 : SUCCESS in 48m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4da9cd0f3f39460d9d44de3706bbec91 : SUCCESS in 37m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/129ac30ccacb460397303c52241a380e : SUCCESS in 37m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5bd3cff6d54245d98900645d0ee5b2d1 : SUCCESS in 20m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aae84eb4f7fc4daabdc073e072a79197 : SUCCESS in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/84778253e9c647c197bc25993c009ebd : SUCCESS in 23m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e20493a3e5d42f78735b4836da706a6 : SUCCESS in 1h 06m 30s","accounts_in_message":[],"_revision_number":2},{"id":"3b054532d2d2213e58f060d37870c6a636c0ef4f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-19 12:13:14.000000000","message":"Patch Set 2:\n\n(5 comments)\n\nJust a few inline comments. It\u0027s late so might need to make them with a grain of salt 😋","accounts_in_message":[],"_revision_number":2},{"id":"d6f8f2c62bcf560601ce62333e34e4ed3b6de994","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-19 23:11:46.000000000","message":"Patch Set 2:\n\n(9 comments)\n\nJust a first pass; wanted to see what happens with network partitions and node unavailability in general.","accounts_in_message":[],"_revision_number":2},{"id":"6abf8ee266f9c80188902b33ee91279ccb80ea6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-20 18:40:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ad3a705245e4f26f9800a9eb3e9a16ab3e6295b1","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-20 18:41:42.000000000","message":"Patch Set 2:\n\n(11 comments)\n\nThanks for the reviews - I have addressed some but not all comments in next version, and will continue working on the rest. Just pushing latest off my machine.","accounts_in_message":[],"_revision_number":2},{"id":"8827eade5a93c71cc57e3dc2b08c028016e6fb99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 20:43:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bb122ab7339471c8e24ab3f82445f5f : FAILURE in 6m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/cf259dd681b54a708edd313a9ec357cd : SUCCESS in 1h 06m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bb3509e15c61433fa97890252f2428f2 : SUCCESS in 50m 10s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0829fd31bf8a4e06bf9bce46ff77aeb4 : SUCCESS in 9m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/642510501eb148a5bcd78a135a65b107 : SUCCESS in 6m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fdecd863b86e4a5096f55f4c7e29f2ca : SUCCESS in 18m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2dc48a15fe51488091a1c001510c9d98 : SUCCESS in 13m 36s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b27b557457a84ffb99fc8ec21b5a4e37 : SUCCESS in 13m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f5d954d3859f4a0c8ac0221ff07b4c14 : SUCCESS in 13m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/14eef7f5b4fe4fe287c392a28a3d1e23 : SUCCESS in 27m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5e58d149d90d4dfdac1baa3407a24d9b : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f0f533db7c5e40b2874b79d0dad9672f : SUCCESS in 21m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d0fcadba605044c1819dc8226274bc37 : SUCCESS in 14m 04s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ab8c913fc4944351ac8c9a3626d4147d : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/688d0f15b77b46df951528497d90768e : SUCCESS in 17m 14s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/65a2e7b2049848e388eb3f9f9ff61e05 : SUCCESS in 33m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2923fe501a274f4db8c5f87f16b57038 : SUCCESS in 9m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4a279aaf8201401d804ed6d17358ffcf : SUCCESS in 25m 17s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0c4a27991d4b40909e042da8a6376d80 : SUCCESS in 2h 00m 45s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/561cb89999c348f9b8a319af949143e1 : SUCCESS in 50m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/72df02a7d8964aa39b7dbd6cdf69e212 : SUCCESS in 48m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8837ad4ca8ec49818f1dedcd05111545 : SUCCESS in 31m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae2a53387d754fada7ad71cb003c59c2 : SUCCESS in 15m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbf5368a94ac4c77a5e9f539e23d7ffb : SUCCESS in 6m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/81417134a6184fe0a9e4d0b4e3bec7e6 : SUCCESS in 33m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/926528ef723c4142bf3a1834b2069811 : SUCCESS in 1h 03m 37s","accounts_in_message":[],"_revision_number":3},{"id":"96f14a3c3c0ef1afdc13266d94c7b904b05fa680","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-21 13:08:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f525c8a956ea77d083c4aebf68cc1ba5c0617c3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 14:05:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26527faa4c2a4c25972b4d4470d4e2a5 : SUCCESS in 9m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/9087a152c3cf416a823357513042447f : SUCCESS in 53m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/114d73140f684322a335faa21a01fb09 : SUCCESS in 31m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/167ae8c290b6463d96df37d993bf84cb : SUCCESS in 9m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7fe27e4595cd4ce9a948cc85808c5d2d : SUCCESS in 7m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d46f2da500a14295bc318e2c106cf708 : SUCCESS in 18m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/62ed07ab528047c7b640a28646e3bf10 : SUCCESS in 21m 31s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8582e628a0694814b80385604c1bc1e7 : SUCCESS in 19m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6400601bf7fa4332836933827263dbac : SUCCESS in 18m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e6481afcf0424d7594b8a17c2c475b80 : SUCCESS in 20m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9a3bd941569742fbae76af79d3d6fbb6 : SUCCESS in 22m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4e50eae5983540bbba199ddc2b72c127 : SUCCESS in 20m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9f5eccb75a244989b3e6d4c5c114f95e : SUCCESS in 20m 55s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4b04c1260d014ff6b3d9b76b8a1cfb07 : SUCCESS in 21m 28s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ffa3003778cd467a9f72eb22c69fc6f0 : SUCCESS in 19m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/e1497d12e40046fe8a2ffc82372e4a9d : SUCCESS in 20m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/62c54f6cf33f4b7abd4242bdb0f39064 : SUCCESS in 9m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/965fcff70896440b8fd87de5581c1aa3 : SUCCESS in 21m 45s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/aa40cd0ba5cd46d084b80079ee2bbf9e : SUCCESS in 49m 20s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/ece86a598c2c4d5ba8c9f3feb310c3c8 : SUCCESS in 49m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ef190d8ff8ca4f1485958b456e1ddec3 : SUCCESS in 42m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3b9cb719732245db9c5d07bc86fb7f88 : SUCCESS in 38m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97834c94aa794443b02c52d62b6e7f7c : SUCCESS in 16m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c0adb3f65884c5291fb2d87ea4431a7 : SUCCESS in 6m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3eaa840edfe7446196017060704a7342 : SUCCESS in 27m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e95959bc410746b18ad48d441dac652a : SUCCESS in 48m 41s","accounts_in_message":[],"_revision_number":4},{"id":"76bade1a93b756cc0e5146910098ab8a9030a4ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-21 18:30:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"50b061c8688e6e10b150b89cd3c53618c8b6eaed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 19:32:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e57eaad47c44f1f9f9becf2d6d76a72 : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/38df8b37f235403ca492720188447762 : SUCCESS in 58m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3e980c50e0f04601a3fef670025499c0 : SUCCESS in 37m 30s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a3604dea82da448896d164645bcf0592 : SUCCESS in 8m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6139c6c6332a455d84a14d53bbef6597 : SUCCESS in 6m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/732f71e987b4496aa5ecffdee0ab9e50 : SUCCESS in 15m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/11490ca3a02c4c3cbfde01703be98ec9 : SUCCESS in 18m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/029d197e938e4985bf1dc24baa77ae2c : SUCCESS in 19m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/023e2d0544ba4f209daa00ab51e840a8 : SUCCESS in 18m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bb45d4d83bf94c76a287b423f3df5391 : SUCCESS in 20m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/be3a438481d341a0bcde7d44e75f5614 : SUCCESS in 20m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e2c1e4f7e24b4a839319ea547841ff72 : SUCCESS in 21m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9d7e0241aaa94f7e9e3d457bd0c70a44 : SUCCESS in 21m 48s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/d2d256e676eb4b8e997c245b480cd9ac : SUCCESS in 21m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/54c3a55015cd4586a68ca72acb55fbd3 : SUCCESS in 20m 55s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8b4fa899a5664cbf9b6376a46053c5bd : SUCCESS in 21m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1d1ef463d52444beabae4962f72a3cd1 : SUCCESS in 7m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/724dc59831ec41b5b85c47839390687a : FAILURE in 24m 22s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4f5772e029f64ad7bd6cc3d27c94cc54 : SUCCESS in 54m 12s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b47878b3a3624bf0ac0c1a5d2fcb5a6d : SUCCESS in 50m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a362ba62e9ea48ef9afdc7c537f4c6a2 : SUCCESS in 43m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6786178e1a664ca69be1b388e2a3c0ef : SUCCESS in 40m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/badd40e21e7442a49883191cd4f166e1 : SUCCESS in 16m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d89d29f063e341428f11b187e7f36311 : SUCCESS in 6m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8bcd594d63d54136b57329ac8d210667 : SUCCESS in 30m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f01fb2244744e8e9d867cd320a069f9 : SUCCESS in 55m 31s","accounts_in_message":[],"_revision_number":5},{"id":"982bbc39e2e250de70fe6d13b199435176b57fb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-26 15:32:10.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"96c1922297f973998c9b2e9ad92fd4222f2553c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 16:37:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9ab60b89ea04e81ad1713ced87868b7 : SUCCESS in 9m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/f94c0a2ce50a46c481ff2e22f6dc2655 : SUCCESS in 58m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/07d378947637471d8bf872f668ed22f4 : SUCCESS in 39m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0efac57693f449008d802de957c3cc9e : SUCCESS in 8m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ee0c3096dfb745a098271fdd7f09b48b : SUCCESS in 7m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e2dac47574444478a4ed93f8d93e99f5 : SUCCESS in 20m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e509754521264551bff0be236b0250db : SUCCESS in 22m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0568d1d032234f19a00e5cd627b9a30d : SUCCESS in 17m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4a69539e11224dae95c2c91e0ee2e877 : SUCCESS in 22m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/46e37177c15048d5a928d61b0cc556be : SUCCESS in 20m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/52bc5622456a46e0844b69a72e5477f7 : SUCCESS in 22m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f51aa325c3f74b4ead2d11cb2f96e0f0 : SUCCESS in 22m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6802d88feb494876b7e6c1174b9a8249 : SUCCESS in 19m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ababe72193104b92bdafa51cd6ec2a35 : SUCCESS in 19m 23s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/4e752d25d8e940caaf04bf80c767afba : SUCCESS in 18m 13s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/146e624a211c41d1a71c499010b5c9e6 : SUCCESS in 19m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b91755873d14476ea006564132ca5d76 : SUCCESS in 7m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d5c998ea224c44518b38d0bad93da9ab : SUCCESS in 18m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/88206c1f0c0f419885a4e1741d556f6b : SUCCESS in 1h 00m 07s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/1d4dc64c17e141cf903bf601888b34e5 : SUCCESS in 52m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5e7e4cd9b6a0428485c5665161bf87e5 : SUCCESS in 31m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/23b04d13cd7f4f59bce44a54a91e0049 : SUCCESS in 41m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0cac77fdfee4f1bb539094f104015ce : SUCCESS in 15m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/679af9c5c6ee457ba4de8ebfe91a3926 : SUCCESS in 8m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0f980c4ebfd74377aa9a0180575b6248 : SUCCESS in 30m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1589cee0ed984f6d88e6e2fdf24a1688 : SUCCESS in 49m 53s","accounts_in_message":[],"_revision_number":6},{"id":"548879437cab7d4186d1346623bf0771b2bc4d64","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-26 16:55:55.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"e7e914490c8f75ea06956548449e681c4d5e15fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-27 11:02:52.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"69bc4c0e4701ef1990d77f7987215f7757b649e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 13:19:26.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40abd873e2964de3b0383cd4c925836a : SUCCESS in 10m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/7fdffab4024b4bd7863f49b58710a0c7 : SUCCESS in 1h 05m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1625837f71e847ea9b2387d504b53384 : SUCCESS in 48m 28s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/67e40314a3f243258eddfa922680cf34 : SUCCESS in 9m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/712a3255219e405bbda0fadd68c15c09 : SUCCESS in 8m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/96460ad538584d6d95671c3d0a126b13 : SUCCESS in 15m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1087b1e6e4854fd08e2a8f5f6bfde06b : SUCCESS in 19m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/109f9f93a23840deafed0b7fb1fe2ff8 : SUCCESS in 19m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/20035b37578e4c49abb2c70f3fa58837 : SUCCESS in 13m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/89f365ac6fd7408caf7711ec4a7ebb53 : SUCCESS in 15m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/75be2e6e1ab24c8ea69467dcaefe1d9a : SUCCESS in 18m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0bc6c5f564144c76b5f304285930b7b4 : SUCCESS in 40m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f55e7f62d0ad4f29a847f1cc9bde3cbf : SUCCESS in 16m 30s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8bc253b78c7a4aa6b695927936f7919b : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/20a3f8323917466080c2a11cc0377f42 : SUCCESS in 16m 47s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/089465fba1b54f9ba2acd625c5a7c47a : SUCCESS in 16m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d1cdc400453849e595c64332e64bdb86 : SUCCESS in 7m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/73d0536882c44ff9a66df115a5db6ec4 : SUCCESS in 20m 06s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/039f6043647c4a41bf7642c527a3c3bb : SUCCESS in 50m 11s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/111ddc8025cb464aac22b229cc31857d : SUCCESS in 47m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/756e2502a9744037aeb537baf5a6eaf8 : SUCCESS in 40m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/afe0d13e8446404ea9e4e3c07e7e4032 : SUCCESS in 40m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07dcaa7f228b40578771ccfd16ebee18 : SUCCESS in 15m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac3a710fa4eb4b6396a4c1bf1ecb96b3 : SUCCESS in 8m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1ec56c43665543ab820f88002c29516e : SUCCESS in 31m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b691e56c9874b798028d69e8638d56e : SUCCESS in 1h 00m 06s","accounts_in_message":[],"_revision_number":7},{"id":"77cb10d0cfefa30a7ea1cf09e045b068e465af00","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-28 07:44:48.000000000","message":"Patch Set 7:\n\n(1 comment)\n\nonly part way through a reivew.. but need to go pick up the wife from the station and then deal with kids. Hope to get back tonight or in the morning to finish the reivew.","accounts_in_message":[],"_revision_number":7},{"id":"12e272ee53bfc7ce182f4b2dd7347217148a2407","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-28 21:49:21.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"83af56ee477c6d0c6a2fe660d0e056dcb46ee8bf","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-28 22:07:07.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"98715a9cb3faedec2669799b2aea77579896eb97","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-29 07:37:27.000000000","message":"Patch Set 7:\n\n(4 comments)\n\nThis is looking really great Al! \n\nI just want to have a play on my SAIO before a properly vote.","accounts_in_message":[],"_revision_number":7},{"id":"fc897a6620c9b29e8d2f7b8dcb0c94790d9e6ac7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-29 22:12:16.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"ad79250921b30d5bfa407917f5e76bf7c32e2841","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-29 22:14:26.000000000","message":"Patch Set 8: Code-Review-1\n\nrebased, but I have tried a different approach over on https://review.opendev.org/c/openstack/swift/+/788833 so -1 for now","accounts_in_message":[],"_revision_number":8},{"id":"eb409a2f74e9cfeb98436ee9493f0c8780f43598","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 23:24:43.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f90bc96d6104f50b71dd8d277acff4d : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/8ab96ece8f094b2a9c82e929ed98b128 : SUCCESS in 1h 06m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a3a930739e854ab58ebb7cce801da199 : SUCCESS in 45m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e6c7cc9c0a7d41d2920d22f2f696bfde : SUCCESS in 9m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/07486fdf17474d6d95f450286e00a984 : SUCCESS in 6m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6dd077ee6983402da0cd1cc1e50a9e72 : SUCCESS in 14m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dc8497ee59cc4804a96813c05748b1cc : SUCCESS in 14m 45s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/06f438f769594209baacc107af5fcb10 : SUCCESS in 14m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/12a6446b9aee48c2943bb0ad6cb0458e : SUCCESS in 14m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4bfaf16929d4466d897887f4bafb2973 : SUCCESS in 14m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/299f74eeafd54cedaf2f2d2478c55bad : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/73b7e50da36648a2997b4f69973900aa : SUCCESS in 17m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2a416586ced04905bf75192c7c349a26 : SUCCESS in 16m 39s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b43fcae0ae0c45539bc12c0ffc1eba47 : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ca720fafccfe4fab989843b15be1df0b : SUCCESS in 16m 36s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/dba07414380043f7a96a123fdbaf4194 : SUCCESS in 17m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/21f3e3effce443948eb1b8478207501c : SUCCESS in 6m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/65d186e7dfe1462aba3aba083fcbf3f1 : SUCCESS in 18m 44s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fc47d0c04ea447b68e9b1b203674cd4c : SUCCESS in 45m 11s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/5ea7bee691ff46dbac0490f46e470704 : SUCCESS in 48m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fc7d817939174e61ba41a2e5c833b1b1 : SUCCESS in 35m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/df3991bbc0ab41a19003d8345f2d06d5 : SUCCESS in 39m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27cedbeb5e094409b870225693db9dd9 : SUCCESS in 14m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae5ed39f7f32442483c32657a495fefb : SUCCESS in 6m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/39de9005a5d1478483d96f3bf1e09602 : FAILURE in 20m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30da4b1b53aa438fa4867a84abf3a4fb : SUCCESS in 50m 10s","accounts_in_message":[],"_revision_number":8},{"id":"6634d33676762aa2314c3170be4673ea6e165756","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-30 07:07:17.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1d9ffc83a8b44c5eed5faf20e76326a319ba3097","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-30 17:24:27.000000000","message":"Patch Set 8: Code-Review-2\n\nmy current is the alternative here https://review.opendev.org/c/openstack/swift/+/788833","accounts_in_message":[],"_revision_number":8},{"id":"21a9f629ae4facfbc10aa9ba7aa22454f72af7da","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-05-11 09:03:07.000000000","message":"Abandoned\n\npreferred fix is https://review.opendev.org/c/openstack/swift/+/788833","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"2eb5ec86fc09f795450fdb93226377ba0b05e9fc","revisions":{"db6ab15605ecf4baafaa8e5879c0d52aa1f89f28":{"kind":"REWORK","_number":1,"created":"2021-04-13 17:43:41.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/84/786084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/786084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/786084/1"}}},"commit":{"parents":[{"commit":"c0fad7714e1c0221b15c8cf3d5f4aa74f18b2f33","subject":"Merge \"Fix reclaim to use RECLAIM_PAGE_SIZE batches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0fad7714e1c0221b15c8cf3d5f4aa74f18b2f33"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-13 15:45:25.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-13 17:43:30.000000000","tz":60},"subject":"WIP: Quarantine solitary EC fragments","message":"WIP: Quarantine solitary EC fragments\n\nCloses-Bug: 1655608\nChange-Id: I552d23ab60954ac32fe31e7f85e6d9e58fa25da3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db6ab15605ecf4baafaa8e5879c0d52aa1f89f28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db6ab15605ecf4baafaa8e5879c0d52aa1f89f28"}]},"branch":"refs/heads/master"},"41915f14add2c1203acd1bbfcc136076e9924867":{"kind":"REWORK","_number":2,"created":"2021-04-16 15:52:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/84/786084/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/786084/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/786084/2"}}},"commit":{"parents":[{"commit":"c0fad7714e1c0221b15c8cf3d5f4aa74f18b2f33","subject":"Merge \"Fix reclaim to use RECLAIM_PAGE_SIZE batches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0fad7714e1c0221b15c8cf3d5f4aa74f18b2f33"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-13 15:45:25.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-16 15:49:00.000000000","tz":60},"subject":"WIP: Quarantine stale EC fragments","message":"WIP: Quarantine stale EC fragments\n\nTODO:\n - add internal client call to check container listing\n - improve logs to detail what mix of bad responses we got\n - unit test mixed timestamps\n - add probe test\n\nCloses-Bug: 1655608\nChange-Id: I552d23ab60954ac32fe31e7f85e6d9e58fa25da3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41915f14add2c1203acd1bbfcc136076e9924867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41915f14add2c1203acd1bbfcc136076e9924867"}]},"branch":"refs/heads/master"},"0aa76dc493eca9e3763621ca27ae13b83cd6e437":{"kind":"REWORK","_number":3,"created":"2021-04-20 18:40:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/84/786084/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/786084/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/786084/3"}}},"commit":{"parents":[{"commit":"be3a069def81c66840a3655bd9a7a6e89c469482","subject":"reconstructor: log more details when rebuild fails","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be3a069def81c66840a3655bd9a7a6e89c469482"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-13 15:45:25.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-20 18:40:30.000000000","tz":60},"subject":"WIP: Quarantine stale EC fragments","message":"WIP: Quarantine stale EC fragments\n\nTODO:\n- make number of handoffs configurable\n- unit test with mixed timestamps\n- add probe test\n\nCloses-Bug: 1655608\nChange-Id: I552d23ab60954ac32fe31e7f85e6d9e58fa25da3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0aa76dc493eca9e3763621ca27ae13b83cd6e437"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0aa76dc493eca9e3763621ca27ae13b83cd6e437"}]},"branch":"refs/heads/master"},"05d3963e3b620cce2122512e1193f5b9ead372e5":{"kind":"REWORK","_number":4,"created":"2021-04-21 13:08:06.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/84/786084/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/786084/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/786084/4"}}},"commit":{"parents":[{"commit":"a7673463a1cb4d64dfabe8f2bb1c420e60094542","subject":"reconstructor: log more details when rebuild fails","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7673463a1cb4d64dfabe8f2bb1c420e60094542"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-13 15:45:25.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-21 13:07:30.000000000","tz":60},"subject":"WIP: Quarantine stale EC fragments","message":"WIP: Quarantine stale EC fragments\n\nTODO:\n- unit test with mixed timestamps\n- add probe test\n\nCloses-Bug: 1655608\nChange-Id: I552d23ab60954ac32fe31e7f85e6d9e58fa25da3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05d3963e3b620cce2122512e1193f5b9ead372e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05d3963e3b620cce2122512e1193f5b9ead372e5"}]},"branch":"refs/heads/master"},"c4540e58aef03352fbe0cc9e637a26144cdf0074":{"kind":"REWORK","_number":5,"created":"2021-04-21 18:30:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/84/786084/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/786084/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/786084/5"}}},"commit":{"parents":[{"commit":"a7673463a1cb4d64dfabe8f2bb1c420e60094542","subject":"reconstructor: log more details when rebuild fails","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7673463a1cb4d64dfabe8f2bb1c420e60094542"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-13 15:45:25.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-21 18:26:36.000000000","tz":60},"subject":"WIP: Quarantine stale EC fragments","message":"WIP: Quarantine stale EC fragments\n\nTODO:\n- unit test with mixed timestamps\n- probe test?\n\nCloses-Bug: 1655608\nChange-Id: I552d23ab60954ac32fe31e7f85e6d9e58fa25da3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4540e58aef03352fbe0cc9e637a26144cdf0074"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4540e58aef03352fbe0cc9e637a26144cdf0074"}]},"branch":"refs/heads/master"},"660c676ac9561d164b450df16841d87e70ed1b6f":{"kind":"REWORK","_number":6,"created":"2021-04-26 15:32:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/84/786084/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/786084/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/786084/6"}}},"commit":{"parents":[{"commit":"a7673463a1cb4d64dfabe8f2bb1c420e60094542","subject":"reconstructor: log more details when rebuild fails","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7673463a1cb4d64dfabe8f2bb1c420e60094542"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-13 15:45:25.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-26 14:28:31.000000000","tz":60},"subject":"Quarantine stale EC fragments","message":"Quarantine stale EC fragments\n\nIf the reconstructor finds a fragment that appears to be stale then it\nwill now quarantine the fragment.  Fragments are considered stale if\ninsufficient fragments at the same timestamp can be found to rebuild\nmissing fragments, and the number found is less than or equal to a new\nreconstructor \u0027quarantine_threshold\u0027 config option.\n\nBefore quarantining a fragment the reconstructor will attempt to fetch\nfragments from handoff nodes in addition to the usual primary nodes.\nThe handoff requests are issued in batches equal to the total number\nof fragments for the policy. The maximum number of batches issued can\nbe set using the new reconstructor \u0027quarantine_handoff_batches\u0027 config\noption.\n\n\u0027quarantine_threshold\u0027 defaults to zero i.e. no fragments will be\nquarantined. \u0027quarantine_handoff_batches\u0027 defaults to 1.\n\nCloses-Bug: 1655608\nChange-Id: I552d23ab60954ac32fe31e7f85e6d9e58fa25da3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/660c676ac9561d164b450df16841d87e70ed1b6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/660c676ac9561d164b450df16841d87e70ed1b6f"}]},"branch":"refs/heads/master"},"60dae671b93a48cdde71e1ed5037d91d0cc79647":{"kind":"REWORK","_number":7,"created":"2021-04-27 11:02:52.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/84/786084/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/786084/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/786084/7"}}},"commit":{"parents":[{"commit":"7960097f02bab5a4568eddb6b98d7b8b47543561","subject":"reconstructor: log more details when rebuild fails","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7960097f02bab5a4568eddb6b98d7b8b47543561"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-13 15:45:25.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-27 11:01:31.000000000","tz":60},"subject":"Quarantine stale EC fragments","message":"Quarantine stale EC fragments\n\nIf the reconstructor finds a fragment that appears to be stale then it\nwill now quarantine the fragment.  Fragments are considered stale if\ninsufficient fragments at the same timestamp can be found to rebuild\nmissing fragments, and the number found is less than or equal to a new\nreconstructor \u0027quarantine_threshold\u0027 config option.\n\nBefore quarantining a fragment the reconstructor will attempt to fetch\nfragments from handoff nodes in addition to the usual primary nodes.\nThe handoff requests are issued in batches equal to the total number\nof fragments for the policy. The maximum number of batches issued can\nbe set using the new reconstructor \u0027quarantine_handoff_batches\u0027 config\noption.\n\n\u0027quarantine_threshold\u0027 defaults to zero i.e. no fragments will be\nquarantined. \u0027quarantine_handoff_batches\u0027 defaults to 1.\n\nCloses-Bug: 1655608\nChange-Id: I552d23ab60954ac32fe31e7f85e6d9e58fa25da3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/60dae671b93a48cdde71e1ed5037d91d0cc79647"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/60dae671b93a48cdde71e1ed5037d91d0cc79647"}]},"branch":"refs/heads/master"},"2eb5ec86fc09f795450fdb93226377ba0b05e9fc":{"kind":"REWORK","_number":8,"created":"2021-04-29 22:12:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/84/786084/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/786084/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786084/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/786084/8"}}},"commit":{"parents":[{"commit":"7cfdb50f93f6780a668c4d6c04b1f1ef25166c69","subject":"Merge \"reconstructor: extract closure for handle_response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7cfdb50f93f6780a668c4d6c04b1f1ef25166c69"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-13 15:45:25.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-29 08:50:33.000000000","tz":60},"subject":"Quarantine stale EC fragments","message":"Quarantine stale EC fragments\n\nIf the reconstructor finds a fragment that appears to be stale then it\nwill now quarantine the fragment.  Fragments are considered stale if\ninsufficient fragments at the same timestamp can be found to rebuild\nmissing fragments, and the number found is less than or equal to a new\nreconstructor \u0027quarantine_threshold\u0027 config option.\n\nBefore quarantining a fragment the reconstructor will attempt to fetch\nfragments from handoff nodes in addition to the usual primary nodes.\nThe handoff requests are issued in batches equal to the total number\nof fragments for the policy. The maximum number of batches issued can\nbe set using the new reconstructor \u0027quarantine_handoff_batches\u0027 config\noption.\n\n\u0027quarantine_threshold\u0027 defaults to zero i.e. no fragments will be\nquarantined. \u0027quarantine_handoff_batches\u0027 defaults to 1.\n\nCloses-Bug: 1655608\nChange-Id: I552d23ab60954ac32fe31e7f85e6d9e58fa25da3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2eb5ec86fc09f795450fdb93226377ba0b05e9fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2eb5ec86fc09f795450fdb93226377ba0b05e9fc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
