)]}'
{"id":"openstack%2Fswift~786575","triplet_id":"openstack%2Fswift~master~I07277e82646b650e5ed81b4b703498dc7fe49d0a","project":"openstack/swift","branch":"master","topic":"p-lonely-frag","hashtags":[],"change_id":"I07277e82646b650e5ed81b4b703498dc7fe49d0a","subject":"Quarantine solitary EC fragments - Followup","status":"ABANDONED","created":"2021-04-16 06:50:58.000000000","updated":"2021-04-21 06:20:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9377da8cc215bc5607e69738d10805c941edb3f2","_number":786575,"virtual_id_number":786575,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-04-16 07:58:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-16 07:58:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-20 08:48:42.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"f5b235561539fd52a26eb2fa763bb95763497cec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-16 06:50:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8eeab50df3a67161134aa7676a93048bc9b797f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 07:58:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bd8529a6e784b1faffb1d1bd44f5895 : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/d567132ac8c74374b156c430aad9843f : SUCCESS in 57m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/93998e5a4c334382a1f00c4ea4715931 : SUCCESS in 37m 58s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/83304dc037bc4144a87f32827c458bf7 : SUCCESS in 8m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/746dc40db0674149a46aae4594d31bf1 : SUCCESS in 8m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/68c330840ce24846bdfe858ae50492a9 : FAILURE in 11m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/da555d0db3e54842af569788d9809745 : FAILURE in 16m 01s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/563ae44cfe6b4eadafa9ab2c7566b993 : FAILURE in 13m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7d1656261a144f6a954dca96ffeb26b5 : FAILURE in 18m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c170e4759a2245d3ae147365221c532d : FAILURE in 16m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bde9d6013e20482e88b17cd790118c0f : SUCCESS in 13m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a1f72d96e95047afabf8bfd2da156a7e : SUCCESS in 16m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d5d41ea0225d4b9c9a0f5b3414f270aa : SUCCESS in 21m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9b7ec84d80364718873bdd5f7496f3e4 : SUCCESS in 18m 08s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/20d12df8b1a64e3f9855232ff59a4d92 : SUCCESS in 19m 25s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/182098ca9ccc47209af731e2f5a6dbc9 : SUCCESS in 14m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d7dab5f33bdd4332952491f00fe7d424 : SUCCESS in 6m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d76a19005e984975ad08f5dd03c68cc5 : FAILURE in 28m 28s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4a12e139b57049d5867a5b67712a3589 : SUCCESS in 36m 19s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/505a6c3518f64f759f2a8c3ff6c8d52a : SUCCESS in 50m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7b4fc33f1b6848baa89513df59647af2 : SUCCESS in 40m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5e7b84012a8a4a939b8b7267f504d233 : SUCCESS in 48m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68710668aa8c45219741f3cabaa011c1 : FAILURE in 21m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1af289aa99e44fb95edb5974a8852fd : SUCCESS in 6m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6946ae765b864e69bcc12dedac75e3ae : SUCCESS in 29m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1fece2aa861449fa539ae2d115009b2 : SUCCESS in 51m 32s","accounts_in_message":[],"_revision_number":1},{"id":"0348242f09af38a959c10ecaa7cff64f0f581eed","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-20 08:48:42.000000000","message":"Patch Set 1:\n\nThanks Matt - I took some of this into the parent, and the parent also changed quite a bit","accounts_in_message":[],"_revision_number":1},{"id":"dd5c5a44022922f3e69148ba80d2eb92903fdb86","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-21 06:19:20.000000000","message":"Patch Set 1:\n\nCools nps. Never expected this patch to last, thought it\u0027ll either demonstrait what the parent was missing or would be squashed. Will abandon it now 😊","accounts_in_message":[],"_revision_number":1},{"id":"9377da8cc215bc5607e69738d10805c941edb3f2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-21 06:20:02.000000000","message":"Abandoned\n\nparent reworked with some of this patch taken into account, so it isn\u0027t needed anymore","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3f884184fc623d336164f2d5f383ade6ce1d32f2","revisions":{"3f884184fc623d336164f2d5f383ade6ce1d32f2":{"kind":"REWORK","_number":1,"created":"2021-04-16 06:50:58.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/75/786575/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/75/786575/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/786575/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/786575/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/75/786575/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/75/786575/1"}}},"commit":{"parents":[{"commit":"db6ab15605ecf4baafaa8e5879c0d52aa1f89f28","subject":"WIP: Quarantine solitary EC fragments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db6ab15605ecf4baafaa8e5879c0d52aa1f89f28"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-04-16 06:48:11.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2021-04-16 06:48:11.000000000","tz":600},"subject":"Quarantine solitary EC fragments - Followup","message":"Quarantine solitary EC fragments - Followup\n\nJust a patch to squash in. It fixes a bunch of tests (but not all I ran\nout of time), adds another lonely frag test which tests 2 - 9 lonely\nfrags. Plus makes the new reconstructor code py3 compliant.\n\nChange-Id: I07277e82646b650e5ed81b4b703498dc7fe49d0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f884184fc623d336164f2d5f383ade6ce1d32f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f884184fc623d336164f2d5f383ade6ce1d32f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
