)]}'
{"id":"openstack%2Fswift~786684","triplet_id":"openstack%2Fswift~master~Ibc9dc5f40e3008d4223af3076382114dabb9165c","project":"openstack/swift","branch":"master","topic":"p-lonely-frag","hashtags":[],"change_id":"Ibc9dc5f40e3008d4223af3076382114dabb9165c","subject":"WIP: check container listing before quarantining EC frag","status":"ABANDONED","created":"2021-04-16 16:03:59.000000000","updated":"2021-04-28 16:08:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e6cc56c1d85e685d951b0f03413001d99995adf5","_number":786684,"virtual_id_number":786684,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-04-16 17:17:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-16 17:17:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2a7adc159f6d2cae30c55e7cc392c3b13366acff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-16 16:03:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7430f64d966668c9f91f5aa5be1c0584b439fe7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 17:17:10.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45120fbba7034a6baa85d0b52e719093 : FAILURE in 8m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/11ffe35725c740d7b8558656e712865b : SUCCESS in 1h 07m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/deac8e392d724f8ba47a5ca13afc0835 : SUCCESS in 54m 12s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/037fa4ed78124dd484fb0367c7fc2900 : SUCCESS in 13m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f2761bd361554de28f93d1a7a392c8f9 : SUCCESS in 9m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b49cf2b29dab421fb1f5e0968b9faef2 : FAILURE in 12m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fe01032213fc49beb09c87a2800f09ce : FAILURE in 11m 50s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/63e5f391a474456c9f16f1d48281205f : FAILURE in 14m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9f5e25ef56ca4ffc9ef909beffed7198 : FAILURE in 18m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0263cf5597874ca1916230338a521cb2 : FAILURE in 13m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7f81f4b7ba294268a71d4496ceb19776 : SUCCESS in 14m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/441ad031999a451ba749893abbe53117 : SUCCESS in 15m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ef12e5f6fe4f413a868bca3c8742785c : SUCCESS in 14m 18s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8de8e8e3319848e08dd72730364f89e6 : SUCCESS in 16m 19s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/15c2f8c16b6d4f4b8a3758f1afccd99d : SUCCESS in 14m 59s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3c6a8abd89534c21b90e4f908778a036 : SUCCESS in 16m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7ec172901b654a6c9a857dd277d191b6 : SUCCESS in 9m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/26bb4441e29d4b7aa394df2ebd572dc1 : SUCCESS in 25m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6c5e87f86dd749f09c63531a39a75e8b : SUCCESS in 41m 57s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/667f24025a2244b084ac5e4d426d1b71 : SUCCESS in 39m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aefe8265fb3c4874b3a10296730ed51a : SUCCESS in 43m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d7d104c2f7364e3ca3b05c3c29bb194d : SUCCESS in 36m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfa50e6824ee46b2b3f576284e83ddbe : FAILURE in 16m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d89caa4e13b84fe6a14261224caf0462 : FAILURE in 6m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c22f81bf439247e6b4c1aca1d3a9c42c : SUCCESS in 39m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f61526577d74fe28b6d853cfc968efd : SUCCESS in 44m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e6cc56c1d85e685d951b0f03413001d99995adf5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-28 16:08:12.000000000","message":"Abandoned\n\nconsensus at PTG was that this isn\u0027t required before quarantining","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ebc9a841db2500fdd354267f30ba24084da9b770","revisions":{"ebc9a841db2500fdd354267f30ba24084da9b770":{"kind":"REWORK","_number":1,"created":"2021-04-16 16:03:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/84/786684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/786684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/786684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/786684/1"}}},"commit":{"parents":[{"commit":"41915f14add2c1203acd1bbfcc136076e9924867","subject":"WIP: Quarantine stale EC fragments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41915f14add2c1203acd1bbfcc136076e9924867"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-16 15:52:40.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2021-04-16 15:58:53.000000000","tz":60},"subject":"WIP: check container listing before quarantining EC frag","message":"WIP: check container listing before quarantining EC frag\n\nTODO: mock internal client\n\nChange-Id: Ibc9dc5f40e3008d4223af3076382114dabb9165c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ebc9a841db2500fdd354267f30ba24084da9b770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ebc9a841db2500fdd354267f30ba24084da9b770"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
