)]}'
{"id":"openstack%2Fswift~787440","triplet_id":"openstack%2Fswift~master~I1d6cc0faa3a43852c46089e509d48cc3ee9f9cf8","project":"openstack/swift","branch":"master","topic":"bug/1911811","hashtags":[],"change_id":"I1d6cc0faa3a43852c46089e509d48cc3ee9f9cf8","subject":"fix not clear cause for invalid username","status":"MERGED","created":"2021-04-21 20:42:29.000000000","updated":"2021-04-23 17:56:56.000000000","submitted":"2021-04-23 17:55:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"787440-bug/1911811","meta_rev_id":"fe1fe8c8084f06500af5ea84ddb83a9d48beadb0","_number":787440,"virtual_id_number":787440,"owner":{"_account_id":33384,"name":"Tiago Octaviano Primini","display_name":"Tiago Primini","email":"tiagooctaviano.primini@windriver.com","username":"tiago-primini"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-23 17:55:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2021-04-23 08:30:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-22 19:46:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-22 19:46:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-21 21:42:52.000000000","updated_by":{"_account_id":33384,"name":"Tiago Octaviano Primini","display_name":"Tiago Primini","email":"tiagooctaviano.primini@windriver.com","username":"tiago-primini"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2021-04-21 21:43:24.000000000","updated_by":{"_account_id":33384,"name":"Tiago Octaviano Primini","display_name":"Tiago Primini","email":"tiagooctaviano.primini@windriver.com","username":"tiago-primini"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-21 23:27:25.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2021-04-22 19:46:48.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"3c08de61b517bc77ba11811f94cef65a83feb460","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33384,"name":"Tiago Octaviano Primini","display_name":"Tiago Primini","email":"tiagooctaviano.primini@windriver.com","username":"tiago-primini"},"date":"2021-04-21 20:42:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4916aeec7f3c76786653a2ce2be7579bfcdd5de7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 22:26:07.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79c14985e3e3459aad258796644675d3 : SUCCESS in 10m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/68902955d8114248abf78410ae31b1f2 : SUCCESS in 1h 00m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/05e45701cde647ea98119cee5765ab01 : SUCCESS in 1h 00m 04s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9c38d7b05a754b4f81039e4748067c1e : SUCCESS in 9m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d5b453e3bcbb435894e30b33659c7125 : SUCCESS in 8m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7092ec622df74f459f3352639eaaad27 : SUCCESS in 16m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ea0f4828070e4b458a2a3c229e30b191 : SUCCESS in 16m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f9f35314cb824d259ece5e5a203e1e18 : SUCCESS in 17m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c112e041217d4e25a359e182ee7ad1b7 : SUCCESS in 20m 25s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a3d622b7d32745dcb9a69536e2ed1c71 : SUCCESS in 16m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f5425b70a8834be1bd35b1d0f3bd9e24 : SUCCESS in 18m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/554776632f4c4544a8b8dfc9524a94ef : SUCCESS in 17m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8a1e7979bfc44e379eff1171e1082d90 : SUCCESS in 21m 50s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1dbc2055bb424d05a6756f29f29d49c1 : SUCCESS in 17m 47s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9809360c2a634ba59a77235aa5188f94 : SUCCESS in 24m 18s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/1f1d40dee83a4c3eae2af86752472112 : SUCCESS in 22m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/10cbbb2f818e4d2595bc25946abce7f1 : SUCCESS in 8m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/87880812e67549b68edd9b1fe7329a60 : SUCCESS in 26m 05s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/6308ccd5dbbe4951a6b4e2c88c69e8be : SUCCESS in 51m 18s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/69eb499f8ed9433a8f0f12a31669f751 : SUCCESS in 48m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eea5bd071636472aa718fb1bc074b803 : SUCCESS in 47m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/90a1f527f9a34284abe691731c358214 : SUCCESS in 34m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/624341758a0e4c418c9dbd05f3727f63 : SUCCESS in 17m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf6dbfd210084ae9a5d9519a60f01299 : FAILURE in 7m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2a744c9370ff409680943a5be4152c67 : SUCCESS in 26m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ac84c976d204705aae9dcfa153904e6 : SUCCESS in 56m 19s","accounts_in_message":[],"_revision_number":1},{"id":"06421be224e910d9453b120415be3ed25d3fc797","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33384,"name":"Tiago Octaviano Primini","display_name":"Tiago Primini","email":"tiagooctaviano.primini@windriver.com","username":"tiago-primini"},"date":"2021-04-21 22:31:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4af63047541b9f46f653f68b0eeb8e412d537ef7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33384,"name":"Tiago Octaviano Primini","display_name":"Tiago Primini","email":"tiagooctaviano.primini@windriver.com","username":"tiago-primini"},"date":"2021-04-21 22:37:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"477295971e65fee3ba816bdf4e0c014c3a180954","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-04-21 23:27:25.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"86680b9688df74a560cde1765904331125fa2455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 23:48:18.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7c3df33fa344a9ba0efbb129f829a22 : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/441e4dcfa91b41e4b2135e0c8c02fc82 : SUCCESS in 1h 04m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2acb7c5168f04c3aa0fa5aaaff600a73 : SUCCESS in 36m 56s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7659fcb08593498ea61588f4d0f59f78 : SUCCESS in 9m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3348ddb2fef348cca12fa8d69ccae59c : SUCCESS in 6m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0c8480583aa14f3786818dc4a3689e1d : SUCCESS in 14m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/09a985ae8726436b93349ff2d23de6fc : SUCCESS in 18m 23s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5cbcd838cea04876aa9139eaf55fffa4 : SUCCESS in 16m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ca746af314134f0b975bb7f710f5505a : SUCCESS in 16m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0b35ac84ed034971a3acf8aab9253201 : SUCCESS in 17m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/afd04a423aff4ee28d167e7abe987530 : SUCCESS in 19m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/20f720c02c504e13b28b8b5e58501180 : SUCCESS in 16m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e0f754918036469ca6b7f4832e7d2b75 : SUCCESS in 18m 57s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/65392360761446c483fe1a9a63097d4c : SUCCESS in 20m 09s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/60c26d89fbd3479bbcacac5c054768b9 : SUCCESS in 19m 23s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/3228a060f0f14ecaa7f3ce1eb9ea47bd : SUCCESS in 21m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d2482eff6cd34b259ab7011aac993514 : SUCCESS in 6m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9273630302174a9b8cef8938800e9341 : SUCCESS in 24m 16s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/43135fa0639247d1994dd55778515973 : SUCCESS in 45m 50s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/f9d674472a3345879e2a29db37e107ac : SUCCESS in 46m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d8c1f91adb3c4fffbba16f5fb7268caa : SUCCESS in 42m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e8ad8cf2071948b3998aa02e0a4acd1b : SUCCESS in 40m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13884f65788d4351abfdefd132f97b7f : TIMED_OUT in 40m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/572c7ef8c8d44b5f9910c872139e487e : SUCCESS in 9m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/115667d7e65e42ecbfa16549fed0831a : SUCCESS in 23m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0afddf4a16f448d095d91c528d7c62fa : SUCCESS in 51m 04s","accounts_in_message":[],"_revision_number":3},{"id":"09a64f6fbb2598c863dae133f1c16fb5094b7cab","author":{"_account_id":33384,"name":"Tiago Octaviano Primini","display_name":"Tiago Primini","email":"tiagooctaviano.primini@windriver.com","username":"tiago-primini"},"date":"2021-04-22 12:06:47.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"30e0a72c9633ea07be485a1b72a8899530d2858b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 13:33:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28e2ee87989c4aa3bafb7c30e2d1595a : SUCCESS in 11m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/2283cc1b41e3423b9ffb739ba01bbfbd : SUCCESS in 1h 00m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/34ffa831dc7e4ed5993fe910cb170306 : SUCCESS in 41m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1cd8c022adda4e12b7f32dd1ba84ab3b : SUCCESS in 9m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/acf1dc235a93485ca702906a096ec4a7 : SUCCESS in 7m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/40a92de74e584f6998f40922740d0530 : SUCCESS in 14m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/34c3f5b9f7824540ab24b83fbd8342e8 : SUCCESS in 14m 32s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7dcd3f58000849759af34cc12aad84bb : SUCCESS in 16m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8a84f47ae56d4a6eb0f23277fc5080da : SUCCESS in 17m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c6572d47f6c04e328d02b2771cbd3390 : SUCCESS in 14m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/70e0a06d5a4c4352a5644cdb6209154a : SUCCESS in 17m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a3ff5a21300a4c7ca791deb7878b6412 : SUCCESS in 16m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a33fef52fdef40bc9ba1bf792ee3e78c : SUCCESS in 19m 31s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8c7b8a96bf1b47fe911d8f6123e4dcce : SUCCESS in 35m 45s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/426e9952191c4e7fb456009d59778cda : SUCCESS in 36m 23s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/d55a2dbd0e5b4086a07188dedfb7f61f : SUCCESS in 20m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1db9fe3ee11f4fe1a318803bf1a4c210 : SUCCESS in 8m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/92cf8549bc404aa7b7e99f26ff3b7770 : SUCCESS in 15m 53s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4b46d55001c64290b285eafe6cdd2f1e : SUCCESS in 59m 20s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/273f591f9edd482ea8a961db03c3cb79 : SUCCESS in 1h 21m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9dfda2dc3cd14118af33e4a48290c0ba : SUCCESS in 37m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/655422c3ed194df1b8f3eeb88ab7dd08 : SUCCESS in 43m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8384a6a82354e69841ec61f533b63e8 : SUCCESS in 15m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/237cdad95dfd4b42931ee5718c4144d7 : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/79e599d826fa4eaabc018f04c27414ae : SUCCESS in 31m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/347d3e02bcea4a0f84694f8b1a94b025 : SUCCESS in 57m 20s","accounts_in_message":[],"_revision_number":3},{"id":"343c4e66a0015e176774e7333833b8108a3cdf08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-22 19:45:53.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"7adab695d02e70954081166e5ec686b5f5633b69","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-22 19:46:48.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nThanks, looks great! Just added a little bit of bug-tracking.","accounts_in_message":[],"_revision_number":4},{"id":"7193050fcdb09b635bc938b608c3f803cf8f20b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 20:58:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cb4daaa417f456db9ff2d6528cd34df : SUCCESS in 11m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/c6c367797c614a1bbb3cd6e5c32099b4 : SUCCESS in 1h 07m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/661e9569f62141619ed38504022396df : SUCCESS in 37m 25s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cc12d0d39d0b4168ac38d9aeceec298a : SUCCESS in 11m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/aebfe1b58acd4228ac7dc1393d728a3e : SUCCESS in 8m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4696b605a68e4f0d83f294cb52a97129 : SUCCESS in 33m 31s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ef8d1a888c28450c941ed3491ab20c6d : SUCCESS in 17m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7d92fe145d67492c9c1737b8a2c99bdd : SUCCESS in 34m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/72b8b5fb30514879ac3bc7fd2e736271 : SUCCESS in 15m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/08efe37d01ac45c2930025537ada24ec : SUCCESS in 33m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/904e271b0ca141dea069a982fd69bc65 : SUCCESS in 17m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5ec646137a9c4fcb90a2e61990cce83b : SUCCESS in 16m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1f905dba858642029e952253cc75a5a5 : SUCCESS in 19m 51s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4e5cbde62085467aa3f914b291b84a15 : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/348bde451ba54f2daca0c32410a21bd1 : SUCCESS in 16m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/aa8cfa560ff34ae6a25a3a022b885579 : SUCCESS in 19m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f1dcc2b2947c4185af374b92ad2b0cb1 : SUCCESS in 9m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d3628ccb4a2644ceb37ddf5a5b0c5f80 : SUCCESS in 19m 11s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ec43fbc7ac3c40f4af37a4cef089771c : SUCCESS in 47m 28s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/22a992de2c1f4f2eb47e5ddff19a31d8 : SUCCESS in 45m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e12611f759dd4b8f82a9df0b793fae45 : SUCCESS in 40m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ae52bb2da9ed4316a9e8ccff6c9a2b82 : SUCCESS in 31m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16c0420980a548d1959dfb5ce9a77579 : SUCCESS in 13m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c23330cd8b0a455995e05127b870666e : SUCCESS in 7m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d4d637cfb7fe40b9b595ddb8c5272e13 : SUCCESS in 29m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc26f564dc5e4594afdaf480f112bc77 : SUCCESS in 45m 54s","accounts_in_message":[],"_revision_number":4},{"id":"0abc3c63b25d4be5b57679ade08142ace5fe4f64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 20:58:24.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7e281f8034361eeccb171668fdac1e3897323674","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 22:46:43.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0632cdf5e5a4f97bc588532ae803f42 : SUCCESS in 9m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/8211be05fbf3419cbfead541516e25a0 : SUCCESS in 51m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cc8e2da9b96841be93a7370b91cc7f2c : SUCCESS in 39m 58s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1b8121023e1845c39822b0a810771f68 : SUCCESS in 10m 46s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/5c5c37d5a1b441f08d54aa3fb3cfbc8b : SUCCESS in 8m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/82cc2d17b11f4bebb5829d89fb28dd6b : SUCCESS in 16m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0ea09c5b7a7f43cf9872d21c4a5bf189 : POST_FAILURE in 18m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/138cd910b0b54807bf8b1b695482cae2 : SUCCESS in 14m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d37f54397f8e4dea8cd73670f2208e0d : SUCCESS in 14m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/457ef0304a1647889d7b291bd0c04eae : SUCCESS in 14m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fe8acbca6b3049c4b6bbb486b70f2823 : SUCCESS in 16m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0f8afe2d0433462f9adb209aecbd57d1 : SUCCESS in 19m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5fb3facd6d594a18b9fc7b9b66ea0ec7 : SUCCESS in 17m 45s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/a7729a1f5cae4fc898d7c12f04e2dad5 : SUCCESS in 30m 45s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/74515e6f20334a63812072713d330548 : SUCCESS in 26m 00s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bdfbb6b14e9a4d3e88fc34ca2f9eec18 : SUCCESS in 17m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/55a6223b240b46aeb0ce640a5b889a58 : SUCCESS in 6m 56s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/9705f1cf06684710ada452d0af4b3635 : SUCCESS in 50m 20s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/a2854942e7b74d689b9b48f233493002 : SUCCESS in 47m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/00987bc9c62548c5a833558d1e1e5ba3 : SUCCESS in 38m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d6a2f324672541dbaf5f24855b396768 : SUCCESS in 38m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af59a28652f041018ee3cfcef8834d81 : SUCCESS in 15m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/766eacdfd78840dc84cd6ce6a12b334d : SUCCESS in 5m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/42d879cf9c1541c09813de439e74bd0b : FAILURE in 28m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ac52264d8f34b5e92290e439440cf4a : SUCCESS in 47m 41s","accounts_in_message":[],"_revision_number":4},{"id":"27f68716cf7a121d7299bbc47c61bbcdae591e18","author":{"_account_id":33384,"name":"Tiago Octaviano Primini","display_name":"Tiago Primini","email":"tiagooctaviano.primini@windriver.com","username":"tiago-primini"},"date":"2021-04-22 23:38:23.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"fdf96830218b2de4e2bc77f5431917a384b49f0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 01:58:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4e018dc710149c4b85d0fbe54a77801 : SUCCESS in 10m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/a0403d8d2f29405cb4aca0b8692cd593 : SUCCESS in 2h 14m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/325bbf9df3524827809c566379e1147f : FAILURE in 1h 22m 59s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/28ef5eb46a954e8d9a00274a73271c61 : SUCCESS in 8m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/117e9431c776404a8191e4829b8b2350 : SUCCESS in 7m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ee337b64f45047f1a8a03d242d95df2a : SUCCESS in 15m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/573013996ab34531b86a3deefc4f9403 : SUCCESS in 14m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b2cde0fe86914f17af9be8ce2f6199c2 : SUCCESS in 15m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d08fde0d66104e16aed352a3a9e997cd : SUCCESS in 16m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b4968436d2104c6688a922368b20e3a1 : SUCCESS in 17m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a095d22b4c49468dad2789257421ab4f : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f94ec73919ec495da2749166ff722590 : SUCCESS in 18m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c6d18a05234149feba75f57384341c5b : SUCCESS in 17m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/acdf5f6965e74e4384bd886d8ec2b794 : SUCCESS in 17m 39s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6c95fb680ad44c5ca688e43ae4c3bb25 : SUCCESS in 18m 11s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/a32f5bca93134e999654dd5a2f9a566d : SUCCESS in 17m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b889e9c2e33a4f3a9378672e942bc53e : SUCCESS in 7m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/04859bd4ab6b49b2a622d9d2e933e923 : FAILURE in 20m 22s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/71b8c2d8e1f84be0b030e8b8a4bf4586 : TIMED_OUT in 2h 01m 07s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/0deb9c795b61408abec1b06be8823dfe : SUCCESS in 52m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c7f4afbf22ba4e2ebfb444bb114aaf0c : SUCCESS in 36m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9cfec3ae228f478ba241d668174485f3 : SUCCESS in 1h 11m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/649aef8e030d4065b84afa535f4aaa0e : SUCCESS in 14m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79a6d0abd8564ab3921e839e9a7be938 : SUCCESS in 7m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a547ec3b4d264388ad32cd0c6b86f121 : SUCCESS in 25m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36bfd7ddce684f62b19f4f0e41e9d09b : SUCCESS in 55m 26s","accounts_in_message":[],"_revision_number":4},{"id":"b86185f85e169791a0c63bee4f9215f6142fd4dc","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-23 08:30:00.000000000","message":"Patch Set 4: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"29ab7c46468310f6d43a6774f8a9ec4405b54998","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 09:54:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59bc1687b13e4fb38b71812fb7745c94 : SUCCESS in 10m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/1db72ee23b784674a582600dc6c2e115 : SUCCESS in 1h 22m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cb382db476ea411e8f040096db94de0a : SUCCESS in 41m 16s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5f1a812c9b64451aa0a96a1f4aee50cd : SUCCESS in 10m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/accb18dbdc134b88aa12d58ab1cd7aa5 : SUCCESS in 7m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d692ca3d1ca54af18f1111eed5e1f326 : SUCCESS in 14m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cb128e03493a43528a62e50721f35dd2 : SUCCESS in 17m 34s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e33984c115f74b258e08c9e2ce892a7d : SUCCESS in 16m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/867a52131fce4974aca3a19c9c010225 : SUCCESS in 14m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0421afd3638947208251b4cee681e50e : SUCCESS in 15m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/71927a5cd1864e38aa798784391188f1 : SUCCESS in 21m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a72f5f528c944446be525bf74e233995 : SUCCESS in 19m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0bac10b1620945748fb5ef26700f33d4 : SUCCESS in 18m 42s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/faa5d08a558f427d9ca8309c0c89982b : SUCCESS in 18m 27s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/3956047d5cda4d2f898472ee9e8de304 : SUCCESS in 20m 00s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/daa3db53064c4a719b3df5187fc77e69 : SUCCESS in 19m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0761e245abd442549fdf754d42c27434 : SUCCESS in 9m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0aed05dcf2f84f6abbc6611a1405d3c8 : SUCCESS in 25m 08s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d0a0cac55e394850b4c0e226a0045801 : SUCCESS in 50m 43s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/e4b2715dc8364ef78c8cc6fea34554dc : SUCCESS in 53m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d6c58af972cd4b9a9dd7db76e2a3f3e9 : SUCCESS in 43m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ea08e71a1e924182bc3c58737fea224e : SUCCESS in 43m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4e9f57f9c4954d70996a16c998f14066 : SUCCESS in 19m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1646211d11be47f99eaad654c817388d : SUCCESS in 6m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ab0ebdf5d6324190ac66f5ab93c25cc3 : SUCCESS in 27m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/095b45f4ddf647c3a3cea1a4ca602d1e : SUCCESS in 52m 06s","accounts_in_message":[],"_revision_number":4},{"id":"c462a8c519e7972a6077741e95ce1cd990e46e5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 09:55:20.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2cfae26ab1c1fdaadc7b21745e847293f603aa20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 16:11:06.000000000","message":"Patch Set 4:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"22d962535c288c2f35e95cdb6688badbfd77f9b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 17:55:14.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89d408eaa8a540beab19ba4d6adbb3f8 : SUCCESS in 10m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/84f1774a0d28439c89643dee1705d12f : SUCCESS in 49m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eee818ebaa814460b3e0b012f552683d : SUCCESS in 33m 20s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5281e4444b964de8ab5338ab5c2206d4 : SUCCESS in 10m 55s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/f4489ba2717d40d08dc560615f5a8fe8 : SUCCESS in 8m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e49aaa86664b498bb67ea39836cb0bf4 : SUCCESS in 15m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2995a80248a8493aaffc42968977c88d : SUCCESS in 15m 32s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/20be908e4d644da0bbe386d25a31afe6 : SUCCESS in 17m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/73d4148e0644461c9ad61989272873fc : SUCCESS in 15m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f0705f6213bd4e649d7c6fc9a5676888 : SUCCESS in 15m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4e20ce44ddc54edc873849148cd1a63e : SUCCESS in 18m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/40e5e669a95d474ab2a24cedc9d4749f : SUCCESS in 16m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5209398b4d6143d3aaa8b96afd6bfced : SUCCESS in 18m 01s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/8557ed6b4c8247b3927ad344cdd6befd : SUCCESS in 19m 41s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9156ff8d2f184264bc3395cfa231b825 : SUCCESS in 17m 12s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/44963fc7ce4546f28518c226e9fa1a4f : SUCCESS in 19m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/90230966ae7a4b05a486c96dd3ce52b4 : SUCCESS in 7m 22s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a26ee533888a424eb7ec7de86f3fdbd2 : SUCCESS in 46m 50s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/0d439389a07241dd8f051a11bb6ccd81 : SUCCESS in 41m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3f6e417ae1584b0d8aaf7c345d3b85ca : SUCCESS in 33m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/57de64b2983d4199a89464ef1c59f280 : SUCCESS in 30m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3227906319934018ba67451c2ae0b921 : SUCCESS in 15m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34793a6dfd8e4de1b87c780224d7b25e : SUCCESS in 5m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0ef05ad2754743babaed064c1319d328 : SUCCESS in 34m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3e8692233a04160b21695358e59b2ed : SUCCESS in 53m 26s","accounts_in_message":[],"_revision_number":4},{"id":"51b7aa62e0e53bb5693194ddaf78c3817fca6ccc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 17:55:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"fe1fe8c8084f06500af5ea84ddb83a9d48beadb0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 17:56:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10836e74a32f4be7aa091d78237adf6c : SUCCESS in 1m 09s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7f7aac543fd5441a9821d20f240d5e0d : SUCCESS in 50s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/deb36f8ae53c4a5b889e9551fe832fed : SUCCESS in 46s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"717d21ccbd84794444a9f8cbcc5864ae46a2ffbb","revisions":{"35d5b17d2ccea9a1c42f09a29f0b596be04c3bc0":{"kind":"REWORK","_number":1,"created":"2021-04-21 20:42:29.000000000","uploader":{"_account_id":33384,"name":"Tiago Octaviano Primini","display_name":"Tiago Primini","email":"tiagooctaviano.primini@windriver.com","username":"tiago-primini"},"ref":"refs/changes/40/787440/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/787440/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/787440/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/787440/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/787440/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/787440/1"}}},"commit":{"parents":[{"commit":"c9b5d44e9e8d4a75e4a1e2d31036823560e9a0a1","subject":"Merge \"Update AUTHORS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9b5d44e9e8d4a75e4a1e2d31036823560e9a0a1"}]}],"author":{"name":"Tiago Primini","email":"primini@gmail.com","date":"2021-04-21 20:39:53.000000000","tz":-180},"committer":{"name":"Tiago Primini","email":"primini@gmail.com","date":"2021-04-21 20:42:26.000000000","tz":-180},"subject":"fix not clear cause for invalid username","message":"fix not clear cause for invalid username\n\n - add a message saying the reason for the value error exception\n - add a unit test to validate the expected message\n\nChange-Id: I1d6cc0faa3a43852c46089e509d48cc3ee9f9cf8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/35d5b17d2ccea9a1c42f09a29f0b596be04c3bc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/35d5b17d2ccea9a1c42f09a29f0b596be04c3bc0"}]},"branch":"refs/heads/master"},"24db73a9e9b0e6061f2a1c3ee508c3ee5b7111a2":{"kind":"REWORK","_number":2,"created":"2021-04-21 22:31:10.000000000","uploader":{"_account_id":33384,"name":"Tiago Octaviano Primini","display_name":"Tiago Primini","email":"tiagooctaviano.primini@windriver.com","username":"tiago-primini"},"ref":"refs/changes/40/787440/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/787440/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/787440/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/787440/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/787440/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/787440/2"}}},"commit":{"parents":[{"commit":"c9b5d44e9e8d4a75e4a1e2d31036823560e9a0a1","subject":"Merge \"Update AUTHORS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9b5d44e9e8d4a75e4a1e2d31036823560e9a0a1"}]}],"author":{"name":"Tiago Primini","email":"primini@gmail.com","date":"2021-04-21 20:39:53.000000000","tz":-180},"committer":{"name":"Tiago Primini","email":"primini@gmail.com","date":"2021-04-21 22:30:24.000000000","tz":-180},"subject":"fix not clear cause for invalid username","message":"fix not clear cause for invalid username\n\n - add a message saying the reason for the value error exception\n - add a unit test to validate the expected message\n\nChange-Id: I1d6cc0faa3a43852c46089e509d48cc3ee9f9cf8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24db73a9e9b0e6061f2a1c3ee508c3ee5b7111a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24db73a9e9b0e6061f2a1c3ee508c3ee5b7111a2"}]},"branch":"refs/heads/master"},"9803b18edf03dc11e5c003d8056ff8b3f11ec099":{"kind":"REWORK","_number":3,"created":"2021-04-21 22:37:32.000000000","uploader":{"_account_id":33384,"name":"Tiago Octaviano Primini","display_name":"Tiago Primini","email":"tiagooctaviano.primini@windriver.com","username":"tiago-primini"},"ref":"refs/changes/40/787440/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/787440/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/787440/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/787440/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/787440/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/787440/3"}}},"commit":{"parents":[{"commit":"c9b5d44e9e8d4a75e4a1e2d31036823560e9a0a1","subject":"Merge \"Update AUTHORS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9b5d44e9e8d4a75e4a1e2d31036823560e9a0a1"}]}],"author":{"name":"Tiago Primini","email":"primini@gmail.com","date":"2021-04-21 20:39:53.000000000","tz":-180},"committer":{"name":"Tiago Primini","email":"primini@gmail.com","date":"2021-04-21 22:37:16.000000000","tz":-180},"subject":"fix not clear cause for invalid username","message":"fix not clear cause for invalid username\n\n - add a message saying the reason for the value error exception\n - add a unit test to validate the expected message\n\nChange-Id: I1d6cc0faa3a43852c46089e509d48cc3ee9f9cf8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9803b18edf03dc11e5c003d8056ff8b3f11ec099"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9803b18edf03dc11e5c003d8056ff8b3f11ec099"}]},"branch":"refs/heads/master"},"717d21ccbd84794444a9f8cbcc5864ae46a2ffbb":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-04-22 19:45:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/40/787440/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/40/787440/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/787440/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/787440/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/40/787440/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/40/787440/4"}}},"commit":{"parents":[{"commit":"c9b5d44e9e8d4a75e4a1e2d31036823560e9a0a1","subject":"Merge \"Update AUTHORS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9b5d44e9e8d4a75e4a1e2d31036823560e9a0a1"}]}],"author":{"name":"Tiago Primini","email":"primini@gmail.com","date":"2021-04-21 20:39:53.000000000","tz":-180},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2021-04-22 19:45:38.000000000","tz":0},"subject":"fix not clear cause for invalid username","message":"fix not clear cause for invalid username\n\n - add a message saying the reason for the value error exception\n - add a unit test to validate the expected message\n\nChange-Id: I1d6cc0faa3a43852c46089e509d48cc3ee9f9cf8\nCloses-Bug: #1911811\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/717d21ccbd84794444a9f8cbcc5864ae46a2ffbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/717d21ccbd84794444a9f8cbcc5864ae46a2ffbb"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
