)]}'
{"id":"openstack%2Fswift~787831","triplet_id":"openstack%2Fswift~master~Id7e828a56c8a62a1f3e9a1dbbff5a56c928ac6b8","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Id7e828a56c8a62a1f3e9a1dbbff5a56c928ac6b8","subject":"Include sharding cycle time in recon","status":"MERGED","created":"2021-04-23 23:14:26.000000000","updated":"2021-04-26 19:15:08.000000000","submitted":"2021-04-26 19:13:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787831","meta_rev_id":"613f4ff4ec5481c1d528f43fe30a64df6501ab39","_number":787831,"virtual_id_number":787831,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-26 19:13:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2021-04-26 09:12:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2021-04-26 09:12:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-24 00:53:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-26 09:12:50.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"dcfa8f4696bae98a606593e7809fa863eb81dec7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-23 23:14:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45458caf8a7c0d063d97a0f14352efad07e8af1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-24 00:53:57.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c8c579d443e4d8eb276793ee030519b : SUCCESS in 9m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/c4f18b561f444fdf9cc26cfbcac0b19a : SUCCESS in 1h 37m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a60ecc3924e54430bcb5f48cbf0a9d6f : SUCCESS in 44m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/62220815ae6b48b18153c95bba0e97ac : SUCCESS in 9m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a3e0037aa33446beac13a9f21329ca54 : SUCCESS in 7m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/61477f2912b143e8b06a0a388b140ff6 : SUCCESS in 15m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/53903c7e4ea547c5ad81eb238c67bd9e : SUCCESS in 15m 25s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9922317705b14b4296082d7eb9d67532 : SUCCESS in 14m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/903fa0377f704aa4a4e5a4018fc691f0 : SUCCESS in 17m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b3dc8e8ff3c2452e87c633f40a809018 : SUCCESS in 12m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7a11bf7d195d4429b4177a779ad02933 : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/96ce31338b8946ee9a7e9161901b2f34 : SUCCESS in 16m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/47e3f4911bcb431baac806c996bd6e72 : SUCCESS in 19m 02s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/f9c17490e2344f01836e1814bc285168 : SUCCESS in 17m 45s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/d24209c8ad774dcf92c683cc8c629761 : SUCCESS in 19m 12s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/77eaa19684b24ba19a5cb9fddb0373de : SUCCESS in 21m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7056aa3175f04d1882ffe5ae4275a4db : SUCCESS in 8m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1303af798b484d02adb7e8f8ef9ffdbe : SUCCESS in 23m 37s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/940638ea90ab45d29f8a333bc541f297 : SUCCESS in 51m 47s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/f1152dda91254a08b08c3c8210010c84 : SUCCESS in 37m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dbe720d23a754b639d4bbd567925d834 : SUCCESS in 42m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5996d982d81f4e0ba9da7e0fd09d5874 : SUCCESS in 37m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f9e1b74bf7b4c70a683f53d617ed7dc : SUCCESS in 17m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d7e17def3c94364bc0c03467ffd2fb5 : FAILURE in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f11ff812984d40b5b762153ba3a45e8a : SUCCESS in 24m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0db63f9a4873498aa8030074765960c3 : SUCCESS in 50m 30s","accounts_in_message":[],"_revision_number":1},{"id":"69dc15f9ef7f77d25067de4d667c267547a03dae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-25 15:12:00.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"48ba66667be53585c909ab258d8c5007d49fceb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-25 16:44:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48fac1741bb247ba9544241eb5322a5f : SUCCESS in 9m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/360a39773a7a427aa5c84d805ed7eab9 : SUCCESS in 1h 04m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cac666d4bd0342e78aef49ab59b002d4 : SUCCESS in 44m 45s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/daa7cfb5c4ba47aa9f1c8eac51ee3268 : SUCCESS in 8m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bb28cc658c1d483ba6c84c0322d481d7 : SUCCESS in 6m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8416736ed8dc4e1cb688aa69239ab9e7 : SUCCESS in 15m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e0d5e0e60cb843468dabab6fb449191d : SUCCESS in 15m 18s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7b4a9499ee2c423598390fa4c49d43f7 : SUCCESS in 15m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3e108e98b7934e2980eca14b0f804d60 : SUCCESS in 15m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f7c00a9728444b159e85faea30d14c43 : SUCCESS in 14m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b0bb42c13fce40e0bde43486d3da2728 : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7db3f4140a434e598419964314a24208 : SUCCESS in 15m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cf41da08c0f04ac6942a42d14cd9a70a : SUCCESS in 16m 08s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e4ea6f92edc847ef844d46458060ebe3 : SUCCESS in 17m 07s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b95e6c4f123b4452bf2f08c89707cd56 : SUCCESS in 16m 30s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/6796a0de04c34daf8466df94268aaf3a : SUCCESS in 17m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e16b3477d20547edb03e1f62c48ce3a6 : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1ca69f50f6bb44af8e43955f12d7ac66 : FAILURE in 16m 04s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fc3ee22c7ca74db6aac148be9f4efd3d : SUCCESS in 49m 57s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/33b911f1c4ef460da54721fc9f9a22b6 : SUCCESS in 40m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f73f34a0ec744d819b58908aeb195c97 : SUCCESS in 37m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e4d7902e875f409c9cbc5f2c92e0de7f : SUCCESS in 29m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/506f010bdaca43f1bb921d95b4bceb31 : SUCCESS in 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/947bef6634944645a10f87836ef3461d : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d191a20b333b453ca418224f715d04cc : SUCCESS in 24m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0f62193cc044881a912cc21eea331a9 : SUCCESS in 43m 07s","accounts_in_message":[],"_revision_number":2},{"id":"0f98260e37b93d02e46f7d131895d7666907ca2a","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-26 09:12:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLooks good. Checked functionality on my SAIO.","accounts_in_message":[],"_revision_number":2},{"id":"8fc7126253587ac2505c21047009893addd9e963","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 09:13:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3ca6f91bce7a5423871863a9e790256028497915","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 12:26:41.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebcdf988a1d44372aedbe6fa9e64662d : SUCCESS in 10m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/9a5386f30faa456cb53e4140dc0cafab : SUCCESS in 57m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7243508ed9b147bfa0fbcf866304121d : SUCCESS in 38m 33s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/eaff882ea6a24721b36b443fc40a9aca : SUCCESS in 10m 45s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1803fe3a69cd478d80e3041d2d922b26 : SUCCESS in 7m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b41781ac36e8489e90eb6996ba01756b : SUCCESS in 16m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fb49b5c681464a15ba97c816b65edfb7 : SUCCESS in 16m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/18b29e2e83fc4353b640652462150803 : SUCCESS in 16m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/102fce70271e4b34a5c2ab86da834633 : SUCCESS in 16m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/298886b2793c416abd65efa7de805093 : SUCCESS in 17m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ebf800cd9ab14a9aae1d85199ca5c20c : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ef4d586209094c7db66710b1841ab482 : SUCCESS in 21m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/26486cab39af48b0b12668ce206c1452 : SUCCESS in 17m 55s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/dccb3eab9af04d54a28fa4cf48d9f6d3 : SUCCESS in 22m 37s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/54144dade78e4dfea883fb9fc2f2dff7 : SUCCESS in 17m 29s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/bcbc22e8854f4a4aacdb4ebaf9adfe5f : SUCCESS in 23m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2af9b4abd08249c4acd0fb0209855d3d : SUCCESS in 7m 50s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b8031590502d48c6ae1f4bdfb7a9ce56 : SUCCESS in 51m 33s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/65c3efec9c604d48a85ca3dc8538b0c8 : SUCCESS in 44m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0de06f439e06455e8ee84aa815f89557 : SUCCESS in 41m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c4f8a2441b414d76a743973dcf34b250 : SUCCESS in 36m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3ceb21af3bb481ca5905dbfd2d76a46 : SUCCESS in 17m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aac903f2929460790f116636ce1ff28 : SUCCESS in 5m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b37deee981b94efcb862c377e53b86ed : FAILURE in 35m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00fe229bc07d49c6af950bcfa3831d31 : SUCCESS in 49m 48s","accounts_in_message":[],"_revision_number":2},{"id":"bef9bf3ad35d6f6d70ac480f8b768f2740b77c6a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-26 16:17:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"76dcc161a4085257c3c9ccd211d3aa25364a6fe3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 17:41:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e72b1d2c22b04b7b8741e41a8f3c0980 : SUCCESS in 9m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/1dc64ddc1bba493395988886e9692ec7 : SUCCESS in 1h 03m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1412b36b571d447b856c92a0499070fd : SUCCESS in 42m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/92f5e334385e4d3f97f49f7b9ff5fda9 : SUCCESS in 9m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7448b154533f4ec8a2b231d1bf34e041 : SUCCESS in 7m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1024b06229a64e25bfcbf0af13c4340d : SUCCESS in 28m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b7353b64531d45038169b4d0036544f1 : SUCCESS in 27m 32s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/71500bccb1b54035a12c335f698eaef3 : SUCCESS in 28m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4bae99ec84174f3a8999639767d1b067 : SUCCESS in 15m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/871e117625154bb28f48920f8fe62bda : SUCCESS in 20m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9feeb5b25f924878a2fbb1971c3e0aed : SUCCESS in 24m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5f39d9515a25450c81d5dfada193e469 : SUCCESS in 24m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/41916f1c296a4fc393019c1b868009c7 : SUCCESS in 24m 02s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/3edc8d2ce7dd452cb3dacad879721577 : SUCCESS in 24m 14s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/6ca4d8c9549b491dba99452b06378b89 : SUCCESS in 25m 42s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/be7a00aa75ef4e59bd1f60afed18c90e : SUCCESS in 20m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/abd0b332df9e40e4be4f1e2367ce6fd9 : SUCCESS in 7m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d4f06abf0696422ba7d975ebec17837e : SUCCESS in 19m 42s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b17a8676fd2f4ec7a3c41e1422bf1cc2 : SUCCESS in 50m 28s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/614a5558406347f096160bb2f78b27a4 : SUCCESS in 50m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0ead14a61fdd454c89d7daf2e2ea523a : SUCCESS in 24m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/56dff7a2539e4745bb426c40b2675742 : SUCCESS in 40m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/403a8c7d801d46db8c294c62ae84b69b : SUCCESS in 14m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceed10788a5843e59136566075d79944 : SUCCESS in 6m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0dd54dd640f142dc9116ac2108f2feae : SUCCESS in 36m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e99763999c540fe83ae0a1e37cec53b : SUCCESS in 40m 11s","accounts_in_message":[],"_revision_number":2},{"id":"12757c32ceba0ea7a41b8d17cd874ccf65f07955","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 17:41:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f332b4dec14b2d8c60902586767ef9565132eaa7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 19:13:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/454d8f84826e46ff89d0d173ba0fe48e : SUCCESS in 10m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/74510008bfb141e7bbf4ca81dacd12c7 : SUCCESS in 51m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ec83ce6a864c41a68e5860ccfac26775 : SUCCESS in 32m 02s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6a016bb6f7db403184028beb476a22a7 : SUCCESS in 11m 37s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/127e5dcd6d3b4b70932e1eb6c77cd872 : SUCCESS in 7m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/38083d5fb70a4e8995f26822989deeb3 : SUCCESS in 15m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3fdc82bbb649426fa116de0701fe5e0d : SUCCESS in 12m 22s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/232bfb2e3a594b09820f7762aebbd78f : SUCCESS in 14m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f22e97b67e56414db092d47b929d7340 : SUCCESS in 11m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/67a966c253b8416e85907af6a8cf035c : SUCCESS in 16m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bc3b3a10273843fab8b313cf243ed65a : SUCCESS in 14m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cbf35e9225df4333b032990bb29c845f : SUCCESS in 15m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6b6e1214553044aabb92df3f26cd1938 : SUCCESS in 18m 43s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/36fe6297ded643c2a3734288ae583580 : SUCCESS in 18m 03s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ac372c500bab4ad8bc0b4abd4931dd91 : SUCCESS in 16m 04s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f35e771b06254eaaae20f91948d7b051 : SUCCESS in 19m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dad2724e82f2447cbb9290c36317dbc9 : SUCCESS in 6m 18s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0c464a95612548cb925407f431aafbd7 : SUCCESS in 48m 44s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/79795d9c11d94e40b70fd23e0c9e8a30 : SUCCESS in 38m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/347fa415fa794464828ce8e2049bcb33 : SUCCESS in 39m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5df652adba4b4af3af7285a8acc03862 : SUCCESS in 27m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1297c8ca65fd43f480c37ee3ae54110a : SUCCESS in 14m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddd12ec47c3d4635beb7ba4941fcc45f : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6a47dac1861b44b08fc5a9c016ae40df : SUCCESS in 28m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71bb9c47fbec411d867b8156a69017ed : SUCCESS in 49m 08s","accounts_in_message":[],"_revision_number":2},{"id":"8590d32b56af5a844b053f6389529cbd9b3b9a35","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 19:13:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"613f4ff4ec5481c1d528f43fe30a64df6501ab39","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 19:15:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8e335236efe45dc802b1b76987b8a1a : SUCCESS in 54s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/faf35f9afa2c462aab432372f60da8cf : SUCCESS in 39s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/d70bc5288c284ca9ad68635a30acddce : SUCCESS in 50s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f2a4c50dce4f8e3322b1a016f7b15afee53f04a0","revisions":{"40243bfcce000b40e6822e12c8999d7611b03582":{"kind":"REWORK","_number":1,"created":"2021-04-23 23:14:26.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/787831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/787831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/787831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/787831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/787831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/787831/1"}}},"commit":{"parents":[{"commit":"c9b5d44e9e8d4a75e4a1e2d31036823560e9a0a1","subject":"Merge \"Update AUTHORS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9b5d44e9e8d4a75e4a1e2d31036823560e9a0a1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-23 23:10:53.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-23 23:10:53.000000000","tz":-420},"subject":"Include sharding cycle time in recon","message":"Include sharding cycle time in recon\n\nChange-Id: Id7e828a56c8a62a1f3e9a1dbbff5a56c928ac6b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40243bfcce000b40e6822e12c8999d7611b03582"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40243bfcce000b40e6822e12c8999d7611b03582"}]},"branch":"refs/heads/master"},"f2a4c50dce4f8e3322b1a016f7b15afee53f04a0":{"kind":"REWORK","_number":2,"created":"2021-04-25 15:12:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/31/787831/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/31/787831/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/787831/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/787831/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/31/787831/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/31/787831/2"}}},"commit":{"parents":[{"commit":"c9b5d44e9e8d4a75e4a1e2d31036823560e9a0a1","subject":"Merge \"Update AUTHORS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9b5d44e9e8d4a75e4a1e2d31036823560e9a0a1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-23 23:10:53.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2021-04-25 15:11:49.000000000","tz":0},"subject":"Include sharding cycle time in recon","message":"Include sharding cycle time in recon\n\nChange-Id: Id7e828a56c8a62a1f3e9a1dbbff5a56c928ac6b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2a4c50dce4f8e3322b1a016f7b15afee53f04a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f2a4c50dce4f8e3322b1a016f7b15afee53f04a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
