)]}'
{"id":"openstack%2Fswift~787868","triplet_id":"openstack%2Fswift~master~I7d5c19304a5c5f83558e91624de356caaf0ab4d5","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I7d5c19304a5c5f83558e91624de356caaf0ab4d5","subject":"auditors: Log and dump recon *before* sleeping","status":"MERGED","created":"2021-04-25 05:22:11.000000000","updated":"2021-04-26 05:51:20.000000000","submitted":"2021-04-26 05:49:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787868","meta_rev_id":"d32d4225bf89f904cf76e0b740ae131cad885998","_number":787868,"virtual_id_number":787868,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-26 05:49:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2021-04-26 03:49:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2021-04-26 03:50:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-25 06:27:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-26 03:49:58.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"36973fec7e17a953c22169e2ffae7535c236fcd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-25 05:22:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15c2af3cf2450fb8a6725339ad467794ccfec1dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-25 06:27:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bf5987695e04b8b9330fb9d69d49fae : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/35e96466bd5140248085757cb75288cc : SUCCESS in 57m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3ae14de2227341b387c98a78a8f0e6d3 : SUCCESS in 38m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e71e8eb536fa4599b577d9878173531e : SUCCESS in 9m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ea2f5fb5ed2446ab82c6992966ece425 : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b27198d7db2245d49ab93e4f132e9329 : SUCCESS in 13m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a8ec136342124ff19ed88825f2008e37 : SUCCESS in 16m 48s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f3e6697601e246bc9c79ecf8c01fb92d : SUCCESS in 15m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/163b4f21fd7e49809d76bcfd3440a2c8 : SUCCESS in 19m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/df07c05b3599404ea7c4bb379767a7c5 : SUCCESS in 18m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b43ec274d24c46818ed73e82ee6691e0 : SUCCESS in 20m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a6b668c187c8466b8d5d56a6940748c0 : SUCCESS in 20m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/879f87c2120646e281fc3be7d5187410 : SUCCESS in 21m 40s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/64b932801772441793fe2f0a1171bfb3 : SUCCESS in 18m 46s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/7369be5fdc3a4ca8b3f2a9e611729f8b : SUCCESS in 20m 53s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/68e5403d83e44fe0a28ba900676ada5f : SUCCESS in 21m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b3a1f4026ae9489a8b1553b35d0b1c9a : SUCCESS in 6m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/448873b21b1e45b5bf0e592dba7af2aa : FAILURE in 24m 23s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ab275191714d4d36941f497db8192d4f : SUCCESS in 52m 20s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/ee8bebf81c4e4ea793263c8914d3155e : SUCCESS in 50m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/37e42bbda1d84e808e08190f6e55a0c5 : SUCCESS in 33m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7b60aec6fed346fea4644682e38d05c8 : SUCCESS in 40m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4891036f570e484f9dde84070a0ee9f6 : SUCCESS in 14m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe35314a3ae94d6f97221bdfffea11ea : SUCCESS in 6m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/094f7ef719374f70bd9dce6a3cc40284 : SUCCESS in 33m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18bc51321cc143bab099e280e8cb4b4a : SUCCESS in 51m 06s","accounts_in_message":[],"_revision_number":1},{"id":"9e6166b9e568174f31d57579bb5e19ea01035148","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-26 03:49:58.000000000","message":"Patch Set 1: Code-Review+2\n\numm, yes! So we\u0027ve been pausing and only updating the status once we reach interval, yeah, let\u0027s fix that!","accounts_in_message":[],"_revision_number":1},{"id":"1d3bedfd0309d99848947469ea695e47c59d7a33","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2021-04-26 03:50:23.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3e8029e3eb8c2fd68c0c8bc6464340d5fe3c4753","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 03:50:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"baf847d64a91fb3b28240e02c0f39c924699dd6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 05:49:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ac872a3a8c44197b2c1e04ed3a03db9 : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/e33b4d1d84c9454d9b49126317ffb4e9 : SUCCESS in 54m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/243ce0820581489eaf6fecfeb8abbc98 : SUCCESS in 46m 50s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5d16eeed22c1468387ccc2726174cb70 : SUCCESS in 8m 50s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/67c56f9aad0f40618d5eb8899414af6c : SUCCESS in 7m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8d1c1a4e480e45d099c17953dec81527 : SUCCESS in 18m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/82b14ffb29e24004b9a34ae74e95c118 : SUCCESS in 18m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/32dc98d3497b486bbb16be1192a901bd : SUCCESS in 18m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c58154005c124467b660ecdb50d68f4f : SUCCESS in 18m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1fbf4aa5d8764d65ae101b20d3fd49c8 : SUCCESS in 19m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c6e4a21b55c745ba8205f17a37a9abd7 : SUCCESS in 20m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/793e73c98ac749b391e66c4b25cc34b5 : SUCCESS in 21m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eed100a030044c259bfb88aa760d5ac3 : SUCCESS in 18m 12s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/9d5cafe4f6284145a787f92b99a759ec : SUCCESS in 20m 38s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/9fd589c7860442dc985059168ff703d8 : SUCCESS in 21m 04s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/0c851cbfe8e44f92bd465b195574cba0 : SUCCESS in 20m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/640053b589b34a7a9cb597a7c192b202 : SUCCESS in 6m 17s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/316064d083e04c3ea6ec2738fad038f7 : SUCCESS in 55m 32s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b7872f435b304fb3a55c8804049b3eae : SUCCESS in 45m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fb2fa164bac9400caec9ec370b2e61cf : SUCCESS in 39m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/59806d16e4ff4503813c11f38e7ec562 : SUCCESS in 49m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b7abc4eb357a4f87ab4987e02626467b : SUCCESS in 15m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69d652dd750c4c5a96ed2ebd45c1846e : SUCCESS in 6m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/07f669bc8bff4282af5edba461949aae : SUCCESS in 25m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d691c27e1d34d7a80df3e24ae03d816 : SUCCESS in 51m 18s","accounts_in_message":[],"_revision_number":1},{"id":"32ea5bd3d47d26e7efeb9109bd5c9bd3ab66ce37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 05:49:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d32d4225bf89f904cf76e0b740ae131cad885998","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 05:51:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40c8fd74505947d2b8df043e62514776 : SUCCESS in 54s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f7c6f0dc37dd4fffb7959f640ccb6d90 : SUCCESS in 40s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/d3f03ebbef944763ac7007d90ba815e6 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9d6006f6463c24404a8b80ed5535a1eff63d39ad","revisions":{"9d6006f6463c24404a8b80ed5535a1eff63d39ad":{"kind":"REWORK","_number":1,"created":"2021-04-25 05:22:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/68/787868/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/68/787868/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/787868/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/787868/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/68/787868/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/68/787868/1"}}},"commit":{"parents":[{"commit":"413732637850623ab7b41dc94483f01babb7712c","subject":"Merge \"Refactor db auditors into a db_auditor base class\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/413732637850623ab7b41dc94483f01babb7712c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-25 05:21:33.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-25 05:21:33.000000000","tz":-420},"subject":"auditors: Log and dump recon *before* sleeping","message":"auditors: Log and dump recon *before* sleeping\n\nNo good reason to delay getting the info about a completed run out there.\n\nChange-Id: I7d5c19304a5c5f83558e91624de356caaf0ab4d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d6006f6463c24404a8b80ed5535a1eff63d39ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d6006f6463c24404a8b80ed5535a1eff63d39ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
