)]}'
{"id":"openstack%2Fswift~788067","triplet_id":"openstack%2Fswift~master~I74adda0385ad4e30cb09fa2604d070694707293b","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I74adda0385ad4e30cb09fa2604d070694707293b","subject":"Recommend setting recon_cache_path in [DEFAULT] section","status":"NEW","created":"2021-04-26 19:33:34.000000000","updated":"2021-04-27 08:05:31.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"76f40a5d2affac884b89ed74b3b423b759d7fb9b","_number":788067,"virtual_id_number":788067,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-26 20:38:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2021-04-27 08:05:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-26 20:38:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-27 08:05:31.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"465dc2e87a836476034f6d23c879bf2da2ddc698","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-26 19:33:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2563fc140f5280387f35eee6db847f23eb293ad4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 20:38:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dee3f2d821554ebebba158cfdbd32aea : SUCCESS in 9m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/8fadf82a944a4803867aa1c6c2c1b209 : SUCCESS in 1h 02m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9660ff253e2b46cb831e6bc96ef029e8 : SUCCESS in 1h 00m 35s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f7a74dad88824505b9a7afeb0ef85467 : SUCCESS in 8m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/428317f9a58f4ca682e08b46c19c9ae3 : SUCCESS in 7m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f737498acfd4491db301562f0f90d6cd : SUCCESS in 12m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/027d58105cfa4323813e169a5df782cf : SUCCESS in 29m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cffd378c42494014be99c1c3cebcb37c : SUCCESS in 13m 36s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d1155756087940b9a54a612497bbab1c : SUCCESS in 20m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5daec218b4dd4aa19c24c2f725c586ab : SUCCESS in 32m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/48eca5473daa47b1927e79dbb0a12e39 : SUCCESS in 13m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/07f8b51306ce4d17a61646133749de8e : SUCCESS in 20m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9b02604e533b478cb805718c78ce83e9 : SUCCESS in 13m 52s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/e60f240a8f17414e896c3c3121d3fbc8 : SUCCESS in 38m 27s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/f28d7db585894131a3821cbed1616ff6 : SUCCESS in 18m 41s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/388eb5ac0f334cfd97c9401e7621f341 : SUCCESS in 18m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/03423ef4fcb14594b93579bd9b9323b5 : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/066e57ec02134a56b16bc0f82600c05c : SUCCESS in 19m 09s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ce12e966b65b477c81314b78fae9f9c5 : SUCCESS in 56m 36s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/b88e1b295f884f0ead270b83322265ea : SUCCESS in 49m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0c8c504e705f4c9d8465937614ae848e : SUCCESS in 45m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/70a22746291f4d7da352153da787441e : SUCCESS in 38m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32978478b16d4bacadace6c4ced8527b : SUCCESS in 18m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/85bc94915fcf4e12b2fb62aa0c8ecf23 : SUCCESS in 26m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1159231a48324ede8953a986e1f6edcc : SUCCESS in 1h 00m 09s","accounts_in_message":[],"_revision_number":1},{"id":"76f40a5d2affac884b89ed74b3b423b759d7fb9b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-04-27 08:05:31.000000000","message":"Patch Set 1: Code-Review+1\n\nProbably worth adding a note that the option can be overridden per-section, or leaving the per-section entries too. Actually, maybe I prefer the latter because it retains the explicit indication as to which daemons dump to recon at all (which is perhaps *all*? but nevertheless, more info might be better).\n\nCf. The logging conf options are repeated in DEFAULTS and in each section with a note about overriding per-section.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d54b432c296acca46afd3411ca2a772cd3006d62","revisions":{"d54b432c296acca46afd3411ca2a772cd3006d62":{"kind":"REWORK","_number":1,"created":"2021-04-26 19:33:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/67/788067/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/788067/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/788067/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/788067/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/788067/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/788067/1"}}},"commit":{"parents":[{"commit":"8bf93f1f40f213428188d17216c0d3af864448fc","subject":"Merge \"Include sharding cycle time in recon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bf93f1f40f213428188d17216c0d3af864448fc"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-26 19:32:02.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-26 19:32:12.000000000","tz":-420},"subject":"Recommend setting recon_cache_path in [DEFAULT] section","message":"Recommend setting recon_cache_path in [DEFAULT] section\n\nYou can still set it per-daemon, of course, but most of the time\noperators seem to want all daemons for a layer dumping to the same\nlocation.\n\nChange-Id: I74adda0385ad4e30cb09fa2604d070694707293b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d54b432c296acca46afd3411ca2a772cd3006d62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d54b432c296acca46afd3411ca2a772cd3006d62"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
