)]}'
{"id":"openstack%2Fswift~788398","triplet_id":"openstack%2Fswift~master~Ieafa72a496328f7a487ca7062da6253994a5a07d","project":"openstack/swift","branch":"master","topic":"bug/1925782","hashtags":[],"change_id":"Ieafa72a496328f7a487ca7062da6253994a5a07d","subject":"Make dark data watcher ignore the newly updated objects","status":"MERGED","created":"2021-04-28 03:19:09.000000000","updated":"2021-07-02 20:56:16.000000000","submitted":"2021-07-02 20:52:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":48,"unresolved_comment_count":13,"has_review_started":true,"submission_id":"788398-bug/1925782","meta_rev_id":"2f18a1e5793dae762c65c6b32a925cfba2ddba4e","_number":788398,"virtual_id_number":788398,"owner":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-02 20:52:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":2,"date":"2021-07-01 08:49:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-01 20:10:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-01 20:10:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-28 04:23:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-19 23:44:24.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2021-05-21 01:12:49.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2021-05-26 12:03:03.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"33ff975e5d270161ead921f9fed139c2a204f15f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-04-28 03:19:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"676884eb4628abd0c2bf7feea1ed69415d5b0caf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 04:23:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26e7bec38b8248f184f865fc62939494 : SUCCESS in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/c19755952fc14d58a8c23809a3a6cf71 : SUCCESS in 1h 02m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6f6134fc588a41658bbcae5fbf03bb61 : SUCCESS in 48m 49s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/285e12792a5b42b09d68d254c54b50c7 : SUCCESS in 9m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/034e8a3dfc004391b5a48a77bef1dcfb : SUCCESS in 6m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8cc3436f3f574c64b15743fc124b0504 : SUCCESS in 14m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/69449dde1f7242b7af4fdfe3856219d1 : SUCCESS in 15m 37s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ddb50e81582a4070893d13161a2b0f97 : SUCCESS in 20m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f012447c29ad4f65af8b0b4b8e8594c1 : SUCCESS in 17m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d8e0518bdfdd4ebaa83e34100e4455ec : SUCCESS in 16m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e25dfa05b3054d839bb845e701c01764 : SUCCESS in 20m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4d33464813b94c84b0eef7a24d0fabca : SUCCESS in 14m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f4425918014140e6b0b515dd438b8c18 : SUCCESS in 18m 47s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/001f2a9be39d4bf5966b8184831c2b5a : SUCCESS in 15m 11s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8d4b1ca516c746ff8bce5bd660f490e5 : SUCCESS in 16m 51s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/7d7771086cbc4e45b97de463f6d013a6 : SUCCESS in 20m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c0414cea2e794e958e67d373d0b3b84c : SUCCESS in 6m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f3cdbc57d2ac48aab47ba5432f6962ac : SUCCESS in 16m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/81027bcb46974ee2910625c031c11592 : FAILURE in 52m 21s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/403daa4141bf43748ce7b016953d32e4 : FAILURE in 50m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b782fd08412a420886af2788f849f4b5 : SUCCESS in 39m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8cbe6204451d4af7bf77ebea7df29bbc : SUCCESS in 54m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd7d6fc169624826a2101fec3a6cd9a7 : SUCCESS in 11m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb6522083c124201943dfdd7607d35c3 : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6a635f8a9d234d8a8ea594f66bdb4314 : SUCCESS in 17m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4fd4c30d66674b35a7e7129cee948956 : SUCCESS in 55m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6abf68193b6ab1bbfe5ccac707000745c8ece720","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-05-13 05:18:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a86ef679342dc995ac9bafbb4c5faa1d34363a61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 06:21:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dfbacef76fa457cba3fae55d7f3c389 : SUCCESS in 7m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/b0ef810a2108478a82150f8c6eebf175 : SUCCESS in 48m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1e90dbcd830f46409fa71f62232eb89c : SUCCESS in 35m 01s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/52a4ba370dbc462bb81e8f4df84a15cc : SUCCESS in 10m 14s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a9eec3fb93974e8cb4bdcaa5823f6d8e : SUCCESS in 7m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e993f6c00f524912bb42dd3f456c2cfa : SUCCESS in 19m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d16287e4164c4214b7a8a597897f5d78 : SUCCESS in 16m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/08767cf863274331ae3cf2087ce322c2 : SUCCESS in 14m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a88bb453c6f84c8e844c5e5fe94dc0c5 : SUCCESS in 16m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/73e41e0dbf3b4d6db7295e61912d11fb : SUCCESS in 14m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7214934b4d77430abe2daa3fa607f96d : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7a315d686a7b447a8dea8902d5229a35 : SUCCESS in 17m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2d502a6490ee4fe9bc32b59501f36971 : SUCCESS in 15m 06s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/c8a650f43276446f8851767d5cf8cc10 : SUCCESS in 27m 31s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/71dfa02227404e08968595e1943ca150 : SUCCESS in 18m 39s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/8704d6b59bdf409f8e665cac57dea1a6 : SUCCESS in 16m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fb78a24a0b1f4f6689ede26c37fe63ea : SUCCESS in 7m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8f5bfa29a05f414b9b1d293fce865121 : FAILURE in 20m 12s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/4f74a6385fef49de82984932d1463ec0 : SUCCESS in 48m 21s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/5a38ca5cd6f744e488b8f75360e3e53a : SUCCESS in 46m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ce000d24422b48c189165ccb6a096b5f : SUCCESS in 40m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/84128748426b4a538b1d0c80b74eacbe : SUCCESS in 32m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be9c64309066441988259576c6ce6ba7 : SUCCESS in 14m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ede2cdd66b5420b8ca4dc9420f0416e : SUCCESS in 5m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1ecfcfaa43ca4a12b947806c88a987b4 : SUCCESS in 21m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb6136fdac1f4ac6bbbc7459c005202a : SUCCESS in 57m 21s","accounts_in_message":[],"_revision_number":2},{"id":"dd5408293b0ce4a8e7d017fef35511f60e31d0f8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-19 23:44:24.000000000","message":"Patch Set 2:\n\n(3 comments)\n\nI\u0027m a bit surprised by the mtime -- why not use the timestamp from the file name?","accounts_in_message":[],"_revision_number":2},{"id":"301889ab33cb89a2fa6d23cb952a6fcba8c48cf2","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-05-20 03:23:10.000000000","message":"Patch Set 2: Workflow-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ebce212c31aafe7aa8076f355ce013868919c8cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-05-20 19:40:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5c92c4f83a43b98cbb23e3e4a4ce432e54fe62c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 20:44:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc87aef888d04d23850828689cd6bb3b : SUCCESS in 8m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/7505dd1e004e413c827770924234d948 : SUCCESS in 1h 02m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2a0f075bfdc24dcfb8e30b290f6aade1 : SUCCESS in 37m 33s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c7fae71ae8414d0799d309b82d245363 : SUCCESS in 8m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/dd3d047dafe14692b2f559301ac16b8a : SUCCESS in 7m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a4477b7a4d11422eaa7cc82d91ae321b : SUCCESS in 16m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9d8ccc91ce574e128efc77f7575d04a9 : SUCCESS in 19m 53s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/11247be3e1624e52924f6a42dd1d35bf : SUCCESS in 19m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5da315d169364decbc72c13ab4eb4e33 : SUCCESS in 19m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/263bfcf7eba049f898a5b4fe9974a411 : SUCCESS in 17m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b36183088372440c81b0365c98192e21 : SUCCESS in 19m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a947052c04f347f48e2129401431f863 : SUCCESS in 20m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d8e4526a81b340e7b5187884b7f1de8b : SUCCESS in 21m 13s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/df21eb12a6274a8d90e3e75a45df0954 : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/cf8d92f8cf4b4c14aba37f4a789c09ce : SUCCESS in 21m 24s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/31732b38e0a643769ecc09217c6bdd2f : SUCCESS in 16m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a8190b0915ee44f19be2d03555a2233e : SUCCESS in 7m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a247d6364f2e46b781d405cde25a4b74 : FAILURE in 45m 07s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1ff4fbddb7b6474b802d3f8caa26b797 : SUCCESS in 54m 41s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/aa702cb89f4b4ca2b8c37b632c76cc75 : SUCCESS in 50m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/68102a58d0ba42e8bcffb3cbf332ffd1 : SUCCESS in 40m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1f01f10bf42649b19bd1784b60ee3fc6 : SUCCESS in 40m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25fa7d7163e545ba84253f4a35ce48d0 : SUCCESS in 16m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ce4522544cb423bae5efa7e09a60480 : SUCCESS in 6m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/12879fa14ab34290b7e3466937a0a891 : SUCCESS in 31m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/441fb5f103624e34ba70344d18f86160 : SUCCESS in 52m 30s","accounts_in_message":[],"_revision_number":3},{"id":"611d94df3cabc72dcd948d240a74b24a6e7fc7cd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-21 01:12:49.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"90e8db9cb9d5e6745bc758cd2bc0ebb1004915d4","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-05-21 02:59:55.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d938680b9ac4da6e1fdc704e18188b306907bbdf","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-21 18:49:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"05c23abe77a46f03f1665ff5b53d95bc90d42479","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-05-22 02:21:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1accc641e8beafb69ca50014593e30f916eacc38","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-05-26 12:03:03.000000000","message":"Patch Set 3: Code-Review-1\n\n(12 comments)\n\nThe change makes sense, but -1 mainly because of the missing test coverage. I left some suggestions here: https://review.opendev.org/c/openstack/swift/+/793148/2","accounts_in_message":[],"_revision_number":3},{"id":"aa4519e88eefa6a6cb30afc34e2a1a745baa4284","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-06-02 05:22:19.000000000","message":"Patch Set 3:\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ca1e4ac0a4d55901ef89c0418be409c699c202b6","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-06-02 05:22:30.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"052b78533c1a8cd05052d4ac5af4383f1dbe3c30","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-06-03 06:13:24.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6e67e44901c7f5e9dadc840ca00bd059c5c04c33","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-06-03 16:25:01.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"dcbf5f2477f3a6ceafe973880965c4dd40f26a4c","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-06-03 17:09:42.000000000","message":"Patch Set 3: -Workflow\n\nOkay, I think I dealt with everything Alistair marked. However, I postponed any changes to the probe test, and I promise to add the unit test for container servers that error out into the \"complete agreement\" review https://review.opendev.org/c/openstack/swift/+/792713 .","accounts_in_message":[],"_revision_number":3},{"id":"36af0e9b1b52827b1d5ba7617a36da49fceb2899","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-06-03 17:10:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f497a5307aa09cbaa39fd31df829c065c5600aae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 20:35:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79239293a66a47648f973af33c1af6fb : SUCCESS in 9m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/0466ffaa9f0340e194e7dfcabe2e669b : SUCCESS in 1h 12m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6630439d748942aca2a2ad6d834d75c7 : SUCCESS in 55m 14s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/80898191106749aca16ca3c47693432a : SUCCESS in 7m 55s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/78b91a7d71ee4539b50539c15e768338 : SUCCESS in 10m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/878ee7509c1d45e995381eb0f8437173 : SUCCESS in 21m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1048491c51c34c918295e8f8019f6f5d : SUCCESS in 19m 07s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d6654512e49843feb1c10b27bf73f542 : SUCCESS in 23m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3a1a4ead6a814278935be45fd311bac7 : SUCCESS in 14m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/93e529264aa64c1ab44518a197a2d24f : SUCCESS in 16m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e2f4c34bd23f4b9ca8eae72e0e113a24 : SUCCESS in 22m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/97e8662b891d45c0a98912218410db90 : SUCCESS in 17m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f8224a28e3ec42319c9bab750309c0fa : SUCCESS in 17m 37s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/4e2a12a143a545218569e4dab2382596 : SUCCESS in 22m 58s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/8f06118f57e64b0c9471656432f9ce0d : SUCCESS in 19m 29s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/b442e9807acc42ea93b256f3defd3674 : SUCCESS in 22m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f1cb9a1f8eaa4b618f45c573cdd1e519 : SUCCESS in 7m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1ef7720b91a34801898c9376b9337ca2 : SUCCESS in 27m 10s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0fda32a7f90f4540b8dd02fce867145a : SUCCESS in 1h 05m 20s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/a15fc7fea6e84863b50e1a14109a4ae5 : SUCCESS in 45m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/84fed0a4350641cab43525e6ced8a965 : SUCCESS in 37m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/74b404e4c7dc4332a0fabcca3b7e31b1 : SUCCESS in 45m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d1c734ed781146bd8057c1ecb22197fa : SUCCESS in 24m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2de6d95625bc47a4b6b07aff24c5e0dc : SUCCESS in 6m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0e70e1fe574f46b390d99d76e8d16eb8 : SUCCESS in 22m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f4a981872d246318857855be3e344ca : SUCCESS in 54m 08s","accounts_in_message":[],"_revision_number":4},{"id":"6a1425bb7fedcbac02ce9362060dae3158322405","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-06-09 20:26:08.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9f859e03265fc676dcab97b3e7b5ec2c552f492f","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-06-10 13:26:19.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"93d60fdc71f6558d23ba654f5ff0ddc5ef4d5ef8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-06-10 13:41:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bd961a28f6b07464aa2adcd5bf8a1c0850a65a91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 14:36:22.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a802b9dddc34a739471da15ca293bbd : SUCCESS in 42m 44s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ca073003fea4498e9f872fcfcb6bd7dd : SUCCESS in 48m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9443542e33f0ff190d3211b3dd83a5a75fce4c3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 15:12:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c23db210634424ca574f803533a17c3 : SUCCESS in 8m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/70a233f3ca6e4818aa234d882aaa2237 : SUCCESS in 1h 28m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f0f419601c4e4c029162bdb1ef63d614 : SUCCESS in 39m 38s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2b8877565b7e47aca715c61bf2145b09 : SUCCESS in 8m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/11537f8222e54f23b4b1882374d3e59b : SUCCESS in 9m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/31189d8a280648958a4380cd74d37049 : SUCCESS in 13m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1562e48d5e044cc4813d3cf6e91d96e3 : SUCCESS in 15m 55s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/83beb8316d1546bcbf2a07b28a5266f6 : SUCCESS in 17m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/29dbbe48f475447d9c2786fe300ca548 : SUCCESS in 16m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e8831ff206dc4c738891a232cd7a4bc4 : SUCCESS in 15m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9d64b804c5d64c67a0bd4fcabf6a68f4 : SUCCESS in 18m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/92e11284ad5046e4aead6042c694d0ae : SUCCESS in 18m 42s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/926ff9da00f74dd19e94474b412781e0 : SUCCESS in 18m 09s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/430aff84dfc4431bbf02d2f846fe5783 : SUCCESS in 18m 52s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/53aff8324d734a6eb7c503e911bd4510 : SUCCESS in 19m 03s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/874b0b43bd7c42b381988e7807e4473b : SUCCESS in 21m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/548b8beb1a564e208ff3758f638b481b : SUCCESS in 7m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2e405fa814544ba78430dd29d93543c1 : SUCCESS in 22m 26s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/06663b4c14004af197c5336839770c45 : SUCCESS in 53m 51s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/123988f1411c4a0db68c2324d93d1d36 : SUCCESS in 49m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1b171c98416247c8b9458c4971dcf87c : SUCCESS in 41m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f3ba1a16547d4a72b70fab6e70c91587 : SUCCESS in 38m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18be192d01004a1aa8060a3720c1d7ac : SUCCESS in 17m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb4955cb6ca648f3bf8c13c11a823f56 : SUCCESS in 7m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bfc59a27d72242bf8efc3067f363bd07 : SUCCESS in 29m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/456c3a79054c44ecba933313efc62dcf : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":5},{"id":"b3927ae50facc6fe3bacb21bf46f06af4051a65e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-06-28 11:47:06.000000000","message":"Patch Set 5: Code-Review+1\n\n(8 comments)\n\nThere\u0027s a couple of doc typos that are worth fixing and I think one more unit test would be useful, both are offered in https://review.opendev.org/c/openstack/swift/+/798287, feel free to squash that in if you agree.\n\nOtherwise this looks good.","accounts_in_message":[],"_revision_number":5},{"id":"4e16252020efc7e07dae4504449999d18d211fe7","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-06-30 21:37:55.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"8d727b04d31f24c7fc781576572d7ab6088bcb4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-06-30 21:39:10.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a9798cb569180d19a9babd3bdb2d62f0a4760548","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 22:34:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/990f340d37d348f6acc2f46d163a6173 : SUCCESS in 36m 56s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eee730579ced4f3c9d2f1155f95d4d6a : SUCCESS in 53m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"331c17df2b89ba3b62ebb8403ea3510dd97afdf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 23:17:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faab16e2dc734e059183d3b7704b7340 : SUCCESS in 9m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/a227da97733742809b296a0fdea63842 : SUCCESS in 1h 00m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/51d2c3b7885c466190ba37b1b5be514e : SUCCESS in 41m 09s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fe04d8067fc24abf8a72510faac9abf2 : SUCCESS in 9m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b59cc1e1b35a47b48f66065c0ecb4bfc : SUCCESS in 9m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4d646cfacce54067b4c0ea5d544e0e8c : SUCCESS in 14m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ef338bd934874f4fb4c4b16654333319 : SUCCESS in 14m 35s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/50e4aa3d92f049078998abc9bc023dcd : SUCCESS in 17m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e163fc9802144cb9b6037c0e9f4fedb9 : SUCCESS in 16m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/16f5a104adf7477abf16c5de16b5655a : SUCCESS in 17m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3c113ff0b950410f80398be31708c055 : SUCCESS in 17m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e5d9262cb6d34a1e9e920ecb6a6e877a : SUCCESS in 18m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4cbdbe415ff943be8b2ce8347b47a1a0 : SUCCESS in 20m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6fdcbd07f55e48d6aaed9e0e114a7024 : SUCCESS in 18m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b6ca0a7c9ae5424aa34f8f90ba3074ac : SUCCESS in 18m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ad24c280b2854e3291c297adaa254df9 : SUCCESS in 45m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ebeade3bb44c4a31810e5429fea55ec4 : SUCCESS in 9m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/481bdda555a246f7912afdaaf6831961 : SUCCESS in 51m 53s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8e87c37961b44f32a07ecf89272cc6ab : SUCCESS in 1h 32m 00s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/52300299ba0e4a75a7b2e2cb7f4204e4 : SUCCESS in 51m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b989228821d14c559cda9caf5f6ec85e : SUCCESS in 37m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8b015956b7eb44238b11ee4cb54562ee : SUCCESS in 41m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ad4c6806cd64eba9b61c4aebe3b8187 : SUCCESS in 15m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41e593e7f4b1474ba7a3013e9459ed2e : SUCCESS in 5m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8fbbca7828f549ebab5171adc089c326 : SUCCESS in 30m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5a3226f821d4afaa28a60553fd3887f : SUCCESS in 46m 50s","accounts_in_message":[],"_revision_number":6},{"id":"f2eee85710abaea430366839e3e32ed20c2a82a3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-01 08:49:15.000000000","message":"Patch Set 6: Code-Review+2\n\nLeaving opportunity for Tim to take another look before +A","accounts_in_message":[],"_revision_number":6},{"id":"8aade1f42f7e94bd980bda03671cd64ee7a1e32e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-01 20:10:42.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"32637702263fef893856d73f5d05ad713efc4f88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 20:11:06.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"debafd03931fc53f934a2735952c176bba903ca3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 21:13:19.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f427b1e025c497d8952191b24019755 : SUCCESS in 10m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/433e6259ee8e4cbbb525a7e0849e33eb : SUCCESS in 1h 01m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6860e18d862f40ef8ab1e90315aa5d74 : SUCCESS in 40m 12s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b030652460da4f9eac77708d7db62037 : SUCCESS in 8m 28s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/1a7c5395bac5437fa68a3db2cc27d5cb : SUCCESS in 9m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/60a67bbf27e4457b81cb93adc5492ecf : SUCCESS in 20m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/20b28d5cdce7471db397feffed059608 : SUCCESS in 21m 04s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4d29a2f973624cdba56751a4ac09bda5 : SUCCESS in 17m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/90c1470d6d3344fcba531ae940135de3 : SUCCESS in 20m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/273ad143caf94f8199915df4a67e768c : SUCCESS in 17m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/67cec0c4b80f44a68aa7f297a7846b84 : SUCCESS in 21m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a41601da9e7c4aee9a6dd0bc2c460ff2 : SUCCESS in 20m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/df446cc6767042baa9c98172207eb443 : SUCCESS in 51m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a388b67487dd4e19a9ed80b8271727d5 : SUCCESS in 19m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/049783a9edc74380b516bed0916eaf67 : SUCCESS in 20m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ca253a130a5e4ea980497223b3d6b452 : SUCCESS in 19m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/68bbfa621f684a6fa543dfe3c095cf73 : SUCCESS in 7m 08s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/771e8e453cdf42b49919457335e5ffa8 : SUCCESS in 50m 23s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/8e2a582dd9cc4c6ead0c93157d7fc0a3 : SUCCESS in 49m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fe72fe2a65874136bb5e294d6dc6ea3c : SUCCESS in 39m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0257d36bae7a4c2f9e47d21da68f1079 : SUCCESS in 41m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fc94edac18b4775ba9088dee49fa696 : SUCCESS in 16m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed815bd31c284f05bd542d9a71113361 : SUCCESS in 6m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/23af203aa823408fb6d4fd8cc79bd6bf : FAILURE in 36m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5a61a07266f43fe9c6fbfa53dd5d19d : SUCCESS in 49m 05s","accounts_in_message":[],"_revision_number":6},{"id":"d5d77c8f2ef10eb5a1d19cbcb66f41b8d1fbd2ad","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2021-07-02 10:12:01.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"3c4db6537fe0f4d42ff6b396134e5a6fffec3762","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 11:20:59.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ecbef2b778a745c8a45bc179a87bf27e : SUCCESS in 49m 56s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5695d2960b6d413eb84bb0700547d2dd : SUCCESS in 1h 06m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5c92603b97e321817139b895549f367c0ae70c78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 11:23:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c0729c31cc74af597993637200fee28 : SUCCESS in 13m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/0d9ff46f971b4a67849b34ff42f82092 : SUCCESS in 1h 03m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b53fdbd497b04862819ed09da9a3d5f4 : SUCCESS in 44m 54s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ac5c219a0b774ed3b49f7873fcb1e449 : SUCCESS in 8m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/eb6e8b44dbf245338fc83f0766762407 : SUCCESS in 9m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ba7ac414dc134aeab9b9ba5bd3562f90 : SUCCESS in 17m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fe562287e20042aab7062df0da709fac : SUCCESS in 17m 14s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/240f7fdb921647f1bc0b25c6cac59158 : SUCCESS in 18m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fea69f94023949bbb37681c3a0c2a273 : SUCCESS in 20m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6f542c00f3ae459c84efa576bffaab48 : SUCCESS in 16m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0d1d9f12983a4deca09d95b41d5465a9 : SUCCESS in 25m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9af1d7318b884797bcb9a738e071744b : SUCCESS in 26m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/565bd928a64e42da918bd12b5bc946da : SUCCESS in 26m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/dc8f7e38a33349d3bee8d8724fa9de6a : SUCCESS in 24m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6761f446cf41437284db0672ff02d37d : SUCCESS in 24m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9f7f2084906c41ab98f36a5db3bfd574 : SUCCESS in 24m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8f73b1fd3e6745be903a2e7744ec3bb2 : SUCCESS in 8m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5fd3b5874c0a44749dbb0dec31bb0e67 : SUCCESS in 26m 57s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f4fcfeefbaef4305b52dac0aa540cab9 : SUCCESS in 51m 35s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/0723a9ca71d8462d92ad44a5a7fce598 : SUCCESS in 52m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8f82652bad564a039008c02399fd941a : SUCCESS in 44m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c8166529684b4565b4db273a1edea376 : SUCCESS in 42m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e9020ded86048f2af340696f889797b : SUCCESS in 17m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86493537226c4f16b3d03c26e738f89c : SUCCESS in 6m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e7a6139036424c1f83a26371fb2fa0d9 : SUCCESS in 29m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54d4d4bcd03444a49b66efd153334602 : SUCCESS in 46m 21s","accounts_in_message":[],"_revision_number":6},{"id":"b48651f2852399dc24748d73ae67a33de92a35e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 11:24:01.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1bc3c101529b107a5df0974f2919ce2c3e168d90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 14:32:00.000000000","message":"Patch Set 6:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e6a0e5c10b3b81f7f194a0d316f7bf870f90e1ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 16:57:24.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e598752f87164ad0acb8b45b35dba582 : SUCCESS in 10m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/19e306be694845c89430006a2f22f47f : SUCCESS in 1h 07m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f9276eb083e9431bb2d051253a3327fa : SUCCESS in 43m 18s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/3b71f79bde98474d9b94044e26fb0a9d : SUCCESS in 11m 38s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/433dca0d4e6f425b937e4833da5ad0e7 : SUCCESS in 12m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4f01ffc002cd44dda0fb2fb52cde0554 : SUCCESS in 19m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5eae99601b6345ebb5ca8896f83ab1e4 : SUCCESS in 18m 47s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bbb1901c0b284124af57d1e373d21c8c : SUCCESS in 19m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0b092a968b79414abe7bfe64a2801ebf : SUCCESS in 17m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7ed88675c1394c50bc22941f777f3bcb : SUCCESS in 19m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b0039401c472428ba0fa9ef960a35b4e : SUCCESS in 19m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0b45e18d9ab447c2bf175e1df80c07f9 : SUCCESS in 21m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/17d29c41a84f49d1be9079692150b048 : SUCCESS in 18m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/aa8f6d982f9c4fd7b748bc80ed1b2ffc : SUCCESS in 17m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/296d388be96148398e117188da6b7b07 : SUCCESS in 17m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a97733a8ad9b410e9c5974c27cb57e8e : SUCCESS in 17m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9bb91d02e0b4482e9234d06b7e3d1f63 : SUCCESS in 7m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/e3aabb4c9a9441b7869810eef79eae78 : SUCCESS in 51m 38s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/851aeadd64ec413ab055b6ead6eca4b0 : SUCCESS in 50m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7fff0db1fdfd415494c2695428aba2bb : SUCCESS in 37m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fb754bd80f1c4681afcce4b63e25e879 : SUCCESS in 46m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6791256d699b43d087b5c919d3d2318e : SUCCESS in 18m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21f8ad1fb80d4646bd0a5e8a37dd07fe : SUCCESS in 7m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b4dcf8d9bb2f47bfb783017dacdb0c09 : FAILURE in 24m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/055ac5c263dd4aec9eecf171afe8fa8a : SUCCESS in 52m 59s","accounts_in_message":[],"_revision_number":6},{"id":"410ab5166699d509e8db24c6bc076060950e706e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-07-02 17:21:14.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"61d6b571cd53013f8ef62616dd51854faabdac99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 18:30:11.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bbfc55b82c974277a6f5dd6e4703289a : SUCCESS in 40m 44s (non-voting)\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c24679f5da4d4ecd8dfdfcae07394b42 : SUCCESS in 55m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d670ea1c3b579f44e21f52cad9dbd23b657585c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 18:32:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfb3029da1cb464b9c3418701bcedc95 : SUCCESS in 10m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/20ca539d50ef4696bb0c5167a1aafb9d : SUCCESS in 1h 06m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8023f62f7263466fbfa531d9ddfd2794 : SUCCESS in 47m 22s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2a587692ee4f49c0802c32aa3dc9dda9 : SUCCESS in 9m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/81a456334109415aba57b302e5cbd06d : SUCCESS in 9m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6fae3279148f41ca977bb581bae51826 : SUCCESS in 15m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c162ca0415d74c7b82b5a7076b50b78d : SUCCESS in 17m 11s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4959324d4b79436eb5b9bb04a11f4f75 : SUCCESS in 16m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c740a46346d74e1194d1e71b033f9938 : SUCCESS in 16m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bfab14368b15451aa3c7b5fcbd199acd : SUCCESS in 15m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/52d9565ada964bc69f95720cb1e3d539 : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/51664c2d80cc4fc286b04dcdc192c72b : SUCCESS in 16m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b56cbe1f7c8b4e42957befb1726d0899 : SUCCESS in 20m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/eddf2fcab175473fb5958cdb22b744ba : SUCCESS in 17m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b4027d4076f44571bbb7518ca460db94 : SUCCESS in 16m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e9badbbb8cbc4c1a8bc2dd722395ff47 : SUCCESS in 16m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/956a0a2c1d484c1aae08d4c46fd44013 : SUCCESS in 7m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/08848f2992434eaaaab440e31df6ce86 : SUCCESS in 19m 47s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/48db92a9e04a4fb3858b8505346f3d2d : SUCCESS in 47m 37s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/9f6b46870e444e059dec69e2bba08d67 : SUCCESS in 47m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2ced6f99129845a5a8e52afebc953c2c : SUCCESS in 37m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/83ddcbbf09164a40abd365be503eebf0 : SUCCESS in 42m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38dfb6d7477945b0892d8601eead6d0e : SUCCESS in 16m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/052ebc599c574873a0fa8e9ebfadbc85 : SUCCESS in 7m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bc52ed7588fa49df888bafaac37d0ec7 : SUCCESS in 21m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b4591e1f5674ed881fa7eeb5bcce8fc : SUCCESS in 54m 17s","accounts_in_message":[],"_revision_number":6},{"id":"113cf409e8f2aa9256d13b22cf7ba9170ceb47f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 18:32:33.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"d3755231ba8bbaaa0709add069f566120eb79e7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 20:52:52.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12febd40e119439fab6eeb73a673b061 : SUCCESS in 13m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/ee9f99a62e504161904a5d039591159a : SUCCESS in 1h 23m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b5ea4959ac8a41328bfb2bedfa2e4865 : SUCCESS in 43m 00s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/1d7fcfdb72fb46caa6094b41c9b3a305 : SUCCESS in 8m 35s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/236ad0c6b2ca4d209676485cc2cecd7c : SUCCESS in 10m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8ee4dd2e4ece45d999b99dac20003e71 : SUCCESS in 17m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/213667390ffc4241998163f49f85e7d4 : SUCCESS in 17m 29s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a80474bd6e5a46868c3b54596010bab9 : SUCCESS in 21m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/10a5f1ceb23b460fbc04d4394c453118 : SUCCESS in 20m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/abdd5ef570e24c168a4dc98b1f527572 : SUCCESS in 21m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4dfc7bb75fe04677a768da0f76235397 : SUCCESS in 21m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9ae8eae6c7b34c3897c09a6cbd78cae3 : SUCCESS in 18m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c6eda06219aa449eafe5f05a15a27b75 : SUCCESS in 22m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d42864de18cf4177866eca0127b6fca7 : SUCCESS in 21m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/af7c452501c849ae858ed3f26845d014 : SUCCESS in 19m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7f8f0d242fd14bf59732c00d599f8a80 : SUCCESS in 22m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6fc48f890a2442f8b7b6934d8bb84a5d : SUCCESS in 7m 14s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f539d1c776ef4fed9d76d5c8002de042 : SUCCESS in 49m 25s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/943b63c542304d69aeb27e50ed0e9086 : SUCCESS in 48m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/24d93039a4bf4b209d77a02e0e7124ef : SUCCESS in 38m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7fc73cfab5b449d68625acc8f20160b1 : SUCCESS in 41m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eeb4ab9273f6499ba9e54f88c5aa6931 : SUCCESS in 19m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0f4bbab3443415fa743a912cfe7010b : SUCCESS in 6m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/749bdb50547c43169d4f4c5352ccee66 : SUCCESS in 30m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c347c48195ba41b7ae6af3518f9bd718 : SUCCESS in 47m 35s","accounts_in_message":[],"_revision_number":6},{"id":"821e8996441ea22cdc0a28301a0408b4c772e04b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 20:52:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"2f18a1e5793dae762c65c6b32a925cfba2ddba4e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 20:56:16.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f7790d2ed324d98892592aff34df1b3 : SUCCESS in 1m 03s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/246632b3c1fa43759b41ecd54020e640 : SUCCESS in 53s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/c382f06f07d348da88abeb9fdf3bca5f : SUCCESS in 48s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"95e031645120d5ffedb4b15127ff701d8bc24f59","revisions":{"8306af1c3e5dc5242a8a3758e221c32fef762aea":{"kind":"REWORK","_number":1,"created":"2021-04-28 03:19:09.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/98/788398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/788398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/788398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/788398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/788398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/788398/1"}}},"commit":{"parents":[{"commit":"8bf93f1f40f213428188d17216c0d3af864448fc","subject":"Merge \"Include sharding cycle time in recon\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8bf93f1f40f213428188d17216c0d3af864448fc"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-04-28 03:15:56.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-04-28 03:18:48.000000000","tz":-300},"subject":"WIP: Make dark data watcher ignore the newly updated objects","message":"WIP: Make dark data watcher ignore the newly updated objects\n\nWhen objects are freshly uploaded, they may take a little time\nto appear in container listings, producing false positives.\n\nCloses-Bug: 1925782\nChange-Id: Ieafa72a496328f7a487ca7062da6253994a5a07d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8306af1c3e5dc5242a8a3758e221c32fef762aea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8306af1c3e5dc5242a8a3758e221c32fef762aea"}]},"branch":"refs/heads/master"},"e14144e75cb6071ff2c95ca0b4e2efa4e71b454d":{"kind":"REWORK","_number":2,"created":"2021-05-13 05:18:23.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/98/788398/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/788398/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/788398/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/788398/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/788398/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/788398/2"}}},"commit":{"parents":[{"commit":"6167dbf7ba15099fef26e6f00160e67ed7ff42a4","subject":"Merge \"relinker: Rehash the parts actually touched when relinking\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6167dbf7ba15099fef26e6f00160e67ed7ff42a4"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-04-28 03:15:56.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-05-13 05:10:05.000000000","tz":-300},"subject":"Make dark data watcher ignore the newly updated objects","message":"Make dark data watcher ignore the newly updated objects\n\nWhen objects are freshly uploaded, they may take a little time\nto appear in container listings, producing false positives.\n\nBecause we needed to test this, we also reworked the tests and\nfixed some issues.\n\nCloses-Bug: 1925782\nChange-Id: Ieafa72a496328f7a487ca7062da6253994a5a07d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e14144e75cb6071ff2c95ca0b4e2efa4e71b454d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e14144e75cb6071ff2c95ca0b4e2efa4e71b454d"}]},"branch":"refs/heads/master"},"9ceae2dfd258b95684449737c34cd171f436dd76":{"kind":"REWORK","_number":3,"created":"2021-05-20 19:40:42.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/98/788398/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/788398/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/788398/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/788398/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/788398/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/788398/3"}}},"commit":{"parents":[{"commit":"0a58574109e0e329ddb6ee78b7ed82b3a9c42c61","subject":"Merge \"sharder: report perfectly overlapping shard ranges\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a58574109e0e329ddb6ee78b7ed82b3a9c42c61"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-04-28 03:15:56.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-05-20 19:31:07.000000000","tz":-300},"subject":"Make dark data watcher ignore the newly updated objects","message":"Make dark data watcher ignore the newly updated objects\n\nWhen objects are freshly uploaded, they may take a little time\nto appear in container listings, producing false positives.\n\nBecause we needed to test this, we also reworked the tests and\nfixed some issues.\n\nCloses-Bug: 1925782\nChange-Id: Ieafa72a496328f7a487ca7062da6253994a5a07d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ceae2dfd258b95684449737c34cd171f436dd76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ceae2dfd258b95684449737c34cd171f436dd76"}]},"branch":"refs/heads/master"},"7d3592a82d96c5f010953bea42974da6a0baeb2e":{"kind":"REWORK","_number":4,"created":"2021-06-03 17:10:08.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/98/788398/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/788398/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/788398/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/788398/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/788398/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/788398/4"}}},"commit":{"parents":[{"commit":"74be74395c7aae6bf80bb3c039b4ceb9ba5a11e3","subject":"Merge \"Use a less bogus credit for Melissa Ma Lei\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74be74395c7aae6bf80bb3c039b4ceb9ba5a11e3"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-04-28 03:15:56.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-06-03 16:57:46.000000000","tz":-300},"subject":"Make dark data watcher ignore the newly updated objects","message":"Make dark data watcher ignore the newly updated objects\n\nWhen objects are freshly uploaded, they may take a little time\nto appear in container listings, producing false positives.\n\nBecause we needed to test this, we also reworked the tests and\nfixed some issues (including adding an EC fragment).\n\nCloses-Bug: 1925782\nChange-Id: Ieafa72a496328f7a487ca7062da6253994a5a07d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d3592a82d96c5f010953bea42974da6a0baeb2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d3592a82d96c5f010953bea42974da6a0baeb2e"}]},"branch":"refs/heads/master"},"a82e66e6ea9f1a96cdb3e6ce9e36047d656d40c6":{"kind":"REWORK","_number":5,"created":"2021-06-10 13:41:40.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/98/788398/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/788398/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/788398/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/788398/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/788398/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/788398/5"}}},"commit":{"parents":[{"commit":"74be74395c7aae6bf80bb3c039b4ceb9ba5a11e3","subject":"Merge \"Use a less bogus credit for Melissa Ma Lei\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74be74395c7aae6bf80bb3c039b4ceb9ba5a11e3"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-04-28 03:15:56.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-06-10 13:41:29.000000000","tz":-300},"subject":"Make dark data watcher ignore the newly updated objects","message":"Make dark data watcher ignore the newly updated objects\n\nWhen objects are freshly uploaded, they may take a little time\nto appear in container listings, producing false positives.\n\nBecause we needed to test this, we also reworked the tests and\nfixed some issues (including adding an EC fragment).\n\nCloses-Bug: 1925782\nChange-Id: Ieafa72a496328f7a487ca7062da6253994a5a07d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a82e66e6ea9f1a96cdb3e6ce9e36047d656d40c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a82e66e6ea9f1a96cdb3e6ce9e36047d656d40c6"}]},"branch":"refs/heads/master"},"95e031645120d5ffedb4b15127ff701d8bc24f59":{"kind":"REWORK","_number":6,"created":"2021-06-30 21:39:10.000000000","uploader":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"ref":"refs/changes/98/788398/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/98/788398/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/788398/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/788398/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/98/788398/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/98/788398/6"}}},"commit":{"parents":[{"commit":"74be74395c7aae6bf80bb3c039b4ceb9ba5a11e3","subject":"Merge \"Use a less bogus credit for Melissa Ma Lei\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/74be74395c7aae6bf80bb3c039b4ceb9ba5a11e3"}]}],"author":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-04-28 03:15:56.000000000","tz":-300},"committer":{"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","date":"2021-06-30 21:38:57.000000000","tz":-300},"subject":"Make dark data watcher ignore the newly updated objects","message":"Make dark data watcher ignore the newly updated objects\n\nWhen objects are freshly uploaded, they may take a little time\nto appear in container listings, producing false positives.\n\nBecause we needed to test this, we also reworked/added the tests\nand fixed some issues, including adding an EC fragment (thanks\nto Alistair\u0027s code).\n\nCloses-Bug: 1925782\nChange-Id: Ieafa72a496328f7a487ca7062da6253994a5a07d\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/95e031645120d5ffedb4b15127ff701d8bc24f59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/95e031645120d5ffedb4b15127ff701d8bc24f59"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
