)]}'
{"id":"openstack%2Fswift~788755","triplet_id":"openstack%2Fswift~master~I8826e6987a8cbaf477a1008957a38b1ac0c7cfab","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I8826e6987a8cbaf477a1008957a38b1ac0c7cfab","subject":"relinker: make total_errors a SSOT","status":"ABANDONED","created":"2021-04-29 14:42:49.000000000","updated":"2021-04-29 20:49:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fedcd28015a0fef1d87632b78cdf231630af3f8b","_number":788755,"virtual_id_number":788755,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-29 15:47:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-29 16:35:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-29 16:35:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-29 15:47:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-29 16:35:32.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"3c911fbc0e3e15db16fc12fa6a73b2a0c6603e66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2021-04-29 14:42:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7020c88d14d48db50651df51202241297138af9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 15:47:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7462530252741c387a75f22bdb861a0 : SUCCESS in 9m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/67edf1c98d2e4a1ea83ddf11b4584af0 : SUCCESS in 1h 00m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/50e6751d68ce41b7b8954a95869985f0 : SUCCESS in 42m 19s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e4ec880de874459f9986cbb2cdb86d4f : SUCCESS in 10m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/04c6585be6184d26a5e296139f68f1af : SUCCESS in 7m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f6fbb6ff81dd47c284694cb8be0cfd8a : SUCCESS in 19m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1be51ee29619426d9a62f3a230e475bc : SUCCESS in 16m 12s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c21c876cb00b480e94027fe25c36fcce : SUCCESS in 21m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c4e9a1ff911f41e6afce06bbf68947b4 : SUCCESS in 18m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5f02faf0ace04bbbaf567d49454cc550 : SUCCESS in 19m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e8e93913dab64cd19ab42239b5fd081a : SUCCESS in 18m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/665f484e3c824cf3acb28533fa27a968 : SUCCESS in 20m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9ad40a8dad554026b9256ed5db911e26 : SUCCESS in 19m 15s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/2f7581dfedc2455d8ba500c16328200b : SUCCESS in 17m 52s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/236dd37cdd4c43f6b0c356082a3edc4f : SUCCESS in 21m 20s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/756e9ec6f9c24d23a48cceadacad3d80 : SUCCESS in 19m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4b884138d9c045548ddf0429efd466f4 : SUCCESS in 6m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8d54b16aa49848a19840b743c895a7ab : FAILURE in 22m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ebea8a059d3547bdb197ffec937fd1bd : SUCCESS in 50m 59s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/dff583781b924bb981978e85be7134a5 : SUCCESS in 55m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e067a714462941af8d5b813fea6039e6 : SUCCESS in 43m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/917ad0dc36834bfb83ff1ba6b8481b4c : SUCCESS in 41m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/398960a115644901b0fa4478ac014bed : SUCCESS in 17m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cda4ca5e03641a59621b441f868359a : SUCCESS in 6m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0db2e479f9d844f98b0f571e4ef668f2 : SUCCESS in 25m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/13dfa8f93b7b496eb0c0651a46d38d10 : SUCCESS in 54m 02s","accounts_in_message":[],"_revision_number":1},{"id":"513f9b3888185e21cfa4872479ba14d0c57e0084","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-29 16:35:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fedcd28015a0fef1d87632b78cdf231630af3f8b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-29 20:49:11.000000000","message":"Abandoned\n\nSquashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c6bff54ff4d4820b333cb474ff13a85a19875c6e","revisions":{"c6bff54ff4d4820b333cb474ff13a85a19875c6e":{"kind":"REWORK","_number":1,"created":"2021-04-29 14:42:49.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/55/788755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/55/788755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/788755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/788755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/55/788755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/55/788755/1"}}},"commit":{"parents":[{"commit":"a8805caeb573d7e8c0da629ae5fbd53af99dc870","subject":"relinker: Only mark partitions \"done\" if there were no (new) errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8805caeb573d7e8c0da629ae5fbd53af99dc870"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-04-29 14:36:48.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2021-04-29 14:41:58.000000000","tz":-300},"subject":"relinker: make total_errors a SSOT","message":"relinker: make total_errors a SSOT\n\nWe mark partitions done iff our relinker has accumulated no additional\n\"errors\" from the previous partition.  Similarly the relinker will exit\nsuccess iff there are no \"errors\" in during the run.  The concept of\n\"errors\" in the relinker is reused in code, but it\u0027s calculation is a\nderived value - the sum of a variety of error types.\n\nWe move the definition of the \"total_errors\" property closer the\n\"zero_stats\" method so it\u0027s easier to see it in the context of the\npre-defined stats error key and also reuse the \"total_errors\"\ncalculation when determining exit status code and log level.\n\nRelated-Change-Id: Ida1c1c0c8a695b1b226121b426b8226a43f3056b\nChange-Id: I8826e6987a8cbaf477a1008957a38b1ac0c7cfab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6bff54ff4d4820b333cb474ff13a85a19875c6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6bff54ff4d4820b333cb474ff13a85a19875c6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
