)]}'
{"id":"openstack%2Fswift~788783","triplet_id":"openstack%2Fswift~master~I9cef4558aeb595e7a16d8aaace000d8455b527c5","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I9cef4558aeb595e7a16d8aaace000d8455b527c5","subject":"WIP - another work with sharded containers","status":"NEW","created":"2021-04-29 14:58:54.000000000","updated":"2021-05-13 23:37:27.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"b3c94bc67dae404ad8c09433d960aebd6d8bbd6f","_number":788783,"virtual_id_number":788783,"owner":{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-04-29 16:23:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-29 16:23:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-03 16:03:58.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"ca5e38c450c532edfd41758ab015141e7a51b2fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},"date":"2021-04-29 14:58:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41a676423e26defe4042dc4f2164613773542a18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 16:23:59.000000000","message":"Patch Set 1: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c26c6d1932a4e19a8bc16bd2166619b : FAILURE in 7m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/13ad44c0297f4fa0a2f7c8cde19f394b : SUCCESS in 1h 15m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a74d0bb292624ef7819dea00db68adcd : SUCCESS in 44m 08s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9117ece056ca42b4aea2623d591a6c40 : SUCCESS in 9m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/4da7c07244dc4149a482a474acc6e66e : SUCCESS in 8m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/01df371469684824912c8818d58ba818 : FAILURE in 17m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ec8129dbe4e74557888e67031fea7542 : FAILURE in 15m 05s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/88c2c5d5b6744953a7f4f99a308cc305 : FAILURE in 17m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/570f045d54d64f2cb72ccf46c4318ef7 : FAILURE in 18m 49s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d7b50bf7631a4c5d819b6777433803fa : FAILURE in 16m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/46ee62c7034c422aa405f84781575d0b : SUCCESS in 37m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1241cc4a9c10495395de20165b91c8bc : SUCCESS in 39m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/70134328a952410ab41a7ebebb2e6153 : SUCCESS in 21m 19s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/069017fae98d47fd8bdd114b77500804 : SUCCESS in 21m 10s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/04d66cede5254c008b463eb9c22a8d42 : SUCCESS in 38m 12s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/398e9a8071ff4db08fa11af3f20fb224 : SUCCESS in 39m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7179c3fd2250498da24051d1a0a43a91 : SUCCESS in 7m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/15f25652a7594b2db5ca095e00669c1f : SUCCESS in 24m 30s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/25f7462e397b4298888fe06f7e354dbd : FAILURE in 54m 38s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/3a4bf0d36beb4664bae6693dfdc72341 : FAILURE in 51m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f8af23989bc4d41bcfda9fcd7186ef2 : SUCCESS in 42m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3826e82db27f4bd991a42de15dd57143 : SUCCESS in 38m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c54e1fb16c14f7db467aee61de45c8b : FAILURE in 18m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0f52f50f82b41b296202741a23e4895 : FAILURE in 7m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c567f0d92aff4604a3b7dcd8ac62f400 : SUCCESS in 34m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ba48ad4314442b8aaf48df2a84c70b3 : SUCCESS in 58m 10s","accounts_in_message":[],"_revision_number":1},{"id":"b3c94bc67dae404ad8c09433d960aebd6d8bbd6f","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2021-05-13 23:37:27.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nI may be missing something, but grafting a large chunk of proxy into auditor may be more involved than it looks. Not saying it\u0027s impossible, but this seems to snowball.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7344d1c88259bdc2dee87e283f39cc21efccdc39","revisions":{"7344d1c88259bdc2dee87e283f39cc21efccdc39":{"kind":"REWORK","_number":1,"created":"2021-04-29 14:58:54.000000000","uploader":{"_account_id":18142,"name":"David Sariel","email":"dsariel@redhat.com","username":"dsariel"},"ref":"refs/changes/83/788783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/788783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/788783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/788783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/788783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/788783/1"}}},"commit":{"parents":[{"commit":"7cfdb50f93f6780a668c4d6c04b1f1ef25166c69","subject":"Merge \"reconstructor: extract closure for handle_response\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7cfdb50f93f6780a668c4d6c04b1f1ef25166c69"}]}],"author":{"name":"David Sariel","email":"dsariel@redhat.com","date":"2021-04-29 14:57:55.000000000","tz":180},"committer":{"name":"David Sariel","email":"dsariel@redhat.com","date":"2021-04-29 14:57:55.000000000","tz":180},"subject":"WIP - another work with sharded containers","message":"WIP - another work with sharded containers\n\nChange-Id: I9cef4558aeb595e7a16d8aaace000d8455b527c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7344d1c88259bdc2dee87e283f39cc21efccdc39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7344d1c88259bdc2dee87e283f39cc21efccdc39"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
