)]}'
{"id":"openstack%2Fswift~789035","triplet_id":"openstack%2Fswift~master~I8227939d04fda8db66fb2f131f2c71ce8741c7d9","project":"openstack/swift","branch":"master","topic":"inherit-children","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-01-16 04:42:36.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},"22042":{"account":{"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},"last_update":"2025-01-16 05:49:41.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-01-16 05:49:41.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-01-16 05:49:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8227939d04fda8db66fb2f131f2c71ce8741c7d9","subject":"wsgi: Reap stale workers (after a timeout) following a reload","status":"MERGED","created":"2021-04-30 21:24:28.000000000","updated":"2025-01-16 05:50:44.000000000","submitted":"2025-01-16 05:49:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":86,"unresolved_comment_count":9,"has_review_started":true,"submission_id":"789035-inherit-children","meta_rev_id":"e25db0cb2eebc951cc25bdb1895b0c72709c36e8","_number":789035,"virtual_id_number":789035,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-16 05:49:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2025-01-16 04:42:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2025-01-16 04:42:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-30 22:27:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-17 15:45:08.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2023-06-09 06:47:48.000000000","updated_by":{"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},"reviewer":{"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},"state":"CC"},{"updated":"2024-12-12 21:52:40.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2024-12-18 18:13:32.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"1706715ee966b10611485fa73ef4aa9e180a24d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-30 21:24:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1308de9fd46e90a85d6a355efbe4b52980bacc90","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-30 21:25:21.000000000","message":"Patch Set 1: Workflow-1\n\nNeeds tests, but I\u0027m digging this so far.","accounts_in_message":[],"_revision_number":1},{"id":"55c5bb87492d780612441ba191f074c0766e2044","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-30 21:33:45.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e1780b1324d3ba547b145d4dab244a37ef168f44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 22:27:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9996116b96434432a8be65f12c202323 : SUCCESS in 11m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/21098c3c89664e8dbe7fb7e792a43024 : SUCCESS in 1h 01m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/479b799c5d75437b8ce3f807e4503dd2 : SUCCESS in 36m 39s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f259ddc67c1540b1a4cae76e80224ff0 : SUCCESS in 10m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/72a13942493b4459b1a82a537f666826 : SUCCESS in 6m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d8dde79c5c084decae0e0cc42cea98e4 : SUCCESS in 20m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fd8cfffed84443eb95c2569f43702d91 : SUCCESS in 20m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2ee0b44bd7004591a1dd209a006e206f : SUCCESS in 18m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a3383733dd904cb99973e33f489a5166 : SUCCESS in 19m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0f8119d54bf94d2abc5c91b609cba539 : SUCCESS in 19m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7927c9bb0c9e470f83f9fbf8550a5797 : SUCCESS in 23m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/06ac507cd87e4f158221ab736dcbd2f8 : SUCCESS in 23m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/71aa28e997d24237918715b93481de15 : SUCCESS in 21m 46s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/1b2dee27770b43d099328b1e6c24238d : SUCCESS in 18m 41s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/38e051deb75b4c06971d57571db46bf2 : SUCCESS in 19m 36s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/df1fe243187b46cab95f1de35a19366b : SUCCESS in 19m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/135c6319bc2b41f988e7bb66d083ed40 : SUCCESS in 6m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/399fa094e69e4beaa329f33fdb04d019 : FAILURE in 24m 53s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/643cddba6cad4938971dfba3524f3f0a : FAILURE in 52m 00s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/a6821e5a836c40d58ab4c0f8b58cd9b9 : FAILURE in 52m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/427daf62361d4e4da514091fbaf7949e : SUCCESS in 40m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cd1f84272c384cd7a565a898eacd62e9 : SUCCESS in 38m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa23d05d838642ca85a04842b85967e6 : SUCCESS in 16m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f68ce2090f2d4ba3beafc6f008b7d7ba : SUCCESS in 5m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f959182798ab4cd68dd44dfa08502060 : SUCCESS in 31m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df0e59eeaa074098aeb70ef43c02e5f7 : SUCCESS in 43m 28s","accounts_in_message":[],"_revision_number":1},{"id":"8d1e65ac80c8d69c453ca867fae34e07cd395534","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-02 04:07:07.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nProbe test failures are legit.","accounts_in_message":[],"_revision_number":1},{"id":"811b5a71426a77dd807e5da2b7c629b926c79a55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-03 21:31:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5c191ec286bea577e448a7e8eef730dd00bf1c74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 22:38:23.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7c5d87b63bc4fdca843c5efd5682a58 : SUCCESS in 9m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/55fae5d4f8714bb5a25042aa166e7641 : SUCCESS in 1h 02m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1256a4be1a86433c87e73cc87cae23fd : SUCCESS in 43m 44s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ab0a6afb3e0043dd8bb73338047005da : SUCCESS in 9m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/585a06ad1fad4d40a81bf1c812e689a2 : SUCCESS in 6m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0af59e59975c4875960bd1e136fb5eea : SUCCESS in 18m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/72733ff29a76415f96c8d561bcc30ce8 : SUCCESS in 13m 28s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/89c80396a1024b68b2b569aec0503786 : SUCCESS in 12m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/62258a5b8e1c4cce9666a16612113606 : SUCCESS in 14m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/23ab91f237ce490fb06aad5f41812bc4 : SUCCESS in 13m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8ef4904a89374014a7cc73b171ca0519 : SUCCESS in 14m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5198c710c5fe4837a2260d54f9270c01 : SUCCESS in 20m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/28bc1321dd334524a3c0d7e3b6f752bc : SUCCESS in 16m 56s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/686e8308e3fd46f09459544f4c17c2ad : SUCCESS in 18m 02s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/0cf35294da4e450dac23656587049d82 : SUCCESS in 20m 37s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/f9946fb233704473a9090b1d9e4fc576 : SUCCESS in 17m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f0f2fd8488ce47da967f2a014e057931 : SUCCESS in 6m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e254d3fd264e4f9dbac64b5ba138a619 : SUCCESS in 22m 53s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/1f3f10ea954644fc94848e3305e289a8 : SUCCESS in 40m 15s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/2cf43ab37c294ce1987974320a9f9107 : SUCCESS in 48m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6ca32da70046462fbd1fa348eecdc52b : SUCCESS in 44m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/473d054a4684418286af81d4a4ae996d : SUCCESS in 41m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4def56eb03b496f855d23e4fe11d100 : SUCCESS in 16m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c8a005267e949e8b165d004af85b415 : FAILURE in 6m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/943f43b73b094aa2a491d9deb2460c3e : SUCCESS in 31m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f2e2d9d85924180b891ae35f96e1981 : SUCCESS in 53m 38s","accounts_in_message":[],"_revision_number":2},{"id":"f1eb47393c907cc9a05d3cc7f227764bdd56d09c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-04 00:01:46.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"d602c217a95a9251cc2c76af74d0af98297d5aef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 01:10:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/713fee409dc643f9807eec05616e0440 : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/66ce96ae03964ab6986c82d27c75072a : SUCCESS in 56m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d58f0d6243584d61b1d2c4c06fcefc2d : SUCCESS in 36m 06s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9d8e5649b47446cdbc1ceebc91254f35 : SUCCESS in 7m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c4435db1152b4058a089db7182b5c314 : SUCCESS in 7m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/02f85c6a51b64895a523cd105e49509a : SUCCESS in 18m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fe908cb4ec204f009af888b9786fc70d : SUCCESS in 16m 20s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0dc18779391e49b8b0ca735719dad6df : SUCCESS in 18m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c5555f3e603b456790e2216acae800ad : SUCCESS in 19m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4e2b8e0447534a8fa30c19318c27c38f : SUCCESS in 15m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b78eb524f5c44e3a9568b9c34fb6b2b7 : SUCCESS in 17m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d1a0d721bd8d470fa912c25a40521f6c : SUCCESS in 20m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7186e6da75b44027ba1edfb86fd1821a : SUCCESS in 18m 53s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/b29b7144e3e94c9b9128160bf7f904ff : SUCCESS in 20m 29s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/b5938582affd41bdb8665ae81b06f7e2 : SUCCESS in 20m 56s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/52432113ec244ec0b63855cfcc00f899 : SUCCESS in 18m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3cf6df0af53a4c54b11985932b01301f : SUCCESS in 7m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6c624760fece4edc8bc2b31169566594 : FAILURE in 20m 00s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/202f41c0e43a4158bff3b6671ac34a6b : SUCCESS in 51m 25s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/22692333ac3b401fa1fe6cb152d33e69 : SUCCESS in 46m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a1e24e75151a444f860b6ec8976045da : SUCCESS in 44m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6f47a79ac7f34b1fb739e980bdf0994b : SUCCESS in 31m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3eca546106541438fa86fbf18ab08de : SUCCESS in 12m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/349316d50d364ca3b1ea8fa06875bbed : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/87971a09f8674bd7a73e7b9656a7e957 : SUCCESS in 23m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/488eec4b701643d0a40fa34ac51f82d9 : SUCCESS in 53m 29s","accounts_in_message":[],"_revision_number":3},{"id":"dcd87aa6362b45cc35429ae106f650614c5d03a2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-03-25 19:33:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"421e054efe8375a6ce08afd168871353d6408734","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-17 04:34:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0447e9695ed6bd56d54f68d40b6c4d6f830d003b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-17 21:40:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"28fe1d9b96e5582287cd934d6bfa6b2fc5bdc580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 22:35:03.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7391fec9720b46d290f9f7655d68a605\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7db56f4c54f4669a30a58661e121dc7 : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/b64cde363bed4742b8d6db5eb91e3825 : SUCCESS in 50m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f93284a94b0a4deea3eaa988080be8bf : SUCCESS in 43m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/653a8345169a48e98e0f10501cff4ef5 : SUCCESS in 35m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c422e1a968c44e3487dfcfc47a3003eb : FAILURE in 18m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cceb9904a87442bc861bcbafe1720ae4 : FAILURE in 17m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f78f8071489403fad0fa383278ded70 : FAILURE in 22m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e9556e78b657491d9b95fee0cb6b7b55 : SUCCESS in 7m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0c1f9ec590174f98a90838ea6cb80ad5 : SUCCESS in 4m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/adcea15fb38b4b0581bf61d8a5f6871c : SUCCESS in 13m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7859f30b17b94db8be42afc22b7b329f : SUCCESS in 12m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/663240025a10495aa215c776490f712e : SUCCESS in 14m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/867bb36b109445fba17ec4bc22673f93 : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cb22937b5c6c483a96b9de44cbc0683f : SUCCESS in 15m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e24745bd8531497ead2b76582884fceb : SUCCESS in 23m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b29bad752ee24aa498495bbf04334b74 : SUCCESS in 15m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/78accc169a0e43c6bf48f958b8a5729d : SUCCESS in 15m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e5639db37b144fdcbe9c81dbb01b16cb : SUCCESS in 14m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/31f09d04789f47f2877fc1c225176dca : SUCCESS in 5m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6d7390eef0164fe9a93b9e788b6b29c4 : SUCCESS in 19m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dad31fd8686349aab8ab7d84e8092671 : SUCCESS in 7m 56s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f0fa07dd8f9f463db9c750677a740dd9 : SUCCESS in 53m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5c4cdb04b83e438b83478c6d2a55c6d4 : SUCCESS in 51m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c390fabf67f5423a8ca2cdc9ce02378c : SUCCESS in 38m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5ef2cdd319044926a761fe05a838c710 : SUCCESS in 32m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b2abe8f1a234e059f315056f1b058ac : SUCCESS in 14m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/124e04aaa0fd43fb862458f7ee7713fd : FAILURE in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e807fa8c9986480e9b38bea6d07fc340 : SUCCESS in 21m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/040528f1ee744dcbb10dd303337cb3e8 : SUCCESS in 43m 52s","accounts_in_message":[],"_revision_number":4},{"id":"7867b0f72e62204e363d76167bf108667a880d27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-17 22:42:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4397be735eff4ab6ae8b58e9033e7a85\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2f2fcba591fb4a8ba620c956d74e25f2 : SUCCESS in 45m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9d4ba021b9fe4a6db057c3f45bb26c13 : SUCCESS in 44m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f39240e42134187b1e2ace050932f3a : SUCCESS in 54m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4568bff1974c4d148bb6d8c440a1a12b : SUCCESS in 26m 45s","accounts_in_message":[],"_revision_number":4},{"id":"a33cfb03e1d024fa45707470b32519fe7f83dc53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-18 03:18:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"df43cb038bfdbf1a1aef42ecc0b952b987a83e72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 04:22:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/230dcd3bc0414048b84db63a7009122b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/44ad102adfee42ce9762c7967584de77 : SUCCESS in 34m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b39b4e13f294815ae0f84af67e90a28 : SUCCESS in 34m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/dc59024d76fb41748dcc7b751b16d677 : SUCCESS in 35m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ab4542a0b23c4f2489b021d7f3b78928 : SUCCESS in 1h 01m 05s","accounts_in_message":[],"_revision_number":5},{"id":"3ad6fb0f9110daf7f6f84acdc89bbc876273a754","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 04:24:31.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/13fa1cdb127b48da9baa9d36c2604c8c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44e556d176364dd29a66264590fb56c3 : SUCCESS in 9m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/6bbb2471a8e54a4a971a31b1a61fb5e8 : SUCCESS in 1h 00m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a03ee184818140249cb739b1b3f81504 : SUCCESS in 43m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e025e0b8b1c746a89765e45ede5bad31 : SUCCESS in 44m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d1577dd9b874ce2b2bc28c81d317fee : FAILURE in 18m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/732973f0f3984893b9183196dfab4c9c : FAILURE in 21m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f3b2a98e01a4e95a684bbd50e2f0b84 : FAILURE in 20m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4183c33e6b25414682fd671e022ec186 : SUCCESS in 9m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f061e3b196654e5683fee25b4b5b7f0b : SUCCESS in 4m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1a26793bbc994116a68aa97b756025b5 : FAILURE in 15m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/de6a0052425d47019f3eeaa8ffb16388 : SUCCESS in 13m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2b1de28448be40fa9bbebc097a8a6612 : SUCCESS in 13m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/24246c51428f44fda68453175c8c71e3 : SUCCESS in 12m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3f4fab27dd93447ab2f85e88dbccb521 : SUCCESS in 17m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a3b4b65b3cc24057b41218544a961af0 : SUCCESS in 15m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a603bec0db6044c78cc139dd31209e00 : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/636a5d2b8fc542928ccc53d2ee34f6fc : SUCCESS in 11m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/58bdbd0ea8ee46cfb4ec022fbe9658b4 : SUCCESS in 12m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/432ce3c6aef04261bde786859bd0da86 : SUCCESS in 6m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3e35de0cfe8c49fc980afabc4a44de37 : SUCCESS in 12m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/95b265130fa84b639cdd3c952df27636 : SUCCESS in 5m 48s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/aa71efe45bd341e180763379eb10ee42 : SUCCESS in 42m 48s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ac0d665427b14685854ccedd62c918ff : SUCCESS in 48m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fb4e0535da084534a2bb6071091a8c97 : SUCCESS in 32m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c3c986481d5d47ebbbef5774095e07a9 : SUCCESS in 29m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/917732bb3e214d2584110028f3f695a3 : FAILURE in 12m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3de6f3db34347719572fcdca5e09836 : SUCCESS in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c373d5b41935420aab3e291ae75abd25 : SUCCESS in 15m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60b494f94e9f448a812cd66fb8b79f1c : SUCCESS in 43m 15s","accounts_in_message":[],"_revision_number":5},{"id":"0be91cd83331eb5701aac9634e1daf5a1db7e4a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-10-18 15:37:28.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d9cceea9d1be0dee3cd5412c9bc73e9a43453061","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 16:26:27.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dbc42372b61401cab5739ca4cef4083\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/38ec279fff74446d877c4c5c7d3f3117 : SUCCESS in 40m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/11e5590b1b4146bf9008ab6ab4e3baef : SUCCESS in 39m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd34dfd2b29c48a79c9129b1a78d34b0 : SUCCESS in 39m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/33ae494368704cc98a8be610bc66a160 : SUCCESS in 40m 39s","accounts_in_message":[],"_revision_number":6},{"id":"68eef0355f7fe6eebb8e8fe694bbacc7aa8fe07e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 16:40:24.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e62452b2b35437e92bead09b17e8fe4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/807a234a99a54bda9d6228913360fd39 : SUCCESS in 7m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/afe97ccb84264f7584b653b9849435af : SUCCESS in 38m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d75f08e8f32744b29e7ddb5cdb28449e : SUCCESS in 47m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/296760984ca4490c83e5b2805b5bdb39 : SUCCESS in 47m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aa904abdf3134d7085afd7d6ed396a6d : FAILURE in 17m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/710fca9275a646b68dcc888c52ccbfe2 : FAILURE in 19m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9efb0ab068c449278bb7fd1445ba2c8c : FAILURE in 21m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2af6a9eee88e4f67b9d1ee7fc5f28cb7 : SUCCESS in 7m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3378c866575f49f0b13df3f9e97968e5 : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9f30ea2e9c254bf3ba75b70fd8a6e3a1 : SUCCESS in 16m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f5379157e68a45b59260169499f1814d : SUCCESS in 16m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2f01c0b4ebc9402e80e8aca6f58d6fde : SUCCESS in 16m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a3af873e72694dd99031d6525f3e8d9d : SUCCESS in 17m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c1095f1e2e2b45728bbb8737a2b0528a : SUCCESS in 20m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/61302bc8453d4d5aa732ddab20b8e5e3 : SUCCESS in 21m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0852d5e444ad464389eb5eaa277bc035 : SUCCESS in 14m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/28e0f0ab7157497aae20ee59300be005 : SUCCESS in 18m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6820bb42f6f04441999770fb3377a5d1 : SUCCESS in 16m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d86496f975164da4b8b91e093e593df4 : SUCCESS in 6m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/25a2366cdab245c090dbed3d28eae338 : SUCCESS in 15m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ad29e9e01b974020a2e4fa706744dfc0 : SUCCESS in 7m 11s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/dc4e2f696d6244a7bf8a833742a4a4f6 : SUCCESS in 42m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2aefe8279fc149f48b5aa0f071f539ac : SUCCESS in 1h 01m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c5baa06453db4baca8a197980f5b6544 : SUCCESS in 40m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cba2af8a59844ca8815ff1942bbc096e : SUCCESS in 43m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05018eaebcb344ae9d16e73fef62fb29 : SUCCESS in 14m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34d02e95f2f943f7b7ce207712c8ca0b : SUCCESS in 5m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0af7e9f7aa524d8ead17288a45bea07a : SUCCESS in 17m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec685178ddf4494984b607c5b77402bc : SUCCESS in 43m 31s","accounts_in_message":[],"_revision_number":6},{"id":"ded8482a9e7d2c4aa8b25435b790466968202871","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-11-02 19:52:36.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4a14733771af82b5a0f2de07747aa1651fdeaa3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-01-06 23:50:47.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"2ddc689d52907754c9b4c180b528f02d26d7bd30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 01:19:58.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d415c0b2a13c4239a37da0ad78457423\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0822bfe29864c5285941d151ef12552 : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/8b6bdf401d4d405e99aeabae4766495d : SUCCESS in 1h 05m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7f0c3b6c802e48b686321b6f2e1e58a9 : SUCCESS in 39m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9184f32b3ea54c43a6c4b32f6c3da2f6 : SUCCESS in 45m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c6c03039bb894880ba22e1424a98e46f : SUCCESS in 17m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc05da207f954a3181cc948c83b4ba50 : SUCCESS in 26m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f6ede64609284f4e9ab904f8f03eb0da : SUCCESS in 15m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/acd075ce12034409b805cb97eaee6a74 : SUCCESS in 9m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a721a2f808a74697b2b7424747736538 : SUCCESS in 4m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3ea7d24e12e342aeadc0274f7db42aaf : SUCCESS in 11m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cde42a850e634805bd93fa6937f1adc6 : SUCCESS in 14m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/03d209f9b99745f4ace2542a39876a21 : SUCCESS in 15m 18s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dc95967fbb904682a823591fae36c310 : SUCCESS in 21m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bea0a8c8b85947a5bd7fc1c9df37fbdf : SUCCESS in 16m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8dda19fcdcad4bf2a1c673ac2db12789 : SUCCESS in 14m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dd6dd5f8d6894d09afca897d85680c9a : SUCCESS in 26m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6b6d05c005ab4b819c91444b65635932 : SUCCESS in 17m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/aa733b0e5cd04925875af562352e04c0 : SUCCESS in 16m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b76d21bc0efa4c9eba563a277f891716 : SUCCESS in 15m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/74d3a987765a4bde8bc97f75b1d65e64 : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e77b9d7c79cb4d36a4cf6532d862cefa : SUCCESS in 18m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e55084c71c9441a59df02b1a0a98358d : SUCCESS in 7m 13s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/950bddb122414285aaf9812b178c1a3f : SUCCESS in 1h 24m 24s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0efcd28ba0cc4268be92fd283c08f7a4 : SUCCESS in 1h 11m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6c9450556153490db9b4daefdd189f4a : SUCCESS in 38m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/61a6976564c54f809af58b610dec38e4 : SUCCESS in 47m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3592f81b1d54753b0048381e5ea51b1 : SUCCESS in 13m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7762b9943ac42c9aeaa8b494b106487 : SUCCESS in 4m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d3e296db85174041b5e4b371fc597357 : FAILURE in 25m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bfb973d1277448c8fd7d65597b36c1d : SUCCESS in 59m 51s","accounts_in_message":[],"_revision_number":7},{"id":"a90b24e975627865c27e2be3b3053936c93f01ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-07 02:03:03.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/823dc02a9bd543239bc8e25852286772\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba42467d3ae44187b17d0b81a7dcc255 : SUCCESS in 1h 28m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fb5eb2d87c914d62823a6417b896de3c : SUCCESS in 53m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2dc5b48655dd47638332947931fbe7b3 : SUCCESS in 1h 17m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c694cf93fdc40dfa6965408d1cacbdd : SUCCESS in 1h 13m 52s","accounts_in_message":[],"_revision_number":7},{"id":"67c2f3dab19fad0e3be6e7c8a4e93056fb97dab2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-06 03:48:11.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"42eb13daf5444d1a438a11bdd7696f3acdaaa139","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 04:11:35.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1b86b24456446639f9520468e6428f6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c96c978fe2334267b9d7db65335f208c : SUCCESS in 18m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0c347427a4c247c0826dcf5ed9c32c28 : SUCCESS in 18m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b07fab7f21d74ee4bcff24d204834fb5 : SUCCESS in 21m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2906e5fdb82e4253833d776a601f7231 : SUCCESS in 19m 15s","accounts_in_message":[],"_revision_number":8},{"id":"bf1fd8c06ba2f6eaddd53f7ea5a1723b70763a7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 05:05:06.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1239c0d3030242a2bf4f402bafc43f41\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bbcaeab620b4e1485516921243eb83e : SUCCESS in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/b07e623b98434cb8a75d7e5de2c1ba2d : SUCCESS in 44m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/19c29a9b026c498ea0e20d3b60aaf75f : SUCCESS in 37m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d39c8d72c944a5ebc258a3a74662c0c : SUCCESS in 46m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d29293b17f84bc287fa03cec5792ad5 : RETRY_LIMIT in 4m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/574b61212cf4419194b9752f7282d395 : RETRY_LIMIT in 5m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f5021146d98a464698d51c6add9846ea : RETRY_LIMIT in 4m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b69bcbc608f742bc88c1275be10e360a : SUCCESS in 7m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e760afafd8c94026be5d2737d57ea984 : SUCCESS in 9m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cecac61ba3d34f3bb48e042229380408 : SUCCESS in 26m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/472b3c5e79c345a48793da9377e287c7 : SUCCESS in 25m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/001b517c216544fcbd8ccbf9c8844af4 : SUCCESS in 12m 48s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/cde7ed6b27f84f4f959ca78551403eb7 : SUCCESS in 15m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8212cdf0487a4d20804d59b18a5daf42 : SUCCESS in 15m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9b633b37d6df41bcb81c1a18373afd1d : SUCCESS in 18m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c1c853490f3b4a8c9db73c093027d057 : SUCCESS in 29m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/760408eaebb642038d0d523af546733a : SUCCESS in 12m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9ad4b1031a6c4f2a9f3e6f17bb8f3be0 : SUCCESS in 14m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d3606b5591ae472a8f016bf2e6d87c18 : SUCCESS in 25m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/edaccc9067354cd3980cdc358a47f500 : SUCCESS in 7m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7bd10964a83545fcbfd81823eb5c75cd : SUCCESS in 15m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1b3701cd329847b7830eed03616c4c25 : SUCCESS in 7m 30s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/41cfcf436dd447f684d18bd10e8f303a : SUCCESS in 59m 02s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9c5cfad2688d437883bc254dba604881 : SUCCESS in 52m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9f98b804916048549d81cc3ba914e992 : SUCCESS in 47m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e2dbaa6301ac4e3d8d9f9ad729421fb7 : SUCCESS in 46m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4df27d38a0a046f7acea2a7c765bd129 : SUCCESS in 15m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/807216cba1d44f09b7b18256f4cb5b21 : SUCCESS in 4m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8416a2e378564eb5b08be93748bfef73 : SUCCESS in 23m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a755a3ff52eb4616977fb9393b2bab1b : SUCCESS in 38m 42s","accounts_in_message":[],"_revision_number":8},{"id":"5a77fe5bfe160442b5a76f832ceac0c5829ff477","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-04-17 15:45:08.000000000","message":"Patch Set 8: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":8},{"id":"21c22828a364ca7ceca1e84b584a78fad440e5ab","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-17 17:10:59.000000000","message":"Patch Set 8:\n\n(8 comments)","accounts_in_message":[],"_revision_number":8},{"id":"c15bb5a2c985782d98447ac711254fd27209186a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-18 16:03:18.000000000","message":"Patch Set 8: Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"a421bb91c71faa88726f31d909e53eb61d507e88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-18 23:40:01.000000000","message":"Uploaded patch set 9: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":9},{"id":"d7868f461e08f4779f5f0f654babd46f16b66571","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 00:30:42.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49493427691c4910b8b9515c1c0d9e94\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e4111cf36c34fc58739199588a4b945 : SUCCESS in 13m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/39101747c2714dcbac1cc5f74c598b1b : SUCCESS in 48m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/54a19628bacd45aeb247ac7361de3e49 : SUCCESS in 44m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca6d27d0e8ef49c6acd94cb91b91aea7 : SUCCESS in 22m 09s","accounts_in_message":[],"_revision_number":9},{"id":"00561de6a56511f301fa536d9d9023375ec34d62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 00:47:40.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea3ec3a6c5cc4e8e9597865fc458ced9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/293137d405854635a01ab01585fc7c69 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/595be0c68c4c48dda1e3ae2606f04934 : SUCCESS in 44m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6abbd08ee33944318c7fa29695383c6b : SUCCESS in 35m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49f1ce7d66464ba2b079e6619617efe4 : SUCCESS in 51m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d31c4eb2c58042d3bf681d328fa0f8ae : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/56b88a452a854de4ba8642ccaa1e5b9d : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a91165cbb1334e5287c1039da180fb2b : SUCCESS in 18m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8dc51cf26f414581af2de77096d9d313 : SUCCESS in 9m 53s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e42b416f61d24c4f8e331ff6f6b73b44 : SUCCESS in 6m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/85430f96c65d49faadbfb09003ccd90e : SUCCESS in 15m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8d78fa85c42347a78e405c2b2bc484ff : SUCCESS in 15m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5661de3f0727406d9694022890119e68 : SUCCESS in 14m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f81675c918ea4573b1fb41ef7bc44f8b : SUCCESS in 13m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/debd05cc942b424a884ee9447bc20374 : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6ede03e3376449bb9ce27d41ca1a1529 : SUCCESS in 19m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f96ef5634b0b4cba8d1deaa80e45cd51 : SUCCESS in 16m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a5e8a019eba546d9a8171ef5bb13f8c7 : SUCCESS in 14m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e3794e1ecd2d4af7b28058ef3ec85412 : SUCCESS in 13m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/dd930bf28d9944149f591d0c054ef18f : SUCCESS in 15m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ae3fe4287f2642a0b7bd692c3ffc533f : SUCCESS in 7m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d032bf98b15d4cd681b5ffc1df1eae99 : SUCCESS in 16m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/84333cbf8f3a447e9b07b0661417d0c3 : SUCCESS in 7m 01s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/0198d0c9474e440dad83b97c6003fb19 : SUCCESS in 53m 00s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a0c6bc9ee77432bb01af56ee2aab571 : FAILURE in 1h 02m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d9814d3e957344ab94ebfe496d959650 : SUCCESS in 45m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e8d94dfb8ae847fb8b682f4165c5cb73 : SUCCESS in 52m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c3bd391d91b4b308cc30d449fbce196 : SUCCESS in 17m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a54cd6b402ff4c8b825ea25cf6d0e126 : SUCCESS in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d86afd2775084221aa85717516985db6 : SUCCESS in 20m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d1e83f7fa024f728038b0d689387fe2 : SUCCESS in 38m 39s","accounts_in_message":[],"_revision_number":9},{"id":"50326a2eb0d19986323078a783a604f20e3e7b94","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-04-28 22:00:06.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"258652b9e66fad3f4aa7d5b5adc407a0ead743e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 23:11:46.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0d9e6052795407b8cf9235584261aaf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e1ed85e12bb42cc9058168a4f8b0bc2 : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/8ab8e2f597f8449cb7eff1ffe303248e : SUCCESS in 43m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ad286d2970ab48d48f72423dabd623a0 : SUCCESS in 37m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9c41f60b6504673be5c8f3182ae327f : SUCCESS in 45m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/474263e8768941a9acf2ff85853544bc : SUCCESS in 23m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/086945e2180749118bc4797c4e72763e : SUCCESS in 23m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99cc78df7a08406bb709aad5fac6db4a : SUCCESS in 23m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1947e3582e0f439a9a81f5ef059defa1 : SUCCESS in 9m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/745e69af6b0d4306a1206ad43d78258c : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1a0947e45bd74feb8f3d024d4efd759e : SUCCESS in 17m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/302227c5e59f42be89c113d60dfb2ade : SUCCESS in 21m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b328317d671647d5be6f6badcaa36fec : SUCCESS in 16m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/875f36f048524723a279e29b09244d64 : SUCCESS in 20m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/02e2ac0d9f9641938c8d2af7a7fb57ee : SUCCESS in 16m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aaa963ead8e049ea92b5ed329c1d5844 : SUCCESS in 18m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f72de0c3505e4d6484aa9b97f6dccd56 : SUCCESS in 17m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/995f9e86797b48da9825a05713d36236 : SUCCESS in 13m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ebce4e3ad838408080b4fe51f752bd15 : SUCCESS in 20m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b5e76941385a4144ad612e42a36bb738 : SUCCESS in 21m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/05525024c95d42eeaaf4bc145eb47144 : SUCCESS in 6m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/240a09b1187142e3a1490c7a98b6e2ee : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b69a1f438f394038968371160949d2f8 : SUCCESS in 7m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8eeba50309f24ef28951420c9566009d : SUCCESS in 1h 02m 13s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/2ffadc1169f54fab92de5aec3c695bfd : SUCCESS in 1h 09m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8be2384cbea84da493a82e0bbf5d66ba : SUCCESS in 45m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/69fd5f519d9f4591bc868c9a27dddc10 : SUCCESS in 45m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/891aca1e565047b6b53be2964b01bb51 : SUCCESS in 23m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e754ff5989842148de4fbdd0e744555 : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/560e72834a994a1fa2a0917eba90261b : SUCCESS in 19m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70a78643d8dc4566bd75f5d1d65e46a9 : SUCCESS in 56m 52s","accounts_in_message":[],"_revision_number":9},{"id":"92ff11203f62f4baf8f0d6f3925681b2723fdb60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 23:33:17.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/684395e109b84f1d8dbac74f1548e2c6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/170376380f6e48b8a5cd8d5d70da74fa : SUCCESS in 1h 13m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/95b8f206346a48f88cf9d0d623a329f6 : TIMED_OUT in 1h 30m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4abd9d8e29d640a8a28f3b7558e7be8b : SUCCESS in 18m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c80015308f5147cdab463fc4d2e40e4c : SUCCESS in 1h 08m 18s","accounts_in_message":[],"_revision_number":9},{"id":"e4fb48a4aeb114c092a928dfb06725ba336156e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-05-23 16:38:16.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"3581772f8767069e4b67759667af34bb446d6190","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 17:08:20.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99d954ef03e1416f86a338b91f113dfd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5cd2c09d9bb47f29d91b537aea743c8 : SUCCESS in 23m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4ace6e36af284d22ac4395d73120fe87 : SUCCESS in 20m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/758dd7487c4744e3b3a95dfeba54a66a : SUCCESS in 28m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e346817a1d354ad49e6dfcaa3c754ad7 : SUCCESS in 27m 19s","accounts_in_message":[],"_revision_number":10},{"id":"6762cf1e0e512ad72d8758a104a8d5b4efa8aaf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 17:44:29.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0508d9bc21b241708dc4c20c993b358a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a8b92faaaca4c0ca96297b77d06f8f1 : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/39de717905a54144ac3e197b5cb69028 : SUCCESS in 53m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d3f2e4ec2c194f4b9d106edad1ba50fc : SUCCESS in 41m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9eaccc934a8b49febd8db8bd5d8d0709 : SUCCESS in 48m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fac6b873fad4405c90c047fa4931c2d9 : SUCCESS in 16m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/26a144c4456b46beb7df32c19fbb74f9 : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d9563aa889c4bda8e6d8dbdf37bf577 : SUCCESS in 21m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b719f8ff925e405c84c9ce2986956a76 : SUCCESS in 8m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/99d8849e16df4467be17166d4ae1b264 : SUCCESS in 5m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c6dcd61f67714fef9679c1a45028f0f6 : SUCCESS in 15m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d7abff0f4dbc4a038b4bab34132ae51c : SUCCESS in 12m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b1c3e6cd655a41faadc77adda3f89618 : SUCCESS in 11m 32s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a91e9dc61e1a41298ee75e93d4183239 : SUCCESS in 15m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/59d73cdacbd545d9bc408fd7f9f16b6d : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b7f15082239943858169715a273827d8 : SUCCESS in 12m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8dcd2d17081347679be4624f280a27f4 : SUCCESS in 17m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d0658a7f53744a1c8cb7c3214c399a96 : SUCCESS in 12m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c8f0d0f394e646c388e732d474ab4191 : SUCCESS in 14m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/737216976b0c49e091979bd417636b0c : SUCCESS in 14m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a1dd49c539894a2b888b32253c5ec4ce : SUCCESS in 6m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f5ecde5dec594e2eb9a424787b290334 : SUCCESS in 18m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ebb78b80d862485f8b38945b30217061 : SUCCESS in 6m 35s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/46058aeec5be449fa74d321f5220525e : SUCCESS in 1h 02m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4c4296f44ea14dd7af8e086300836307 : SUCCESS in 1h 01m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/37a96e0d2ea9468289f46c1449e4dd44 : SUCCESS in 51m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ba2c80a932a645668bdcbb315b95918c : SUCCESS in 53m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c21aad9713a44fe4b8e5a4a9f39931f6 : SUCCESS in 18m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b21816e3d6fb4fcc8b4746a9e02f47e8 : SUCCESS in 4m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c64d61d98d9d4dba960ce10461ac53c5 : SUCCESS in 23m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b93f963d63c241f183b77ab8954d069d : SUCCESS in 50m 17s","accounts_in_message":[],"_revision_number":10},{"id":"7843a6d8edb5b7f5c400e9efc094994bbbb766ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-06 00:16:29.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"3e21c06740194dee72c8a4cf02dbb5b533b4384a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 01:08:42.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e86dfcdad8448cb93c07e3165973c66\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/57e4728b0c284a598acb290674a6169b : SUCCESS in 12m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ed6db710639f4e0481d225d5b16d2f15 : SUCCESS in 15m 06s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/06ce416e5b4f4429896ca33dc36ad0c0 : SUCCESS in 50m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d3ff018c097a4bb1919038996d234f51 : SUCCESS in 22m 41s","accounts_in_message":[],"_revision_number":11},{"id":"176aacb72ce6c59968255d4bfbe24c80241bb0fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-06 01:24:57.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2caa81b0e7442ce9babdce3afb807e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9506a2d9751f47b0a8ae12828c46c495 : SUCCESS in 7m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/c4acd55fcdb849c2b8c6594d4dbd012e : SUCCESS in 39m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ae623706d6b54596b603d6eaf0645424 : SUCCESS in 35m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3afff38f547047a7945714937c8fbe28 : SUCCESS in 1h 01m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ccf23ecb5f7147b299fa9e84b6e71601 : SUCCESS in 17m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac78e3f73a6f42f1bb2d9fa89a9aeef4 : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a2343e7ecddf448b81b9449f026d9416 : SUCCESS in 20m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a1f27665ae974585bd4861a33eb9663d : SUCCESS in 7m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ad9a2c62fa3846469695763569622dc8 : SUCCESS in 4m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f1d18ffeed2243deb09470830d406336 : SUCCESS in 17m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e430fee3e3a54a63a2797b42b0ce4de9 : SUCCESS in 16m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dcfd73c17ce54f118725bfe5caf17711 : SUCCESS in 11m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4724803fa69f49d2b70d0d1bf308d765 : SUCCESS in 12m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/685b352c3b7e44d9ba0563f97e7bbdb8 : SUCCESS in 13m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7b3a8864da8746caabbab40c0fada103 : SUCCESS in 13m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2a450548d84c4d618c8619041fb802ff : SUCCESS in 17m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/08fa3a67808149639d90698b7fbdd99f : SUCCESS in 14m 04s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a88def4788ca4ae7b92b810e9fca44e4 : SUCCESS in 14m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9929b3fbf7354b9da23f12e9925aade7 : SUCCESS in 14m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/534f079c448f4ebb936e1c7095aa7bf2 : SUCCESS in 5m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/61f49bc9077d4a7995bc27248ddf2900 : SUCCESS in 17m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c55618b4242c4ba4903a990db060a89f : SUCCESS in 6m 02s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/c04d080f8b8b4c05860dfe03cfbaf7aa : SUCCESS in 50m 50s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e7a122c04fc47e880812060bb32813c : SUCCESS in 54m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f1d3315257d049d59db4d493186f01d0 : SUCCESS in 39m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d5c5c5dc7d9b47b5bdd43a72bad7c990 : SUCCESS in 31m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f59138b53878454999d09f11632496c3 : SUCCESS in 12m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db4e86175aac4ce4955daf5ca872ef40 : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b63a79913b17446f84a7b09945fd2375 : SUCCESS in 15m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/803f8cf2ee8c4ba993e5f07b8fb40e2d : SUCCESS in 38m 49s","accounts_in_message":[],"_revision_number":11},{"id":"969e935521bd2d1d458cbb30caaa02a5df72e596","author":{"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},"date":"2023-06-09 06:47:48.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"8f3731b120e045980e6b5c5633c1084359ccf668","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-09 21:55:47.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"1ea2a0be08e6555d2ec5d29ab118b45a49d77b3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-09 23:04:11.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"a9701c2cbee3debeb62b0437a60accc50bb11848","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-10 00:24:44.000000000","message":"Patch Set 12: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3a425cdc9a44176887e179e652cb9ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16d05efb77264a7abe2d123b1ebdbb47 : SUCCESS in 9m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/51426ccf023f43d8a2463cbf81fa22f3 : SUCCESS in 38m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/86eb37b7a49946b5b13ff5603acc4a58 : SUCCESS in 50m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9fb9e553982413aacde49f981acf5f3 : SUCCESS in 41m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/18b591217044489ba1ff8ba06eee9278 : SUCCESS in 17m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e89d7c27dd204f849a0478748b649d49 : SUCCESS in 17m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d7c1b3170264145a9ef2e1bd970ec84 : SUCCESS in 17m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0ca303c1dee24f67ad876f1c79f184a9 : SUCCESS in 7m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/546a7bd1fdb2449a9416098d657d32e5 : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ce4398e0c5c34033a87792379913a523 : SUCCESS in 15m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4fea5ac29a6b4d788c7bd7cfab2b0d7f : SUCCESS in 11m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/82f6c03b0ede4019ad7f61785751129c : SUCCESS in 12m 50s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a85f18df8d2f40b197ec6409c3980f23 : SUCCESS in 14m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e907647496484d63851e5e7002afd57c : SUCCESS in 13m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/77cc61ee83184f4c995c473c274d5295 : SUCCESS in 14m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2dba93a3602d4e91b44f073e6c139d40 : SUCCESS in 16m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9cb219a9434f4d54bca41b7c19c70550 : SUCCESS in 11m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f8e53f72d8d647e8a741ef976c480d22 : SUCCESS in 12m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/028e9e36511542d4aa3f3f4f0732f194 : SUCCESS in 12m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3d7c31068562415b90ee1442c147220d : SUCCESS in 6m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/df842faff8644d5ba2bab895da20b8d0 : SUCCESS in 16m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/323a23dc433b417dbb6b7b9c2587ef53 : SUCCESS in 6m 33s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/b26658a216df44c08b1edcfad2b43074 : SUCCESS in 1h 02m 32s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/7c01b2d63d5741618b3e9fe4f0505871 : SUCCESS in 1h 05m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8294e0cba75540ba818a405bb227c803 : SUCCESS in 53m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/09ad5735b59b4398aff7fac62b80c25a : SUCCESS in 48m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/373eaf50c66f459da8621aa635b371e9 : SUCCESS in 13m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b711d1d8bbf54e37bf8c8b95a2d46c20 : FAILURE in 3m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d1a4a0c908f3416e9e6ed8d56be8a457 : SUCCESS in 22m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/42e1a1c997a743ff935488689b5926c1 : SUCCESS in 42m 27s","accounts_in_message":[],"_revision_number":12},{"id":"9e468b5ce26d09720e7ee8121adbfd2a7f410b0a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-10 00:37:07.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/415237edfcda44dc9ad4a0640954c801\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9965e55571bf4760a103669267e83804 : SUCCESS in 1h 16m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1b3f7ca267e1438bb42826ebc0392662 : TIMED_OUT in 1h 30m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8393e9b358f04ee7860e77c20e66d621 : SUCCESS in 17m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0621e2d0682841248bf53438ae666e83 : SUCCESS in 18m 08s","accounts_in_message":[],"_revision_number":12},{"id":"baaf6c083c4f789bd8fcee43beabb4d08d769782","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-06-10 03:41:10.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"e4637a5595838bd4a2d61d200f9ff1920b75d781","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-10 05:09:30.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/419ee4816b964ef3a89767c63452390b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f8437d1a8634fe3a2833fa18401b8f8 : SUCCESS in 7m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/8c492579af4c430a97070cbdb4076cbe : SUCCESS in 39m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/adc02ed3d45447738c0ff7230c3f8589 : SUCCESS in 35m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83031c0b16184820b0d2dbfda0baf549 : SUCCESS in 45m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/042a6999eb614006af06a6c8c78cfccb : SUCCESS in 18m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca70af342c734c9781d528a5e29c9849 : SUCCESS in 21m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6cd9dd59080248829387bffd2fa2a84b : SUCCESS in 21m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3220bc64b1a04bb8930a2fadef5b2991 : SUCCESS in 9m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2e31a25e4af84b228c9a1697f93feb5d : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7cefb72b1c4640abba22a0219e65cfa3 : SUCCESS in 14m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0d06a70d83ed44b2a8f8112c14578edb : SUCCESS in 16m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6015718090ad470c91e397a3837589d5 : SUCCESS in 14m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9f38add7580449a98b69160714652c00 : SUCCESS in 15m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1554703978234c6b9e1e17fd5366611d : SUCCESS in 15m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e2ccb52f6a6448a88b7199707e9e8c18 : SUCCESS in 16m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ffa3f28540464933a82a15272f16dc72 : SUCCESS in 18m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0659c3c7c8284d1faa4b7bd54df9cb7a : SUCCESS in 13m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0de88e163c08466fa5dcdc372adea3ca : SUCCESS in 15m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b8d2547c06684ccda2d9b6345412d03a : SUCCESS in 14m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/71bfd39f176f4da1927f3ea729f0b082 : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3476c1c1d1fa44bd9922005bb2e4dc4f : SUCCESS in 18m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/28217d5f1f3a4f71a58a5cb76405a17a : SUCCESS in 6m 51s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/fcbf3d9ffb6b4fe5a8f77a8b7e4d0472 : SUCCESS in 50m 33s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c3dc9125357a40e58a68631227cade7e : SUCCESS in 50m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e2a8beec90ab4089a601e27a80b974a5 : SUCCESS in 52m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a7ffffd45592459eb800445907a4a5ef : SUCCESS in 36m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13f87c1da71e412b87f5a7e5501092b4 : SUCCESS in 18m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cfe460e4d514436b48359b5f0ca3a95 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b1969c44ccc54bf289abd9f8e30b8d76 : SUCCESS in 17m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e82aab6079004ae495df27efd12b3f75 : SUCCESS in 51m 36s","accounts_in_message":[],"_revision_number":13},{"id":"0bacff8580993b9136576a90c00cd663cb807616","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-10 05:14:15.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d54bee1fca4843b8a7abcc3b3f7d6bc2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9777b0e762af4c359ffbbec57cd0b22b : SUCCESS in 15m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/40e6a73e7c4945fc839011621cc2547a : SUCCESS in 1h 13m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f513f77168fb43ceb392550818788dc2 : SUCCESS in 23m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/098ec619876640c18aa18bfd46ddfcb0 : TIMED_OUT in 1h 30m 55s","accounts_in_message":[],"_revision_number":13},{"id":"17902fe7ab53867324ee4a1cca354dd61ef0708d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-06-27 00:46:46.000000000","message":"Patch Set 13: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"b02604a53d275001b7426f2f6cc0fd8dd8239317","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-21 22:24:43.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"959f6af57336e73a2726753f5937631c036da643","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-08-21 22:26:27.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"d23235f16cfdb612cb9dadd23b20f719fa2fe16a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 23:21:43.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e4e61d00fe34ff893620c61086ae63e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6be2901b5f1345c9bd09bdb53d28d656 : SUCCESS in 55m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b8498b3d3da9440389d38ed7604d1c45 : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/04cd5173ccfd45e98c6731103ef15dfa : SUCCESS in 22m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/dee37fa4e1c24457b5019bc9968c1b2e : SUCCESS in 26m 24s","accounts_in_message":[],"_revision_number":14},{"id":"39534e6b4320368276bdc41f77ad75ea6f62ae73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 23:35:34.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42a24c98ed164edf92ed8887dedc1a02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcd0cf601bed48eeaa28e93f1e97ebe4 : SUCCESS in 9m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/42408a926239475f80c63f7f37eb90af : SUCCESS in 50m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/83b15793b1994936a426a7b49167e8fc : SUCCESS in 51m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1c1757e984d48eb92c94298856932d8 : SUCCESS in 47m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/641909ad63b04778b1f12bdfe60b703f : SUCCESS in 17m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ff8d7144d1c43efb00681b6e30fc61b : SUCCESS in 22m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad7e7457d6ff40f1b756b5b4d81ab8ef : SUCCESS in 18m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e220282b4087499099d195197626005e : SUCCESS in 8m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0908e5e9852f4cb08474fbae8b560cf1 : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/853f2b97c59049b2a44157924ad25946 : SUCCESS in 19m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c819b3a3721547fc812ecdc7b25bdfd0 : SUCCESS in 20m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/aa813f9357794b24afee70f8c3bf5a1f : SUCCESS in 17m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/17f94f8055c64db09aaab44269988082 : SUCCESS in 19m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a441415a408e4eefbfba79df472b55e8 : SUCCESS in 18m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6520e4a1d17149048d777c3170763e6b : SUCCESS in 16m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/52a25f6d76df46d18ae7b4687123ae80 : SUCCESS in 18m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7b873ebb9c4e4fcda8a2090aca461d1f : SUCCESS in 23m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/08530f5153754a3d96ca67bb284542de : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e73bc31904f448b78eab959bf25a5a36 : SUCCESS in 20m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5f9846bd3e7242d9b893100dc4f3cc7e : SUCCESS in 16m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4a77f3a3d42049ac9e711a09b60e00b7 : SUCCESS in 6m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/764df2d4a8ef4aafb52d0394bb37be4c : SUCCESS in 16m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0d00ff830b76464bb7e9cc93b04cb181 : SUCCESS in 7m 10s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f6e7e35d14cc44d88adddfa86d22dce2 : SUCCESS in 53m 17s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/0d7835cbdf6044f9928c266c761558a1 : SUCCESS in 59m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/74e1cd196cc5449397e9aace49b2a9c1 : SUCCESS in 1h 04m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3fdf12b1b22a4807b395b47ae15b4dbe : SUCCESS in 52m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be2058520b144984904b216cb7ddc3cd : SUCCESS in 22m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c7ce1ed8f8746aab88ab5cc18b1d360 : SUCCESS in 3m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ca6ab50cc5334d25ab4f918b1c726a74 : SUCCESS in 19m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/beda796366574d7387323f8590361059 : SUCCESS in 44m 12s","accounts_in_message":[],"_revision_number":14},{"id":"3f1933f88b26f11c91382bfda04459df83426574","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-03 21:32:19.000000000","message":"Patch Set 14:\n\n(15 comments)","accounts_in_message":[],"_revision_number":14},{"id":"5eb46e85e443875a642c9fa327b0000ce451cfd3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-11-07 00:02:34.000000000","message":"Patch Set 14:\n\n(4 comments)","accounts_in_message":[],"_revision_number":14},{"id":"1a23b204d3b080f202892a6adebd972dc5dccd02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-11-07 00:26:37.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"e59b93b00e0feca47a5b6b2f0eb01218725853f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 01:35:09.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/647a6e02f3554d5ca1af55edf378dc1b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dd7ed9890c44751831e9862e0fa84eb : FAILURE in 5m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/925d38cfab1040729ca7c98b97fa2ebf : SUCCESS in 1h 02m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6e8c2b2afe704502b99c0623e058a698 : SUCCESS in 43m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2f17504e9e7646fa8612a918c374da79 : SUCCESS in 38m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/797b7a54f9734416b426642b19481d69 : SUCCESS in 45m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c7387e47743c451a83a052b87f1d8a61 : SUCCESS in 20m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae65b0990470409d8da65f173507c932 : SUCCESS in 22m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d0e51ac939d7491b883e831925b021b8 : SUCCESS in 15m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/01f23c1b57d644e29b73dd9b6b5fb572 : SUCCESS in 7m 58s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9238960ffa694a6e94aa62d012240aab : SUCCESS in 4m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8bd68ee113814d7c869d6cb1fac05e85 : SUCCESS in 17m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ddd669798ac44d959f9abaf9b13f3143 : SUCCESS in 15m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/721fa785ab3541f28c803d647231f001 : SUCCESS in 12m 09s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5d6d492b71f94813a221003f6e49a286 : SUCCESS in 19m 52s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b288c14947d84385a2f4b3930e01711c : SUCCESS in 16m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/65b3f55025fd4e86bba2a51aff64804a : SUCCESS in 21m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f0c4421da06645fca9e21f1e9daa03b1 : SUCCESS in 17m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/54f9f27813eb447890a2c27fc64720f3 : SUCCESS in 18m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/be8d83c253664155a612f6df2229e745 : SUCCESS in 12m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/88ace4818e7d461b8c7286dd5741e882 : SUCCESS in 19m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1f8019b64f884e1380cac8bd7c662e27 : SUCCESS in 16m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3ea3981dda004df2b60c021a17ce7183 : SUCCESS in 7m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0b4194a9e15949a091191a3ff28d0f18 : SUCCESS in 19m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b767fc11a4c84db2b06d40435082702f : SUCCESS in 7m 27s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/591d06f43fa94e44ad0ab01b3a315d90 : FAILURE in 1h 00m 25s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/9a6cfea972a44cc5aefedf1e846814ed : SUCCESS in 52m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cc2520c014954dabb0a49215c373b8a0 : SUCCESS in 52m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cb8b476e8e624b20893d9f87328bcc2d : SUCCESS in 55m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9565a81127440ea81000fff5b642397 : SUCCESS in 21m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/290667caa6604818b06900f0f31865e9 : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cc06b7e67bd344a797d9b1a697beec93 : SUCCESS in 19m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc19ff77f0824da3a5984066b44bff1d : SUCCESS in 40m 54s","accounts_in_message":[],"_revision_number":15},{"id":"2702d01aa2f0db2cb92747a6e344e756899ee45d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 01:40:47.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1757d8d240ae47339a00d652ddf35dfb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d1f7dbc5eb044d6c83a1f9257596817f : SUCCESS in 52m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d178e9d109db4fc79dc1752d71c57310 : SUCCESS in 1h 12m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c76eb3ef0ff4926bd7b33390d112c51 : SUCCESS in 57m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2900214c238f405f8aa0c4f6792e2ddc : SUCCESS in 1h 10m 21s","accounts_in_message":[],"_revision_number":15},{"id":"39cd3846d002ef18e79b818d1c5ce6af58b9ea1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-11-07 16:53:52.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"61cad929c51582ab234ae047423ca231e090cb15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 17:47:12.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3f7339f105f42cea6542795e7869130\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/07b35dc617934e3ab94648fc9d59c04a : SUCCESS in 14m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/89878238e44a4c288a26b5892f4cc917 : SUCCESS in 17m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/328b5a30f7b54c20a58881080f300f36 : SUCCESS in 17m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ae9d876f4264b3ba20ecf73e9db4ebb : SUCCESS in 49m 42s","accounts_in_message":[],"_revision_number":16},{"id":"ea4ee8398719bc513f888d0c8466cb74159fded6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 18:10:36.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ea678e57e174c4bb3827f817da0bef2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/163c256c765f4cceb7920e33c20a3e98 : SUCCESS in 10m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/cb1d2ab075584dac8bd8213c68fd8601 : SUCCESS in 52m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fc4dca340274478fa1d3fce0915c04f6 : SUCCESS in 1h 09m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0a11da566bb54ecc98319ca6c42e0238 : SUCCESS in 55m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ea9941ccade43a490154c107f5133b5 : SUCCESS in 1h 00m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/403b5fb4749040ea80edfd4e8d851ab6 : SUCCESS in 16m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0cd0caa86df349aabb6723ea89b99b8e : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/01bab805b42e45deb0e7abffde88d3c8 : SUCCESS in 20m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bdc27b666bc74e1fbacf661ce0ea7b03 : SUCCESS in 11m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b1dcd13288574c65bcf8b397bd2bbcbb : SUCCESS in 7m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bd3bdbb083d64ce3bda686796563a28c : SUCCESS in 17m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/724ad09544d04117a7c7be0dd8a0c5a9 : SUCCESS in 16m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/af73965dc97f4de5b22bd460d1b584d2 : SUCCESS in 19m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/675f60faa97a4a33a0d4a85af5d30e4f : SUCCESS in 16m 54s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/52c486b20c7e4007a66891785d0865fc : SUCCESS in 23m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f53951f682834b77ae2366262374b3a5 : SUCCESS in 18m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/03eb2be1896e449abfb3a139d99715d3 : SUCCESS in 20m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ee127cd7a73b484199e827481cd6c836 : SUCCESS in 21m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6741ce6c4b0a4fbb8134899fea3172d8 : SUCCESS in 19m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7f885ca0d22543ecb94f2ac276374909 : SUCCESS in 17m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/22933375a6bf46e1989893dd43ec4b1a : SUCCESS in 22m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3e14ad9f4e564c4cb4d374c51aaedf70 : SUCCESS in 7m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b06fc752e103464893bb77ff1787e536 : SUCCESS in 16m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/466641125fcf4688bb6a091bf35834f2 : SUCCESS in 8m 13s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/cf636a859d5f4d8d9acbd5216189c267 : SUCCESS in 1h 02m 53s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/1d17d6ac147441168c7531a0faff27e2 : SUCCESS in 1h 10m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a8d48374be674af497ef22a6aba919e8 : SUCCESS in 55m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/61b00e43957942ec8d0c6cc62a131b83 : SUCCESS in 58m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af1623dcdbab4dd3846ec0ea0f15f685 : SUCCESS in 20m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e540c504d5c4a47917c6fc844c77812 : SUCCESS in 5m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d7927291957a4c1bbae9bef8468bd920 : SUCCESS in 26m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c137ed2c08f44d9a80051275892c37b : SUCCESS in 50m 02s","accounts_in_message":[],"_revision_number":16},{"id":"20fcd7921037434fd1541dd4f0388261ff944e26","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-07 21:57:12.000000000","message":"Patch Set 16:\n\n(5 comments)","accounts_in_message":[],"_revision_number":16},{"id":"1ff9b803cf4dd8cadf536746d2d1652a7e8a18cd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-11-08 17:41:17.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"c383dd45bfc85c49e24f119ba0f46699eeeabdc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-11-09 22:10:13.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"95be951c694ae6ac0bdbd914da2d559102e8ddb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 22:32:55.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7fb590a33414f18b3675da66a31d038\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/95f3cab2a8064420a8f066485fcbaa06 : SUCCESS in 16m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9e094c8ac4ad4677a0ab94b8447ea706 : SUCCESS in 14m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e528af2f6ad44f309935ddb3207f0106 : SUCCESS in 21m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/83c5eb8f6f1a44e1bcf86c41c28f4091 : SUCCESS in 19m 11s","accounts_in_message":[],"_revision_number":17},{"id":"a1aee20935d1338442e3411cc0e837985e9b52cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 23:32:19.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ed843dc696047b289bd1e8273cc269a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa90df4af2844e50a4889a95f33a71bf : SUCCESS in 9m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/7ee42b8be376490588a8752195207af4 : SUCCESS in 1h 07m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bf269d2acb3f48f9a6f5fc12349d4613 : FAILURE in 48m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2c8f0844d924426ea4079b7e078bd0b6 : SUCCESS in 55m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d2a736c246544d329cb083dca8d87cc5 : SUCCESS in 47m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4eac2df8ef544aad9759f6520fbee957 : SUCCESS in 17m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef664dd07a8348b0b427cf995260c1c0 : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de4a62499baa4e66ade59dd336af5c30 : SUCCESS in 20m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/694de9e0b09f4fd0ad6dace785527d69 : SUCCESS in 9m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c217df5c7bc04cf793a17d06ba61e29b : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5820cabd17ed4df9935735f43ca5fd0a : SUCCESS in 17m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/30b25651282648cda7c3f3609cb45e65 : SUCCESS in 16m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ed62ec9436874fb590cfa756a0a57f43 : SUCCESS in 14m 58s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b9c23886a7e647afaf6660ac0c3a7824 : SUCCESS in 16m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7c0d1fb719e14ca9b1160f7103980ca2 : SUCCESS in 12m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2f5a9fcb83974ae8b1508f23026c4159 : SUCCESS in 18m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0b199b9e298845348edf9eed617d4c02 : SUCCESS in 18m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1b327fe4b30540878164b9e792afc784 : SUCCESS in 19m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d7fa324d347747d8a4fcbad9e0f80f44 : SUCCESS in 13m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/00ccc236295a42d58c6839895a048a0c : SUCCESS in 16m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/84d06cb664074723b78860f4d8f10a17 : SUCCESS in 17m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4b370e60a43b44bea6e99840efc0ffb0 : SUCCESS in 6m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9f4d36c244d54866b58ea6441f7d041b : SUCCESS in 16m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/467f31baef2a4c4f9e174d2c16575aec : SUCCESS in 9m 00s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8a4eb73a229c408cbb977ab663a70c79 : SUCCESS in 56m 38s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/59b23652577e420bb92c99f1e2d8675b : SUCCESS in 50m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a8837216c22f49c39f6b658d72d1f2d4 : SUCCESS in 48m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5ffdb36292ff4c09813d6c513d738c6b : SUCCESS in 58m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c9e87a09ac24c8889f61ee14dec1d57 : SUCCESS in 16m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3e506f7cf7a4415832a325eecedb6ac : SUCCESS in 5m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a87929a6a3464660b22ce264b85147c2 : SUCCESS in 20m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/526d7cbd7d8441b898e5d59d23100aea : SUCCESS in 39m 37s","accounts_in_message":[],"_revision_number":17},{"id":"390551729fb1d65df5c5726b4f80b931729ab7d1","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2023-11-10 17:43:19.000000000","message":"Patch Set 17:\n\n(9 comments)","accounts_in_message":[],"_revision_number":17},{"id":"c35c7759f7ea89c9761d8ce53c1d28e8f0e5a91e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-11-13 18:20:44.000000000","message":"Patch Set 17:\n\n(4 comments)","accounts_in_message":[],"_revision_number":17},{"id":"0f82746726feee4cd8c4f43ca397c0e6f7397b96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2023-11-30 23:12:25.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"6f3b428911aef5659816dcfac4447f5218370d77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 00:29:42.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56443e17479d441893689163ee3b9351\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79879a17e3a84d1eacb549acf5659a0c : SUCCESS in 9m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/62e93ecce1e0487da97c1a9b3fb0d5c8 : SUCCESS in 1h 03m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0410188902bc40b3a6dbfc7ec2b7ba28 : SUCCESS in 43m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/45db6111f7a34a13996a0131cd1f8f2d : SUCCESS in 46m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2fb1f07d2c0e4bb6b95105c7672ec4ca : SUCCESS in 1h 01m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1acead1d279f4229872c4236e2d244f8 : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/633b312af15a4ea8af1efae627e333de : SUCCESS in 22m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ddef9d783914824998b0407ad715d22 : SUCCESS in 21m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/87e64af833f048e7b9e4ab0f2d916c0e : SUCCESS in 8m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ba4b6431f0304900ac7614a64eae0f9c : SUCCESS in 8m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b759f88227e9494ab72312ba73d34d60 : SUCCESS in 18m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d53c8b45937f44f1a54b88ab4083187d : SUCCESS in 17m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/12c06982cece4450be7764b6f6851676 : SUCCESS in 16m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6c317395e99a43a5894ec9374b128ac6 : SUCCESS in 16m 36s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8b13347684004e6980bef152672cb2e4 : SUCCESS in 14m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fabaa213277148f99e9875c4c006c551 : SUCCESS in 20m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/db14e457eb8749df8c9a392efbb7e91b : SUCCESS in 18m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/99dd9a858bc740c89fd644c32e5c4368 : SUCCESS in 18m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d450187ac33f480f840d70d4999dad2b : SUCCESS in 15m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/82627fcf22124c158230e9540c661fe6 : SUCCESS in 14m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2755147c0e4247fea53dae4dbbf293e5 : SUCCESS in 19m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/60f4f7f86cf445ef9a0b7403a024a10b : SUCCESS in 6m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8350c9c0c2744515bf4a9f8d4ef9421d : SUCCESS in 12m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/584346cf9d4344a488903042e414175a : SUCCESS in 8m 07s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/ffaa560940834cfc83f84584d2d58e27 : SUCCESS in 59m 34s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/93f59330908c45cea74b15907597d6d3 : SUCCESS in 1h 00m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4a1e1e2e7e5f4ac3b9bc856a86753f43 : SUCCESS in 1h 02m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/77659cd7d2fd4c7994e085e014377814 : SUCCESS in 55m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6aad411803d4a6089feb80cc24afe75 : SUCCESS in 19m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6346e4e46bcf4412b0041587e358939c : SUCCESS in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/549e949b86574298b2414407e7ea9e2f : SUCCESS in 20m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fe98a081cdc4b8ca3493a44eeb86968 : SUCCESS in 1h 06m 47s","accounts_in_message":[],"_revision_number":18},{"id":"1cfd1e09260e33de911b602c39ca43ea38153160","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 00:45:58.000000000","message":"Patch Set 18:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a537686fd23c40298dd4e84878a3d302\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe34e3fefe234146aa11aa76705d08d4 : SUCCESS in 16m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0a5c67b2a464224ab40a1f5559cabc9 : SUCCESS in 1h 10m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/709e2b7454764a8694e4c6ee863c2765 : SUCCESS in 22m 14s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e7f76000a7e246b4af97378696d3d1a2 : TIMED_OUT in 1h 30m 54s","accounts_in_message":[],"_revision_number":18},{"id":"fe441d1adf47e694d586a63a2b0b7381bc543792","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-02-26 08:13:57.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"5d0f202d8f7dd2a38b41fa3d7e460f021490e8b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 09:08:19.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd13c57645b44dc5b5249abcddd5f1e7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1aca67b86d794270b5eb648e953cead8 : SUCCESS in 19m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6af0571053b2404aa71c445904a674ab : SUCCESS in 52m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f3fe0fbd4ba348dea1ccae14944181d7 : SUCCESS in 21m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0034875e802f4472bcaa3eca7ccb11a4 : SUCCESS in 42m 07s","accounts_in_message":[],"_revision_number":19},{"id":"92f897deae8fafd5378954a9fa8a7bb126779e24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 09:45:53.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eadb57ee41674861b1882c7097bbe28f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23978cc854174f05b8510d646a155103 : SUCCESS in 8m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/9ad4e4b6fa4b48f1b2dc8c8bcc4001f4 : SUCCESS in 56m 51s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/203137e947f1413c904e19e2a2531d62 : SUCCESS in 1h 07m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2b1c35aa45c140d1a5a78da41fffc34f : SUCCESS in 45m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/663def08bf414a3797268457e0fcb117 : SUCCESS in 55m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/26d648ef1a80440d972762a6463eb3bb : SUCCESS in 30m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6c3e656c6e554246bcea898b9d95c729 : SUCCESS in 29m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/23f38e6709454215b4d2d2f4925d5013 : SUCCESS in 21m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7d399fb7f172497fbe559e6d5b7d0ada : SUCCESS in 10m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a0d73ea9cbdf4864a39dc57ae8a654e8 : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ddf1623f438946c3a394b0273ee9e810 : SUCCESS in 31m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0a72bdc698664728be946c53131b6e6a : SUCCESS in 14m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/085e68b29e834b0db2972dc881c44a64 : SUCCESS in 19m 33s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/84500aee09224721b837a04e0ca13804 : SUCCESS in 15m 44s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e0291259d6674d16ad9b6de5247cf547 : SUCCESS in 14m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7e75f3c732e24b22ba4b542bd76e48d7 : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9be8693a8d8e471eb9f893d5069acb19 : SUCCESS in 16m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2d7fdbe1a76c450481688d20249a4d0f : SUCCESS in 17m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1bab9f96694f4a2087314bbe03a254ac : SUCCESS in 14m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/88258f9cff7a46e5a302f7d088c8bd9c : SUCCESS in 16m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/577d91459ffd4f45a32f2fed4e7d1148 : SUCCESS in 15m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6db72a386b344e028f52085c2e79fbdf : SUCCESS in 8m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/878f8ed2187547a6b8fffbf90413e8c6 : SUCCESS in 18m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2768f4fdcdce478b8937c4cc0246ff8e : SUCCESS in 10m 12s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/8b88e366a31d4a9e98bd75a53e261a5a : SUCCESS in 57m 28s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c9d77a4c96a146aa863470973e4f8a74 : SUCCESS in 1h 30m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/394f898f05594e869329a1ab2ca34aaf : SUCCESS in 52m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e549b0984c5740a9af56b0935aa7a29b : SUCCESS in 54m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f08c5ac20ac43e18a24fb8c6b336d47 : SUCCESS in 29m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/321debf0ebe64b7ba35d2fb85f535cca : SUCCESS in 3m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/041a18656546430cb37d39103ee9381b : SUCCESS in 32m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ebb4e96d4d54c60b0e5c5dab99e8976 : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":19},{"id":"51a75f5b8b49190712eec42e7415936c223de5b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-12 21:15:07.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"789e9dbdd634756149299b5b52fda7fb079416b1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-12 21:16:17.000000000","message":"Patch Set 20:\n\n(9 comments)","accounts_in_message":[],"_revision_number":20},{"id":"d30b4d3d27ac431deba320efee4f1b1abedbe1eb","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-12-12 21:52:40.000000000","message":"Patch Set 20: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"c0389c3768d8087326dc2ec6df4ba135765f87f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 22:30:36.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8406e95539554ce4b75eb5f3407aee51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a434301dd0ea4b5c87fb9d828d910131 : SUCCESS in 7m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/7023522e52cf4be2ba971e03b480eb6f : SUCCESS in 51m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/de9f22e8d4934cc3b39c2b9d1fca7930 : SUCCESS in 54m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f4d4f31c31b8450a93217a21eea5e223 : SUCCESS in 48m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26d7e4cec6df427eb2fc7154060d213a : SUCCESS in 55m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a17985f06d504cd4a08d112d5771eb1e : SUCCESS in 23m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/050eee9ad0b9408399b5ecbcf90e1e79 : SUCCESS in 20m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/12cade65afea44418c2484fd1938278d : SUCCESS in 19m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9b8195188fa743328261136a2598b2d3 : SUCCESS in 6m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3c4a0c3df1104862894d32f1c66771c4 : SUCCESS in 22m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6df0e65bccd64ca0b3cc3f4c91ec4f4c : SUCCESS in 20m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/324ad116c6d84c8690ce9a8207c55ad7 : SUCCESS in 22m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/81003ec88f2c4d3286b65c1028ccfc3a : SUCCESS in 27m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/51cd4f3345774b55bb2ca7aa0460271a : SUCCESS in 17m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7ae0f8cc41e64c69a1e8b52e36aec651 : SUCCESS in 18m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/438eeab8f157431fbb5b080bc674671b : SUCCESS in 18m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/47f1e6c325bd403696d013d32b77d931 : SUCCESS in 14m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d594acf2c0324a468f8a4c3bd16b4b86 : SUCCESS in 14m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f8593f4799dd49a79e1da6a826dc0c4e : SUCCESS in 18m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6bd4bf0afba94f1ea418be751c78a96c : SUCCESS in 6m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3a2cf46fce224e67aff8574f0d70c9ee : SUCCESS in 17m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/052c2a25993141f9bf49eb84ee7b967f : SUCCESS in 6m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ecd4bf9c475c4b7c98b8dce22f83f989 : SUCCESS in 53m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5fe368ab2edd4f93a5e6904504004213 : SUCCESS in 25m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4c5f0441ee544ca1a34f3a1307c46fb6 : SUCCESS in 38m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ee6f5a139d54364a4879c41b69f4146 : SUCCESS in 19m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d308343514447e680cebc2f2f1a0554 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4e925205cb7641beb808fede2aa7639b : SUCCESS in 10m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c55f496c2cf246c0a79cc418075e3be2 : SUCCESS in 43m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4ff22d707cc4dbd8e593fa75b7fa769 : SUCCESS in 50m 17s","accounts_in_message":[],"_revision_number":20},{"id":"aebbd67458100dbe9833d9ff3c86106ae91c8c07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-12 23:03:14.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74a3ece66de44092a20e9c9d5b84c990\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1c23e76d7b9b46fd8ee10979e1aa858a : SUCCESS in 1h 46m 13s","accounts_in_message":[],"_revision_number":20},{"id":"81bd58630f61edefa1f3e3e80a350ebfe6990f56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-12-17 19:32:16.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"8f0d80a8d7ac5b5140ec8315e19ed069f81fdd3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 20:44:51.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30eb502636084c06ad70b638062a0ccc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43114483dcae4c53befb9dcd6d6d01da : SUCCESS in 9m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/595efc9b02594373932d0942f66cac66 : SUCCESS in 1h 05m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7eef94b366b94e8c9591874b14225fc5 : SUCCESS in 42m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d0c5c072927a4f00a217306d2e08657e : SUCCESS in 52m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05089c84119a40c794b0972df2d5f3fb : SUCCESS in 42m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b0114bfc02d14365b9effdbe1d9cce86 : SUCCESS in 20m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/19873d7bcb39478dadf41699d9e9cfa3 : SUCCESS in 19m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/491fa06d36a94bc8836a929fbdf7f331 : SUCCESS in 13m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b8548a9cde4e4c078b077cc4a62b1998 : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dd6927723f07469fa6e8bdb0d7e57ba9 : SUCCESS in 17m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7283fa553ec9455389d45fa491758237 : SUCCESS in 16m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e013c7a17a2f4a2d8dbbde5fec4cd808 : SUCCESS in 15m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/03564a65b7ac4ece9b455e1faa49b9d2 : SUCCESS in 21m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3d4c21490b714238b5389f43dcf011b3 : SUCCESS in 14m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/af74aa89e3a449a585a1cc10882e1f4f : SUCCESS in 15m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1c8ac80a8c994839b9abe6e32bb3bdb8 : SUCCESS in 17m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/005ca42a479c418e8ec79e78a9f9e800 : SUCCESS in 12m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/762f98e6fe23451ea3e6b21a10ec9826 : SUCCESS in 13m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/03e4ca445ddb48e686fe7e61d72142b3 : SUCCESS in 14m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b93d08a8d8214e39b961b9b157e563df : SUCCESS in 5m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cac20fda63f64674ad280e2d46e07077 : SUCCESS in 14m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/623821f8816a401ab98edb6d71fcb665 : SUCCESS in 6m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b1e69c1f9f545e7a4c7bb64e2379e79 : SUCCESS in 51m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6586e8071e3940bca8e0782aa05c55d6 : SUCCESS in 49m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/67f4d10243764c6bbcf72377c3b0cd50 : SUCCESS in 44m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c0f75269cb941ccaf87996f48419ff9 : SUCCESS in 19m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97665ae2289b4e1bbd907cf25ebc9d31 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/952f8c3a7e59478c81786056b3dae9ea : SUCCESS in 18m 28s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d4ee7d3e429458b8df5fffaa0bfead0 : SUCCESS in 54m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/969c6a606a1a487db3f498e266d45106 : SUCCESS in 50m 59s","accounts_in_message":[],"_revision_number":21},{"id":"40d0d32e51c1a006d4787ed61a6655b987337803","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-17 22:33:41.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ecbfc5294474a748d22ec46ed7632c6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b6d941f5e88d46399fd3b8981f7d7162 : SUCCESS in 1h 42m 30s","accounts_in_message":[],"_revision_number":21},{"id":"aeb336aabe8fc6519e03618340982fbbb6b5eddb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-01-16 03:00:21.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"671e927d13ffb91195b1b0db5dc8b62b73696d3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 04:05:08.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69e2ca381de343608dd1c8df97362256\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c36a0a39dae48c3a896c80c70777b0d : SUCCESS in 6m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/126e3b4fbf4f45afbb896d261d2bb37b : SUCCESS in 56m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/94ea58bb0efc436696569a60b87c776a : SUCCESS in 39m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8e6612e58e7a4e95a250addc6674fdaf : SUCCESS in 37m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f81aa737f55488aa0519748ec4745fe : SUCCESS in 1h 02m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1739f7ca2e3b41f4b4d6a09c1d60c027 : SUCCESS in 15m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2711f7d48b9a4f099d7956a34ee4fe36 : SUCCESS in 21m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/98abdb07bc1f405b99f1bcf12294802d : SUCCESS in 19m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3b462eb8b7034f898c713e889d607cdd : SUCCESS in 4m 37s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/98b96ab94bec48f9bcde3b3d805aedaa : SUCCESS in 19m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d2d832b9a606472290d3fc89afd7d284 : SUCCESS in 18m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/13ff2daad3c84a0fb4671864e4283929 : SUCCESS in 12m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b6f3338456ea434baac23f8cd64597b5 : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a6fc1b79b6ad4d7fbf109398ee6b6f50 : SUCCESS in 15m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e6ae1633ab5340a99b59caf08cf916db : SUCCESS in 16m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/59fa952f38ed4977a9a1b9a18af5ef4d : SUCCESS in 5m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1ecd660bf21c488d99196105f91629c6 : FAILURE in 19m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2163590451df40aea40ab65acdf51169 : SUCCESS in 5m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d72c4d1087ae428c8076fe8d027d728e : SUCCESS in 53m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fadef1141ce1419f864fa04703d40a4b : SUCCESS in 46m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6446fbc7d3f148fa9c31302219a15fb0 : SUCCESS in 50m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a981ce8624948aa835f0e250212a003 : SUCCESS in 13m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ac9c99cdd30426eac8de5fe19976eb0 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ac12b93019cc4456bfc7f74242e76acb : SUCCESS in 15m 57s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c56bc1e6988041848306f635728ceccb : SUCCESS in 45m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1be6202bd8294ea8a111a80b76477bc0 : SUCCESS in 42m 32s","accounts_in_message":[],"_revision_number":22},{"id":"2ed716ddef4cd02fa99d16b46787f6e4f7cb8d75","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-01-16 04:42:36.000000000","message":"Patch Set 22: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":22},{"id":"d4c9e1e676db259c02af17a8fdb2c4d294cae633","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 04:43:13.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"4b417bf2d7cc46aa0db04b5172912d7066e6b72e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 05:04:28.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da25ee529b4c4ec894e967fa33c56809\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cff12de30e034710995a60c9924202d2 : SUCCESS in 2h 01m 48s","accounts_in_message":[],"_revision_number":22},{"id":"d5a08250f3100044c084c9f4957a3f4ab7df9126","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 05:49:40.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82a549792d9e47a19db3b739fd509077\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ccf3a59f9df43a6a09b147ec1f0b34e : SUCCESS in 7m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/8499cf5fa94545d7afca8314621f4066 : SUCCESS in 45m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cfa5fb6b3e074f7cb11e4ac0d75a12a2 : SUCCESS in 1h 02m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0feacaf3c2ba4ba3bad34dfe28008f91 : SUCCESS in 40m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d26e4f3dea4d481ebf761745b103fdc4 : SUCCESS in 1h 06m 23s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/fc2ec303f14844828fd9bc0ce28c8b0a : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/aa2aa223837d4dedbcab4c25ddcef31b : SUCCESS in 18m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/04aefc0be6004b0796b977a3f2510031 : SUCCESS in 10m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5006f52b400440c599755d0e8e6d92fe : SUCCESS in 20m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cd29764fd76a4759a985cdac398a40a4 : SUCCESS in 13m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f0ed90517d694355aa0f678915417cd2 : SUCCESS in 15m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/644222e629d0452fb7317db593072d5f : SUCCESS in 17m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/58817cc1ff484ace98869fa54ae1efb5 : SUCCESS in 6m 01s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/82d35afd69e44f33ad90361ee31f7b85 : SUCCESS in 6m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/71b7bb3059e2405f8a183a813d7e4e02 : SUCCESS in 59m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8341eddf94e4408483a51d3835cccf3f : SUCCESS in 27m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/127e855ebe714eca8a67843fb297a271 : SUCCESS in 36m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9529005a393b4ccb8112e67f3d5f5bdd : SUCCESS in 20m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64941875003f4f60be2eb7b79c844e61 : SUCCESS in 2m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/29a5fd2686254e63897becc68bb797cb : SUCCESS in 18m 23s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d80bc4740ca3415892418ec25366d6e4 : SUCCESS in 38m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c20e1aa39ab84968bbc6b82b80ba5baf : SUCCESS in 48m 38s","accounts_in_message":[],"_revision_number":22},{"id":"7a1e3f17f7ddf8888539f4c70822b64815e6fffc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 05:49:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":22},{"id":"e25db0cb2eebc951cc25bdb1895b0c72709c36e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 05:50:44.000000000","message":"Patch Set 22:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbc3be1736b74e568194b9730aadd563\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d1ebeba44c94f4488a3187b511f9a4f : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/bd9ab0a9673145d7b08c6eb4e09ebcb9 : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"ae6300af86c292b4ded1b4606f7e978bf7474f8a","revisions":{"d07a7f8a38176769fc51644df021ac560cf394cc":{"kind":"REWORK","_number":1,"created":"2021-04-30 21:24:28.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/789035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/789035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/789035/1"}}},"commit":{"parents":[{"commit":"443677f104c0181490cc1ff04f95efbed0a6a0de","subject":"Merge \"object: Plumb logger_thread_locals through _finalize_put\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/443677f104c0181490cc1ff04f95efbed0a6a0de"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 20:33:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 21:23:55.000000000","tz":-420},"subject":"wsgi: Pass worker pids to new manager when reloading","message":"wsgi: Pass worker pids to new manager when reloading\n\nThis allows the new manager to recognize them as they exit instead of\nlogging\n\n   Ignoring wait() result from unknown PID ...\n\nWith the improved knowledge of subprocesses, we can kick the log level\nfor the above message up to warning; we no longer expect it to trigger\nin practice.\n\nDrive-by: Add logging to ServersPerPortStrategy.register_worker_exit\nthat\u0027s comparable to what WorkersStrategy does.\n\nChange-Id: I8227939d04fda8db66fb2f131f2c71ce8741c7d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d07a7f8a38176769fc51644df021ac560cf394cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d07a7f8a38176769fc51644df021ac560cf394cc"}]},"branch":"refs/heads/master"},"30ed8b8bd53f6d9e719039d2b86a00e2b5125ac0":{"kind":"REWORK","_number":2,"created":"2021-05-03 21:31:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/789035/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/789035/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/789035/2"}}},"commit":{"parents":[{"commit":"443677f104c0181490cc1ff04f95efbed0a6a0de","subject":"Merge \"object: Plumb logger_thread_locals through _finalize_put\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/443677f104c0181490cc1ff04f95efbed0a6a0de"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 20:33:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-05-03 21:31:03.000000000","tz":-420},"subject":"wsgi: Pass worker pids to new manager when reloading","message":"wsgi: Pass worker pids to new manager when reloading\n\nThis allows the new manager to recognize them as they exit instead of\nlogging\n\n   Ignoring wait() result from unknown PID ...\n\nWith the improved knowledge of subprocesses, we can kick the log level\nfor the above message up to warning; we no longer expect it to trigger\nin practice.\n\nDrive-by: Add logging to ServersPerPortStrategy.register_worker_exit\nthat\u0027s comparable to what WorkersStrategy does.\n\nChange-Id: I8227939d04fda8db66fb2f131f2c71ce8741c7d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30ed8b8bd53f6d9e719039d2b86a00e2b5125ac0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30ed8b8bd53f6d9e719039d2b86a00e2b5125ac0"}]},"branch":"refs/heads/master"},"fa6c6b4efaed74543076f7a9124e3e9467fd65c0":{"kind":"REWORK","_number":3,"created":"2021-05-04 00:01:46.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/789035/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/789035/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/789035/3"}}},"commit":{"parents":[{"commit":"443677f104c0181490cc1ff04f95efbed0a6a0de","subject":"Merge \"object: Plumb logger_thread_locals through _finalize_put\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/443677f104c0181490cc1ff04f95efbed0a6a0de"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 20:33:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2021-05-04 00:01:29.000000000","tz":0},"subject":"wsgi: Pass worker pids to new manager when reloading","message":"wsgi: Pass worker pids to new manager when reloading\n\nThis allows the new manager to recognize them as they exit instead of\nlogging\n\n   Ignoring wait() result from unknown PID ...\n\nWith the improved knowledge of subprocesses, we can kick the log level\nfor the above message up to warning; we no longer expect it to trigger\nin practice.\n\nDrive-by: Add logging to ServersPerPortStrategy.register_worker_exit\nthat\u0027s comparable to what WorkersStrategy does.\n\nChange-Id: I8227939d04fda8db66fb2f131f2c71ce8741c7d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa6c6b4efaed74543076f7a9124e3e9467fd65c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa6c6b4efaed74543076f7a9124e3e9467fd65c0"}]},"branch":"refs/heads/master"},"292208b2a8e7620fd5fec129a0cc1695fe0e95b3":{"kind":"REWORK","_number":4,"created":"2022-10-17 21:40:56.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/789035/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/789035/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/789035/4"}}},"commit":{"parents":[{"commit":"443677f104c0181490cc1ff04f95efbed0a6a0de","subject":"Merge \"object: Plumb logger_thread_locals through _finalize_put\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/443677f104c0181490cc1ff04f95efbed0a6a0de"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 20:33:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-17 21:25:43.000000000","tz":-420},"subject":"wsgi: Pass worker pids to new manager when reloading","message":"wsgi: Pass worker pids to new manager when reloading\n\nThis allows the new manager to recognize them as they exit instead of\nlogging\n\n   Ignoring wait() result from unknown PID ...\n\nWith the improved knowledge of subprocesses, we can kick the log level\nfor the above message up to warning; we no longer expect it to trigger\nin practice.\n\nAdditionally, add a new tunable, `stale_worker_timeout`. Once this many\nseconds pass following a reload, the manager process will issue SIGKILLs\nto the remaining stale workers. This ensures there is some limit to how\nlong old code and configs may still be running.\n\nDrive-by: Add logging to ServersPerPortStrategy.register_worker_exit\nthat\u0027s comparable to what WorkersStrategy does.\n\nChange-Id: I8227939d04fda8db66fb2f131f2c71ce8741c7d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/292208b2a8e7620fd5fec129a0cc1695fe0e95b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/292208b2a8e7620fd5fec129a0cc1695fe0e95b3"}]},"branch":"refs/heads/master"},"9c05f41b0582909ae2544badd31dd0753bbfa6f3":{"kind":"REWORK","_number":5,"created":"2022-10-18 03:18:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/789035/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/789035/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/789035/5"}}},"commit":{"parents":[{"commit":"675769bc5e3eb9a263651f67cb68a483f391be7a","subject":"Merge \"swift-drive-audit: reload systemd after editing fstab\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/675769bc5e3eb9a263651f67cb68a483f391be7a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 20:33:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-18 03:17:51.000000000","tz":-420},"subject":"wsgi: Pass worker pids to new manager when reloading","message":"wsgi: Pass worker pids to new manager when reloading\n\nThis allows the new manager to recognize them as they exit instead of\nlogging\n\n   Ignoring wait() result from unknown PID ...\n\nWith the improved knowledge of subprocesses, we can kick the log level\nfor the above message up to warning; we no longer expect it to trigger\nin practice.\n\nAdditionally, add a new tunable, `stale_worker_timeout`. Once this many\nseconds pass following a reload, the manager process will issue SIGKILLs\nto the remaining stale workers. This ensures there is some limit to how\nlong old code and configs may still be running.\n\nDrive-by: Add logging to ServersPerPortStrategy.register_worker_exit\nthat\u0027s comparable to what WorkersStrategy does.\n\nChange-Id: I8227939d04fda8db66fb2f131f2c71ce8741c7d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c05f41b0582909ae2544badd31dd0753bbfa6f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c05f41b0582909ae2544badd31dd0753bbfa6f3"}]},"branch":"refs/heads/master"},"d5b6fee1612947af346d56a56207ef0c725c92c7":{"kind":"REWORK","_number":6,"created":"2022-10-18 15:37:28.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/789035/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/789035/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/789035/6"}}},"commit":{"parents":[{"commit":"675769bc5e3eb9a263651f67cb68a483f391be7a","subject":"Merge \"swift-drive-audit: reload systemd after editing fstab\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/675769bc5e3eb9a263651f67cb68a483f391be7a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 20:33:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-10-18 15:37:22.000000000","tz":-420},"subject":"wsgi: Pass worker pids to new manager when reloading","message":"wsgi: Pass worker pids to new manager when reloading\n\nThis allows the new manager to recognize them as they exit instead of\nlogging\n\n   Ignoring wait() result from unknown PID ...\n\nWith the improved knowledge of subprocesses, we can kick the log level\nfor the above message up to warning; we no longer expect it to trigger\nin practice.\n\nAdditionally, add a new tunable, `stale_worker_timeout`. Once this many\nseconds pass following a reload, the manager process will issue SIGKILLs\nto the remaining stale workers. This ensures there is some limit to how\nlong old code and configs may still be running.\n\nDrive-by: Add logging to ServersPerPortStrategy.register_worker_exit\nthat\u0027s comparable to what WorkersStrategy does.\n\nChange-Id: I8227939d04fda8db66fb2f131f2c71ce8741c7d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5b6fee1612947af346d56a56207ef0c725c92c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5b6fee1612947af346d56a56207ef0c725c92c7"}]},"branch":"refs/heads/master"},"ada5d346658347527977c106c3975024d59ae50d":{"kind":"REWORK","_number":7,"created":"2023-01-06 23:50:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/789035/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/789035/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/789035/7"}}},"commit":{"parents":[{"commit":"b4124e0cd25b7f50b7d6be284fe2b936d7191acc","subject":"Memcached: add timing stats to set/get and other public functions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4124e0cd25b7f50b7d6be284fe2b936d7191acc"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 20:33:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-01-06 23:50:07.000000000","tz":-480},"subject":"wsgi: Pass worker pids to new manager when reloading","message":"wsgi: Pass worker pids to new manager when reloading\n\nThis allows the new manager to recognize them as they exit instead of\nlogging\n\n   Ignoring wait() result from unknown PID ...\n\nWith the improved knowledge of subprocesses, we can kick the log level\nfor the above message up to warning; we no longer expect it to trigger\nin practice.\n\nAdditionally, add a new tunable, `stale_worker_timeout`. Once this many\nseconds pass following a reload, the manager process will issue SIGKILLs\nto the remaining stale workers. This ensures there is some limit to how\nlong old code and configs may still be running.\n\nDrive-by: Add logging to ServersPerPortStrategy.register_worker_exit\nthat\u0027s comparable to what WorkersStrategy does.\n\nChange-Id: I8227939d04fda8db66fb2f131f2c71ce8741c7d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ada5d346658347527977c106c3975024d59ae50d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ada5d346658347527977c106c3975024d59ae50d"}]},"branch":"refs/heads/master"},"5d1395585f005c90533fe052f75038846008e16c":{"kind":"REWORK","_number":8,"created":"2023-04-06 03:48:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/789035/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/789035/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/789035/8"}}},"commit":{"parents":[{"commit":"57ce156a7f5cecd58fb4d83acbe1416d583275d1","subject":"Merge \"Start common.utils refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57ce156a7f5cecd58fb4d83acbe1416d583275d1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 20:33:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-06 03:47:47.000000000","tz":-420},"subject":"wsgi: Reap stale workers after a (configurable) timeout following a reload","message":"wsgi: Reap stale workers after a (configurable) timeout following a reload\n\nAdd a new tunable, `stale_worker_timeout`, defaulting to 7200 (i.e. two hours).\nOnce this time elapses following a reload, the manager process will issue\nSIGKILLs to any remaining stale workers.\n\nThis gives operators a way to configure a limit for how long old code and\nconfigs may still be running in their cluster.\n\nTo enable this, the temporary reload child (which waits for the reload to\ncomplete then closes the accept socket on all the old workers) has grown\nthe ability to send state to the re-exec\u0027ed manager. Currently, this is\nlimited to just the set of pre-re-exec child PIDs, though it was designed\nto be reasonably extensible.\n\nThis allows the new manager to recognize stale workers as they exit instead\nof logging\n\n   Ignoring wait() result from unknown PID ...\n\nWith the improved knowledge of subprocesses, we can kick the log level\nfor the above message up from info to warning; we no longer expect it to\ntrigger in practice.\n\nDrive-by: Add logging to ServersPerPortStrategy.register_worker_exit\nthat\u0027s comparable to what WorkersStrategy does.\n\nChange-Id: I8227939d04fda8db66fb2f131f2c71ce8741c7d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d1395585f005c90533fe052f75038846008e16c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d1395585f005c90533fe052f75038846008e16c"}]},"branch":"refs/heads/master"},"72add5ee9b47593fd1887ba4bd4a730406cb5e6c":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2023-04-18 23:40:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/789035/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/789035/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/789035/9"}}},"commit":{"parents":[{"commit":"57ce156a7f5cecd58fb4d83acbe1416d583275d1","subject":"Merge \"Start common.utils refactor\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57ce156a7f5cecd58fb4d83acbe1416d583275d1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 20:33:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-04-18 22:32:09.000000000","tz":-420},"subject":"wsgi: Reap stale workers (after a timeout) following a reload","message":"wsgi: Reap stale workers (after a timeout) following a reload\n\nAdd a new tunable, `stale_worker_timeout`, defaulting to 7200 (i.e. two\nhours). Once this time elapses following a reload, the manager process\nwill issue SIGKILLs to any remaining stale workers.\n\nThis gives operators a way to configure a limit for how long old code\nand configs may still be running in their cluster.\n\nTo enable this, the temporary reload child (which waits for the reload\nto complete then closes the accept socket on all the old workers) has\ngrown the ability to send state to the re-exec\u0027ed manager. Currently,\nthis is limited to just the set of pre-re-exec child PIDs, though it\nwas designed to be reasonably extensible.\n\nThis allows the new manager to recognize stale workers as they exit\ninstead of logging\n\n   Ignoring wait() result from unknown PID ...\n\nWith the improved knowledge of subprocesses, we can kick the log level\nfor the above message up from info to warning; we no longer expect it\nto trigger in practice.\n\nDrive-by: Add logging to ServersPerPortStrategy.register_worker_exit\nthat\u0027s comparable to what WorkersStrategy does.\n\nChange-Id: I8227939d04fda8db66fb2f131f2c71ce8741c7d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/72add5ee9b47593fd1887ba4bd4a730406cb5e6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/72add5ee9b47593fd1887ba4bd4a730406cb5e6c"}]},"branch":"refs/heads/master"},"5bdcfca028f4b0c47cb2d9c2661816ffbf1c1c2b":{"kind":"REWORK","_number":10,"created":"2023-05-23 16:38:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/789035/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/789035/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/789035/10"}}},"commit":{"parents":[{"commit":"ef64b63fdfd3a3c644cd7f64a94f83e952091d10","subject":"Merge \"proxy: add periodic zero-time sleep during object PUT\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ef64b63fdfd3a3c644cd7f64a94f83e952091d10"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 20:33:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-05-23 16:37:46.000000000","tz":-420},"subject":"wsgi: Reap stale workers (after a timeout) following a reload","message":"wsgi: Reap stale workers (after a timeout) following a reload\n\nAdd a new tunable, `stale_worker_timeout`, defaulting to 7200 (i.e. two\nhours). Once this time elapses following a reload, the manager process\nwill issue SIGKILLs to any remaining stale workers.\n\nThis gives operators a way to configure a limit for how long old code\nand configs may still be running in their cluster.\n\nTo enable this, the temporary reload child (which waits for the reload\nto complete then closes the accept socket on all the old workers) has\ngrown the ability to send state to the re-exec\u0027ed manager. Currently,\nthis is limited to just the set of pre-re-exec child PIDs, though it\nwas designed to be reasonably extensible.\n\nThis allows the new manager to recognize stale workers as they exit\ninstead of logging\n\n   Ignoring wait() result from unknown PID ...\n\nWith the improved knowledge of subprocesses, we can kick the log level\nfor the above message up from info to warning; we no longer expect it\nto trigger in practice.\n\nDrive-by: Add logging to ServersPerPortStrategy.register_worker_exit\nthat\u0027s comparable to what WorkersStrategy does.\n\nChange-Id: I8227939d04fda8db66fb2f131f2c71ce8741c7d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bdcfca028f4b0c47cb2d9c2661816ffbf1c1c2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bdcfca028f4b0c47cb2d9c2661816ffbf1c1c2b"}]},"branch":"refs/heads/master"},"1cabdd0aea4ab847c35e5d5d8643c07bbec18730":{"kind":"REWORK","_number":11,"created":"2023-06-06 00:16:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/789035/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/789035/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/789035/11"}}},"commit":{"parents":[{"commit":"eb3fe260d51d021674b2bccc4b9d1b4cc0757450","subject":"Merge \"memcached: add method and time_spent to exception logs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb3fe260d51d021674b2bccc4b9d1b4cc0757450"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 20:33:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-05 23:46:19.000000000","tz":-420},"subject":"wsgi: Reap stale workers (after a timeout) following a reload","message":"wsgi: Reap stale workers (after a timeout) following a reload\n\nAdd a new tunable, `stale_worker_timeout`, defaulting to 7200 (i.e. two\nhours). Once this time elapses following a reload, the manager process\nwill issue SIGKILLs to any remaining stale workers.\n\nThis gives operators a way to configure a limit for how long old code\nand configs may still be running in their cluster.\n\nTo enable this, the temporary reload child (which waits for the reload\nto complete then closes the accept socket on all the old workers) has\ngrown the ability to send state to the re-exec\u0027ed manager. Currently,\nthis is limited to just the set of pre-re-exec child PIDs, though it\nwas designed to be reasonably extensible.\n\nThis allows the new manager to recognize stale workers as they exit\ninstead of logging\n\n   Ignoring wait() result from unknown PID ...\n\nWith the improved knowledge of subprocesses, we can kick the log level\nfor the above message up from info to warning; we no longer expect it\nto trigger in practice.\n\nDrive-by: Add logging to ServersPerPortStrategy.register_worker_exit\nthat\u0027s comparable to what WorkersStrategy does.\n\nChange-Id: I8227939d04fda8db66fb2f131f2c71ce8741c7d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1cabdd0aea4ab847c35e5d5d8643c07bbec18730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1cabdd0aea4ab847c35e5d5d8643c07bbec18730"}]},"branch":"refs/heads/master"},"9a00ddd6ccc4ab7e2f3ecdd49b2cb25cf2793f95":{"kind":"REWORK","_number":12,"created":"2023-06-09 23:04:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/789035/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/789035/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/789035/12"}}},"commit":{"parents":[{"commit":"144b69d901ea676f6935e089bcdadaaa34e73b34","subject":"Merge \"Add --test-config option to WSGI servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/144b69d901ea676f6935e089bcdadaaa34e73b34"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 20:33:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-09 23:03:57.000000000","tz":-420},"subject":"wsgi: Reap stale workers (after a timeout) following a reload","message":"wsgi: Reap stale workers (after a timeout) following a reload\n\nAdd a new tunable, `stale_worker_timeout`, defaulting to 86400 (i.e. 24\nhours). Once this time elapses following a reload, the manager process\nwill issue SIGKILLs to any remaining stale workers.\n\nThis gives operators a way to configure a limit for how long old code\nand configs may still be running in their cluster.\n\nTo enable this, the temporary reload child (which waits for the reload\nto complete then closes the accept socket on all the old workers) has\ngrown the ability to send state to the re-exec\u0027ed manager. Currently,\nthis is limited to just the set of pre-re-exec child PIDs and their\nreload times, though it was designed to be reasonably extensible.\n\nThis allows the new manager to recognize stale workers as they exit\ninstead of logging\n\n   Ignoring wait() result from unknown PID ...\n\nWith the improved knowledge of subprocesses, we can kick the log level\nfor the above message up from info to warning; we no longer expect it\nto trigger in practice.\n\nDrive-by: Add logging to ServersPerPortStrategy.register_worker_exit\nthat\u0027s comparable to what WorkersStrategy does.\n\nChange-Id: I8227939d04fda8db66fb2f131f2c71ce8741c7d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a00ddd6ccc4ab7e2f3ecdd49b2cb25cf2793f95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a00ddd6ccc4ab7e2f3ecdd49b2cb25cf2793f95"}]},"branch":"refs/heads/master"},"4e38af269558bb7f8c3b38f3f1d448172d392d30":{"kind":"REWORK","_number":13,"created":"2023-06-10 03:41:10.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/789035/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/789035/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/789035/13"}}},"commit":{"parents":[{"commit":"144b69d901ea676f6935e089bcdadaaa34e73b34","subject":"Merge \"Add --test-config option to WSGI servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/144b69d901ea676f6935e089bcdadaaa34e73b34"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 20:33:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-06-10 03:26:56.000000000","tz":-420},"subject":"wsgi: Reap stale workers (after a timeout) following a reload","message":"wsgi: Reap stale workers (after a timeout) following a reload\n\nAdd a new tunable, `stale_worker_timeout`, defaulting to 86400 (i.e. 24\nhours). Once this time elapses following a reload, the manager process\nwill issue SIGKILLs to any remaining stale workers.\n\nThis gives operators a way to configure a limit for how long old code\nand configs may still be running in their cluster.\n\nTo enable this, the temporary reload child (which waits for the reload\nto complete then closes the accept socket on all the old workers) has\ngrown the ability to send state to the re-exec\u0027ed manager. Currently,\nthis is limited to just the set of pre-re-exec child PIDs and their\nreload times, though it was designed to be reasonably extensible.\n\nThis allows the new manager to recognize stale workers as they exit\ninstead of logging\n\n   Ignoring wait() result from unknown PID ...\n\nWith the improved knowledge of subprocesses, we can kick the log level\nfor the above message up from info to warning; we no longer expect it\nto trigger in practice.\n\nDrive-by: Add logging to ServersPerPortStrategy.register_worker_exit\nthat\u0027s comparable to what WorkersStrategy does.\n\nChange-Id: I8227939d04fda8db66fb2f131f2c71ce8741c7d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e38af269558bb7f8c3b38f3f1d448172d392d30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e38af269558bb7f8c3b38f3f1d448172d392d30"}]},"branch":"refs/heads/master"},"c242f1520bc741a24822a712e4a9758a61832227":{"kind":"REWORK","_number":14,"created":"2023-08-21 22:24:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/789035/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/789035/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/789035/14"}}},"commit":{"parents":[{"commit":"144b69d901ea676f6935e089bcdadaaa34e73b34","subject":"Merge \"Add --test-config option to WSGI servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/144b69d901ea676f6935e089bcdadaaa34e73b34"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 20:33:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-08-21 22:24:34.000000000","tz":-420},"subject":"wsgi: Reap stale workers (after a timeout) following a reload","message":"wsgi: Reap stale workers (after a timeout) following a reload\n\nAdd a new tunable, `stale_worker_timeout`, defaulting to 86400 (i.e. 24\nhours). Once this time elapses following a reload, the manager process\nwill issue SIGKILLs to any remaining stale workers.\n\nThis gives operators a way to configure a limit for how long old code\nand configs may still be running in their cluster.\n\nTo enable this, the temporary reload child (which waits for the reload\nto complete then closes the accept socket on all the old workers) has\ngrown the ability to send state to the re-exec\u0027ed manager. Currently,\nthis is limited to just the set of pre-re-exec child PIDs and their\nreload times, though it was designed to be reasonably extensible.\n\nThis allows the new manager to recognize stale workers as they exit\ninstead of logging\n\n   Ignoring wait() result from unknown PID ...\n\nWith the improved knowledge of subprocesses, we can kick the log level\nfor the above message up from info to warning; we no longer expect it\nto trigger in practice.\n\nDrive-by: Add logging to ServersPerPortStrategy.register_worker_exit\nthat\u0027s comparable to what WorkersStrategy does.\n\nChange-Id: I8227939d04fda8db66fb2f131f2c71ce8741c7d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c242f1520bc741a24822a712e4a9758a61832227"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c242f1520bc741a24822a712e4a9758a61832227"}]},"branch":"refs/heads/master"},"c22a03a2e2165ba9ed3580d2e51f76cce9f701ad":{"kind":"REWORK","_number":15,"created":"2023-11-07 00:26:37.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/789035/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/789035/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/789035/15"}}},"commit":{"parents":[{"commit":"ed335b272b3d8ed850b1baf432602dfe372589c9","subject":"docs: Document WSGI reload process","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed335b272b3d8ed850b1baf432602dfe372589c9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 20:33:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-11-07 00:26:23.000000000","tz":-480},"subject":"wsgi: Reap stale workers (after a timeout) following a reload","message":"wsgi: Reap stale workers (after a timeout) following a reload\n\nAdd a new tunable, `stale_worker_timeout`, defaulting to 86400 (i.e. 24\nhours). Once this time elapses following a reload, the manager process\nwill issue SIGKILLs to any remaining stale workers.\n\nThis gives operators a way to configure a limit for how long old code\nand configs may still be running in their cluster.\n\nTo enable this, the temporary reload child (which waits for the reload\nto complete then closes the accept socket on all the old workers) has\ngrown the ability to send state to the re-exec\u0027ed manager. Currently,\nthis is limited to just the set of pre-re-exec child PIDs and their\nreload times, though it was designed to be reasonably extensible.\n\nThis allows the new manager to recognize stale workers as they exit\ninstead of logging\n\n   Ignoring wait() result from unknown PID ...\n\nWith the improved knowledge of subprocesses, we can kick the log level\nfor the above message up from info to warning; we no longer expect it\nto trigger in practice.\n\nDrive-by: Add logging to ServersPerPortStrategy.register_worker_exit\nthat\u0027s comparable to what WorkersStrategy does.\n\nChange-Id: I8227939d04fda8db66fb2f131f2c71ce8741c7d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c22a03a2e2165ba9ed3580d2e51f76cce9f701ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c22a03a2e2165ba9ed3580d2e51f76cce9f701ad"}]},"branch":"refs/heads/master"},"9de6a504f40b0523b6413f4e556c99873d234323":{"kind":"REWORK","_number":16,"created":"2023-11-07 16:53:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/789035/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/789035/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/789035/16"}}},"commit":{"parents":[{"commit":"c05c2f798d28258a60fa4e91ff2428c9c68dda81","subject":"docs: Document WSGI reload process","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c05c2f798d28258a60fa4e91ff2428c9c68dda81"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 20:33:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-11-07 16:53:39.000000000","tz":-480},"subject":"wsgi: Reap stale workers (after a timeout) following a reload","message":"wsgi: Reap stale workers (after a timeout) following a reload\n\nAdd a new tunable, `stale_worker_timeout`, defaulting to 86400 (i.e. 24\nhours). Once this time elapses following a reload, the manager process\nwill issue SIGKILLs to any remaining stale workers.\n\nThis gives operators a way to configure a limit for how long old code\nand configs may still be running in their cluster.\n\nTo enable this, the temporary reload child (which waits for the reload\nto complete then closes the accept socket on all the old workers) has\ngrown the ability to send state to the re-exec\u0027ed manager. Currently,\nthis is limited to just the set of pre-re-exec child PIDs and their\nreload times, though it was designed to be reasonably extensible.\n\nThis allows the new manager to recognize stale workers as they exit\ninstead of logging\n\n   Ignoring wait() result from unknown PID ...\n\nWith the improved knowledge of subprocesses, we can kick the log level\nfor the above message up from info to warning; we no longer expect it\nto trigger in practice.\n\nDrive-by: Add logging to ServersPerPortStrategy.register_worker_exit\nthat\u0027s comparable to what WorkersStrategy does.\n\nChange-Id: I8227939d04fda8db66fb2f131f2c71ce8741c7d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9de6a504f40b0523b6413f4e556c99873d234323"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9de6a504f40b0523b6413f4e556c99873d234323"}]},"branch":"refs/heads/master"},"c8e775a94eb1350c1650e91f7615ae83b2e5081f":{"kind":"REWORK","_number":17,"created":"2023-11-09 22:10:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/789035/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/789035/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/789035/17"}}},"commit":{"parents":[{"commit":"c94e1c15130ef3689ba99337fdab39a80bb3fb29","subject":"docs: Document WSGI server process management","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c94e1c15130ef3689ba99337fdab39a80bb3fb29"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 20:33:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-11-09 22:09:48.000000000","tz":-480},"subject":"wsgi: Reap stale workers (after a timeout) following a reload","message":"wsgi: Reap stale workers (after a timeout) following a reload\n\nAdd a new tunable, `stale_worker_timeout`, defaulting to 86400 (i.e. 24\nhours). Once this time elapses following a reload, the manager process\nwill issue SIGKILLs to any remaining stale workers.\n\nThis gives operators a way to configure a limit for how long old code\nand configs may still be running in their cluster.\n\nTo enable this, the temporary reload child (which waits for the reload\nto complete then closes the accept socket on all the old workers) has\ngrown the ability to send state to the re-exec\u0027ed manager. Currently,\nthis is limited to just the set of pre-re-exec child PIDs and their\nreload times, though it was designed to be reasonably extensible.\n\nThis allows the new manager to recognize stale workers as they exit\ninstead of logging\n\n   Ignoring wait() result from unknown PID ...\n\nWith the improved knowledge of subprocesses, we can kick the log level\nfor the above message up from info to warning; we no longer expect it\nto trigger in practice.\n\nDrive-by: Add logging to ServersPerPortStrategy.register_worker_exit\nthat\u0027s comparable to what WorkersStrategy does.\n\nChange-Id: I8227939d04fda8db66fb2f131f2c71ce8741c7d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8e775a94eb1350c1650e91f7615ae83b2e5081f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c8e775a94eb1350c1650e91f7615ae83b2e5081f"}]},"branch":"refs/heads/master"},"4231b6e4f89209156d82a9b94672bb2e7c3b3a57":{"kind":"REWORK","_number":18,"created":"2023-11-30 23:12:25.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/789035/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/789035/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/789035/18"}}},"commit":{"parents":[{"commit":"90419df78594ee08eb047d44d52b4675e33cf378","subject":"Merge \"s3api: return 503 if mpu complete gets 409 deleting marker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90419df78594ee08eb047d44d52b4675e33cf378"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 20:33:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2023-11-30 23:11:00.000000000","tz":-480},"subject":"wsgi: Reap stale workers (after a timeout) following a reload","message":"wsgi: Reap stale workers (after a timeout) following a reload\n\nAdd a new tunable, `stale_worker_timeout`, defaulting to 86400 (i.e. 24\nhours). Once this time elapses following a reload, the manager process\nwill issue SIGKILLs to any remaining stale workers.\n\nThis gives operators a way to configure a limit for how long old code\nand configs may still be running in their cluster.\n\nTo enable this, the temporary reload child (which waits for the reload\nto complete then closes the accept socket on all the old workers) has\ngrown the ability to send state to the re-exec\u0027ed manager. Currently,\nthis is limited to just the set of pre-re-exec child PIDs and their\nreload times, though it was designed to be reasonably extensible.\n\nThis allows the new manager to recognize stale workers as they exit\ninstead of logging\n\n   Ignoring wait() result from unknown PID ...\n\nWith the improved knowledge of subprocesses, we can kick the log level\nfor the above message up from info to warning; we no longer expect it\nto trigger in practice.\n\nDrive-by: Add logging to ServersPerPortStrategy.register_worker_exit\nthat\u0027s comparable to what WorkersStrategy does.\n\nChange-Id: I8227939d04fda8db66fb2f131f2c71ce8741c7d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4231b6e4f89209156d82a9b94672bb2e7c3b3a57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4231b6e4f89209156d82a9b94672bb2e7c3b3a57"}]},"branch":"refs/heads/master"},"9f6dfdbe692217bfe4a9a355033c5d6073255227":{"kind":"REWORK","_number":19,"created":"2024-02-26 08:13:57.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/35/789035/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/789035/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/789035/19"}}},"commit":{"parents":[{"commit":"50336c509877e7c325823498cd67bf6cf4f4bef3","subject":"Merge \"test: all primary error limit is error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/50336c509877e7c325823498cd67bf6cf4f4bef3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 20:33:37.000000000","tz":-420},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-02-26 08:13:38.000000000","tz":660},"subject":"wsgi: Reap stale workers (after a timeout) following a reload","message":"wsgi: Reap stale workers (after a timeout) following a reload\n\nAdd a new tunable, `stale_worker_timeout`, defaulting to 86400 (i.e. 24\nhours). Once this time elapses following a reload, the manager process\nwill issue SIGKILLs to any remaining stale workers.\n\nThis gives operators a way to configure a limit for how long old code\nand configs may still be running in their cluster.\n\nTo enable this, the temporary reload child (which waits for the reload\nto complete then closes the accept socket on all the old workers) has\ngrown the ability to send state to the re-exec\u0027ed manager. Currently,\nthis is limited to just the set of pre-re-exec child PIDs and their\nreload times, though it was designed to be reasonably extensible.\n\nThis allows the new manager to recognize stale workers as they exit\ninstead of logging\n\n   Ignoring wait() result from unknown PID ...\n\nWith the improved knowledge of subprocesses, we can kick the log level\nfor the above message up from info to warning; we no longer expect it\nto trigger in practice.\n\nDrive-by: Add logging to ServersPerPortStrategy.register_worker_exit\nthat\u0027s comparable to what WorkersStrategy does.\n\nChange-Id: I8227939d04fda8db66fb2f131f2c71ce8741c7d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f6dfdbe692217bfe4a9a355033c5d6073255227"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f6dfdbe692217bfe4a9a355033c5d6073255227"}]},"branch":"refs/heads/master"},"76ef9c3d7fe24787139d4b2ebcd22e8bb9c33cef":{"kind":"REWORK","_number":20,"created":"2024-12-12 21:15:07.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/789035/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/789035/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/789035/20"}}},"commit":{"parents":[{"commit":"fe7928ea8aa45b15fde3154cd8ccdfabc43053a0","subject":"Merge \"Add unit test for object-updater recon dump\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe7928ea8aa45b15fde3154cd8ccdfabc43053a0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 20:33:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-12 21:14:54.000000000","tz":-480},"subject":"wsgi: Reap stale workers (after a timeout) following a reload","message":"wsgi: Reap stale workers (after a timeout) following a reload\n\nAdd a new tunable, `stale_worker_timeout`, defaulting to 86400 (i.e. 24\nhours). Once this time elapses following a reload, the manager process\nwill issue SIGKILLs to any remaining stale workers.\n\nThis gives operators a way to configure a limit for how long old code\nand configs may still be running in their cluster.\n\nTo enable this, the temporary reload child (which waits for the reload\nto complete then closes the accept socket on all the old workers) has\ngrown the ability to send state to the re-exec\u0027ed manager. Currently,\nthis is limited to just the set of pre-re-exec child PIDs and their\nreload times, though it was designed to be reasonably extensible.\n\nThis allows the new manager to recognize stale workers as they exit\ninstead of logging\n\n   Ignoring wait() result from unknown PID ...\n\nWith the improved knowledge of subprocesses, we can kick the log level\nfor the above message up from info to warning; we no longer expect it\nto trigger in practice.\n\nDrive-by: Add logging to ServersPerPortStrategy.register_worker_exit\nthat\u0027s comparable to what WorkersStrategy does.\n\nChange-Id: I8227939d04fda8db66fb2f131f2c71ce8741c7d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/76ef9c3d7fe24787139d4b2ebcd22e8bb9c33cef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/76ef9c3d7fe24787139d4b2ebcd22e8bb9c33cef"}]},"branch":"refs/heads/master"},"04865c60eade25898b0903ff59e94a73dcc56c5b":{"kind":"REWORK","_number":21,"created":"2024-12-17 19:32:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/35/789035/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/789035/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/789035/21"}}},"commit":{"parents":[{"commit":"4faa3523c9c0522c4f28a3028398949be4934483","subject":"Merge \"CI: Configure bandit better\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4faa3523c9c0522c4f28a3028398949be4934483"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 20:33:37.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-12-17 19:32:03.000000000","tz":-480},"subject":"wsgi: Reap stale workers (after a timeout) following a reload","message":"wsgi: Reap stale workers (after a timeout) following a reload\n\nAdd a new tunable, `stale_worker_timeout`, defaulting to 86400 (i.e. 24\nhours). Once this time elapses following a reload, the manager process\nwill issue SIGKILLs to any remaining stale workers.\n\nThis gives operators a way to configure a limit for how long old code\nand configs may still be running in their cluster.\n\nTo enable this, the temporary reload child (which waits for the reload\nto complete then closes the accept socket on all the old workers) has\ngrown the ability to send state to the re-exec\u0027ed manager. Currently,\nthis is limited to just the set of pre-re-exec child PIDs and their\nreload times, though it was designed to be reasonably extensible.\n\nThis allows the new manager to recognize stale workers as they exit\ninstead of logging\n\n   Ignoring wait() result from unknown PID ...\n\nWith the improved knowledge of subprocesses, we can kick the log level\nfor the above message up from info to warning; we no longer expect it\nto trigger in practice.\n\nDrive-by: Add logging to ServersPerPortStrategy.register_worker_exit\nthat\u0027s comparable to what WorkersStrategy does.\n\nChange-Id: I8227939d04fda8db66fb2f131f2c71ce8741c7d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04865c60eade25898b0903ff59e94a73dcc56c5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/04865c60eade25898b0903ff59e94a73dcc56c5b"}]},"branch":"refs/heads/master"},"ae6300af86c292b4ded1b4606f7e978bf7474f8a":{"kind":"REWORK","_number":22,"created":"2025-01-16 03:00:21.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/35/789035/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/789035/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/789035/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/789035/22"}}},"commit":{"parents":[{"commit":"734ed9cdd8fcc26e6b80c74da55ee1f72935d490","subject":"Merge \"Remove py2-only code paths\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/734ed9cdd8fcc26e6b80c74da55ee1f72935d490"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 20:33:37.000000000","tz":-420},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-01-16 02:44:21.000000000","tz":660},"subject":"wsgi: Reap stale workers (after a timeout) following a reload","message":"wsgi: Reap stale workers (after a timeout) following a reload\n\nAdd a new tunable, `stale_worker_timeout`, defaulting to 86400 (i.e. 24\nhours). Once this time elapses following a reload, the manager process\nwill issue SIGKILLs to any remaining stale workers.\n\nThis gives operators a way to configure a limit for how long old code\nand configs may still be running in their cluster.\n\nTo enable this, the temporary reload child (which waits for the reload\nto complete then closes the accept socket on all the old workers) has\ngrown the ability to send state to the re-exec\u0027ed manager. Currently,\nthis is limited to just the set of pre-re-exec child PIDs and their\nreload times, though it was designed to be reasonably extensible.\n\nThis allows the new manager to recognize stale workers as they exit\ninstead of logging\n\n   Ignoring wait() result from unknown PID ...\n\nWith the improved knowledge of subprocesses, we can kick the log level\nfor the above message up from info to warning; we no longer expect it\nto trigger in practice.\n\nDrive-by: Add logging to ServersPerPortStrategy.register_worker_exit\nthat\u0027s comparable to what WorkersStrategy does.\n\nChange-Id: I8227939d04fda8db66fb2f131f2c71ce8741c7d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae6300af86c292b4ded1b4606f7e978bf7474f8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae6300af86c292b4ded1b4606f7e978bf7474f8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
