)]}'
{"id":"openstack%2Fswift~789036","triplet_id":"openstack%2Fswift~master~I787b9d5a35ce760450bc7389b53f0540f70c8d76","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2022-05-02 16:10:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I787b9d5a35ce760450bc7389b53f0540f70c8d76","subject":"Log signal handling at notice","status":"MERGED","created":"2021-04-30 22:43:27.000000000","updated":"2022-05-02 16:11:55.000000000","submitted":"2022-05-02 16:10:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"789036","meta_rev_id":"d61bc8b3af90ac69a15d733c1b8eae37db436df3","_number":789036,"virtual_id_number":789036,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-02 16:10:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2022-05-02 14:36:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2022-05-02 14:36:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-01 00:05:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-28 19:18:11.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"61204a54680132cee52314e203b82985a4be3ef3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-04-30 22:43:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c61ae92d402607139626206b134353ddfeed0234","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 00:05:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d252aea1c46b491ba18aa9fe0732df4b : SUCCESS in 8m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/5d3f68dbaf2344e6891b63f0ade1b2bf : SUCCESS in 1h 14m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b0f999a5051b45be8d6970263158931e : SUCCESS in 32m 49s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/70c577b671a74a878754ec61eacda86f : SUCCESS in 8m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b03c6ce0e48341a99bd38e89090df575 : SUCCESS in 6m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/39ea69af0b5c46b29777fbaaf79e151e : SUCCESS in 15m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b90b6446c4834e2f8b354ae4f42f6c8c : SUCCESS in 15m 52s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e79ca035d8aa47e8a3c3d48aa96a539c : SUCCESS in 15m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/023665770154499b9ffff2601ddef8e6 : SUCCESS in 15m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/be51e20feba14b508a875f71e44f3de9 : SUCCESS in 16m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/da3b0b60d97c42bea23eb40d8fc8d4dd : SUCCESS in 16m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/773a1f4951a34b04a456970ac6ee5a3f : SUCCESS in 17m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/51086b2d994548c3878387f34cc96804 : SUCCESS in 19m 06s\n- swift-tox-func-py37 https://zuul.opendev.org/t/openstack/build/ef4cb66901234f4787449575e66ad1a9 : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py37 https://zuul.opendev.org/t/openstack/build/ef013ad6198a4b83a544e91be3ba003a : SUCCESS in 17m 37s\n- swift-tox-func-ec-py37 https://zuul.opendev.org/t/openstack/build/725156ebb99c4ef0a12a41062b66ae3e : SUCCESS in 17m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2e487c7124f3465a9dbe527e5f90301b : SUCCESS in 6m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/47154139cc314528a217c391135ff0dc : SUCCESS in 19m 51s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/02e9a2f65088485598e5c6466a702ead : SUCCESS in 44m 56s\n- swift-probetests-centos-8 https://zuul.opendev.org/t/openstack/build/22e237a5cbac495194fccb9b140b765e : SUCCESS in 49m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b6bc9f54329541dd93d843ebf5eca1c7 : SUCCESS in 35m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b0c4272f61ab46edb03869e427810a4b : SUCCESS in 39m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dfa5f6f171ae468e93f18caea0c347ae : SUCCESS in 16m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3575c0e30e6f476395c6fc6801d4af47 : SUCCESS in 5m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8410e4a462014c24942dcc45607f4b53 : SUCCESS in 22m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a597907aa7a479485ab487e8ae36827 : SUCCESS in 53m 03s","accounts_in_message":[],"_revision_number":1},{"id":"d2c72bb01194541252950dc374739e4322d865d1","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-04-28 19:18:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3890080d0a145180e36e171e776c5ab18b0ed20a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 19:18:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"39dd91af52cae917f248d2961d0c3bff209f4602","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-28 19:46:11.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cdeca02c60b4e283240dbbf9a43890271b6b78f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 22:31:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ebf710c50cc469399e00a62a6be5d03 : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/f6d8bd82d6e1474b8b7dc7f2c6723a8f : SUCCESS in 1h 06m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3463458d065b464681fef870f96cbf4c : SUCCESS in 50m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7a4867240f74bdfad3b684abc9e6fb3 : SUCCESS in 53m 11s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a764596af470465086eddd8ef2e95e3d : SUCCESS in 8m 22s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/2d52fa0068594f998a94c63be79b8652 : SUCCESS in 6m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3bd8616abc424b86b247c39a9e0cde90 : SUCCESS in 16m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/01807d5fbe8448cdbc998ec53df40582 : SUCCESS in 15m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fc72b64d1bbf42f0baff64c98b4d9d5d : SUCCESS in 15m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/085df00222a1461cba088ae38b958213 : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c20022de6c9243788c2e8d3d7eec5cae : SUCCESS in 18m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b69c91ffe9324603bda92d636991ab51 : SUCCESS in 19m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a27714b260594d789d7ad65c26048b38 : SUCCESS in 17m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ee95be9c8ca94c8abe8f7c62bb55a60e : SUCCESS in 15m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b6d4910de39b4487a5ade2d7ec021b46 : SUCCESS in 16m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4aadccd86bab4535b3c0451fdfb624eb : SUCCESS in 7m 49s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/f4c2cea1af444e49b80c3fac7a74eeeb : SUCCESS in 53m 47s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/e89a3fd1dc2f4405973c74bbd0d3bb62 : TIMED_OUT in 2h 01m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5a8af3f79443476b88bd3090de3addaa : SUCCESS in 46m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ba8b5eb4dd33457191c557647c8a2494 : SUCCESS in 47m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18f3c922dd814baf8fc65b7f676e4438 : SUCCESS in 16m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94032188e71b44609ae9e61e777faed4 : SUCCESS in 6m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d8b8ba79ff75418c8c5b37fe2b9c3c4b : SUCCESS in 20m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6859a5b159fd4b679a79f89bec559a43 : SUCCESS in 1h 03m 57s","accounts_in_message":[],"_revision_number":1},{"id":"12f866e3a7987ad4819d164e860d540eb7756f11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-29 15:49:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"234f22bca16cb279b1cb160dec6e6ec971216806","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-29 16:27:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e028ca08f3126d6616499854ca0c5ac9bcc836fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 17:53:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ef601605a404b95bbb470f95cc33972 : SUCCESS in 9m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/4ff1ea8075bd47848954058b72384262 : SUCCESS in 53m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/40138eacba334c34bee515f21daf2e63 : SUCCESS in 42m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c93d6d7fef8b49ce88ca982b7b844f97 : SUCCESS in 43m 08s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/49b7b1f3eed04d47b6c734f290afda17 : SUCCESS in 22m 17s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/6aa9d225d496410bab74e5af2f57b23a : SUCCESS in 21m 16s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/9a30d5829e254bef9bff484a6df9fce5 : SUCCESS in 20m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ac81009a38714b0f88e46bd1092fc9a3 : SUCCESS in 8m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b33277f5cc784233b39716ec311e81f4 : SUCCESS in 5m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dbf420f95a314020867996581a3bb84d : SUCCESS in 13m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/eb1353eea1dd4f8195be4ee61468d411 : SUCCESS in 19m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/160d3b9e88344578a4d51908e7bee019 : SUCCESS in 19m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7b065401f37a4aa98f6438d777051bfb : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/adf03de485a646829f0bf9ed5121ffc7 : SUCCESS in 17m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/da7806fcc11e4997a3d0a5f91b4bec06 : SUCCESS in 15m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cc1843ff7d7f49bda1886f64fe57bdb2 : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2df17d67f74e446f8f5fe5218deaddd7 : SUCCESS in 17m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7b45e148a4344982ba434b1e574ae3cc : SUCCESS in 14m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/198e25d153684d4597bde14047cfc7e5 : SUCCESS in 7m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8cf501a3686e404e81e29b932b7329ca : SUCCESS in 17m 29s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/a031736900734b22a55ad6eb1c76c9de : FAILURE in 47m 01s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/ff7dcfd8183f44b190bf0cd11331faab : SUCCESS in 1h 23m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/22d18905b82948188ead5972f1a1f2da : SUCCESS in 38m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/da051ef6e6f64d4fb0578a276d30db44 : SUCCESS in 41m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d0aa512b3fb48aa98f3fea4c3113bfb : SUCCESS in 15m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df1c82253695443b859362e53f161c49 : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1d68d427079847d1836b216c357f7330 : SUCCESS in 27m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2806077620e4815b95a6c35d263e49e : SUCCESS in 1h 07m 26s","accounts_in_message":[],"_revision_number":3},{"id":"7daf079be9b72f94c1e49ebc98b5b9988fd166c8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2022-04-29 20:31:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aed3473cac73ba0bd07e18d106e254337e38d499","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 20:34:49.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/938124c4beb242eaaa490f8779afc615 : SUCCESS in 48m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d32da4912eb14297af7d24981aff6757 : SUCCESS in 52m 45s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/410b28eba00743bbb726c6badcc07095 : TIMED_OUT in 3h 01m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/27b32beb710a4b66b12ea2d76b8b563f : SUCCESS in 1h 07m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2e5bf35cdae40b4a3cd09f6fbdc9e20 : SUCCESS in 1h 12m 38s","accounts_in_message":[],"_revision_number":3},{"id":"762e97875879e31f89bd41af833f62036d885d88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 21:54:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05304a17753b464f9e14cd48b26db500 : SUCCESS in 9m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/a9e922d16ec14399b6851691788329d8 : SUCCESS in 44m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e100c12860cf4f3b97babf8ea26cbe41 : SUCCESS in 50m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/265f6db9574f4b668ad0889d775142e9 : SUCCESS in 49m 12s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/df66607599b74191a8b39791c2c078a8 : SUCCESS in 22m 31s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/ccc69d45119d4612841435d5a93ef83c : SUCCESS in 22m 32s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/47c22dd627f34148bf9b5ce10f095087 : SUCCESS in 19m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/89b02f7a5ce141c396cee74f44a9df6d : SUCCESS in 8m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/46668cbea11040f78ec2d0dd48563363 : SUCCESS in 5m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1f3f71268cbd498aa307f0701b9c813e : SUCCESS in 18m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7df4c11e9e0e47b28fe993631bb5e611 : SUCCESS in 12m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d5075fb591a24395a2c60dba1cf4231a : SUCCESS in 14m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b178cbc6a9bd4a65b21ce423012ec33c : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2b2a26ca6293432e9ba17306fbf290e9 : SUCCESS in 20m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dd5a3ad53d6f40fa9a0812b7ed651008 : SUCCESS in 17m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4e985815bd3a473fa4e8c2b3cd2fdea7 : SUCCESS in 15m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1d567ae759ac48d0aecc8e1b4a4ba62c : SUCCESS in 12m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3c7e5e22607a4c09a7c22a4a5cfed0c2 : SUCCESS in 19m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9a56d7ebc7994cce9c6e515730fe7d0e : SUCCESS in 7m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/535962e58bf84a73a0287a38afba67e8 : SUCCESS in 21m 05s (non-voting)\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/d4961e289fad45d4ae7932eba9c33670 : SUCCESS in 44m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/f38348eb43004f56aa21dcaee2f62234 : SUCCESS in 56m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/96ad8da49cf147b8b0f26d59beb58211 : SUCCESS in 50m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0ae5991b0fa64fbb99aedba502b2270d : SUCCESS in 42m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c87b3552506d4046a134eab8b66b04dd : SUCCESS in 22m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24983fe746b742d69651f2d99d22bdf9 : SUCCESS in 6m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c0450b5fea6848669c6a96d31d077079 : SUCCESS in 26m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f423d99fbf614d1fb4c51e1fa86ac26b : SUCCESS in 1h 06m 47s","accounts_in_message":[],"_revision_number":3},{"id":"ff1267ab8f52ee0a450c38f4a85be08c8f73e221","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2022-05-02 14:36:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3be15e45edd4f9d69309e01891e4ab7c709aadc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 14:37:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"85cca2bd4575a3801ccbed5a707152383752d6fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 16:10:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11b700f4d09b48719ba50c975ed04a71 : SUCCESS in 9m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/3df7a0778b1d4e7b841ac41d34165682 : SUCCESS in 59m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d8375bd1e27e4a9a8d8ac423f5c3d2df : SUCCESS in 45m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/afedf0bc8137495db88960c323b238d2 : SUCCESS in 48m 02s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ef76e35d1cdc41f986a88e1a52d9ab21 : SUCCESS in 7m 29s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/7d2e96e7ecc04bbd98283b3c88f646e0 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cfb0d22f63904d4a84869ee009e47454 : SUCCESS in 13m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e6ee58892b6d4ce6b40abf804ad24406 : SUCCESS in 20m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c2898f632c8f4659b63479ec23bce762 : SUCCESS in 21m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b39cf1f18ddd4db58751c45ea51c311f : SUCCESS in 27m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a9e2e67628794c1b9e9e1e737760bf0e : SUCCESS in 27m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/177e4557175140838cf3330c456d0228 : SUCCESS in 28m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f200d55a16ba4a0c9f57b097b5628354 : SUCCESS in 17m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4e7d2329ea4548239920ce424d3a9588 : SUCCESS in 12m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/14493ff698c846439a7446adb0d0ea99 : SUCCESS in 18m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c4999d30698a43c8a46b937612d63645 : SUCCESS in 6m 24s\n- swift-probetests-centos-7 https://zuul.opendev.org/t/openstack/build/92234b2f3c324c509dc0a55da77c3549 : SUCCESS in 1h 23m 22s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/63fcf552d8474b95bc19849c310a9132 : SUCCESS in 1h 19m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/18507873590144aa835c3d4c122c04ca : SUCCESS in 40m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4fee4d381952482e864b255dde3fd4e2 : SUCCESS in 32m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d0bddd4e5ce4c0985de3da1ee7f0a85 : SUCCESS in 15m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0de9657fd27e4f68960d65bf916ada72 : SUCCESS in 6m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/49ce0051426c44da8fcc94d6e02dd5ba : SUCCESS in 26m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce93186680f0479381cef9309126495e : SUCCESS in 46m 23s","accounts_in_message":[],"_revision_number":3},{"id":"95d436ec6f1d3c25366c8c1163e7130a8f9dacd3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 16:10:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d61bc8b3af90ac69a15d733c1b8eae37db436df3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 16:11:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5eb4558611cf470285c79676904ea754 : SUCCESS in 1m 06s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2746c665e2eb44c8af36aff9ad826b14 : SUCCESS in 41s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/aa1738ded9aa48c4a42ce322d79b02b8 : SUCCESS in 44s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"05b2e894a92423a0e34ce2ea4b92f48ce1924c17","revisions":{"73e1e8c8761b6761aa01d48484a368f37598084e":{"kind":"REWORK","_number":1,"created":"2021-04-30 22:43:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/36/789036/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/789036/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/789036/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/789036/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/789036/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/789036/1"}}},"commit":{"parents":[{"commit":"443677f104c0181490cc1ff04f95efbed0a6a0de","subject":"Merge \"object: Plumb logger_thread_locals through _finalize_put\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/443677f104c0181490cc1ff04f95efbed0a6a0de"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 22:42:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 22:42:52.000000000","tz":-420},"subject":"Log signal handling at notice","message":"Log signal handling at notice\n\nSignals feel like they belong with the other process-management info\nwe already log at notice -- more important than a simple info(), but\ndefinitely not an error().\n\nChange-Id: I787b9d5a35ce760450bc7389b53f0540f70c8d76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73e1e8c8761b6761aa01d48484a368f37598084e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/73e1e8c8761b6761aa01d48484a368f37598084e"}]},"branch":"refs/heads/master"},"ce32cf3024dd33557d24def44d00331dd2ca2b20":{"kind":"REWORK","_number":2,"created":"2022-04-29 15:49:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/36/789036/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/789036/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/789036/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/789036/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/789036/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/789036/2"}}},"commit":{"parents":[{"commit":"b621a6f932edcda1cdba02534e382b962e759f9e","subject":"Merge \"tests: Fix swiftclient/requests log level adjustment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b621a6f932edcda1cdba02534e382b962e759f9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 22:42:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-29 15:48:29.000000000","tz":-420},"subject":"Log signal handling at notice","message":"Log signal handling at notice\n\nSignals feel like they belong with the other process-management info\nwe already log at notice -- more important than a simple info(), but\ndefinitely not an error().\n\nWhile we\u0027re at it, have Daemon include the receiving pid when logging\nabout SIGTERM.\n\nChange-Id: I787b9d5a35ce760450bc7389b53f0540f70c8d76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ce32cf3024dd33557d24def44d00331dd2ca2b20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ce32cf3024dd33557d24def44d00331dd2ca2b20"}]},"branch":"refs/heads/master"},"05b2e894a92423a0e34ce2ea4b92f48ce1924c17":{"kind":"REWORK","_number":3,"created":"2022-04-29 16:27:44.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/36/789036/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/789036/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/789036/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/789036/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/789036/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/789036/3"}}},"commit":{"parents":[{"commit":"b621a6f932edcda1cdba02534e382b962e759f9e","subject":"Merge \"tests: Fix swiftclient/requests log level adjustment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b621a6f932edcda1cdba02534e382b962e759f9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2021-04-30 22:42:52.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2022-04-29 16:27:38.000000000","tz":-420},"subject":"Log signal handling at notice","message":"Log signal handling at notice\n\nSignals feel like they belong with the other process-management info\nwe already log at notice -- more important than a simple info(), but\ndefinitely not an error().\n\nWhile we\u0027re at it, have Daemon include the receiving pid when logging\nabout SIGTERM.\n\nChange-Id: I787b9d5a35ce760450bc7389b53f0540f70c8d76\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05b2e894a92423a0e34ce2ea4b92f48ce1924c17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05b2e894a92423a0e34ce2ea4b92f48ce1924c17"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
